)]}'
{"id":"openstack%2Foctavia~858498","triplet_id":"openstack%2Foctavia~stable%2Fwallaby~Icc7dc9bb22187f908852c20415227574ded347aa","project":"openstack/octavia","branch":"stable/wallaby","attention_set":{},"removed_from_attention_set":{"11628":{"account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"last_update":"2022-11-17 13:02:42.000000000","reason":"Change was submitted"},"34429":{"account":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"last_update":"2022-11-17 13:02:42.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Icc7dc9bb22187f908852c20415227574ded347aa","subject":"Handle feature compatibility of HAProxy server-state-file option","status":"MERGED","created":"2022-09-20 08:49:28.000000000","updated":"2022-11-17 13:04:20.000000000","submitted":"2022-11-17 13:02:42.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"858498","meta_rev_id":"858b68c3d943bdf8edbb210a1828be70da173522","_number":858498,"virtual_id_number":858498,"owner":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-11-17 13:02:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"all":[{"value":0,"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":2,"date":"2022-11-17 10:04:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":2,"date":"2022-11-16 17:43:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"all":[{"value":0,"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":1,"date":"2022-11-17 10:04:33.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-09-20 10:42:19.000000000","updated_by":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"reviewer":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"state":"REVIEWER"},{"updated":"2022-09-22 12:29:17.000000000","updated_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"REVIEWER"},{"updated":"2022-11-16 17:43:24.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2022-11-16 20:02:01.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"91605e60eab68b724d571ead038a114065afa276","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-09-20 08:49:28.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b449d866eb3f0477be7b4273e36b4e70c7ed0bce","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2022-09-20 10:42:19.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"5c53cde5f30f43ac2d141a030050f87da8566ce3","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2022-09-22 12:29:17.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"2aee695813604587e2fd2fac3b0055bbfe651357","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2022-11-16 17:43:14.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"ac3c960f9ea61d9af20b7bbee0ad19809ae5e7bb","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2022-11-16 17:43:24.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"8b02bae8beeab374a36da0d1b280319e6cde29a7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-16 20:02:01.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7623c59cd0b84dd7ad3032db7d1b3ea8\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/42604852764b4abeac445520939a99f0 : SUCCESS in 9m 08s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a926150bfacf4abb84b5cca070e1973f : SUCCESS in 7m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c15cba15e60e4e68801ab0ddd9fd5663 : SUCCESS in 5m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fc62398bb04f4e59a9a70bc372de1da0 : SUCCESS in 4m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f3a8511049904a9aa514fff7de764942 : SUCCESS in 4m 53s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6371f18cf2aa4d3eb3649bd1699a188c : SUCCESS in 4m 30s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5fd3ebfc9b354837b813beb2f6db242c : SUCCESS in 12m 06s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/01345107b9d4405db387e144890e96fc : SUCCESS in 4m 49s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/7f35d148d10148f8b3061c68d871fd0f : SUCCESS in 4m 54s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/8ce75daa10764247ab59af3c8ca23025 : SUCCESS in 7m 39s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/393508c6aac24e6dac9965037becadde : SUCCESS in 4m 15s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/62b7d7a4592f47229ee34244991e99d6 : SUCCESS in 6m 29s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/208b0a7128d54a64a210682725386dbd : SUCCESS in 1h 05m 10s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/7f96d529408041dc903941548d198733 : SUCCESS in 1h 11m 03s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/2757df436ab04b1d8f910c4fdf74f4dc : SUCCESS in 32m 03s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/c5a48788dc02473ba6d14094d18c03df : SUCCESS in 28m 49s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/d84933f426054af4a286bbb678f82269 : SUCCESS in 30m 13s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/90abb1db34ea4991b25252f8808c5d56 : RETRY_LIMIT in 8m 54s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/86595e132d8a404498b8ef04349e21d8 : SUCCESS in 1h 55m 35s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/bb34047b2b2949448caced5b36834497 : FAILURE in 28m 33s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/a5abbe32fab549f381172a175b6e77eb : SUCCESS in 2h 11m 10s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/2e0e839f3df14a22b84cd58ba6ee8085 : SUCCESS in 1h 55m 25s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/5905a5a777d54534a547f5d071d39bca : SUCCESS in 2h 00m 51s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/823a7ff2f376491897513be6e8421843 : NODE_FAILURE Node request 300-0019742278 failed in 0s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"c7130a8bbbbe93507ccc897c326e426075b74b2a","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2022-11-17 10:04:33.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"935040ae4122005558633b3ce1e1ec4a26b92c85","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-17 10:05:37.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"b5283297d810618cfc79558f34144998ea6d8194","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-17 13:02:42.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/30336669b2074a34b54b40d7a210067e\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4a43082f00b54dd49f604e95ba01bb47 : SUCCESS in 7m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1f14c06cddc545a0856d7c63bbebbc62 : SUCCESS in 6m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/268e93adaa6b4773a91d5b9eb3fa0145 : SUCCESS in 5m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e08bf5401c60457a8c23c6ddebcf2a7f : SUCCESS in 4m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2f3a59d9464b4c73a49088f973076ba4 : SUCCESS in 11m 05s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5d08938d22754f67b5fcecba48ec4a87 : SUCCESS in 5m 28s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/dd648386d5554b08888a388e00a1540d : SUCCESS in 5m 06s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/6bebba056af54b50b9c35c237cf60554 : SUCCESS in 8m 00s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/865204cc38334364a02b926171d51d36 : SUCCESS in 57m 55s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/dcbc1eba10b84f9da67e8c2b47a35b82 : SUCCESS in 1h 25m 48s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/dc198320eb2444529da57a0d85747691 : SUCCESS in 34m 50s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/196722a199f94a67900e532381cc6a83 : SUCCESS in 30m 25s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/81361d698d7247d49d1f52fbce1f397e : SUCCESS in 31m 37s","accounts_in_message":[],"_revision_number":2},{"id":"6f0ee32edccc0cc90ded198ffd03faeb8ecce200","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-17 13:02:42.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"858b68c3d943bdf8edbb210a1828be70da173522","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-17 13:04:20.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/81ea4c5817f049e7835f57e77d154c4d\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/08ced3973f76453cb09ec9890b5da85e : SUCCESS in 57s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/93346d9753f74f38ac1bae56d8d67740 : SUCCESS in 43s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"5c50cbe1abdc9dd7ed8f752f543ecfe44b8511cc","revisions":{"5e8331676e043ebaad8998e7c99b9e82f9800642":{"kind":"REWORK","_number":1,"created":"2022-09-20 08:49:28.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/98/858498/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/98/858498/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/98/858498/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/98/858498/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/98/858498/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/98/858498/1"}}},"commit":{"parents":[{"commit":"d1877698badc296654631c256bcf109c31086eb4","subject":"Merge \"Apply openstack-selinux policies in Centos amphorae\" into stable/wallaby","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d1877698badc296654631c256bcf109c31086eb4"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-08-25 10:47:38.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-09-20 08:49:04.000000000","tz":120},"subject":"Handle feature compatibility of HAProxy server-state-file option","message":"Handle feature compatibility of HAProxy server-state-file option\n\nhttps://review.opendev.org/c/openstack/octavia/+/805955 introduced\na new feature that requires HAProxy 1.6 or higher to work. This change\nensures that with older HAProxy version this feature does not get\nenabled to avoid errors.\n\n Conflicts:\n\toctavia/common/jinja/haproxy/combined_listeners/jinja_cfg.py\n\toctavia/tests/unit/common/jinja/haproxy/combined_listeners/test_jinja_cfg.py\n\nStory: 2010254\nTask: 46145\nChange-Id: Icc7dc9bb22187f908852c20415227574ded347aa\n(cherry picked from commit 69ecdb190880d0c49e9f0ef92fd53de14cf66720)\n(cherry picked from commit 7f220d41adaab75471e551f89b0c2b97e830bab7)\n(cherry picked from commit 6346341b8ebb60c33b53f81e8c39f9c24d7de6df)\n(cherry picked from commit de3e6c4ef098f7e0cf29122c56dbedb96252b8d6)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/5e8331676e043ebaad8998e7c99b9e82f9800642"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/5e8331676e043ebaad8998e7c99b9e82f9800642"}]},"branch":"refs/heads/stable/wallaby"},"5c50cbe1abdc9dd7ed8f752f543ecfe44b8511cc":{"kind":"REWORK","_number":2,"created":"2022-11-16 17:43:14.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/98/858498/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/98/858498/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/98/858498/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/98/858498/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/98/858498/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/98/858498/2"}}},"commit":{"parents":[{"commit":"d1877698badc296654631c256bcf109c31086eb4","subject":"Merge \"Apply openstack-selinux policies in Centos amphorae\" into stable/wallaby","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d1877698badc296654631c256bcf109c31086eb4"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-08-25 10:47:38.000000000","tz":120},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2022-11-16 17:43:14.000000000","tz":0},"subject":"Handle feature compatibility of HAProxy server-state-file option","message":"Handle feature compatibility of HAProxy server-state-file option\n\nhttps://review.opendev.org/c/openstack/octavia/+/805955 introduced\na new feature that requires HAProxy 1.6 or higher to work. This change\nensures that with older HAProxy version this feature does not get\nenabled to avoid errors.\n\n Conflicts:\n\toctavia/common/jinja/haproxy/combined_listeners/jinja_cfg.py\n\toctavia/tests/unit/common/jinja/haproxy/combined_listeners/test_jinja_cfg.py\n\nStory: 2010254\nTask: 46145\nChange-Id: Icc7dc9bb22187f908852c20415227574ded347aa\n(cherry picked from commit 69ecdb190880d0c49e9f0ef92fd53de14cf66720)\n(cherry picked from commit 7f220d41adaab75471e551f89b0c2b97e830bab7)\n(cherry picked from commit 6346341b8ebb60c33b53f81e8c39f9c24d7de6df)\n(cherry picked from commit de3e6c4ef098f7e0cf29122c56dbedb96252b8d6)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/5c50cbe1abdc9dd7ed8f752f543ecfe44b8511cc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/5c50cbe1abdc9dd7ed8f752f543ecfe44b8511cc"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[]}
