)]}'
{"id":"openstack%2Foctavia~858503","triplet_id":"openstack%2Foctavia~stable%2Fvictoria~Icc7dc9bb22187f908852c20415227574ded347aa","project":"openstack/octavia","branch":"stable/victoria","attention_set":{},"removed_from_attention_set":{"11628":{"account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"last_update":"2022-11-17 13:02:55.000000000","reason":"Change was submitted"},"34429":{"account":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"last_update":"2022-11-17 13:02:55.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Icc7dc9bb22187f908852c20415227574ded347aa","subject":"Handle feature compatibility of HAProxy server-state-file option","status":"MERGED","created":"2022-09-20 08:58:39.000000000","updated":"2022-11-17 13:04:38.000000000","submitted":"2022-11-17 13:02:55.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"858503","meta_rev_id":"8f3dcae981a9bddb68d1d17e92fffc5074f40c35","_number":858503,"virtual_id_number":858503,"owner":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-11-17 13:02:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":2,"date":"2022-11-16 17:42:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":2,"date":"2022-11-17 10:05:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"all":[{"value":0,"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":1,"date":"2022-11-17 10:05:18.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-09-20 10:39:44.000000000","updated_by":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"reviewer":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"state":"REVIEWER"},{"updated":"2022-09-22 12:30:02.000000000","updated_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"REVIEWER"},{"updated":"2022-11-16 17:42:53.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2022-11-16 19:38:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"4eb7bf8929be8b7a692219f374e47d6dd2e1863d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-09-20 08:58:39.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"dd34d3fcc27ae152ee0e02cbd4a8e11143eb279d","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2022-09-20 10:39:44.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"1efe193983c4d40d8719debbbe88dd0aa1d34caf","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2022-09-22 12:30:02.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"b3697372c4bd179b5a0b48e778086a0836dee44c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2022-11-16 17:42:42.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"3644085b0ace050ea9c89b0e355da628ce53838b","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2022-11-16 17:42:53.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"13b47b0a450b2d739dc2d6448868aa3244fe3b8b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-16 19:38:34.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a634f9af17754080aa3baf9c7d6209db\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/632249336fbb4197b15c24d856ab920e : SUCCESS in 9m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3e397baafe3f43919504610e313784d3 : SUCCESS in 5m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e6c76eb17a9042b288cb9b531b527fcf : SUCCESS in 4m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8ee46e19c2824fe5ac193855b0f1bb99 : SUCCESS in 5m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/61980c74959d404f92db82cec9dc889a : SUCCESS in 10m 06s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/976ed40facf44c09bd3a03e5fc78ab80 : SUCCESS in 5m 33s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/78e72f134e844a62a7cb8aa30e3ac917 : SUCCESS in 5m 57s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/bb1bc6b618174180ad29c5bd269155a9 : SUCCESS in 9m 10s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/6617c632207c4f30a31f3df5228c9e57 : SUCCESS in 5m 06s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/e932de92c9d1463d9e1f84b653ca0dd2 : SUCCESS in 6m 36s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/a4036afcfd744adcb02a1a2f1efdeec0 : SUCCESS in 1h 05m 42s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/df779523f21945e0818ae5ad1747c5be : SUCCESS in 1h 40m 39s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/4b1c159012214913a780851a6f9f6478 : SUCCESS in 31m 58s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/7b05e1b8abfe4762bb7205f28878fdc3 : SUCCESS in 39m 39s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/747368d2b56e45a693a94f644a3ec6e2 : SUCCESS in 40m 20s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/9c5a9072c98b4d12ae93696ae55859bb : SUCCESS in 1h 37m 58s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/23de3aa196a9451195f0f37acc0e9f86 : FAILURE in 29m 10s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/ac1a933500b24bc09d13e5c57ad56e2a : SUCCESS in 1h 50m 37s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/cc33adf8e3c54fe09fbe5aa12845cd13 : FAILURE in 1h 19m 39s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/da04dde3e19e4a5e983e7166c0b56d64 : FAILURE in 1h 35m 49s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/d5cff985892f4efa83650968d27850d0 : NODE_FAILURE Node request 300-0019742254 failed in 0s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"6f9af493e89c51a280e96f2097479e58ec4fb74e","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2022-11-17 10:05:18.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"bf14fb622c526869f647fc45ec471780da052d3d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-17 10:06:30.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"c070d2b56825a77c121b558cc90568aa09727b83","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-17 13:02:55.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cc227be135314e47ac46d5a6f795d194\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b444ca1fa25a4609867016eaa51180c6 : SUCCESS in 5m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0080c77b08f844a9bf3b1203d6d4fe6e : SUCCESS in 5m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3b4bade96bb348728b76e913b87178a9 : SUCCESS in 4m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0562d95251d6476a9b932c75b7fc2213 : SUCCESS in 10m 49s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7ed2b8d006d14b84b0a2c7681062a076 : SUCCESS in 6m 02s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/98a8c1ea481a41aa949978f049e9a7ed : SUCCESS in 4m 50s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/832ae7d212c14b85b724ce88438b8b75 : SUCCESS in 7m 11s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/dfb801abcc234f81bbc5f6dfe3a612a6 : SUCCESS in 58m 40s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/0889b36a6faa41f2bc6c586b985b6539 : SUCCESS in 1h 39m 17s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/5859210b44234942a2b423c41a51efa5 : SUCCESS in 32m 30s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/dc9f16391973473b99360604e8ee1196 : SUCCESS in 29m 04s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/354417a4e6bd45d3be261c9621714967 : SUCCESS in 29m 45s","accounts_in_message":[],"_revision_number":2},{"id":"bea84d0165a7d67d060d2076ca189c93a491694d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-17 13:02:55.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"8f3dcae981a9bddb68d1d17e92fffc5074f40c35","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-17 13:04:38.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/53e90f9406b943e4bd005a63e80f37a7\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bd4767a3a3404cb2b3c382ac307c139b : SUCCESS in 1m 04s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/61c9ec56f6cc4547a2817bd5a2676b6f : SUCCESS in 46s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"2fa09640153bec78ab50d78bb68f6ba04117a243","revisions":{"3fc9d3923c6330c8fe4d39ed225179ad5802e076":{"kind":"REWORK","_number":1,"created":"2022-09-20 08:58:39.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/03/858503/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/03/858503/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/03/858503/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/03/858503/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/03/858503/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/03/858503/1"}}},"commit":{"parents":[{"commit":"7b4f4ea201b2d7c68ee31a53d0e4e2a7844e5b21","subject":"Apply openstack-selinux policies in Centos amphorae","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/7b4f4ea201b2d7c68ee31a53d0e4e2a7844e5b21"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-08-25 10:47:38.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-09-20 08:58:17.000000000","tz":120},"subject":"Handle feature compatibility of HAProxy server-state-file option","message":"Handle feature compatibility of HAProxy server-state-file option\n\nhttps://review.opendev.org/c/openstack/octavia/+/805955 introduced\na new feature that requires HAProxy 1.6 or higher to work. This change\nensures that with older HAProxy version this feature does not get\nenabled to avoid errors.\n\n Conflicts:\n\toctavia/common/constants.py\n\toctavia/common/jinja/haproxy/combined_listeners/jinja_cfg.py\n\toctavia/tests/unit/common/jinja/haproxy/combined_listeners/test_jinja_cfg.py\n\nStory: 2010254\nTask: 46145\nChange-Id: Icc7dc9bb22187f908852c20415227574ded347aa\n(cherry picked from commit 69ecdb190880d0c49e9f0ef92fd53de14cf66720)\n(cherry picked from commit 7f220d41adaab75471e551f89b0c2b97e830bab7)\n(cherry picked from commit 6346341b8ebb60c33b53f81e8c39f9c24d7de6df)\n(cherry picked from commit de3e6c4ef098f7e0cf29122c56dbedb96252b8d6)\n(cherry picked from commit 5e8331676e043ebaad8998e7c99b9e82f9800642)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/3fc9d3923c6330c8fe4d39ed225179ad5802e076"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/3fc9d3923c6330c8fe4d39ed225179ad5802e076"}]},"branch":"refs/heads/stable/victoria"},"2fa09640153bec78ab50d78bb68f6ba04117a243":{"kind":"REWORK","_number":2,"created":"2022-11-16 17:42:42.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/03/858503/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/03/858503/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/03/858503/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/03/858503/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/03/858503/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/03/858503/2"}}},"commit":{"parents":[{"commit":"7b4f4ea201b2d7c68ee31a53d0e4e2a7844e5b21","subject":"Apply openstack-selinux policies in Centos amphorae","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/7b4f4ea201b2d7c68ee31a53d0e4e2a7844e5b21"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-08-25 10:47:38.000000000","tz":120},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2022-11-16 17:42:42.000000000","tz":0},"subject":"Handle feature compatibility of HAProxy server-state-file option","message":"Handle feature compatibility of HAProxy server-state-file option\n\nhttps://review.opendev.org/c/openstack/octavia/+/805955 introduced\na new feature that requires HAProxy 1.6 or higher to work. This change\nensures that with older HAProxy version this feature does not get\nenabled to avoid errors.\n\n Conflicts:\n\toctavia/common/constants.py\n\toctavia/common/jinja/haproxy/combined_listeners/jinja_cfg.py\n\toctavia/tests/unit/common/jinja/haproxy/combined_listeners/test_jinja_cfg.py\n\nStory: 2010254\nTask: 46145\nChange-Id: Icc7dc9bb22187f908852c20415227574ded347aa\n(cherry picked from commit 69ecdb190880d0c49e9f0ef92fd53de14cf66720)\n(cherry picked from commit 7f220d41adaab75471e551f89b0c2b97e830bab7)\n(cherry picked from commit 6346341b8ebb60c33b53f81e8c39f9c24d7de6df)\n(cherry picked from commit de3e6c4ef098f7e0cf29122c56dbedb96252b8d6)\n(cherry picked from commit 5e8331676e043ebaad8998e7c99b9e82f9800642)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/2fa09640153bec78ab50d78bb68f6ba04117a243"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/2fa09640153bec78ab50d78bb68f6ba04117a243"}]},"branch":"refs/heads/stable/victoria"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[]}
