)]}'
{"id":"openstack%2Foctavia~858506","triplet_id":"openstack%2Foctavia~stable%2Ftrain~Icc7dc9bb22187f908852c20415227574ded347aa","project":"openstack/octavia","branch":"stable/train","attention_set":{},"removed_from_attention_set":{"11628":{"account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"last_update":"2022-11-17 13:03:02.000000000","reason":"Change was submitted"},"34429":{"account":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"last_update":"2022-11-17 13:03:02.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Icc7dc9bb22187f908852c20415227574ded347aa","subject":"Handle feature compatibility of HAProxy server-state-file option","status":"MERGED","created":"2022-09-20 09:09:44.000000000","updated":"2022-11-17 13:04:34.000000000","submitted":"2022-11-17 13:03:02.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"858506","meta_rev_id":"425bfa258fa6cd3a5764fb8283843151342b414a","_number":858506,"virtual_id_number":858506,"owner":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-11-17 13:03:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":2,"date":"2022-11-16 17:41:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":2,"date":"2022-11-17 10:05:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"all":[{"value":0,"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":1,"date":"2022-11-17 10:05:37.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-09-20 10:28:44.000000000","updated_by":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"reviewer":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"state":"REVIEWER"},{"updated":"2022-09-21 17:30:37.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-09-22 12:30:32.000000000","updated_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"REVIEWER"},{"updated":"2022-11-16 17:41:36.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"}],"messages":[{"id":"b81319fd41b2903744a971d54aa7abed685fff8b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-09-20 09:09:44.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ae59948a5397a6cf0bc37ef5b67e6cf53e6c7c65","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2022-09-20 10:28:44.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"584bc178505d1d4ad120a6fe1583bfe2635d156d","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-09-21 11:11:50.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"47c6d68533a7dca99a354be72eb71dce5fda4561","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-09-21 14:52:44.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"6b6fc16cf796b9473e84fde52cf23807a23142b6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-21 17:30:37.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bde5b707ee124ce3a40f45e6c3465de9\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/00d9c15bdca74f18b4e12c6c1a0a7c53 : SUCCESS in 9m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1fd9924bf3104556aa563092c1dcc47e : SUCCESS in 6m 41s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/ecd5b71695e94a60acaf860fddfacc57 : SUCCESS in 5m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fad52e3a3d714a61ae4f8c8876aebd50 : SUCCESS in 5m 42s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/fff89babb4c749868c3d4ab897aa09fe : SUCCESS in 6m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3d197b446f544050b2e34f62d29dff70 : SUCCESS in 11m 39s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/df015c3481a74e599aedd1c313ba847e : SUCCESS in 5m 44s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/f4af6cae5bff456f85c63c47040ba80f : SUCCESS in 5m 51s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/e946e811af68431ba02ae09dbbd1e86e : SUCCESS in 8m 56s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/75038c67fb364165902ac7294f329b4c : SUCCESS in 7m 06s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/7676104ecacd4a0192ef804984524bac : SUCCESS in 7m 02s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/486cadcabab34f3bb369acdda4541f59 : SUCCESS in 1h 31m 44s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/48711b9891514b8e8f389652cacf186f : SUCCESS in 1h 18m 15s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/fe574e97889446f0b2bb4b1017c016b3 : SUCCESS in 1h 43m 38s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/a12b40b965b24f32bcd5261f14f3167d : SUCCESS in 2h 13m 38s (non-voting)\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/ad1db0a68cc04dc8863b8b7b7d22b5fa : FAILURE in 20m 28s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/fad4c927fc15449eb3d870c933959249 : RETRY_LIMIT in 10m 49s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/44aff7ba07174f88b8fd779299619243 : RETRY_LIMIT in 9m 29s (non-voting)\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/fac1991dac8a419288f1a45e645720b1 : SUCCESS in 56m 43s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/32ffd60386d3487d9ee0cc35e0151dab : SUCCESS in 2h 26m 20s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/4c8c1e5d96cc44ef80e014e3fd097de0 : FAILURE in 1h 03m 12s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/f33bc409ae88432bad603008c87213f9 : FAILURE in 38m 28s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/7bd3f6c3e93f440cb07d2bdb1775bf4d : FAILURE in 18m 35s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"c5702e768a250ab359939d184b9ab996471651a5","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2022-09-22 12:30:32.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"65a0f1df65c9444476af6b188ec96c6588e4a032","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2022-11-16 17:41:27.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"1760c37755f4f95139e07d7a1bf8a07117f5c676","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2022-11-16 17:41:36.000000000","message":"Patch Set 3: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"8b9d2051620ebd1491fad0cf16920133a70adef3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-16 19:37:47.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4e83c4b6b94147d6b81afd032ceb0ae0\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6baf7a546d9b4169979818cede8df701 : SUCCESS in 8m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0a1660246f65450b9eabd7e009886e35 : SUCCESS in 6m 16s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/110fd2659032402ea1ea39c5c6a0e91c : SUCCESS in 4m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/47bcd039c64b40c1a8429a2061dd024f : SUCCESS in 4m 40s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/abe0d04b2b8a48579e23399c3c52a8c7 : SUCCESS in 5m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/328b15fe02844e1f8a510a29c4fe27f1 : SUCCESS in 11m 42s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9a53d62e835746f2abb48088b402d974 : SUCCESS in 4m 58s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/5b5ef604ba1b4d8783a3a4c3f1c30992 : SUCCESS in 6m 12s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/ec70e4c849eb473b9ea1c613d82ddac0 : SUCCESS in 8m 45s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/7ccb8df361a34426b8f90e6415fc635d : SUCCESS in 6m 14s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/9d917a3c1f664db1bd884932a5ae114c : SUCCESS in 6m 44s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/a41c163db6fa436283e3358b307f44b2 : SUCCESS in 51m 03s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/daa7654e23a44b4790a035bd1d51f24a : SUCCESS in 52m 25s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/963a7e41b71a4a479dd0f2a4422344eb : SUCCESS in 1h 24m 18s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/fa2c90020e054d1abb4fcd2d18256e68 : SUCCESS in 1h 29m 40s (non-voting)\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/c6033cd3a4a04a0a9c5a267ca241c27e : FAILURE in 21m 14s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/70a7d6a4f82a49a7a132c624cb6a73d2 : RETRY_LIMIT in 5m 25s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/ed01ef592ebd46878c5ebeec3ee2246a : RETRY_LIMIT in 5m 53s (non-voting)\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/0b723a7ed59e44cba3d1c6ff92664adb : FAILURE in 26m 20s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/f382ec10f24b48e591ea357897a916f6 : SUCCESS in 1h 53m 26s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/102b161ad26045e5a62b748f3cf310f2 : FAILURE in 29m 34s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/8be553705b9c47aca87128fb491f0a19 : FAILURE in 29m 53s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/4452e867da6741d59895a457401397f4 : FAILURE in 14m 19s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"ee853bc768e72ad57eecbde5182fa427dd8b4437","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2022-11-17 10:05:37.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"b5d62041a3b0caf8aa5c1b9c26782d7d993bca32","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-17 10:06:44.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"e427190045020ebcce14e329ab0aa62befd346c4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-17 13:03:02.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/52b01f50933b4c16a969b564efafc085\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/259d939b5a6948d5acc934a11216f083 : SUCCESS in 6m 46s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/8aa807824f4c4e0cbb55072975095898 : SUCCESS in 5m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/77572e0e2774460197404e8e21b9306b : SUCCESS in 6m 11s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/5e90a6d6eff04862bee1ebcaf86c4335 : SUCCESS in 6m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/481a7352c36947fe8d0c6c18e973b9ce : SUCCESS in 11m 50s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/71b18fde6ed741f891f39b01b77b8102 : SUCCESS in 5m 36s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/629e934b47b54dd6bb52756af84cf107 : SUCCESS in 6m 29s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/94b5120432d04d4dbfc818debc900796 : SUCCESS in 7m 35s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/adbe8a6af9814a4ca4840b5c6278c426 : SUCCESS in 40m 07s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/c09d1d0cb34e4580af65235c988ce33d : SUCCESS in 49m 51s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/8ce615fbe1644a12b58398b554d3f09c : SUCCESS in 1h 28m 41s","accounts_in_message":[],"_revision_number":3},{"id":"cb1bf1b183ea678bae2a3bf585dfb0f2ac135d97","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-17 13:03:02.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"425bfa258fa6cd3a5764fb8283843151342b414a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-17 13:04:34.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/42fe19e6d3df46fabcd6ec6821ef5dcf\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ccb93336f63c44408706a9ed1ed51698 : SUCCESS in 58s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/00d96bedfd6a4fa29fb57200752b5120 : SUCCESS in 42s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"4a54898153df5120f3174da6e002120af8152099","revisions":{"ace0dd52585df11c3ab4139d6cf5b4b157eb764e":{"kind":"REWORK","_number":1,"created":"2022-09-20 09:09:44.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/06/858506/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/06/858506/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/06/858506/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/06/858506/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/06/858506/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/06/858506/1"}}},"commit":{"parents":[{"commit":"2e0a650571065b9a8590e84c2e68d7d948bd6d4d","subject":"Merge \"Apply openstack-selinux policies in Centos amphorae\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/2e0a650571065b9a8590e84c2e68d7d948bd6d4d"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-08-25 10:47:38.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-09-20 09:09:23.000000000","tz":120},"subject":"Handle feature compatibility of HAProxy server-state-file option","message":"Handle feature compatibility of HAProxy server-state-file option\n\nhttps://review.opendev.org/c/openstack/octavia/+/805955 introduced\na new feature that requires HAProxy 1.6 or higher to work. This change\nensures that with older HAProxy version this feature does not get\nenabled to avoid errors.\n\n Conflicts:\n\toctavia/common/constants.py\n\toctavia/common/jinja/haproxy/combined_listeners/templates/haproxy.cfg.j2\n\toctavia/common/jinja/haproxy/combined_listeners/templates/macros.j2\n\toctavia/common/jinja/haproxy/combined_listeners/jinja_cfg.py\n\toctavia/tests/unit/common/jinja/haproxy/combined_listeners/test_jinja_cfg.py\n\nStory: 2010254\nTask: 46145\nChange-Id: Icc7dc9bb22187f908852c20415227574ded347aa\n(cherry picked from commit 69ecdb190880d0c49e9f0ef92fd53de14cf66720)\n(cherry picked from commit 7f220d41adaab75471e551f89b0c2b97e830bab7)\n(cherry picked from commit 6346341b8ebb60c33b53f81e8c39f9c24d7de6df)\n(cherry picked from commit de3e6c4ef098f7e0cf29122c56dbedb96252b8d6)\n(cherry picked from commit 5e8331676e043ebaad8998e7c99b9e82f9800642)\n(cherry picked from commit 3fc9d3923c6330c8fe4d39ed225179ad5802e076)\n(cherry picked from commit f9910f866ecf683e45d9e61643b01808af5d8d3b)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/ace0dd52585df11c3ab4139d6cf5b4b157eb764e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/ace0dd52585df11c3ab4139d6cf5b4b157eb764e"}]},"branch":"refs/heads/stable/train"},"367ca4106b04122e61e3774db6123f899f0fd563":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2022-09-21 14:52:44.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/06/858506/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/06/858506/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/06/858506/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/06/858506/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/06/858506/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/06/858506/2"}}},"commit":{"parents":[{"commit":"2e0a650571065b9a8590e84c2e68d7d948bd6d4d","subject":"Merge \"Apply openstack-selinux policies in Centos amphorae\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/2e0a650571065b9a8590e84c2e68d7d948bd6d4d"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-08-25 10:47:38.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-09-21 14:52:44.000000000","tz":0},"subject":"Handle feature compatibility of HAProxy server-state-file option","message":"Handle feature compatibility of HAProxy server-state-file option\n\nhttps://review.opendev.org/c/openstack/octavia/+/805955 introduced\na new feature that requires HAProxy 1.6 or higher to work. This change\nensures that with older HAProxy version this feature does not get\nenabled to avoid errors.\n\n Conflicts:\n\toctavia/common/constants.py\n\toctavia/common/jinja/haproxy/combined_listeners/templates/haproxy.cfg.j2\n\toctavia/common/jinja/haproxy/combined_listeners/templates/macros.j2\n\toctavia/common/jinja/haproxy/combined_listeners/jinja_cfg.py\n\toctavia/tests/unit/common/jinja/haproxy/combined_listeners/test_jinja_cfg.py\n\nDepends-on: Ie02b5164f3c1089f0d48101aab9e007bb4d132ea\n\nStory: 2010254\nTask: 46145\nChange-Id: Icc7dc9bb22187f908852c20415227574ded347aa\n(cherry picked from commit 69ecdb190880d0c49e9f0ef92fd53de14cf66720)\n(cherry picked from commit 7f220d41adaab75471e551f89b0c2b97e830bab7)\n(cherry picked from commit 6346341b8ebb60c33b53f81e8c39f9c24d7de6df)\n(cherry picked from commit de3e6c4ef098f7e0cf29122c56dbedb96252b8d6)\n(cherry picked from commit 5e8331676e043ebaad8998e7c99b9e82f9800642)\n(cherry picked from commit 3fc9d3923c6330c8fe4d39ed225179ad5802e076)\n(cherry picked from commit f9910f866ecf683e45d9e61643b01808af5d8d3b)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/367ca4106b04122e61e3774db6123f899f0fd563"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/367ca4106b04122e61e3774db6123f899f0fd563"}]},"branch":"refs/heads/stable/train","description":"Edit commit message"},"4a54898153df5120f3174da6e002120af8152099":{"kind":"REWORK","_number":3,"created":"2022-11-16 17:41:27.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/06/858506/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/06/858506/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/06/858506/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/06/858506/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/06/858506/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/06/858506/3"}}},"commit":{"parents":[{"commit":"2e0a650571065b9a8590e84c2e68d7d948bd6d4d","subject":"Merge \"Apply openstack-selinux policies in Centos amphorae\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/2e0a650571065b9a8590e84c2e68d7d948bd6d4d"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-08-25 10:47:38.000000000","tz":120},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2022-11-16 17:41:27.000000000","tz":0},"subject":"Handle feature compatibility of HAProxy server-state-file option","message":"Handle feature compatibility of HAProxy server-state-file option\n\nhttps://review.opendev.org/c/openstack/octavia/+/805955 introduced\na new feature that requires HAProxy 1.6 or higher to work. This change\nensures that with older HAProxy version this feature does not get\nenabled to avoid errors.\n\n Conflicts:\n\toctavia/common/constants.py\n\toctavia/common/jinja/haproxy/combined_listeners/templates/haproxy.cfg.j2\n\toctavia/common/jinja/haproxy/combined_listeners/templates/macros.j2\n\toctavia/common/jinja/haproxy/combined_listeners/jinja_cfg.py\n\toctavia/tests/unit/common/jinja/haproxy/combined_listeners/test_jinja_cfg.py\n\nDepends-on: Ie02b5164f3c1089f0d48101aab9e007bb4d132ea\n\nStory: 2010254\nTask: 46145\nChange-Id: Icc7dc9bb22187f908852c20415227574ded347aa\n(cherry picked from commit 69ecdb190880d0c49e9f0ef92fd53de14cf66720)\n(cherry picked from commit 7f220d41adaab75471e551f89b0c2b97e830bab7)\n(cherry picked from commit 6346341b8ebb60c33b53f81e8c39f9c24d7de6df)\n(cherry picked from commit de3e6c4ef098f7e0cf29122c56dbedb96252b8d6)\n(cherry picked from commit 5e8331676e043ebaad8998e7c99b9e82f9800642)\n(cherry picked from commit 3fc9d3923c6330c8fe4d39ed225179ad5802e076)\n(cherry picked from commit f9910f866ecf683e45d9e61643b01808af5d8d3b)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/4a54898153df5120f3174da6e002120af8152099"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/4a54898153df5120f3174da6e002120af8152099"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[]}
