)]}'
{"id":"openstack%2Foctavia~858743","triplet_id":"openstack%2Foctavia~stable%2Fxena~Ie02b5164f3c1089f0d48101aab9e007bb4d132ea","project":"openstack/octavia","branch":"stable/xena","attention_set":{},"removed_from_attention_set":{"34429":{"account":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"last_update":"2022-09-21 18:44:47.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ie02b5164f3c1089f0d48101aab9e007bb4d132ea","subject":"Update zuul queue configuration","status":"MERGED","created":"2022-09-21 14:34:34.000000000","updated":"2022-09-21 18:47:00.000000000","submitted":"2022-09-21 18:44:47.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"858743","meta_rev_id":"58b959d3fe600ce044ab51a6b9542bb84e0da15d","_number":858743,"virtual_id_number":858743,"owner":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":33408,"name":"Pierre-Yves Jourel","email":"pierre-yves.jourel@ovhcloud.com","username":"PYJ"},{"value":0,"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-09-21 18:44:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"recommended":{"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},"all":[{"value":1,"date":"2022-09-21 16:15:21.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":33408,"name":"Pierre-Yves Jourel","email":"pierre-yves.jourel@ovhcloud.com","username":"PYJ"},{"value":1,"date":"2022-09-21 16:15:05.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},{"value":2,"date":"2022-09-21 14:49:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":2,"date":"2022-09-21 16:14:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"_account_id":33408,"name":"Pierre-Yves Jourel","email":"pierre-yves.jourel@ovhcloud.com","username":"PYJ"},{"value":0,"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":1,"date":"2022-09-21 16:14:58.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":33408,"name":"Pierre-Yves Jourel","email":"pierre-yves.jourel@ovhcloud.com","username":"PYJ"},{"value":0,"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"_account_id":33408,"name":"Pierre-Yves Jourel","email":"pierre-yves.jourel@ovhcloud.com","username":"PYJ"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-09-21 14:34:34.000000000","updated_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"CC"},{"updated":"2022-09-21 14:49:15.000000000","updated_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"REVIEWER"},{"updated":"2022-09-21 16:14:58.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2022-09-21 16:15:05.000000000","updated_by":{"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},"reviewer":{"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},"state":"REVIEWER"},{"updated":"2022-09-21 16:15:21.000000000","updated_by":{"_account_id":33408,"name":"Pierre-Yves Jourel","email":"pierre-yves.jourel@ovhcloud.com","username":"PYJ"},"reviewer":{"_account_id":33408,"name":"Pierre-Yves Jourel","email":"pierre-yves.jourel@ovhcloud.com","username":"PYJ"},"state":"REVIEWER"},{"updated":"2022-09-21 16:30:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"8114cc82ca9431f0ff520232273993430047a105","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-09-21 14:34:34.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"255fa7d9529d4856fd3075980340923f257350f4","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2022-09-21 14:49:15.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"b12981c805b36376b1d5c87225525b0bcdea032c","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2022-09-21 16:14:58.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"dc52eca14a5dfbf7c15330041b3998022ddc4c5d","author":{"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},"date":"2022-09-21 16:15:05.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"bccdec41f26d1ff0c08ede6eb4f919c88ff7e038","author":{"_account_id":33408,"name":"Pierre-Yves Jourel","email":"pierre-yves.jourel@ovhcloud.com","username":"PYJ"},"date":"2022-09-21 16:15:21.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"1981b86c9a0b11c2876cb9ad30e21fb1a412e54d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-21 16:30:30.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ed0fb87148084aeb860ecaafa2d03a2e\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/836a19bc958f4b9a82bd5ebfc1b50cb0 : SUCCESS in 3m 36s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1f4eeb56e0c54dd4b4fc1ce3bdf8f959 : SUCCESS in 10m 53s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/37e1c06a7e9e4bd7aa1614eb59fbfe7a : SUCCESS in 9m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/30d7ef2052f84debac9388c69d5a3a65 : SUCCESS in 6m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5b86d292d1784e509bdcd8cfeb080436 : SUCCESS in 6m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2842c8fad234411c82589cd91d04fbb4 : SUCCESS in 5m 44s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9da0371218f343eca2fc6e0260210eaa : SUCCESS in 6m 23s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8a5226191a654ca09df0e4fffc944a30 : SUCCESS in 14m 11s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/54717b109e1b459394c897e88e5d10af : SUCCESS in 7m 07s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/561b3d2c475c4893b0aeb7c190e2a1d0 : SUCCESS in 7m 44s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/5d05bf45700f4fa0a8e576288378af7b : SUCCESS in 10m 02s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/856d311d6f0246f7b7105bd8c048e70f : SUCCESS in 6m 52s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/576c9f9f8cdc46f6af59432afad4b89a : SUCCESS in 9m 47s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/40d344370c2e4958915cfd0ed85b3768 : SUCCESS in 1h 32m 07s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/27f6d52d5b0d48f6871a4258c86b3302 : SUCCESS in 1h 39m 24s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/cb2204233f874a3e8a92e57d253df0e3 : SUCCESS in 47m 32s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/f4d8b1c67c004e6494974b2eed15ce02 : SUCCESS in 59m 34s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/01ed7f044a68417393c8a155fc8e0f91 : RETRY_LIMIT in 14m 16s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/8cace5e953a74e85b6b34717d54cf82c : SUCCESS in 43m 03s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/aab8d0172bfb4f8e929582f75970520e : SUCCESS in 1h 41m 25s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/b3e2eea80b6d44c287a55375e69425ab : SUCCESS in 1h 52m 59s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/d79cd6d2751848229852e2f6e7e5749e : SUCCESS in 1h 41m 16s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/8236207b36904a69a694014174f7efeb : SUCCESS in 1h 38m 13s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/6bb18c2673a84d54a4b0143f8ac8ff32 : SUCCESS in 1h 28m 48s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/5023c65f4df84cb6b099aa0842f9abf6 : NODE_FAILURE in 0s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d4fcc9daf207bb159270910b0d5623446c5cccf9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-21 16:31:13.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"0fadb83f4ad2d69b4f6f348eda68ff144d183e1d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-21 18:44:47.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8a1c28f5957d4c21af97131a80ebc7e3\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/b5eebd74cda0481c9cf116d220d1c614 : SUCCESS in 2m 41s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/057db2fbe5a0444482211bbdd59c5264 : SUCCESS in 7m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b0d465e903794bcca024c882f14cc68b : SUCCESS in 5m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6b8cf74c406946c3a6202b096c9df5bf : SUCCESS in 5m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/27ac869d008d4339aee665ff4efb5cc5 : SUCCESS in 5m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f757b4df3e0f414d8acbb178b94da496 : SUCCESS in 10m 40s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/18cb3be0dc534c478a0654952e74f2f3 : SUCCESS in 5m 04s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/0a6fd20abcf24cde90d4a7dfe45b96a5 : SUCCESS in 5m 08s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/43b7378dcd8a4e4394e150a44bbb0e36 : SUCCESS in 7m 14s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/3670a0a8667142a48fb40b10e08ee019 : SUCCESS in 1h 24m 06s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/4e8706b0179e4e33809ff545bdb38cf5 : SUCCESS in 2h 11m 37s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/95111db48aad4a21afc833b6cc574e8b : SUCCESS in 52m 13s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/36d453710058419ba75b525bcbf14032 : SUCCESS in 1h 18m 48s","accounts_in_message":[],"_revision_number":1},{"id":"d2d18e3f692f2b03f0e3bdae6b40d241fe01c27e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-21 18:44:47.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"58b959d3fe600ce044ab51a6b9542bb84e0da15d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-21 18:47:00.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fdd26a93b1c24f6d9c46d0cc7e572c8b\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a46447634b3d4261a767327a12fd4a68 : SUCCESS in 56s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"cde823eb631a4bfae93825580e53b0ee69d1102d","revisions":{"cde823eb631a4bfae93825580e53b0ee69d1102d":{"kind":"REWORK","_number":1,"created":"2022-09-21 14:34:34.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/43/858743/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/43/858743/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/43/858743/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/43/858743/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/43/858743/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/43/858743/1"}}},"commit":{"parents":[{"commit":"28a6f28702399561ea5ddd99ce55d540e9b0f02c","subject":"Merge \"Apply openstack-selinux policies in Centos amphorae\" into stable/xena","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/28a6f28702399561ea5ddd99ce55d540e9b0f02c"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2022-05-19 16:18:02.000000000","tz":0},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-09-21 14:34:26.000000000","tz":120},"subject":"Update zuul queue configuration","message":"Update zuul queue configuration\n\nZuul is changing how queues are declared[1]. This patch updates Octavia for the new way to define the queue.\n\n[1] https://lists.zuul-ci.org/pipermail/zuul-discuss/2022-May/001801.html\n\nChange-Id: Ie02b5164f3c1089f0d48101aab9e007bb4d132ea\n(cherry picked from commit 81105bbaed9e83e98c46e70c02b1bd43c4a9d875)\n(cherry picked from commit 62e1d1fa18c9d0800dcc9534eda39cf608f11327)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/cde823eb631a4bfae93825580e53b0ee69d1102d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/cde823eb631a4bfae93825580e53b0ee69d1102d"}]},"branch":"refs/heads/stable/xena"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[]}
