)]}'
{"id":"openstack%2Foctavia~858744","triplet_id":"openstack%2Foctavia~stable%2Fwallaby~Ie02b5164f3c1089f0d48101aab9e007bb4d132ea","project":"openstack/octavia","branch":"stable/wallaby","attention_set":{},"removed_from_attention_set":{"34429":{"account":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"last_update":"2022-09-21 20:55:42.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ie02b5164f3c1089f0d48101aab9e007bb4d132ea","subject":"Update zuul queue configuration","status":"MERGED","created":"2022-09-21 14:34:42.000000000","updated":"2022-09-21 20:58:40.000000000","submitted":"2022-09-21 20:55:42.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"858744","meta_rev_id":"de29f08420aaa93bb58841faef704db093424cf3","_number":858744,"virtual_id_number":858744,"owner":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":33408,"name":"Pierre-Yves Jourel","email":"pierre-yves.jourel@ovhcloud.com","username":"PYJ"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-09-21 20:55:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"recommended":{"_account_id":33408,"name":"Pierre-Yves Jourel","email":"pierre-yves.jourel@ovhcloud.com","username":"PYJ"},"all":[{"value":1,"date":"2022-09-21 16:15:13.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":33408,"name":"Pierre-Yves Jourel","email":"pierre-yves.jourel@ovhcloud.com","username":"PYJ"},{"value":2,"date":"2022-09-21 14:49:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":2,"date":"2022-09-21 16:15:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"_account_id":33408,"name":"Pierre-Yves Jourel","email":"pierre-yves.jourel@ovhcloud.com","username":"PYJ"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":1,"date":"2022-09-21 16:15:05.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":33408,"name":"Pierre-Yves Jourel","email":"pierre-yves.jourel@ovhcloud.com","username":"PYJ"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"_account_id":33408,"name":"Pierre-Yves Jourel","email":"pierre-yves.jourel@ovhcloud.com","username":"PYJ"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-09-21 14:34:42.000000000","updated_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"CC"},{"updated":"2022-09-21 14:49:29.000000000","updated_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"REVIEWER"},{"updated":"2022-09-21 16:15:05.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2022-09-21 16:15:13.000000000","updated_by":{"_account_id":33408,"name":"Pierre-Yves Jourel","email":"pierre-yves.jourel@ovhcloud.com","username":"PYJ"},"reviewer":{"_account_id":33408,"name":"Pierre-Yves Jourel","email":"pierre-yves.jourel@ovhcloud.com","username":"PYJ"},"state":"REVIEWER"},{"updated":"2022-09-21 17:13:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"b7db657c1c92732389b64bce4a4075a240494c27","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-09-21 14:34:42.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"83b307e2f776a5e77b8bfeb769fc3920f9296b84","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2022-09-21 14:49:29.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"2f171192b08ef95f06511d1a2fde792c251e04d1","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2022-09-21 16:15:05.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"80f3b045f22345e68ae6d7fe7246ca608856bc1b","author":{"_account_id":33408,"name":"Pierre-Yves Jourel","email":"pierre-yves.jourel@ovhcloud.com","username":"PYJ"},"date":"2022-09-21 16:15:13.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"fc88ecfa88af2053efc29e6c8c771de01ff1a262","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-21 17:13:53.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b0e5b4daac1344989fc754cb0269b9a3\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/d917c57a6f9146738308d1c6d2076585 : SUCCESS in 3m 44s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b2bae470d4b0469b8b5c475927c73b59 : SUCCESS in 10m 20s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2e222ee8b8be41659b24fcd505131517 : SUCCESS in 9m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/30ad9fb62d4b47c9874f2f9957b1ccef : SUCCESS in 6m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/26e8a9a8e85f42b38f5f72ab6952c9b8 : SUCCESS in 7m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5b6e8df63971427899392a1e6c7691c2 : SUCCESS in 6m 16s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/93ce1de5487647c7a04bb0d559c46657 : SUCCESS in 6m 15s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4f1532b4228a43f2afc386216bcee580 : SUCCESS in 12m 58s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5b1522bacf654530bc4f76b3cbe213be : SUCCESS in 6m 40s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/bade5ce993e74e75817542c23df5f0c7 : SUCCESS in 7m 08s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/b7f39cd6f7304452b9f101a805cd623f : SUCCESS in 8m 45s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/8e162343e95d418c8aaff454ad784d0c : SUCCESS in 5m 42s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/db123c4a677e4a8bab06f704114fb1c8 : SUCCESS in 7m 59s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/11442e12e68a4184a9f9113de8b7906d : SUCCESS in 1h 28m 28s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/8c30d6a1a9c548c1b369f2e628856808 : SUCCESS in 1h 41m 18s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/9b3dffd2481e4a598ff761b1c7677093 : SUCCESS in 45m 04s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/248b689b54bd43f5a6a0758a672c7f26 : SUCCESS in 43m 26s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/d51f9d9a331740af97b90d6adc127ac0 : SUCCESS in 40m 00s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/6c7ce6178d3d4ac09d1a1a84e11081b3 : RETRY_LIMIT in 13m 54s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/9c284762b14d46318c38e09504dcd0a7 : SUCCESS in 1h 28m 27s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/966749e1682546a89730db208b6ea22d : FAILURE in 37m 37s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/c645afb3b68f46a3b7dba0cd079fe8b8 : SUCCESS in 1h 34m 39s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/ba4bdba99d7245d5a7143a038bb51de0 : SUCCESS in 1h 28m 06s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/64955d7d6e6244f9a8a82f9add5d9846 : SUCCESS in 2h 25m 30s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/86d7be01db8a4207a811307cccbd16df : NODE_FAILURE in 0s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"5a3abdc4154b30847cecd2f8c38c78ec2e538119","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-21 17:15:00.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"8a433eab516f5128995934e1d932d2126b24b0ae","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-21 20:55:42.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7dc9f9a68f4c4040bddefc3c01196434\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/c1d2e6032bda4319b0d8653b9fee0a5b : SUCCESS in 5m 47s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a9ecdbd8cabc4087b176bd90f309f8a0 : SUCCESS in 8m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/30c4605e26f7453486583dbc29a3f9c4 : SUCCESS in 4m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9a98274ecc544da9b1bb06ee95a7f10e : SUCCESS in 6m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c0ac65af26ba4523981129345761bc4b : SUCCESS in 5m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/31b031aa07f748abb2ec9b938cfc874d : SUCCESS in 10m 50s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2b3ec79b9c2a4dca92c1b23c04dccc37 : SUCCESS in 6m 27s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/94100e5d197d4120b90d654e8da9513e : SUCCESS in 5m 11s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/d822ef84a2ad45c5b909290909d40e5a : SUCCESS in 7m 24s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/55e37edb4371429a834f64c98a77af6d : SUCCESS in 52m 53s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/03abc81dcfc74623b651c48becfa3e08 : SUCCESS in 1h 25m 54s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/c44132428f6f4273868580bee3c315c0 : SUCCESS in 40m 32s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/6bc4222d2b3540b9ba7812dfea9dfc4c : SUCCESS in 39m 25s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/771fd489d6d742f5991d43ac62ba95b5 : SUCCESS in 34m 57s","accounts_in_message":[],"_revision_number":1},{"id":"03ea48269c6d05e3d1389ff7efec8688d60f89b4","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-21 20:55:42.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"de29f08420aaa93bb58841faef704db093424cf3","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-21 20:58:40.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/af4d0247967f47c486f932b35fa9a358\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/21d4b6b0f19e4afa87f31549cdb1d20b : SUCCESS in 47s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"1cbc0d75aa98ccf495adaa12bb10e4d9f330edd6","revisions":{"1cbc0d75aa98ccf495adaa12bb10e4d9f330edd6":{"kind":"REWORK","_number":1,"created":"2022-09-21 14:34:42.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/44/858744/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/44/858744/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/44/858744/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/44/858744/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/44/858744/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/44/858744/1"}}},"commit":{"parents":[{"commit":"d1877698badc296654631c256bcf109c31086eb4","subject":"Merge \"Apply openstack-selinux policies in Centos amphorae\" into stable/wallaby","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d1877698badc296654631c256bcf109c31086eb4"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2022-05-19 16:18:02.000000000","tz":0},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-09-21 14:34:35.000000000","tz":120},"subject":"Update zuul queue configuration","message":"Update zuul queue configuration\n\nZuul is changing how queues are declared[1]. This patch updates Octavia for the new way to define the queue.\n\n[1] https://lists.zuul-ci.org/pipermail/zuul-discuss/2022-May/001801.html\n\nChange-Id: Ie02b5164f3c1089f0d48101aab9e007bb4d132ea\n(cherry picked from commit 81105bbaed9e83e98c46e70c02b1bd43c4a9d875)\n(cherry picked from commit 62e1d1fa18c9d0800dcc9534eda39cf608f11327)\n(cherry picked from commit cde823eb631a4bfae93825580e53b0ee69d1102d)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/1cbc0d75aa98ccf495adaa12bb10e4d9f330edd6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/1cbc0d75aa98ccf495adaa12bb10e4d9f330edd6"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[]}
