)]}'
{"id":"openstack%2Foctavia~859710","triplet_id":"openstack%2Foctavia~master~Id328b27db1ec9c8e17bf18120259e41e75dab3b9","project":"openstack/octavia","branch":"master","topic":"story/2010335","attention_set":{},"removed_from_attention_set":{"28722":{"account":{"_account_id":28722,"name":"Anton Kurbatov","email":"anton.kurbatov@acronis.com","username":"akurbatov"},"last_update":"2022-11-10 15:14:45.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Id328b27db1ec9c8e17bf18120259e41e75dab3b9","subject":"Fix full graph loadbalancer creation if jobboard is disabled","status":"MERGED","created":"2022-09-28 19:08:27.000000000","updated":"2022-11-10 15:16:09.000000000","submitted":"2022-11-10 15:14:45.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"859710-story/2010335","meta_rev_id":"a4a307cc7b235e1fa7448782748e30cd20f8204a","_number":859710,"virtual_id_number":859710,"owner":{"_account_id":28722,"name":"Anton Kurbatov","email":"anton.kurbatov@acronis.com","username":"akurbatov"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-11-10 15:14:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"recommended":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"all":[{"value":1,"date":"2022-11-03 10:02:17.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":2,"date":"2022-11-04 14:53:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-11-10 13:17:34.000000000","_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"all":[{"value":0,"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-11-10 13:17:34.000000000","_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-09-28 21:29:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-09-30 06:50:10.000000000","updated_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"REVIEWER"},{"updated":"2022-11-03 10:02:17.000000000","updated_by":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"reviewer":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"state":"REVIEWER"},{"updated":"2022-11-10 13:17:34.000000000","updated_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"reviewer":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"state":"REVIEWER"}],"messages":[{"id":"91b022429ab5b7f463e707b16c5aa6ccde03d7fc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28722,"name":"Anton Kurbatov","email":"anton.kurbatov@acronis.com","username":"akurbatov"},"date":"2022-09-28 19:08:27.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"de06b6c5c2c81e057496a669e5d96d6da4d31212","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-28 21:29:53.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/2be22cfa488e466b9337835f65667d0b\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/eebc5ab967d84b1abb0b83edfa5575c1 : SUCCESS in 10m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ba5419edd44f4e92a13e563457e6b338 : FAILURE in 4m 52s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7e1c88859f0b4019aa292937baba392d : SUCCESS in 5m 52s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2933dbc6f56b42dcaa0726d605eca0a2 : SUCCESS in 6m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/46c45775e7994382bbf8141ba6030720 : SUCCESS in 15m 39s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1be4d31310e742e99b193fda34a6c8e8 : SUCCESS in 5m 24s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/3dde9288f37846128881d1397c624544 : SUCCESS in 5m 56s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/2e7393d703d54ca295e950ff1784100d : SUCCESS in 9m 10s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/94ddc2e20ea24579882b79bf451b3862 : SUCCESS in 5m 03s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/dfbf4df1167d4c84aadb0393a5073ca7 : SUCCESS in 9m 01s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/4d66f29613314e5bbe650de92d8cb869 : SUCCESS in 9m 54s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/2f6bc423deba4220a86734a7651e149c : SUCCESS in 11m 23s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/5f59be1d145549ba8fae91498fd1c78e : SUCCESS in 1h 54m 38s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/38ed517111484547a9d03ef6d7628b35 : SUCCESS in 1h 48m 26s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/c2d1bf33c6774588863f269ba8b8a508 : SUCCESS in 46m 44s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/55cbd07f50314dc7a51fe4adc7a3feca : SUCCESS in 57m 00s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/694773dc750b4b7182465007916d1afd : SUCCESS in 1h 21m 20s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/4a3b4189c27042a7a875a946b7691460 : SUCCESS in 41m 02s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/f6546307823b4b448fb907070f9d2f96 : SUCCESS in 1h 59m 09s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/64d8e948e6bc44b5896359b751b11c45 : SUCCESS in 2h 20m 01s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/65bfacb8a33e4598b0d2a567b0a28af5 : SUCCESS in 1h 59m 23s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/3916bb9648a4473f93790741f2163d11 : SUCCESS in 1h 56m 44s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/861b12f9758a43808b6a959365b5b692 : SUCCESS in 1h 40m 07s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/3e1b536fd6aa45a9b393299bbae9f6bf : SUCCESS in 2h 11m 47s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"e6eb568b916c1bbc32bdc3c1c2b8a2e77d34719d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28722,"name":"Anton Kurbatov","email":"anton.kurbatov@acronis.com","username":"akurbatov"},"date":"2022-09-29 07:39:01.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"e825386005e79f90a37cbfaf2879c83100ee4fb1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-29 10:54:37.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/6a4ace97735e48648643e1f6e6fd948a\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f1715970179c4fa98493ac983878cf54 : SUCCESS in 10m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/17cf0c66904f4551a0340eadfd3a9c7a : SUCCESS in 6m 19s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ffa4717404a94c1894d525e3756bbcca : SUCCESS in 5m 56s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/eb06fc00de484604810b5a2a0d903d8f : SUCCESS in 5m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/266838ddb3c94f0c8784338846ebf8a1 : SUCCESS in 12m 14s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e87cf1ea2b994a0896fdaa72b7de149a : SUCCESS in 5m 35s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/d7d0a375810b4948bf9432a0e6c29458 : SUCCESS in 5m 32s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/5e92c83997094ab6a94c630edf06eed8 : SUCCESS in 7m 52s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/485a3ed1fbd24bda9a6c9b681d23a10c : SUCCESS in 5m 06s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/0bf5164cb4294417824b0d38f23e5854 : SUCCESS in 7m 22s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/88c986b8adf94566af04e2986cfba083 : SUCCESS in 8m 03s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/7dccab433b6142bcae877f520b0ee1b4 : SUCCESS in 8m 34s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/e452b109edd9468bbeffba25973c8e82 : SUCCESS in 1h 37m 47s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/978fc8d8767540d8820ed0648dacf6bc : TIMED_OUT in 2h 39m 48s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/08e9da6f165a4c89a9e34361226fbfe8 : SUCCESS in 53m 44s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/e0092004806d4984ac4a6e8d73813b54 : SUCCESS in 1h 24m 28s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/ef1ca257a19846d3b97c63acb748fe47 : SUCCESS in 1h 39m 10s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/05fd7ba7b89a4ec496c6c030f61cfb11 : SUCCESS in 51m 40s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/40a77c52e9df47c48a746d687ebd0f4e : SUCCESS in 2h 26m 08s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/c3614d8194984c5a8cf428b3c902c1b3 : TIMED_OUT in 3h 09m 40s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/9dcfc55b86be401c85be3e61fe46e669 : SUCCESS in 2h 15m 34s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/7a75864b1a8045d1908cd83ad7a03335 : SUCCESS in 2h 39m 05s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/13a18eca37c542cda40bcc55860aa4c8 : SUCCESS in 2h 13m 20s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/4997232c3c2c4271b8cf7dee95c961ce : SUCCESS in 2h 12m 09s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"65b14e457dcca318911e30a5b62f06c095c85545","author":{"_account_id":28722,"name":"Anton Kurbatov","email":"anton.kurbatov@acronis.com","username":"akurbatov"},"date":"2022-09-29 12:40:32.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"350e1a1d59566b95b828d7af61c4b5c375c29090","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-29 15:20:41.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9b273b39175b40f193f4efe12e8ba965\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/76b2f4daf9dc4111953cb7cb72e026c6 : SUCCESS in 11m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0823bd2f946d464eb5fe09c1426a9fca : SUCCESS in 8m 02s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d9191025ea1944b4a9c8eddb6ca67217 : SUCCESS in 6m 58s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0f4f19336d554a958ad92cd2b31a4ec9 : SUCCESS in 5m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/468e75ed9c014a5f82633537fc258971 : SUCCESS in 11m 44s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2c7661eb0a274ef39a201b644f3f5eef : SUCCESS in 5m 28s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/4378492af361432380df9fb0763a738a : SUCCESS in 6m 05s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/16901fe0e894473b95e0dfeb5dee73fa : SUCCESS in 8m 54s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/370f295041db4d1581753435f3398024 : SUCCESS in 5m 05s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/6fb7fde056eb468f93e31c41764a191c : SUCCESS in 7m 44s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/78ed1ee2a3bc48e9a16a8b852238d671 : SUCCESS in 8m 43s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/93149ff5a2b646febd267390c886b29f : SUCCESS in 8m 13s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/1fe98b36b44f4c57881d6811670c3b57 : SUCCESS in 1h 25m 22s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/245cc42a13464419b53b7ff3d0c71e73 : SUCCESS in 2h 07m 25s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/c0cdd19bc46f41a3af419a0e65ce3c34 : SUCCESS in 44m 19s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/1bd045388ee348849b3d07a18ed0a816 : SUCCESS in 1h 04m 38s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/642c217a28a741a3a1b74ea525ea7214 : SUCCESS in 1h 24m 35s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/840eee0e1701436db6777ff35c847c28 : SUCCESS in 39m 20s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/d80f96a88d884948a87cc748c04fd7cb : SUCCESS in 2h 12m 58s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/8c6044f2e8444b49903ef980cf21e169 : SUCCESS in 2h 23m 19s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/055f4f6a9a4a4e14a993262fb99408d3 : SUCCESS in 2h 20m 39s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/412a62776cbe4589b2c5ed2cafc55df1 : SUCCESS in 1h 37m 51s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/8b3aeb39150044659c96134251ed9412 : SUCCESS in 1h 37m 56s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/d9fa80e82b704d79862d17f736acb535 : TIMED_OUT in 2h 37m 06s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"bdd14c2b7e96f60396af3593659cf7a0bdb8ac25","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2022-09-30 06:50:10.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"30b9ec4bd593c6e1991aab4134b530cb06dd675a","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2022-11-03 10:02:17.000000000","message":"Patch Set 2: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"494ebf0c02a92b02cb24419e39937010bc45c40d","tag":"autogenerated:gerrit:deleteVote","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2022-11-04 14:26:08.000000000","message":"Removed Code-Review+2 by \u003cGERRIT_ACCOUNT_29244\u003e\n","accounts_in_message":[{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}],"_revision_number":2},{"id":"83fbb7aa576a8f3df0a549b2da0b088577cc2f7d","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2022-11-04 14:53:24.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"84c1109389a90bd15d9250da8ba6263d52bf7bfb","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-11-10 13:17:34.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"05b7b04b1b8d04cd119a75c4a90bde2302de004d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-10 13:18:09.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"fb428151e3c2efb20264f39b5055f35382d329ed","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-10 15:14:45.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4a0cb60a0568447e81699f9abfa6f787\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/43da30badb0c4fe2b7bbc8247c0eefa4 : SUCCESS in 6m 20s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/be79a67944f5475cab16ba50ec7f85e8 : SUCCESS in 4m 54s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6b16748ee42648aa826d8afa1b593b9a : SUCCESS in 5m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1deee4848aa948bd83d519ae6992e6fe : SUCCESS in 11m 21s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/918bd89a2b1241a8ab97829342241e32 : SUCCESS in 5m 26s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/d25e4998143f4309b40c2bd5e6ccc71c : SUCCESS in 4m 46s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/f8a0d757de064a9fba790d2aaa6c2ef8 : SUCCESS in 7m 30s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/eae5273533ce40908f896238b9b2cc26 : SUCCESS in 1h 49m 27s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/c1b752e633ab4386938e1813f31c277d : SUCCESS in 1h 39m 22s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/f1f936b2b6e048c495eed45a0606d5d8 : SUCCESS in 42m 31s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/2eb6b6ba89684b388c249195835e629f : SUCCESS in 48m 32s","accounts_in_message":[],"_revision_number":2},{"id":"15a5247af7996a34b7ad71865213d3949071e5bf","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-10 15:14:45.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"a4a307cc7b235e1fa7448782748e30cd20f8204a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-10 15:16:09.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/19533e1e8fea45fc8ca6707fe932f1fe\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/87bb8877cc1d4fa9be1e8a5d90241318 : SUCCESS in 1m 04s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/07eaeaaa44284b99812c3b5c25a014aa : SUCCESS in 42s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"216cce39c9c22831d725703f41c54b9305d65dc8","revisions":{"aa716d5b1bbd213074fbe148f51690b37216dc04":{"kind":"REWORK","_number":1,"created":"2022-09-28 19:08:27.000000000","uploader":{"_account_id":28722,"name":"Anton Kurbatov","email":"anton.kurbatov@acronis.com","username":"akurbatov"},"ref":"refs/changes/10/859710/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/10/859710/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/10/859710/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/10/859710/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/10/859710/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/10/859710/1"}}},"commit":{"parents":[{"commit":"409c4bca77287c2d9207001db2d7073d7e402fc5","subject":"Merge \"Switch to 2023.1 Python3 unit tests and generic template name\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/409c4bca77287c2d9207001db2d7073d7e402fc5"}]}],"author":{"name":"Anton Kurbatov","email":"Anton.Kurbatov@acronis.com","date":"2022-09-28 18:48:45.000000000","tz":0},"committer":{"name":"Anton Kurbatov","email":"Anton.Kurbatov@acronis.com","date":"2022-09-28 19:05:06.000000000","tz":0},"subject":"Fix full graph loadbalancer creation if jobboard is disabled","message":"Fix full graph loadbalancer creation if jobboard is disabled\n\nThis patch fixes the creation of a full loadbalancer graph in case the\njobboard_enabled config value is False, when no \"listeners\" parameter was\npassed to the get_create_load_balancer_flow function call: use kwargs as the\nflow creation function parameters and handle \"store\" separately.\n\nStory 2010335\nTask 46462\n\nSigned-off-by: Anton Kurbatov \u003cAnton.Kurbatov@acronis.com\u003e\nChange-Id: Id328b27db1ec9c8e17bf18120259e41e75dab3b9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/aa716d5b1bbd213074fbe148f51690b37216dc04"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/aa716d5b1bbd213074fbe148f51690b37216dc04"}]},"branch":"refs/heads/master"},"216cce39c9c22831d725703f41c54b9305d65dc8":{"kind":"REWORK","_number":2,"created":"2022-09-29 07:39:01.000000000","uploader":{"_account_id":28722,"name":"Anton Kurbatov","email":"anton.kurbatov@acronis.com","username":"akurbatov"},"ref":"refs/changes/10/859710/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/10/859710/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/10/859710/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/10/859710/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/10/859710/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/10/859710/2"}}},"commit":{"parents":[{"commit":"409c4bca77287c2d9207001db2d7073d7e402fc5","subject":"Merge \"Switch to 2023.1 Python3 unit tests and generic template name\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/409c4bca77287c2d9207001db2d7073d7e402fc5"}]}],"author":{"name":"Anton Kurbatov","email":"Anton.Kurbatov@acronis.com","date":"2022-09-28 18:48:45.000000000","tz":0},"committer":{"name":"Anton Kurbatov","email":"Anton.Kurbatov@acronis.com","date":"2022-09-29 07:38:15.000000000","tz":0},"subject":"Fix full graph loadbalancer creation if jobboard is disabled","message":"Fix full graph loadbalancer creation if jobboard is disabled\n\nThis patch fixes the creation of a full loadbalancer graph in case the\njobboard_enabled config value is False, when no \"listeners\" parameter was\npassed to the get_create_load_balancer_flow function call: use kwargs as the\nflow creation function parameters and handle \"store\" separately.\n\nStory 2010335\nTask 46462\n\nSigned-off-by: Anton Kurbatov \u003cAnton.Kurbatov@acronis.com\u003e\nChange-Id: Id328b27db1ec9c8e17bf18120259e41e75dab3b9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/216cce39c9c22831d725703f41c54b9305d65dc8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/216cce39c9c22831d725703f41c54b9305d65dc8"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[]}
