)]}'
{"id":"openstack%2Foctavia~859816","triplet_id":"openstack%2Foctavia~stable%2Fzed~Ifff0a4a5f78e35ecd722ecdb15edd08347230842","project":"openstack/octavia","branch":"stable/zed","attention_set":{},"removed_from_attention_set":{"34429":{"account":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"last_update":"2022-11-17 12:14:51.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ifff0a4a5f78e35ecd722ecdb15edd08347230842","subject":"Add *.orig to .gitignore","status":"MERGED","created":"2022-09-29 09:06:11.000000000","updated":"2022-11-17 12:16:56.000000000","submitted":"2022-11-17 12:14:51.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"859816","meta_rev_id":"44ddd0e9deb9fe2e158260c2ce48479e29e20e8d","_number":859816,"virtual_id_number":859816,"owner":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-11-17 12:14:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":2,"date":"2022-11-16 17:35:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":2,"date":"2022-11-17 09:57:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":1,"date":"2022-11-17 09:57:36.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-09-29 11:58:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-11-16 17:35:25.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2022-11-17 09:57:36.000000000","updated_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"REVIEWER"}],"messages":[{"id":"faecc10bd2d59bbbd252721f608e4c373da9b52d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-09-29 09:06:11.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2be2d6dd64d998ca8b4978bad160ab14d87fb65c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-29 11:58:06.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a8587290fd8f4ab2abd90ea2fb680a03\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cf33a7a4e8434aeb97bb597bf82d1e2f : SUCCESS in 8m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/12fa3716f9fc46118e387cba5dee94c6 : SUCCESS in 6m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9d8dd539dd374f0085f4c6dceba0f463 : SUCCESS in 5m 38s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8dab252e36d241a49f116dce4b07346a : SUCCESS in 6m 00s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7c157fdcf9f24e04839d4072b12c9563 : SUCCESS in 5m 57s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/03ed36ac5a4842a9afc929937029850c : SUCCESS in 11m 37s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/b90907ce39044896a1f9834d79fe4ea9 : SUCCESS in 6m 03s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/543f88f562da4e8eaa616eea7d7fb091 : SUCCESS in 8m 54s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/b3067d63f49549d59ca796bf45c223a2 : SUCCESS in 5m 39s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/af972e3d0a894c07addec4e9dc0f44e9 : SUCCESS in 8m 48s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/e9d07d91af544306b77eca9f78e10ff0 : SUCCESS in 9m 12s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/bc0b8d8f4b0a446c851e37c9ca6a54ad : SUCCESS in 8m 35s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/e3709c58ca1149b1bb625599fb75eb71 : SUCCESS in 1h 12m 08s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/a6919a95d5f54101accb3193d7f492a3 : SUCCESS in 1h 52m 07s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/f382c981ae104afcb5a4412a244a720c : SUCCESS in 41m 30s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/32eec45acc3e44cc80c1fda7554801ae : SUCCESS in 53m 47s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/9ea54b78e3ca4f1d98bd4f492005c836 : SUCCESS in 1h 20m 23s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/04af5cbe4bfe44d5bb51c8a21650cf19 : SUCCESS in 40m 01s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/0e36ec463ca543e4a3485ef2bbd3356c : SUCCESS in 1h 55m 42s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/c8fb22c2d57344a69151a71930d6a2b3 : SUCCESS in 2h 15m 34s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/e174b454c0c34153a9a4c6269b919877 : SUCCESS in 1h 43m 52s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/e18248541f7940c0b91308d2590c7be7 : SUCCESS in 2h 30m 56s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/16531dc1905147a7a1008661843023e6 : SUCCESS in 1h 36m 37s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/3c868aa9e6e64eb09a08f486f67b2a53 : SUCCESS in 2h 37m 32s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ff7c0d84449287963e734e9a78f2d30859555a3f","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2022-11-16 17:35:25.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"8c7a4c265a8ba2b78a9ffa9f10ea8621a86178b0","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2022-11-17 09:57:36.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"78a2a4fb34d966132ce97356fb3b94801414a736","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-17 09:58:11.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"11717cd51cca9dc0d735c01d27d6d1081cbb2dad","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-17 12:14:51.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a20cd820ee204353871f572cf7bb27e9\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c82156fbb5fd467e91f33b9077311f8d : SUCCESS in 6m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d8949968bd784eb6a80de86fae67af75 : SUCCESS in 4m 33s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f57d8707638043d2a563f9a09c8b88ad : SUCCESS in 4m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7b5a1335385648e48a051ddeb313e637 : SUCCESS in 11m 21s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/df9eb8b9d5f24dc0b32a5f498aedde87 : SUCCESS in 4m 27s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/3bde0f593fbb4615b46188e567cd5e3e : SUCCESS in 8m 10s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/68fa4204e7d74fdc99d68e93c25a8e23 : SUCCESS in 1h 51m 44s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/a2967105311e4eaa9912844fd1602232 : SUCCESS in 2h 15m 05s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/0cf566343dca4493a3f8764fb6b8b456 : SUCCESS in 35m 45s","accounts_in_message":[],"_revision_number":1},{"id":"a517129c824e80af7f8c41345a045eff5a750c2c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-17 12:14:51.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"44ddd0e9deb9fe2e158260c2ce48479e29e20e8d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-17 12:16:56.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/df75410260f04fd1b2511a9d8a088ac6\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a86de3bea86c4bd3a8a30dd11276c0fd : SUCCESS in 1m 34s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"b7fc5b10b2787ee84588646a5590de4a43306e98","revisions":{"b7fc5b10b2787ee84588646a5590de4a43306e98":{"kind":"REWORK","_number":1,"created":"2022-09-29 09:06:11.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/16/859816/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/16/859816/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/16/859816/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/16/859816/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/16/859816/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/16/859816/1"}}},"commit":{"parents":[{"commit":"4f2a019db769ed3fd95e3cd2c0fa69458cf77187","subject":"Update TOX_CONSTRAINTS_FILE for stable/zed","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/4f2a019db769ed3fd95e3cd2c0fa69458cf77187"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-08-21 15:02:11.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-09-29 09:06:01.000000000","tz":120},"subject":"Add *.orig to .gitignore","message":"Add *.orig to .gitignore\n\n.orig files get generated by Git when a merge conflict happens. Nobody\nshould ever want to add those files to the Git tree. By including them\nin .gitignore we prevent that those files get added by accident.\n\nChange-Id: Ifff0a4a5f78e35ecd722ecdb15edd08347230842\n(cherry picked from commit 54a778e8a598e5eca7269305ea1017d55d220fb5)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b7fc5b10b2787ee84588646a5590de4a43306e98"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b7fc5b10b2787ee84588646a5590de4a43306e98"}]},"branch":"refs/heads/stable/zed"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[]}
