)]}'
{"id":"openstack%2Foctavia~859818","triplet_id":"openstack%2Foctavia~stable%2Fxena~Ifff0a4a5f78e35ecd722ecdb15edd08347230842","project":"openstack/octavia","branch":"stable/xena","attention_set":{},"removed_from_attention_set":{"34429":{"account":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"last_update":"2022-11-17 12:14:57.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ifff0a4a5f78e35ecd722ecdb15edd08347230842","subject":"Add *.orig to .gitignore","status":"MERGED","created":"2022-09-29 09:06:29.000000000","updated":"2022-11-17 12:16:57.000000000","submitted":"2022-11-17 12:14:57.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"859818","meta_rev_id":"6f93bb356aa7cc447166ac79619d17819946d434","_number":859818,"virtual_id_number":859818,"owner":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-11-17 12:14:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":2,"date":"2022-11-16 17:35:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":2,"date":"2022-11-17 09:58:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":1,"date":"2022-11-17 09:58:18.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-09-29 11:42:37.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-11-16 17:35:13.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2022-11-17 09:58:18.000000000","updated_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"REVIEWER"}],"messages":[{"id":"45cc0a11bdf7b518c87871ef858b4934d505e290","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-09-29 09:06:29.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9e2732f0af7ecb976061cfbd89091b375ea444c6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-29 11:42:37.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/42d0224ad84b4c3c996da2824e5d4286\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f4817ddcace5417b81cc7c172bc891ea : SUCCESS in 9m 00s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/02c813b113df4610b49f4c8076adbec2 : SUCCESS in 8m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4ef800bee6af432dad9e8f8e0c3db782 : SUCCESS in 5m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ef609ed4fb374c48bcf31d5cf66dcdb5 : SUCCESS in 5m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3d83ae17173e489e9fc1fdd10868eb94 : SUCCESS in 5m 19s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7d19b39bd06b44e0b6c3a464852db80e : SUCCESS in 6m 13s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7b5460dae7614bff8d5ad8199b30a83c : SUCCESS in 12m 55s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/52590e60365b465a88e917621730d8c6 : SUCCESS in 6m 19s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/a96c788583904b2fa03f6752cb6241e3 : SUCCESS in 9m 51s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/e18d1b46c1d14bfca226c834b06ed9bd : SUCCESS in 5m 39s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/ed533078d00f490996c33a8393af2491 : SUCCESS in 10m 31s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/fcf44a23747f46609242eb0810e6382a : SUCCESS in 1h 35m 02s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/8afecc859d8c47caafb29a69945e4079 : FAILURE in 1h 24m 18s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/d655f20e3c98458886f2b3f25c310d4e : SUCCESS in 39m 28s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/0237aea3d0fb40059f39bf829371f00a : SUCCESS in 50m 59s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/5c5157209d894ba493c8a5a163ee92f5 : RETRY_LIMIT in 13m 02s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/5fd9b2b58e8342ecb6f451fd46ca353a : SUCCESS in 32m 34s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/861f5e0092aa45f59b3081cecb9cade1 : SUCCESS in 1h 41m 50s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/592f19730f4c4e5f87c18f5ac77d18ec : SUCCESS in 1h 55m 29s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/002ce13766714feaab02c13f68366fce : SUCCESS in 2h 32m 18s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/db338ddff6b14264a9951ff4af3bba25 : SUCCESS in 1h 23m 55s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/fd1c0ebea9174e39951e6d1f0ad057ed : SUCCESS in 1h 46m 06s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/79bccc39d6e3490d9ce65f046273966d : NODE_FAILURE in 0s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"6b62b94126464a01dddf874f94ea60b37bcd44a9","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-09-29 12:55:11.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"0b23ab965fae7fda60c253874be381151524b9d5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-29 15:16:46.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9129f8e9fd4e4de090866742dd8261b2\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/33d67c56452a40158f2d8138a17445ec : SUCCESS in 11m 51s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2675d04f6f544a2b9a817e8385935599 : SUCCESS in 8m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a5c441c5b86a45c5b983cb7b9f83d0bc : SUCCESS in 5m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8e9cee22b17f4e0c8b93add6d76c81bb : SUCCESS in 6m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d633fe6223d84b9596836ccc48c07963 : SUCCESS in 5m 46s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/75ace2138958406ba2d92fbcc4219988 : SUCCESS in 6m 10s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dcc126edde4344c2bd90ac834bfbdfe3 : SUCCESS in 12m 53s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/f42e4b9ab07b46eeb56c1e0cbc967269 : SUCCESS in 6m 02s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/523b17116c5c449385238afd87fadee0 : SUCCESS in 8m 36s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/e95e69ec78e5413cace817cc4949a30e : SUCCESS in 5m 30s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/397b5b1d66344ba99529ef367c8e4cee : SUCCESS in 8m 55s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/6a12a97c2eab49afb39d0c46e35c2050 : SUCCESS in 1h 18m 02s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/1347342896484b91b98fb68ce2243fc4 : SUCCESS in 1h 37m 59s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/c8bacf7591914e9391df877959a2dec2 : SUCCESS in 40m 26s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/9f856395cc434fbcb9aa05c1e529d322 : SUCCESS in 57m 18s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/1c51e842cb6c43dc9ef759db4fa68cdf : RETRY_LIMIT in 13m 47s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/c8116116f4854b2191ab6fece6ecbbad : SUCCESS in 41m 51s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/8514c5faaa9948c793a20201d42820b4 : SUCCESS in 1h 39m 09s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/0ea32be8d5d9413689963d6e8aebccfc : SUCCESS in 2h 19m 24s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/b579b3382ecc46fda1ccaeb550e44f21 : SUCCESS in 1h 50m 42s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/a9219c9a6226462d8d3ae510771b497f : SUCCESS in 1h 35m 46s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/82228a3b79b7460e8aba1d1d788d44d2 : SUCCESS in 1h 38m 03s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/6ce9371270324b39881e8520580d5cdf : NODE_FAILURE in 0s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"fdbc328f71f0d7ec3c1837c6f07c936593d3935f","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2022-11-16 17:35:13.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f5a41198fdfb053a368e35c6e7021eed9dc8733d","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2022-11-17 09:58:18.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"2c8a29425186cefb35e94f6d8d5dcf1e60ab3a73","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-17 09:59:03.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"5693fecb65979be95499bc79ee80de0e32ef7a39","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-17 12:14:57.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/87703ce199c64efb8fc72ad82e24c881\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/20cabede61294400a96ba0cca7557c03 : SUCCESS in 7m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8c74cfae8cf24beda0c52615982fa283 : SUCCESS in 5m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3abb3dceeb4b439f839229d222229f13 : SUCCESS in 5m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c869b7aefcc649d98e3d9e249f010fbb : SUCCESS in 4m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ca19b51002294b0f9e6e2ddd16e30813 : SUCCESS in 12m 25s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/76e9c41705d0467cb6983800add9c1a6 : SUCCESS in 4m 54s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/a5579e2e082c4943b8967f05b237309e : SUCCESS in 8m 10s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/59afd1999a2b4166a13eb5d40eafe5eb : SUCCESS in 1h 25m 05s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/41e313e3ac9a4b60afee1ceab7e3c1f7 : SUCCESS in 1h 26m 53s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/6dd29db25def4193b42288b4a68a59ff : SUCCESS in 34m 53s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/56fd73091da94494b3704e6444e71430 : SUCCESS in 58m 32s","accounts_in_message":[],"_revision_number":1},{"id":"dbd66afed945b1f4818959f82a6f62ed3c437981","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-17 12:14:57.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"6f93bb356aa7cc447166ac79619d17819946d434","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-17 12:16:57.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4667ef514e3443b3957f9c46f94519d5\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5a551c7a3cdf4c02b023b82ce0eb96ba : SUCCESS in 1m 26s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"c87ff96d0b02a31f349967e82a78b740adb6a309","revisions":{"c87ff96d0b02a31f349967e82a78b740adb6a309":{"kind":"REWORK","_number":1,"created":"2022-09-29 09:06:29.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/18/859818/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/18/859818/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/18/859818/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/18/859818/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/18/859818/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/18/859818/1"}}},"commit":{"parents":[{"commit":"f4f1cd27946efb0a5af9107e76b261fc9b9d5718","subject":"Merge \"Fix sporadic unit test failure\" into stable/xena","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/f4f1cd27946efb0a5af9107e76b261fc9b9d5718"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-08-21 15:02:11.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-09-29 09:06:21.000000000","tz":120},"subject":"Add *.orig to .gitignore","message":"Add *.orig to .gitignore\n\n.orig files get generated by Git when a merge conflict happens. Nobody\nshould ever want to add those files to the Git tree. By including them\nin .gitignore we prevent that those files get added by accident.\n\nChange-Id: Ifff0a4a5f78e35ecd722ecdb15edd08347230842\n(cherry picked from commit 54a778e8a598e5eca7269305ea1017d55d220fb5)\n(cherry picked from commit b7fc5b10b2787ee84588646a5590de4a43306e98)\n(cherry picked from commit 10042f76d6f7855d348c1680c1f54f5cff045cc4)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c87ff96d0b02a31f349967e82a78b740adb6a309"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c87ff96d0b02a31f349967e82a78b740adb6a309"}]},"branch":"refs/heads/stable/xena"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[]}
