)]}'
{"id":"openstack%2Foctavia~859912","triplet_id":"openstack%2Foctavia~stable%2Fzed~Id77ce36f59b71d309f153e5c1d44059f162ee440","project":"openstack/octavia","branch":"stable/zed","topic":"barbican-acl-stable/zed","attention_set":{},"removed_from_attention_set":{"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2022-10-25 10:23:56.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Id77ce36f59b71d309f153e5c1d44059f162ee440","subject":"Fix barbican client with application credentials/trusts","status":"MERGED","created":"2022-09-30 07:40:15.000000000","updated":"2022-10-25 10:26:03.000000000","submitted":"2022-10-25 10:23:56.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"859912-barbican-acl-stable/zed","cherry_pick_of_change":854563,"cherry_pick_of_patch_set":4,"meta_rev_id":"7beb8a80055c2d5a62d0f01df41b4ba90acdd687","_number":859912,"virtual_id_number":859912,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-10-25 10:23:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"all":[{"value":2,"date":"2022-10-25 08:15:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-09-30 08:09:36.000000000","_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"all":[{"value":1,"date":"2022-10-25 08:15:28.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-09-30 07:40:15.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"CC"},{"updated":"2022-09-30 08:09:36.000000000","updated_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"reviewer":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"state":"REVIEWER"},{"updated":"2022-09-30 10:54:41.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-10-25 08:15:28.000000000","updated_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"REVIEWER"}],"messages":[{"id":"994d6d8e33e96df4bab1fc3e7e5704cf4a3664c6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2022-09-30 07:40:15.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"1a4e2c7c57065af783b483ffb8fcfe667ad57415","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-09-30 08:09:36.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"b0cf89cff8f974630d82dc11e41f213bafa87d09","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-30 10:54:41.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/81b934c6ea56475e96d6a2e2facf9a24\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3fc32406bb414fa2ae6d385a1ccf5ed0 : SUCCESS in 9m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8eb03fbcacf745a3aef52412fffdad4c : SUCCESS in 6m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1dbd0b9827f7447db48b77f237f9d687 : SUCCESS in 3m 58s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b230e364e7904cb18458c021909f04ff : SUCCESS in 4m 33s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6f9aaf1f193f4f4a9029418a9a7d27ac : SUCCESS in 5m 01s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d27cd4a688bb42bba9afe4a5e03ffb03 : SUCCESS in 10m 19s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/51fcf1825d164212a13c5f990c5893c1 : SUCCESS in 4m 08s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/13c7c18841bf447f87cdc92da695b2ba : SUCCESS in 4m 44s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/f242e9a08bd844ab828f8b39ab394905 : SUCCESS in 6m 35s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/85676f34c83044f69fe836dc2c1e11f4 : SUCCESS in 4m 03s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/54ca3dadfe3f49f2b298c04d89b616e6 : SUCCESS in 6m 50s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/a15994922b1944ff9f10406c6d8a1347 : SUCCESS in 6m 42s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/7e9cc5590cbf4a9e9bcd73fe231854ce : SUCCESS in 6m 22s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/d035b9921944479bb2813ada5ab7a42e : SUCCESS in 1h 18m 43s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/28f2911663ac4f02b67b63c6809f392c : SUCCESS in 2h 23m 23s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/2b351252ff324d07b6058b80736474a5 : SUCCESS in 59m 25s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/f6c5e3a1421745069000e681413c6e6c : SUCCESS in 1h 23m 05s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/9c7139cb34194f24977ba7fe3d196ecd : SUCCESS in 1h 35m 55s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/62bc447c73ba4a44b4cf6fb0009db396 : SUCCESS in 50m 18s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/2d6fe981a8ec408c95a7c793daebb5ab : SUCCESS in 2h 35m 04s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/91593059093545fda9366dc831aaa26f : SUCCESS in 3h 06m 27s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/0a2d9a0abee246d88907dbe037432cdd : TIMED_OUT in 2h 39m 26s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/4e5c7235e19a42c393a14aecefbb246e : SUCCESS in 2h 18m 32s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/aa5594e5215142a1a7bc3b2423a5ce7a : TIMED_OUT in 2h 40m 50s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/96e9052c82c04fe28c8e70da98f7bde9 : SUCCESS in 2h 19m 13s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"2dee646e60ee5320943910c1f4c6dccb29c1568e","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2022-10-25 08:15:28.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"871c70a68d763a85ffbb234897edbc5187946e8e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-25 08:16:08.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"412590352be73790998fd1a601596dfc09b71aa4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-25 10:23:56.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/32cff9b099c64bc5b7cfcf3ce88cd24b\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/18ccb265eb6f405f82d5a9b68023c53a : SUCCESS in 5m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/eeb508e8a1104e19a6e7544c70e5021b : SUCCESS in 5m 19s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d760731a63b84a3ab820315edd421789 : SUCCESS in 4m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/190dad48bd004cb3b3e60e4c0b435520 : SUCCESS in 12m 33s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e2eac8fb2d9f48e08d1cf2f3334abb1e : SUCCESS in 4m 43s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/b60f7e7388ef409eb3a1ab1242748201 : SUCCESS in 4m 40s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/5614900d66d24a0f9cb923baab43cc9f : SUCCESS in 7m 18s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/22f66cb1c1ca4bfc941c2279a723401a : SUCCESS in 1h 04m 10s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/e0193995f7194139a48535c4da533371 : SUCCESS in 2h 06m 26s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/ee638dd6c2194b0f81325bbf81ba24fb : SUCCESS in 39m 17s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/1d5efccb06cd49e8b2caa98ca98d97bd : SUCCESS in 50m 38s","accounts_in_message":[],"_revision_number":1},{"id":"4b6958ca47d16195c292edb493c49ec170efcefa","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-25 10:23:56.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"7beb8a80055c2d5a62d0f01df41b4ba90acdd687","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-25 10:26:03.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c2afb6d134724ba08ae909813d8028c5\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bb07a7c6fb82453fad84b45db7e37768 : SUCCESS in 1m 50s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e2c07db7443643f49a84408071ba5ab3 : SUCCESS in 1m 28s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"960977cc76025fb165ac3b9952bfb4880db07aaf","revisions":{"960977cc76025fb165ac3b9952bfb4880db07aaf":{"kind":"REWORK","_number":1,"created":"2022-09-30 07:40:15.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/12/859912/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/12/859912/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/859912/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/859912/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/859912/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/12/859912/1"}}},"commit":{"parents":[{"commit":"4f2a019db769ed3fd95e3cd2c0fa69458cf77187","subject":"Update TOX_CONSTRAINTS_FILE for stable/zed","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/4f2a019db769ed3fd95e3cd2c0fa69458cf77187"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2022-08-25 08:33:21.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","date":"2022-09-30 07:40:15.000000000","tz":0},"subject":"Fix barbican client with application credentials/trusts","message":"Fix barbican client with application credentials/trusts\n\nIt seems that keystoneauth1.identity.generic.token doesn\u0027t handle\nproperly the application credential/trust tokens passed by the context\nof the requests. When using app credentials, Octavia failed to retrieve\nthe certificates from barbican.\nSwitching to keystoneauth1.token_endpoint fixes the issue, the\nauth tokens are correctly passed to the barbican client.\n\nStory: 2007619\nTask: 39737\n\nChange-Id: Id77ce36f59b71d309f153e5c1d44059f162ee440\n(cherry picked from commit ce7f27e3b7ef6a94501ce975fb0e9dadcffb822b)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/960977cc76025fb165ac3b9952bfb4880db07aaf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/960977cc76025fb165ac3b9952bfb4880db07aaf"}]},"branch":"refs/heads/stable/zed"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[]}
