)]}'
{"id":"openstack%2Foctavia~861145","triplet_id":"openstack%2Foctavia~master~Ie739cd582103d1aab3021c56654e395b5919358f","project":"openstack/octavia","branch":"master","attention_set":{},"removed_from_attention_set":{"32962":{"account":{"_account_id":32962,"name":"Jakob Meng","email":"code@jakobmeng.de","username":"jakobmeng@web.de"},"last_update":"2022-10-13 16:36:15.000000000","reason":"Change was submitted"},"29244":{"account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"last_update":"2022-10-13 16:36:15.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ie739cd582103d1aab3021c56654e395b5919358f","subject":"Fix prometheus-proxy service name in Red Hat-based distros","status":"MERGED","created":"2022-10-13 07:48:56.000000000","updated":"2022-10-13 16:37:38.000000000","submitted":"2022-10-13 16:36:15.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"861145","meta_rev_id":"d2d6311d6171631881bdc780b1ea028b452392c7","_number":861145,"virtual_id_number":861145,"owner":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":32962,"name":"Jakob Meng","email":"code@jakobmeng.de","username":"jakobmeng@web.de"},{"value":0,"_account_id":32458,"name":"Ananya","display_name":"Ananya Banerjee","email":"anbanerj@redhat.com","username":"frenzyfriday"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-10-13 16:36:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},{"value":0,"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"recommended":{"_account_id":32458,"name":"Ananya","display_name":"Ananya Banerjee","email":"anbanerj@redhat.com","username":"frenzyfriday"},"all":[{"value":1,"date":"2022-10-13 09:45:56.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32962,"name":"Jakob Meng","email":"code@jakobmeng.de","username":"jakobmeng@web.de"},{"value":1,"date":"2022-10-13 13:50:40.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32458,"name":"Ananya","display_name":"Ananya Banerjee","email":"anbanerj@redhat.com","username":"frenzyfriday"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":2,"date":"2022-10-13 15:01:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-10-13 09:45:44.000000000","_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},{"value":1,"date":"2022-10-13 13:50:17.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"_account_id":32962,"name":"Jakob Meng","email":"code@jakobmeng.de","username":"jakobmeng@web.de"},{"value":0,"_account_id":32458,"name":"Ananya","display_name":"Ananya Banerjee","email":"anbanerj@redhat.com","username":"frenzyfriday"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":1,"date":"2022-10-13 15:01:56.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},{"value":0,"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"approved":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"recommended":{"_account_id":32962,"name":"Jakob Meng","email":"code@jakobmeng.de","username":"jakobmeng@web.de"},"all":[{"value":1,"date":"2022-10-13 09:45:56.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32962,"name":"Jakob Meng","email":"code@jakobmeng.de","username":"jakobmeng@web.de"},{"value":0,"_account_id":32458,"name":"Ananya","display_name":"Ananya Banerjee","email":"anbanerj@redhat.com","username":"frenzyfriday"},{"value":2,"date":"2022-10-13 12:15:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},{"value":0,"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"_account_id":32458,"name":"Ananya","display_name":"Ananya Banerjee","email":"anbanerj@redhat.com","username":"frenzyfriday"},{"_account_id":32962,"name":"Jakob Meng","email":"code@jakobmeng.de","username":"jakobmeng@web.de"},{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-10-13 09:45:44.000000000","updated_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"reviewer":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"state":"REVIEWER"},{"updated":"2022-10-13 09:45:56.000000000","updated_by":{"_account_id":32962,"name":"Jakob Meng","email":"code@jakobmeng.de","username":"jakobmeng@web.de"},"reviewer":{"_account_id":32962,"name":"Jakob Meng","email":"code@jakobmeng.de","username":"jakobmeng@web.de"},"state":"REVIEWER"},{"updated":"2022-10-13 10:09:44.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-10-13 13:50:17.000000000","updated_by":{"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"},"reviewer":{"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"},"state":"REVIEWER"},{"updated":"2022-10-13 13:50:40.000000000","updated_by":{"_account_id":32458,"name":"Ananya","display_name":"Ananya Banerjee","email":"anbanerj@redhat.com","username":"frenzyfriday"},"reviewer":{"_account_id":32458,"name":"Ananya","display_name":"Ananya Banerjee","email":"anbanerj@redhat.com","username":"frenzyfriday"},"state":"REVIEWER"},{"updated":"2022-10-13 15:01:56.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"}],"messages":[{"id":"e455e4bc1e881b1c9c431a57c28fdee112c13041","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2022-10-13 07:48:56.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"114ced1c1115de2440fe4a2cc4e6b2c40472bc76","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-10-13 09:45:44.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"5429b06b055c018f46c9eb41aa4fc03c1db02a5d","author":{"_account_id":32962,"name":"Jakob Meng","email":"code@jakobmeng.de","username":"jakobmeng@web.de"},"date":"2022-10-13 09:45:56.000000000","message":"Patch Set 1: Backport-Candidate+1 Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"d816380f51862a9fb004c2e986c255dfc4233f95","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-13 10:09:44.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f454fb783f2743589d696d0752b183eb\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/15296b518d33473a9abcfc34bba54569 : SUCCESS in 8m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/84194ff6cec64bc79bfa4ff7af718c0b : SUCCESS in 5m 48s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c0f2248fe5b14cec9de5e536df76aeb7 : SUCCESS in 4m 38s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/1893e5343afc495794ce011a3dcc7bd5 : SUCCESS in 4m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/631096c538ac4cf3b4d8a70e2eaecd9d : SUCCESS in 11m 14s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/02c2c1a993e949849db13344e7d69eaf : SUCCESS in 4m 23s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/491616b04f264cfe8a449b4cfb7f5985 : SUCCESS in 8m 17s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/8293a1e04fe64d12a7889c963746cd30 : SUCCESS in 4m 17s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/2947e17c6cfc4fa7a734fe9618175862 : SUCCESS in 6m 49s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/b2bca858086045b3829427ddca473749 : SUCCESS in 6m 30s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/7a34d35fb0e044a19d665b9119c4b53f : SUCCESS in 6m 53s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/b1c62ed67e94438891a7001327002b19 : SUCCESS in 1h 26m 44s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/f14188fc06ac4340b8e45b193594d6e9 : SUCCESS in 1h 37m 48s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/648aa8b0072f42f3beb599dcb69bfbec : SUCCESS in 42m 03s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/2fd4fbc1621c47de90f8f1af59eb16d6 : SUCCESS in 56m 39s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/ca7cc3e721084a2682f0ab2336023608 : SUCCESS in 1h 07m 57s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/980fe1e226d14947bd7eb579fa2595dd : SUCCESS in 40m 43s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/136f608aa1d541118111d65e98f0816a : SUCCESS in 1h 56m 23s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/b4e6bf2c07224fc58c7fe5ff6a10a76a : SUCCESS in 2h 05m 17s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/4b41e76338e94f37a89f00c1968ec8f6 : SUCCESS in 2h 05m 18s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/d26db1cb75eb481bbdeb20e364372d95 : SUCCESS in 2h 18m 19s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/1d46f7874714421a871f842b4a372609 : SUCCESS in 1h 55m 55s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/c6e1f5a55a0749f199c17e724fdfe722 : SUCCESS in 2h 13m 00s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"e9ae5591890054906129cf69ad02f4b62698a74b","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2022-10-13 12:15:15.000000000","message":"Patch Set 1: Backport-Candidate+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"fded5bab8f354a02dca7a0ce91ccb81c854e5b5d","author":{"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"},"date":"2022-10-13 13:50:17.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"e6aa4bc1bc509df982b6f2e4314db329482acefe","author":{"_account_id":32458,"name":"Ananya","display_name":"Ananya Banerjee","email":"anbanerj@redhat.com","username":"frenzyfriday"},"date":"2022-10-13 13:50:40.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"e425eac5da4ee0d156221865ec66c4a0b69a9b61","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2022-10-13 13:51:21.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"7e80c32918504362a3cdbdfce7e4312357a16ab9","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2022-10-13 15:01:56.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"38c2e480c313f3f8087a35db1d570c652011baaa","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-13 15:02:36.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"ae202552cb8ee234bf942a29d4cff491d4ae301c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-13 16:36:14.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e4f5348307494f07a3239364f83d2a20\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8e2dd45e9e4d4518953fbe5be8893223 : SUCCESS in 5m 12s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/98a1ff5b87454913b8db48f32cbdf63e : SUCCESS in 4m 30s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c387c494ab1f4b47ab66ca7db38a38d0 : SUCCESS in 4m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c08b7ce6bdda49afb39b74675c8e4c47 : SUCCESS in 9m 29s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/8b8e544314594001ab1672cd568ebf31 : SUCCESS in 4m 00s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/7328a63b82024ad3a7a00135087b67bf : SUCCESS in 6m 56s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/42d26480a1764abb9358c96fb0774f4f : SUCCESS in 1h 01m 43s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/d1a421f578c04d8eb99887d241c809c1 : SUCCESS in 1h 32m 59s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/8b8f54e9c4334a3d8ecf8030c542b2e3 : SUCCESS in 38m 43s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/89aded4208e941e3b55134b502ea15c3 : SUCCESS in 50m 59s","accounts_in_message":[],"_revision_number":1},{"id":"077619f52d60e052b86ce21c521e8a6cf5031235","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-13 16:36:15.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"d2d6311d6171631881bdc780b1ea028b452392c7","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-13 16:37:38.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3f5ffbfd6a83419bb194c66020fcbdbc\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/75295dd10bb442068f2d8c622944f895 : SUCCESS in 1m 03s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"344d6d76ce4481f30fe295114551243d61e7eea8","revisions":{"344d6d76ce4481f30fe295114551243d61e7eea8":{"kind":"REWORK","_number":1,"created":"2022-10-13 07:48:56.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/45/861145/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/45/861145/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/45/861145/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/45/861145/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/45/861145/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/45/861145/1"}}},"commit":{"parents":[{"commit":"95adbbe618a2fb5ffe0f0e833dcfce5a9d85870c","subject":"Merge \"Add *.orig to .gitignore\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/95adbbe618a2fb5ffe0f0e833dcfce5a9d85870c"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2022-10-13 07:46:11.000000000","tz":120},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2022-10-13 07:48:17.000000000","tz":120},"subject":"Fix prometheus-proxy service name in Red Hat-based distros","message":"Fix prometheus-proxy service name in Red Hat-based distros\n\nThe prometheus-proxy service is called octavia-prometheus-proxy in RDO\u0027s\nRPMs [0].\nAdd a svc-map for the service name to fix the amphora-agent element for\nthese distros.\n\n[0] https://review.rdoproject.org/r/c/openstack/octavia-distgit/+/44152\n\nPartial-Bug: 1992668\n\nChange-Id: Ie739cd582103d1aab3021c56654e395b5919358f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/344d6d76ce4481f30fe295114551243d61e7eea8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/344d6d76ce4481f30fe295114551243d61e7eea8"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}}]}],"submit_requirements":[]}
