)]}'
{"id":"openstack%2Foctavia~861288","triplet_id":"openstack%2Foctavia~master~Iedcfc826962ad28446e91d44eea3071c11d16973","project":"openstack/octavia","branch":"master","topic":"sqlalchemy2","attention_set":{},"removed_from_attention_set":{"31664":{"account":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"last_update":"2023-06-21 16:27:33.000000000","reason":"Change was submitted"},"32761":{"account":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"last_update":"2023-06-21 16:27:33.000000000","reason":"Change was submitted"},"29244":{"account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"last_update":"2023-06-21 11:24:22.000000000","reason":"\u003cGERRIT_ACCOUNT_29244\u003e replied on the change","reason_account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}},"34429":{"account":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"last_update":"2023-04-28 12:45:03.000000000","reason":"\u003cGERRIT_ACCOUNT_34429\u003e replied on the change","reason_account":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}}},"hashtags":[],"change_id":"Iedcfc826962ad28446e91d44eea3071c11d16973","subject":"[sqlalchemy2] Turn on RemovedIn20Warnings in tests","status":"MERGED","created":"2022-10-14 06:19:17.000000000","updated":"2023-06-21 16:28:47.000000000","submitted":"2023-06-21 16:27:33.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":7,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"861288-sqlalchemy2","meta_rev_id":"6a57b5caa70ce8c2e628bd1261e43c2fe13f8c4f","_number":861288,"virtual_id_number":861288,"owner":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":0,"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-06-21 16:27:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},{"value":0,"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":0,"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-06-15 17:10:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":2,"date":"2023-06-09 14:29:03.000000000","_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},{"value":1,"date":"2023-06-17 05:04:16.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":0,"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-06-15 17:10:44.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},{"value":0,"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":0,"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},{"value":0,"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-10-14 08:31:36.000000000","updated_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"reviewer":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"state":"REVIEWER"},{"updated":"2022-10-14 08:42:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-10-24 12:52:30.000000000","updated_by":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"reviewer":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"state":"REVIEWER"},{"updated":"2022-10-26 11:36:06.000000000","updated_by":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"reviewer":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"state":"REVIEWER"},{"updated":"2023-06-15 17:10:44.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2023-06-17 05:04:16.000000000","updated_by":{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},"reviewer":{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},"state":"REVIEWER"}],"messages":[{"id":"f7a28c7dabfb1f28e248eb7a051fd3ca8e4411de","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2022-10-14 06:19:17.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7f45e3f3f6cfb17dc5753df923357c2d5c1a7ad5","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-10-14 08:31:36.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"6ab5b883e41409d10ee5518178646461a349ae85","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-14 08:42:12.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/39a7321a4cce44d6b58cb294eee4b486\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/5ced7b9684344b6fb13bf2f466e51feb : SUCCESS in 7m 58s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bec32a49de7b4efeac6b873eb39e7117 : SUCCESS in 11m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1e797090cdd9468da31a6ee28a4a976b : SUCCESS in 6m 24s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6a9a3d2518104cc783a5c8b0a8d4da03 : SUCCESS in 5m 21s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f8f100ed7e7a473a8e964af552b4043e : SUCCESS in 5m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/62ef16509f35493d816d3ddb7e435654 : SUCCESS in 12m 33s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/28110605821e4860bd9d9edf60c4ee81 : SUCCESS in 5m 17s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/60440df6cdb2468f8ad8b581492a34ba : SUCCESS in 5m 04s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/d083e5e8b68b4469908fa67af0652952 : SUCCESS in 9m 01s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/7061082f2d734cbebc3a5eb6a3c0cc74 : SUCCESS in 5m 14s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/07fb33d3b357449a90e95d906cb75f02 : SUCCESS in 9m 35s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/7a39927b57e44ab7b7fc9a01d8dae35f : SUCCESS in 8m 52s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/9bb0bc40813a405eba04820dde80dce9 : SUCCESS in 7m 35s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/c472e1b6542a4de6bccf6a8639edd44e : SUCCESS in 1h 17m 55s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/be0b1aabd79143d5aed1dfed65d28141 : SUCCESS in 2h 05m 27s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/df4838ed449d43d082b54bd2ea662b34 : SUCCESS in 34m 09s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/0efa9302802c4bf0865acf59fc1a6f50 : SUCCESS in 1h 17m 47s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/291d508f02294b9b8141ec5567ea25e3 : SUCCESS in 38m 01s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/460ba60911744e52a100df461af9cc53 : SUCCESS in 1h 57m 25s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/937340f01f7646759b6b18be23f0aecb : SUCCESS in 1h 48m 45s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/c005f31ffb8a47f3a4e65323ddd36959 : SUCCESS in 1h 55m 12s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/48aefdef5566467282cabc293419bbb8 : SUCCESS in 1h 43m 00s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/a1270375688446fa89ca7d2c5f67132c : SUCCESS in 2h 13m 01s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/c69b6eebecb64012a3c5c9086f4f7805 : SUCCESS in 2h 20m 27s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"57af627320ceb91025006ed7c101a2b22ffe4325","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2022-10-24 12:52:30.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f000b8fe078ce573a1819c5e2ffcaf0a29e90935","author":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"date":"2022-10-26 11:36:06.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"d320d381b794f779717206f7f4f5ddfb89daa3f3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-02-27 07:52:40.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"b3116c496633e2671810b03a061dc7491e714422","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-27 10:21:41.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/768bfd49e93542ada4269b4590230f86\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/10161bdbbed541828de89a93ea88eb39 : SUCCESS in 4m 44s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fd36d344bd6e48c89ab50d1f4f8999b6 : SUCCESS in 9m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5fe050685f274a0faae5c60d7bc0d110 : SUCCESS in 5m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e56a10fc1e4948e19775e138cb7de0f2 : SUCCESS in 4m 58s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/23902d3de37346ce97f914543a5bc757 : SUCCESS in 3m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3591d3b4e4ae4a8b8722fe81879cf311 : SUCCESS in 12m 32s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/01db191ea4d34477aeb6458ef429bf1d : SUCCESS in 4m 45s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/207a54fa94dd4086a71b364cce69bcd7 : SUCCESS in 4m 31s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/b42794c8ecfd45f4b0b04fc3319eb87c : SUCCESS in 7m 49s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/45aa6375eff746ea9675e75a9515248e : SUCCESS in 4m 18s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/6e81032724f84a1fa042c0e66ab29716 : SUCCESS in 7m 38s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/6eee325ad1134a58b60d9cf4e742d1f5 : SUCCESS in 7m 43s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/0c971bcd40fa4c0eb2fe165720b37324 : SUCCESS in 6m 56s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/a661420e305d440daf28fbf477cace84 : SUCCESS in 1h 19m 39s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/61dacffe93914366bf7b423a1c215fd4 : SUCCESS in 2h 20m 40s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/84772293b041488a9b42c902f9e4bbf5 : SUCCESS in 35m 50s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/8af78fd598064f958ffd93dca5bcbd66 : SUCCESS in 53m 43s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/f31bd5836e914aa594b0fc4f3e020ac8 : SUCCESS in 33m 50s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/888a513286814353b65399b47c18bcd9 : SUCCESS in 1h 53m 54s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/dd828ce839b84345937941f1191d1e6b : SUCCESS in 2h 25m 48s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/76465685891d4937b47c8636e6221760 : SUCCESS in 2h 03m 38s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/9de642d583ef40a1a1e4de454efe6fbc : SUCCESS in 1h 54m 42s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/7a6bb59a93634d0899557492d94f021f : SUCCESS in 1h 36m 19s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/1a98edcc16d846dda8ae0f5a11bac714 : SUCCESS in 2h 23m 32s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"908dde6da01822973698a60cbd18d72b2e4d7230","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-03-29 08:44:58.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.","accounts_in_message":[],"_revision_number":3},{"id":"94eca7054124ffd9947580e09da3c9e959b04808","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-29 11:23:24.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9c4fb90edc4540fa87ea5f8820f2deb2\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/c979499546604121b88bea082b6f2134 : SUCCESS in 5m 21s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/72c4dd5fcf6b4251a1d4bde368d3e8ba : SUCCESS in 10m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/671d7c7e085b4739a1cdcc509dfc3ff2 : SUCCESS in 5m 17s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c7e28c601ad74e39bbbe263ca562c710 : SUCCESS in 5m 34s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f792b2cc638548e2aeb1afc79fad4e4a : SUCCESS in 5m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f2f89cf45391487b9de6b16018b1a6a5 : SUCCESS in 12m 14s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/82f358f67be840bc9ea982d48b61d3e0 : SUCCESS in 6m 15s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/3499606722314efd92516c5ad65b582e : SUCCESS in 6m 55s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/2e6e2d1783b94fa7a319ad8b73b1460b : SUCCESS in 8m 32s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/c1980dbe333e4e3382c1e9188246e9e2 : SUCCESS in 5m 07s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/4d64b189b31c45bc9c8a41050a85c069 : SUCCESS in 11m 51s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/1e2947fc443a43658197df427114ad37 : SUCCESS in 8m 22s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/2ac12efe362a4ed7af27bebe35503488 : SUCCESS in 8m 28s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/38fa1be082e34397ac5c5776c5b43b02 : SUCCESS in 1h 46m 58s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/001d55550e154413a6fdc00c6d42183d : SUCCESS in 2h 04m 34s\n- octavia-v2-dsvm-scenario-jobboard https://zuul.opendev.org/t/openstack/build/f965a0b81e7148c8893407bad998d9e7 : SUCCESS in 1h 54m 20s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/cb09d8921c9f4cdd9fa1bbc505f290a5 : SUCCESS in 33m 16s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/4ff095a895614dd4b7cb07773fa5ad52 : SUCCESS in 1h 03m 59s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/ec7c1f868386469fb40b62a094be4b08 : SUCCESS in 32m 55s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/b0372ee905134ec2a92ad331664dc85f : SUCCESS in 2h 26m 36s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/cec876d12aa4427095c51189aa1e4343 : SUCCESS in 2h 22m 24s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/79db65c5526d414a993de55d8ec283c2 : SUCCESS in 2h 21m 09s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/9cab61ff9853452fa24efefc0399db97 : SUCCESS in 2h 29m 27s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"9915489d49b68efe55381f932af2db182a4671b3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-04-24 06:50:24.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"6ff27d14a0fb928cf4a95eeb9198ddd0c1f56ff9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-24 08:59:47.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/3352b3296b614dfe8ff7b369107b4b01\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/ff58c3606a014a98a6375a3edba8a93c : SUCCESS in 4m 58s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f5db22e7650143f6bdcfaa94409c31b3 : SUCCESS in 10m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dc7a2d7be894427489fb8a1a695aab72 : SUCCESS in 8m 25s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d2395b00f91742a7856aabaf899fb60d : SUCCESS in 4m 15s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/37c4ecb2f3974740956e1600d263a752 : SUCCESS in 7m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4ef1ab695c5c46709bc5d318563f2426 : SUCCESS in 11m 42s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0db3bd17046849fcbab8ecaefdb68665 : SUCCESS in 7m 09s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/cbe5efbab3354b23b35040eb46d786e6 : SUCCESS in 8m 38s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/203b778cf4ac4eba83aa36bed022b4b0 : SUCCESS in 10m 38s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/2bb6662120f7428fb9c97579c46bf219 : SUCCESS in 4m 33s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/10f85506bed54dc3868b7261a5952d18 : SUCCESS in 9m 02s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/4ce421d0b741458e9f90ec31480ed87a : SUCCESS in 7m 01s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/433243f81fa04069aa47c122a1a8b78e : SUCCESS in 7m 51s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/49d0cccd80174059a8009a5e58270603 : SUCCESS in 1h 21m 54s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/fb62c6ec831d404bb68c43d7be017727 : FAILURE in 7m 44s\n- octavia-v2-dsvm-scenario-jobboard https://zuul.opendev.org/t/openstack/build/84860e373c15462587da55d85fc08b41 : FAILURE in 7m 46s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/8bc4e1d4a3d9411498a7dce78dce893d : FAILURE in 13m 15s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/9fb93436cba744bdabd5a19cfb296fd7 : SUCCESS in 54m 50s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/81eb49e0aa4f4e008834584d84b454a3 : FAILURE in 12m 39s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/197f1c4e56e04824acee18e5c0c568c7 : FAILURE in 10m 17s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/5a692b3aea5d47288bf5fbae84e88944 : FAILURE in 14m 20s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/fbe644adb7ee40a582ae381554f8806c : FAILURE in 5m 36s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/e0dd83096b154952bc95b0d27e2bdb02 : SUCCESS in 2h 07m 08s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"c66a2e13b807acd085e4d872ce547900ee6b930d","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2023-04-28 12:45:03.000000000","message":"Patch Set 4: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"91d2ad46f701576d2760e20dcc72859672496d4c","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2023-04-28 12:50:54.000000000","message":"Patch Set 4: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"244b97b571f1d882c0c0868fc8050907666fd654","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-05-03 07:07:19.000000000","message":"Uploaded patch set 5: Patch Set 4 was rebased.\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"7ed0b441b1754a96e968a3188f1b04ba14644c40","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-03 10:23:48.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/7fd2a8d007e44676a285a6545f782640\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/538555009ded4835a790e8042884486e : SUCCESS in 5m 42s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cab1ec6403f8417bac678e99773eea77 : SUCCESS in 9m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fedea038a0634e01a1c6c2bab7707ee7 : SUCCESS in 5m 02s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/16d0b9b30f5a481196c9020bab0306cc : SUCCESS in 4m 50s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/50e6fd99bd3546d5aa089c80a98e62a6 : SUCCESS in 4m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/399a5ea46cb4461c8e0733aa6006e4c7 : SUCCESS in 11m 24s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/54f025f1acdf45799b8973b0ba62c9ab : SUCCESS in 4m 48s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/456d1c8b696b4c2c9c1c0cc99e521dc8 : SUCCESS in 5m 11s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/e954cd4c4e694be991bd47c01a48f76d : SUCCESS in 9m 01s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/33ba6527a1664598aed0554457fc96b2 : SUCCESS in 4m 46s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/beea4a0a46c642389ac446e28eb59ab4 : SUCCESS in 6m 53s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/99626517b3ab4ddf9ed8ab132fadd9e0 : SUCCESS in 6m 54s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/8bdea14ae93e45e49b545bdd5229eee3 : SUCCESS in 6m 44s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/21f4cb4bf8154cb4829331f9d4a03d34 : SUCCESS in 1h 31m 40s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/27801ff8a4d34f7183be6faf477dcaa0 : NODE_FAILURE Node request 300-0021120483 failed in 0s\n- octavia-v2-dsvm-scenario-jobboard https://zuul.opendev.org/t/openstack/build/3217016ec74b4473b712378a009cf3c4 : TIMED_OUT in 2h 32m 45s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/9bfb4df0a1ab49b899b4c22fbde64adf : NODE_FAILURE Node request 300-0021120485 failed in 0s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/b5f5896287f64f08b0c4bd90a2e62fbd : SUCCESS in 1h 09m 05s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/8f9b7329c2544ee6a9a584b4453afabe : SUCCESS in 42m 43s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/8a93133e14f343c2a5bd25e9f1078305 : NODE_FAILURE Node request 300-0021120488 failed in 0s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/1aae38bb228b40acbd416651bbd0cec6 : TIMED_OUT in 3h 03m 57s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/c495d162800549418c2305d279749f84 : NODE_FAILURE Node request 300-0021120490 failed in 0s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/164f41af58ad43b5a2297c18e06fe963 : NODE_FAILURE Node request 300-0021120491 failed in 0s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"d839a4c3e66d43143d4e6782e4de254ac64dc6f4","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2023-06-09 14:29:03.000000000","message":"Patch Set 5: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"d65bd2e530a2d801d8f525fa459ec88ea43267d0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-09 18:00:30.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/6fa89347f1f94a75acc6512821a048f9\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/fca7435080b0411ea920d0fdcda62726 : SUCCESS in 5m 20s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7194f27af83d4519a9a97c8c7bf4a942 : SUCCESS in 13m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a92c1735aa7e490a8ae3554fa65f1485 : SUCCESS in 5m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/01d38e5c8bc14fe6ba23eac09a02a2d8 : SUCCESS in 5m 48s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c4c606407c6f4cc09e1db3ea4fbfcaaa : SUCCESS in 5m 18s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/eb861ac457a44d5bafa0f93073d60cfa : SUCCESS in 5m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1ff09e3bdea14f60ac95d9b62c58ae7a : SUCCESS in 11m 24s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b5e46e2f523544bcbde63c30c8f551e5 : SUCCESS in 4m 37s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/5a6500573c9844dd9432c6d2741c212f : SUCCESS in 6m 35s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/1df8741f93ca4b81a308a29b175a64f5 : SUCCESS in 8m 38s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/3e18fdf6f2884237b8346cce868ec071 : SUCCESS in 5m 08s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/fb27406ec5eb40a48a7a7b508b17297d : SUCCESS in 7m 47s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/897ef758f6c845cbab9e64ddd509f6a0 : SUCCESS in 6m 42s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/4c84e20f7d564db7abd2680e8c258600 : SUCCESS in 9m 21s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/cbaf3f071b434f60bc82e10e8c4e2c3f : SUCCESS in 1h 59m 30s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/dbb4bf221efd49a0a230e65c5a74c1e9 : FAILURE in 40m 20s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/35a532f1606b4980bdb89f111a5cfdca : SUCCESS in 1h 37m 48s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/f15af2e1627d40b49c8b6d1464f17795 : FAILURE in 36m 19s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/691d436d26604cec946d87c245269a81 : SUCCESS in 1h 46m 40s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/ac945b568fe64f37af19c46ee39449a7 : SUCCESS in 35m 21s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/ec947be791124a039304ba4e2e7f0342 : FAILURE in 1h 04m 04s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/d054bac9ecb04c3bb7ddb36e5589f73e : SUCCESS in 31m 04s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/4bb1f0edca6643209156310d4d8e4cc3 : TIMED_OUT in 2h 32m 52s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/abfe2c2e07f043178d8730bbec035744 : FAILURE in 1h 40m 12s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/1fc5a6698956431e82a19e5465dd1f78 : FAILURE in 1h 22m 27s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/15a64e72ef4e46ae9b79255e47edffce : FAILURE in 1h 26m 49s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"51b2a1d4fd6b8d6c22c8f9e345750121767162ef","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-06-13 06:59:54.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"fcf21f8e8edec06b7848b73f76f8188297f1a68f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-13 09:57:07.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/20e33242c6414f0aacafbdab74d0f56c\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/623eefead090431e832dcb11d8ba73ac : SUCCESS in 4m 59s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/29599337a1af4186a53c57b5cd2c360d : SUCCESS in 10m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/01002972fc4f4439b7febd087e3d23c9 : SUCCESS in 5m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/12088ff2e4a84ff98733bea90139b1bf : SUCCESS in 5m 18s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3be54857fa7c4c41b0e36cc5de82803b : SUCCESS in 5m 06s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/da661bb6352341c49eb250ef68c61c23 : SUCCESS in 4m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d6072f6f8d744d82a3d4dd1d58dd5c24 : SUCCESS in 10m 42s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ec3134ce8f5341808e9a56ae1e26b324 : SUCCESS in 4m 47s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/9acce8b949d5474c97f7b270108812b1 : SUCCESS in 4m 45s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/a64becadf38a4aa69cd0d57263856177 : SUCCESS in 6m 55s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/d8b8b44ba26249baaf81ad9a719dcc3e : SUCCESS in 4m 31s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/407fa7a81aa2462aba7c7c71fe13134e : SUCCESS in 6m 44s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/7f4829574df2440b9c90de6bbac9b0dd : SUCCESS in 7m 03s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/7561e3fb7e21494ea62da0c37d235680 : SUCCESS in 7m 20s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/4c30bfc7671a4e5e990b7cda44fc0d20 : SUCCESS in 1h 43m 58s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/ac6dec9225544aa4a6da036a194afdec : FAILURE in 34m 59s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/f95721c7b3dd41469b3719e335c03f97 : SUCCESS in 1h 16m 06s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/be01e3ee9f414fdc9b1cb42b688f626c : FAILURE in 35m 14s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/973e3802cb2f4931bf08bda4b77beaf2 : SUCCESS in 1h 07m 53s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/21790a92c17047de92f3909b726658db : FAILURE in 38m 46s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/fac484496cc7496087daa34384c0f7aa : FAILURE in 59m 00s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/3e02a4a4aa9c4ddca619d83530ec673d : SUCCESS in 30m 49s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/27f65f04eedc48b3b8f2176e2090d6c9 : TIMED_OUT in 2h 32m 41s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/bd4679619e27467da526d8b51157b641 : SUCCESS in 2h 48m 45s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/d61a0c50d0cf44b998e9f7af2941cc8a : FAILURE in 1h 19m 51s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/b6ef640e7c034bea8b0cc89888a51710 : FAILURE in 1h 33m 07s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"2379fe8589ea3cb68aba3f6a64b363c39bf8136d","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-06-15 17:10:44.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"723cd57d202a0126ca0abfa2e06e2767f6e70103","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-15 20:15:51.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/3ef6fb552606405fb233221515379285\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/c84e247df6454441b7177da749e72115 : SUCCESS in 4m 12s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/230f0e404f274ccfbf1436fa4a291e17 : SUCCESS in 9m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d53cd55acac841908b68096200e15a86 : SUCCESS in 5m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e281f8e15e17406ea5414e83927cb18b : SUCCESS in 5m 03s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9a465f780d8b4d3da1ed867f02123c8f : SUCCESS in 4m 52s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/632964e298d546f38edaad7786f3eb24 : SUCCESS in 4m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e79064eaa6df4b3e8376396411accf9c : SUCCESS in 12m 24s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5be54b9282d34b578e7e91e3b1d54a91 : SUCCESS in 4m 45s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/13e69776864f460fa7a0a45cce08db3d : SUCCESS in 5m 10s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/83f6a3da17234da5bcc018508fd4bb7a : SUCCESS in 9m 41s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/e9ac3e2866a54428a04b0289c5cd6bd9 : SUCCESS in 6m 10s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/ec00daa271514021b6ae54309dbe4abe : SUCCESS in 9m 07s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/45c7d74930e24a598baa2fe4046dcf57 : SUCCESS in 7m 50s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/a75fa64707ac4345bda22d82e8c7da04 : SUCCESS in 7m 32s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/e58fb4db09a24efc9823ac051a143dfc : SUCCESS in 1h 21m 18s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/5a2f95bad6dd4e81bc02ccc8ac99f711 : FAILURE in 38m 59s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/088cfd0fd2144c6ab7a0b089024c7509 : SUCCESS in 1h 11m 07s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/913505076935469eb9aac9bbdc22b5b6 : FAILURE in 40m 01s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/465cd9598a1d4cbcb024f149723e6cc9 : SUCCESS in 1h 25m 42s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/a55b43cdf5b8487ab488c084dfaa323b : FAILURE in 36m 55s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/8bf31eda61354037a5bea75cbb6593ab : FAILURE in 55m 53s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/dc75d26f5cd5405aa46134b7570a90f3 : SUCCESS in 39m 34s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/12178243b871434d8e1a89c9873475f0 : TIMED_OUT in 2h 32m 58s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/5b2c80933fcc4dfc9310cf88c422282c : TIMED_OUT in 3h 02m 55s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/ab4f1cfca5b546eca9b3ed0161962dc0 : TIMED_OUT in 2h 32m 47s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/04f57ded429848d4a912b6f1316a762c : FAILURE in 1h 16m 54s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"d3aac613014e4a8488736b2061c9172fd987e6dc","author":{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},"date":"2023-06-17 05:04:16.000000000","message":"Patch Set 5: Code-Review+1","accounts_in_message":[],"_revision_number":5},{"id":"1f353f17963b11141715acc9d40d48f13ce8b923","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-06-21 11:24:22.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"c543384966dc3f59d668c91a1d8d96b9be698609","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-21 14:11:16.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c54a6bb2ddb8408890e2a7f2a72d49c3\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/b6df7aef1bdb4a6f8c45998a632d6013 : SUCCESS in 4m 41s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c970b411bc064679b036015f457fbece : SUCCESS in 10m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c16e002b11924f829f005b81bb3cdfcb : SUCCESS in 6m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8df0797d70214958b8e5d4d07b3675bf : SUCCESS in 6m 05s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a9934a6f5de346dcbec04a704ceabf10 : SUCCESS in 5m 01s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/01c97b98ae394168814bf5235fbf31b5 : SUCCESS in 5m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ad558e67bcc3480991caaef8bbd59289 : SUCCESS in 9m 36s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/aa6e8b4e3356422e8856cda509a15cd0 : SUCCESS in 6m 20s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/236cf3c40905494ba9c8f2cca115ba52 : SUCCESS in 5m 26s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/a969bf2d46d0475f87ea55a8b8cc339a : SUCCESS in 8m 15s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/5aef2206c5364d9a912d9b925280fa07 : SUCCESS in 4m 47s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/ad19eaeb68a94db38ea4faa87d9d6f35 : SUCCESS in 6m 52s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/c53672ddddbe4cc1ba36c7466ac420b8 : SUCCESS in 7m 32s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/e9e0105f23ed4044a81cf1a093efc393 : SUCCESS in 7m 44s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/58fc10e9976647a38a0eb6d79e8912a2 : SUCCESS in 1h 25m 12s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/48045b4a2c3f46faa2f3c7db2aa772db : SUCCESS in 2h 00m 23s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/1982cd013d3b4f25babf1b64c55623d6 : SUCCESS in 1h 09m 03s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/0a1762a20d8f4ce4938b372840546059 : SUCCESS in 2h 06m 29s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/a2bdaee2179d424d88b4959b4fc12790 : SUCCESS in 1h 20m 48s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/7b945290e90a45ed9f9b4073eeb9c258 : SUCCESS in 34m 24s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/9bde481098fd42f6aa067d073c0d52fc : NODE_FAILURE Node request 300-0021506378 failed in 0s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/a0ada6a88fa5496ea1460be93cdce092 : SUCCESS in 32m 57s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/fefb789a56e44633b553b2af6b835bd1 : TIMED_OUT in 2h 32m 49s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/871a84a8cb904cea902c47189e0c9b4d : SUCCESS in 2h 45m 21s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/4743b7de99274da5b3ea620b8abf46d3 : TIMED_OUT in 2h 33m 46s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/86a6903864a941438fde602756379764 : NODE_FAILURE Node request 300-0021506383 failed in 0s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"f9426ded418544cfc853fa9ecbad74eb0ab3a390","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-21 14:11:54.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"bf77c529f42c2283be7f344f70dcd0ed4ea95347","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-21 16:27:33.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ae738cefe7f04930984320bf188edc58\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/03ce2adc61ba499a82ce2a6420a76099 : SUCCESS in 4m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fa0a5aebb1cf46d6a00d74adeb294176 : SUCCESS in 4m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1254b0a5ac5b4b4cbd360692d0a69c0d : SUCCESS in 11m 33s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d0cb76330d4d44bc84c047ffbe3a7e25 : SUCCESS in 4m 58s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/aed4f9e22cd34a1ea8b5a71cb709a3d1 : SUCCESS in 12m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e1e4505627f243158d84e6c36df8eefe : SUCCESS in 10m 25s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e070293a0eb341ba8687abaa1ec7b63f : SUCCESS in 5m 23s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/7391b1de0d03464abafac40d8c86c4dd : SUCCESS in 4m 39s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/a3b71121bbd44ff28be6b1838fcaf26e : SUCCESS in 7m 14s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/858ee564ed064b5c8ecade4456a38f0a : SUCCESS in 2h 04m 56s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/d4513eabb8ee4644a4f4f7a5c5b273ef : SUCCESS in 2h 08m 06s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/128aaea63d0c47d2af4f51e99c7063c9 : SUCCESS in 1h 22m 59s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/6060761669804b828142d230dfbb0cf4 : SUCCESS in 46m 37s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/85f524c7848a42ab8fa73b7056745093 : SUCCESS in 41m 13s","accounts_in_message":[],"_revision_number":5},{"id":"ec51318e4fcf6b4b1ccddfa233128f067f2386b1","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-21 16:27:33.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":5},{"id":"6a57b5caa70ce8c2e628bd1261e43c2fe13f8c4f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-21 16:28:47.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4fbde51d7dd34f23be23635a53e02355\n\n- promote-api-ref https://zuul.opendev.org/t/openstack/build/92e2965ce0e2463eb50c61c040720e2b : SUCCESS in 45s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8dc5db5321ee4c2caf8951d8ddc48798 : SUCCESS in 53s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0cdf589a6a424ef98bdb26d1d65caf07 : SUCCESS in 43s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"f21fee9a72460dde701f7b39e89be3d2ad6a967c","revisions":{"631cfa8e5d2bcf841e5f5509f4276ceaba466f33":{"kind":"REWORK","_number":1,"created":"2022-10-14 06:19:17.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/88/861288/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/88/861288/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/88/861288/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/88/861288/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/88/861288/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/88/861288/1"}}},"commit":{"parents":[{"commit":"95adbbe618a2fb5ffe0f0e833dcfce5a9d85870c","subject":"Merge \"Add *.orig to .gitignore\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/95adbbe618a2fb5ffe0f0e833dcfce5a9d85870c"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2022-10-07 11:51:43.000000000","tz":120},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2022-10-13 17:38:57.000000000","tz":120},"subject":"[sqlalchemy2] Turn on RemovedIn20Warnings","message":"[sqlalchemy2] Turn on RemovedIn20Warnings\n\nMark all sqlachemy deprecation warnings as \u0027error\u0027, but ignore each\nknown deprecation at the moment.\nEach followup commit will remove one or more \u0027ignored\u0027 deprecation\nwarnings and will also fix them.\n\nChange-Id: Iedcfc826962ad28446e91d44eea3071c11d16973\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/631cfa8e5d2bcf841e5f5509f4276ceaba466f33"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/631cfa8e5d2bcf841e5f5509f4276ceaba466f33"}]},"branch":"refs/heads/master"},"56f463afd7a0fa379079ea32300127f02b86ef15":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2023-02-27 07:52:40.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/88/861288/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/88/861288/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/88/861288/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/88/861288/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/88/861288/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/88/861288/2"}}},"commit":{"parents":[{"commit":"b4b6e07fe07f2b1103ebe733db3f76f05d7acf0c","subject":"Merge \"Fix typo in lvs-masquerade.sh\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b4b6e07fe07f2b1103ebe733db3f76f05d7acf0c"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2022-10-07 11:51:43.000000000","tz":120},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-02-24 14:44:49.000000000","tz":-300},"subject":"[sqlalchemy2] Turn on RemovedIn20Warnings","message":"[sqlalchemy2] Turn on RemovedIn20Warnings\n\nMark all sqlachemy deprecation warnings as \u0027error\u0027, but ignore each\nknown deprecation at the moment.\nEach followup commit will remove one or more \u0027ignored\u0027 deprecation\nwarnings and will also fix them.\n\nChange-Id: Iedcfc826962ad28446e91d44eea3071c11d16973\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/56f463afd7a0fa379079ea32300127f02b86ef15"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/56f463afd7a0fa379079ea32300127f02b86ef15"}]},"branch":"refs/heads/master"},"5da06bc79ed0b94f0cff9630d9f333be6bcf66b4":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2023-03-29 08:44:58.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/88/861288/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/88/861288/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/88/861288/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/88/861288/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/88/861288/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/88/861288/3"}}},"commit":{"parents":[{"commit":"cfe19cd9f52c64cd180360360ca66acfc9a82873","subject":"Fix octavia-status with amphorav2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/cfe19cd9f52c64cd180360360ca66acfc9a82873"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2022-10-07 11:51:43.000000000","tz":120},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-03-29 08:12:57.000000000","tz":-240},"subject":"[sqlalchemy2] Turn on RemovedIn20Warnings","message":"[sqlalchemy2] Turn on RemovedIn20Warnings\n\nMark all sqlachemy deprecation warnings as \u0027error\u0027, but ignore each\nknown deprecation at the moment.\nEach followup commit will remove one or more \u0027ignored\u0027 deprecation\nwarnings and will also fix them.\n\nChange-Id: Iedcfc826962ad28446e91d44eea3071c11d16973\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/5da06bc79ed0b94f0cff9630d9f333be6bcf66b4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/5da06bc79ed0b94f0cff9630d9f333be6bcf66b4"}]},"branch":"refs/heads/master"},"fbb7c5efbed8a811681dc9e086b70d412a7959c5":{"kind":"TRIVIAL_REBASE_WITH_MESSAGE_UPDATE","_number":4,"created":"2023-04-24 06:50:24.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/88/861288/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/88/861288/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/88/861288/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/88/861288/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/88/861288/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/88/861288/4"}}},"commit":{"parents":[{"commit":"70f7998f3bf5dc2fe0dcadcec2049fadbc2bdbc4","subject":"Merge \"Update amphorav2 jobs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/70f7998f3bf5dc2fe0dcadcec2049fadbc2bdbc4"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2022-10-07 11:51:43.000000000","tz":120},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-04-24 06:41:48.000000000","tz":-240},"subject":"[sqlalchemy2] Turn on RemovedIn20Warnings in tests","message":"[sqlalchemy2] Turn on RemovedIn20Warnings in tests\n\nMark all sqlachemy deprecation warnings as \u0027error\u0027, but ignore each\nknown deprecation at the moment.\nEach followup commit will remove one or more \u0027ignored\u0027 deprecation\nwarnings and will also fix them.\n\nChange-Id: Iedcfc826962ad28446e91d44eea3071c11d16973\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/fbb7c5efbed8a811681dc9e086b70d412a7959c5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/fbb7c5efbed8a811681dc9e086b70d412a7959c5"}]},"branch":"refs/heads/master"},"f21fee9a72460dde701f7b39e89be3d2ad6a967c":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2023-05-03 07:07:19.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/88/861288/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/88/861288/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/88/861288/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/88/861288/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/88/861288/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/88/861288/5"}}},"commit":{"parents":[{"commit":"db03617acbc9b1ebf90fae6db79d5a10254eee2b","subject":"Merge \"Change Pydevd settrace configuration\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/db03617acbc9b1ebf90fae6db79d5a10254eee2b"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2022-10-07 11:51:43.000000000","tz":120},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-05-03 07:02:50.000000000","tz":-240},"subject":"[sqlalchemy2] Turn on RemovedIn20Warnings in tests","message":"[sqlalchemy2] Turn on RemovedIn20Warnings in tests\n\nMark all sqlachemy deprecation warnings as \u0027error\u0027, but ignore each\nknown deprecation at the moment.\nEach followup commit will remove one or more \u0027ignored\u0027 deprecation\nwarnings and will also fix them.\n\nChange-Id: Iedcfc826962ad28446e91d44eea3071c11d16973\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/f21fee9a72460dde701f7b39e89be3d2ad6a967c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/f21fee9a72460dde701f7b39e89be3d2ad6a967c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
