)]}'
{"id":"openstack%2Foctavia~861289","triplet_id":"openstack%2Foctavia~master~Iea874c9f220c1ae6baa52f24d6f538f747ba087f","project":"openstack/octavia","branch":"master","topic":"sqlalchemy2","attention_set":{},"removed_from_attention_set":{"29244":{"account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"last_update":"2023-06-21 11:24:42.000000000","reason":"\u003cGERRIT_ACCOUNT_29244\u003e replied on the change","reason_account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}}},"hashtags":[],"change_id":"Iea874c9f220c1ae6baa52f24d6f538f747ba087f","subject":"[sqlalchemy2] move to Session.get()","status":"MERGED","created":"2022-10-14 06:19:17.000000000","updated":"2023-06-21 16:57:49.000000000","submitted":"2023-06-21 16:56:35.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"861289-sqlalchemy2","meta_rev_id":"c5751de2e0641a6cf1d9bb5518ec55cf48b1b1e7","_number":861289,"virtual_id_number":861289,"owner":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-06-21 16:56:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},{"value":0,"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"recommended":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"all":[{"value":1,"date":"2023-05-03 07:07:19.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-06-15 17:12:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":2,"date":"2023-05-03 07:07:19.000000000","_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},{"value":1,"date":"2023-06-16 07:58:29.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-06-15 17:12:02.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},{"value":0,"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},{"value":0,"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-10-14 09:29:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-10-24 15:30:23.000000000","updated_by":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"reviewer":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"state":"REVIEWER"},{"updated":"2023-04-28 12:52:51.000000000","updated_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"reviewer":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"state":"REVIEWER"},{"updated":"2023-06-15 17:12:02.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2023-06-16 07:58:29.000000000","updated_by":{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},"reviewer":{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},"state":"REVIEWER"}],"messages":[{"id":"384a51f35130aad9d6a6b142b5dffbc29406ed0e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2022-10-14 06:19:17.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"beeca7b92dd878bfae50d5fa143876fbb27b07cc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-14 09:29:36.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/db548b165aae4d92b27aa9f1b3395455\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5bdf68e772a14842be74c33c741b16e3 : SUCCESS in 7m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f0205df24d7944c4bcd02b0462000cb5 : SUCCESS in 6m 13s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/da8ab993580e49eba391b42cf1662399 : SUCCESS in 4m 33s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6e6fa39201594e94ade243b3e35a6b93 : SUCCESS in 4m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2823d6b3473b4de9a1e046beb3174bc6 : SUCCESS in 10m 55s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/4e5a2c808b494eb3950915c4c71263df : SUCCESS in 5m 21s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/75be46cede4b40ce8533c82ad2614b79 : SUCCESS in 7m 02s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/92df1d2e078d4bec8283798e744bb45b : SUCCESS in 3m 56s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/aaa6444045b34b6bbf2ef572a76ba294 : SUCCESS in 6m 53s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/b3ce0de8b4df421b8a8932cabc4418e7 : SUCCESS in 6m 42s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/a5901dcb2aa147739d4ad3bd7da7fae6 : SUCCESS in 6m 46s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/b24e0065241f4fd6a2c8a503d0545c06 : SUCCESS in 1h 20m 40s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/8c58b456274740c7aa350151eba3e0c4 : SUCCESS in 1h 33m 37s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/53045eceb9d04111bd4b8a796cc62862 : SUCCESS in 39m 16s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/c3157e7817d94bbda4c8f90526fb0652 : SUCCESS in 52m 17s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/6d42d8fb8ad64e92a9ae05b7fa56cf34 : SUCCESS in 1h 11m 37s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/5fa3621922764346b104f4af6c343a8b : SUCCESS in 36m 33s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/a129d5d8e0c449f78e0fccd5efe33f0a : SUCCESS in 1h 48m 43s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/9668cdcfce0c497ca74d949fef1f4f9a : SUCCESS in 2h 28m 42s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/852a6b466f1546d5af4a70abe65c407b : SUCCESS in 2h 18m 07s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/f112fc45102f47bc92381fdc1b243df2 : SUCCESS in 1h 58m 36s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/00bcde6d6068431794cdd4bb09706a9e : SUCCESS in 1h 49m 18s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/d821f6a465964c45b8a981a7daafaf95 : TIMED_OUT in 2h 33m 27s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b3251e915498ae4e443edcf7deb9574aa9281a5c","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2022-10-24 15:30:23.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"936a94ac47929a5b621a7a4aa697c9a51a1d21e2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-02-27 07:52:40.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"4db5385753078c81a3a8b6f628a21479411b22f9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-27 10:11:17.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/abe13753e832448086dcf7f5add81ee9\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bf88460e647d4092a1d316f205df20a2 : SUCCESS in 11m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3604b2390af647d1acb088a9628c637b : SUCCESS in 5m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fe62c706d84f425789e142b706e08587 : SUCCESS in 4m 20s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/40f62a9d67d94ac0801b416cfa164360 : SUCCESS in 5m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/acb87d938ef94d358a281da849fff2bc : SUCCESS in 12m 32s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/d07253a4cc6c487e8c05342994086930 : SUCCESS in 5m 39s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/0782bdcb679f4909b24a01a142d33e89 : SUCCESS in 7m 57s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/1169186c5d9a43fa84dd4330748b9691 : SUCCESS in 4m 16s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/cd2ce321e3bd4aefa0d84f1b2d56f54b : SUCCESS in 7m 24s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/573ee6f9f6de476ba0ebb8f2de9288e9 : SUCCESS in 6m 53s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/b2a14995410f402a8a37a8a75d423298 : SUCCESS in 6m 48s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/eefdedd0cda64d2ba5c5fb010a8b31be : FAILURE in 5m 57s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/426f3974703d4700b50cc065075468ff : SUCCESS in 2h 04m 32s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/f7df6c4a7c014de09dc80c05d647884b : SUCCESS in 45m 59s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/202f44e3cbf3432583402c7276d9807d : SUCCESS in 40m 26s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/3f9d3863efae4d55bd29415fbc4f7740 : SUCCESS in 1h 22m 34s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/09afebce7e9c4acaa92a069d1affa000 : SUCCESS in 31m 30s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/3ea67b59e67f487f8e4e3bbb6269d9fe : SUCCESS in 1h 55m 52s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/27366b9e21d049a29eae8e93f5fb9772 : SUCCESS in 2h 15m 55s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/91014fae1a304071ae895e051f8fb85f : SUCCESS in 1h 39m 37s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/d803b3b1a0a6495da113bc7e18ce38a8 : SUCCESS in 2h 04m 24s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/5f7f408fe0174fc4a10fd09c4c74b87c : SUCCESS in 2h 07m 59s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/002925fc8bc04cb898c6399ad8c6486e : SUCCESS in 2h 07m 09s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"ffd121c9116b4e040bfdecdaaea75d86802a6819","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-03-29 08:44:58.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.","accounts_in_message":[],"_revision_number":3},{"id":"c071c934bdfc1b2eaa369df1beecd9c02b0eb1f6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-29 12:08:24.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1bf8a9dbb9cc445fa2a5af1761769b16\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a962cea5364b42a38e94679a5f34dc74 : SUCCESS in 10m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7ec37ddc3ed748cdbe8a0ee25cd0eb65 : SUCCESS in 6m 07s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/29633defeb0440e3b5c3be4c79ae457c : SUCCESS in 5m 01s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c8a953b103a446318de3be0bd74b3a0d : SUCCESS in 4m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7855050e0134436ea770f5378c953344 : SUCCESS in 12m 43s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/8440bfe21e014fb882f560354debc1fa : SUCCESS in 5m 33s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/8ea7178559254ef0941fd983289798c1 : SUCCESS in 9m 08s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/c4f74ea90ea14b81acccdbcd566e7046 : SUCCESS in 5m 39s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/697c6eb22a9846d4bee663c034c4d4aa : SUCCESS in 7m 44s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/25571e82871a4372a415468938975531 : SUCCESS in 6m 58s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/6ab45ddea510418d98fdfc012308ecbe : SUCCESS in 7m 24s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/98915446c84545ca9677831ccbe3acf4 : SUCCESS in 1h 20m 44s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/55d7a4e4f5a24d63bb4e8e9dbc236865 : SUCCESS in 1h 45m 06s\n- octavia-v2-dsvm-scenario-jobboard https://zuul.opendev.org/t/openstack/build/12598b09c68a4cbfaf42839262189d9f : SUCCESS in 1h 48m 26s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/6f89794207244a1186326507bbbb1bf6 : SUCCESS in 44m 27s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/ca6edd463f4b4c2090388a6408fb3421 : SUCCESS in 37m 52s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/c5751f13921b40efb54e5fe6efc42802 : SUCCESS in 1h 03m 38s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/7ef9af83bbb245f785e2fb34aeb617db : SUCCESS in 31m 23s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/4502b2696b31464cb602fac4d0da5842 : SUCCESS in 2h 31m 34s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/80863c1e782f437890d2a90100fa5226 : SUCCESS in 2h 11m 16s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/080dd1591ac04b608bfd0787bf11704c : TIMED_OUT in 2h 33m 06s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/cd4a6b6ab13d440aa4be3002d2529a9d : SUCCESS in 2h 30m 48s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"fea9e38b1f793057eb1f96fd3ed2264e33e7afe7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-04-24 06:50:24.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.\n\nCopied Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"cd940db72cbe77ba12e1797e5e0074b5a85a4cd0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-24 09:27:11.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/0376b5b43f0d4566a983de31f429bfec\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2011be3b1d334e7eac156fa1e4675f9c : SUCCESS in 11m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/18453f29cc974f58bba2554de1ffea3f : SUCCESS in 6m 24s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ad77469fab254178816bdc313900de32 : SUCCESS in 5m 22s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/44452a5cad77478a90a33ff3111940e4 : SUCCESS in 5m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1a363f92abd54f67a5148d60d2a0556e : SUCCESS in 11m 53s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/cbe20e1ad9264b928c64515ca946ec92 : SUCCESS in 5m 45s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/16e59687058547b186d26e4b6f699101 : SUCCESS in 8m 10s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/ae9588dedaab48ea870b9cabe96ab2c5 : SUCCESS in 5m 09s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/a7195549a6d943c6a34f3b9b4eb740de : SUCCESS in 6m 50s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/354706e3e0b146fab95948ef3cf40914 : SUCCESS in 7m 36s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/6857821e73fd42e491b0d175317b219c : SUCCESS in 9m 16s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/58fe6288772546619d92a2a95f666b90 : SUCCESS in 1h 21m 10s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/0781faaef0b6421a88dc046dfd6d259d : FAILURE in 11m 49s\n- octavia-v2-dsvm-scenario-jobboard https://zuul.opendev.org/t/openstack/build/18ab785dca6f4fd28f0299319c17c7b1 : FAILURE in 9m 25s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/57daf0f4c62f4bbca85bcc0c6c4d6041 : FAILURE in 9m 52s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/545eb71ca81449ad936ff0f9c06060d1 : FAILURE in 44m 56s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/710a4093e9544ba6a2f08e057533bad9 : SUCCESS in 1h 07m 52s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/3428b4b39c144006b6aa34f283b5dfe6 : FAILURE in 7m 31s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/fcc97060a28740db8343e1cc3f95e3e6 : FAILURE in 7m 41s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/cd30ce2bf3694f9fa63b0264459e14b1 : FAILURE in 13m 25s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/b0062baa38154384a0727f2f8fa3c697 : FAILURE in 11m 38s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/747e56fcdef64d9b8115b940a1802778 : TIMED_OUT in 2h 34m 31s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"cb0d6968c244a38f1591f14c864eab604deac68e","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2023-04-28 12:52:51.000000000","message":"Patch Set 4: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"e714dc0d02f423a8a4f7a68367fe5a9f14db8bec","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-28 15:55:25.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/015377366a774467941e1c34f357ab6b\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2e256d15e3a74ecea0e05e8ff4ee3fb1 : SUCCESS in 11m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/753bfa8cbee742429f563caf86c0a4e1 : SUCCESS in 5m 57s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0282ffdb693140e79fe2e124a3a96b7e : SUCCESS in 4m 50s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/912f0670499842de98211be401ac9f97 : SUCCESS in 4m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/82afa1d91e874425863a47076884262f : SUCCESS in 11m 37s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/340157fee82b4ebea882a9c16efe10a3 : SUCCESS in 5m 00s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/ee45fb8ed4224cc0b3cdcf51e719244b : SUCCESS in 9m 39s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/ea0b66d0b724441cb653589754d4734c : SUCCESS in 5m 10s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/e2014de4fee2475ab02fd6f3fdfc2011 : SUCCESS in 7m 30s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/a191b586c42d441fb02dc1650be74c3c : SUCCESS in 6m 44s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/ae6e06f1f8644422bb60ff377edfb159 : SUCCESS in 7m 36s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/3e795d1a4a4a4996adaf694d3a6c0aaa : SUCCESS in 1h 20m 44s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/fa46dd1667e24458a4ff3ad71299979c : SUCCESS in 2h 27m 55s\n- octavia-v2-dsvm-scenario-jobboard https://zuul.opendev.org/t/openstack/build/e283260c08964ccd865c404a687053ef : FAILURE in 7m 54s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/92e67c7e546f4af7b88eee01f7a6e038 : FAILURE in 7m 57s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/0a851208cdd8408aa4e790128847a5a8 : SUCCESS in 46m 43s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/707fe51151e84c73b30d749294a68c3b : SUCCESS in 1h 07m 16s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/364b17e704994df79c5a9e33e091e220 : SUCCESS in 34m 06s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/05d0c95cf044497eb09d6d9595a648c7 : FAILURE in 7m 15s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/b86caf2611a6435fa3964dde395e9ae9 : SUCCESS in 2h 54m 37s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/c4fde0bf0b304933be2519be9e4ddc17 : TIMED_OUT in 2h 32m 32s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/319806503c074258b80e1b914ac582c6 : TIMED_OUT in 2h 33m 38s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"9210f0586b6588db520762eae0b705f76b1ebbe8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-05-03 07:07:19.000000000","message":"Uploaded patch set 5: Patch Set 4 was rebased.\n\nCopied Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"7e88d798377bc4f322f06b64759b85a3605f6ac0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-03 09:52:55.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/bd10dba2705242d0a4aea84a1b217665\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c9d107d5d585445a95d036b48fa563a1 : SUCCESS in 9m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bb67dce0a9f9448ba3d6885d92f0700a : SUCCESS in 5m 09s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ace51ad1d7ed47e8a7fad22b7d9eb0eb : SUCCESS in 5m 30s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/fd89c65f408f402fb0cdac096275cc26 : SUCCESS in 4m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c9c97933a952474dafa2777be4457f2c : SUCCESS in 10m 31s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/788a7cab3d3a4b6d998983111601ab96 : SUCCESS in 5m 15s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/aee0ffaac2a24d4bb607b980e5f9c76a : SUCCESS in 6m 50s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/435b1f5cb6594906b09b6fca75415330 : SUCCESS in 4m 38s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/a8878557ea0848acbefb0e175a53e08f : SUCCESS in 7m 17s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/d8a7b44fa7e6497e9e147d7331c0acbc : SUCCESS in 8m 12s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/833cd1a55cf248febaff73d67e82d7f0 : SUCCESS in 6m 59s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/3df0203f8aba45ec8a14b2f7bb4fd5af : SUCCESS in 1h 46m 40s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/c39788fe79cd48acb9ad734cc22b3dc9 : SUCCESS in 2h 16m 38s\n- octavia-v2-dsvm-scenario-jobboard https://zuul.opendev.org/t/openstack/build/ec09bffa2c504bb6b48aaf8349fcb74f : TIMED_OUT in 2h 33m 11s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/81707e710b7845188a136dfd0d62fba5 : SUCCESS in 39m 16s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/4690319f77904a1d84f386aafb0f087a : NODE_FAILURE Node request 300-0021120507 failed in 0s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/0573436b17974c5f9a6ce7db46435db6 : NODE_FAILURE Node request 300-0021120508 failed in 0s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/9dcf9143f68f4daaae04ae2e0f7cb754 : NODE_FAILURE Node request 300-0021120509 failed in 0s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/4db5613747f74cd9b4ce8083bb434526 : NODE_FAILURE Node request 300-0021120510 failed in 0s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/0808b0b0637242a29bc60d55b0d3c85a : NODE_FAILURE Node request 300-0021120511 failed in 0s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/0ffe86aef4df45cdbcad48d6f53ee921 : NODE_FAILURE Node request 300-0021120512 failed in 0s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/334b956d1f8a4db08f788fc4feca21a2 : NODE_FAILURE Node request 300-0021120513 failed in 0s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"b89555c2e5b96897710101c8f034e780b6029ca2","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-06-15 17:12:02.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"5be7ec4940a720ebb1733637d35ced6a7fee7671","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-15 20:30:55.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/8e0db73d8f5e45b2b51ffdf54d904ea9\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a2822b3e3ec04680b7390d92f4fdcc10 : SUCCESS in 12m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1ce74f47e53f4f71a63e3b69365c7496 : SUCCESS in 4m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1694f08c874643de897f1c8aa6875d86 : SUCCESS in 4m 46s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3ff28e5296ff495b8d6df4bbe0d730d4 : SUCCESS in 4m 17s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/1f258b08cfc94888be8f8bb620a8de57 : SUCCESS in 4m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9948c7ff20354aeeb2933d08c5f5e327 : SUCCESS in 9m 54s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/0d842eceb2d24845864e7a8fed7d7b30 : SUCCESS in 5m 14s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/aa117a4091ba4c1da8b1fdf4b926ded2 : TIMED_OUT in 31m 15s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/62f308ac4b294c27a7b2785bced90a3f : SUCCESS in 4m 16s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/62b6de5cccb24f05b59ecc517ff7721a : SUCCESS in 6m 58s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/6bb3feab3ab8476b86f03fca943f7b26 : SUCCESS in 8m 11s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/dadbe1e83efe40f3845d28c3a3d94df7 : SUCCESS in 8m 01s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/0b2a7a179bf34debab40b3332e2f6385 : TIMED_OUT in 3h 14m 33s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/951f22b26dea48378b4943f1d933c1de : FAILURE in 35m 27s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/d4ca875d00fa45b4be124fe968e610c9 : SUCCESS in 1h 18m 02s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/25bfcb4c85ea423f8f43fdd6505ad6b8 : SUCCESS in 2h 00m 27s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/6386545d57b646f19255cdcfdf7f183d : SUCCESS in 1h 17m 44s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/5b299a86b67d45b19779875dbeda2dec : SUCCESS in 38m 23s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/e48ba0c598a94eccb80a912fd595158b : SUCCESS in 37m 51s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/eee2be29cdd146cd8d8c64f7133a4a8a : FAILURE in 1h 00m 15s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/ffc3026b7b4d41159c98e5ab01c21a30 : FAILURE in 35m 03s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/32b67eeaf5a14b9cb60e4b11cf3a0aad : FAILURE in 1h 21m 28s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/27baf963346f4e65b38a4a069e7c78d8 : FAILURE in 1h 24m 20s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/50526238ab98429190697cea7dca3824 : FAILURE in 1h 19m 17s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/3c23cea64c9a4402a2284ad857dc4c78 : FAILURE in 1h 32m 53s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"b4ac4b7935ddc87999203a4ba7b16984c157bcf9","author":{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},"date":"2023-06-16 07:58:29.000000000","message":"Patch Set 5: Code-Review+1","accounts_in_message":[],"_revision_number":5},{"id":"31f7abeee4b9985af804ce9d807629c6bab08dc5","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-06-21 11:24:42.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"cb91eebddf24df40cec638243fb215b074fdc92c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-21 14:32:40.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4094abe32ea94717a0d86e11e546b54f\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/faa184a7ae1a41899c5fc6dccac96867 : SUCCESS in 8m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9bc70558b35e4554a86ad2c2ed9963fb : SUCCESS in 4m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d3d17e1b6ef44919904c92de23916513 : SUCCESS in 4m 34s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c2758d4c476b46718107a6e4edce1130 : SUCCESS in 5m 06s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5b3f0448671144ac9e4d466ec709f422 : SUCCESS in 4m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c4b7bf8574a5407ab37d3d110ae3005e : SUCCESS in 16m 19s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/f782ad8bbdb34b82b0e5b9130ff7eab6 : SUCCESS in 5m 20s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/b967322f60b84cb98d57cf8278013951 : SUCCESS in 7m 10s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/f1ab581b628a43ddaefffd81ad70b3ea : SUCCESS in 4m 15s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/2018445def414fda9a78848e4950ad2a : SUCCESS in 7m 54s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/095143adc7864a84bdf4906cc55ee246 : SUCCESS in 8m 17s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/10372dd710e94403a4f232cbd92d8e25 : SUCCESS in 8m 53s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/7400d85d4b2f48f69f575e82fe693ad8 : SUCCESS in 1h 31m 31s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/a453e37c30944505acde0565311976f4 : SUCCESS in 2h 03m 53s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/1c898915b5a642aa83da9da07c96a824 : SUCCESS in 1h 17m 46s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/3c08990880a942479b732002ec6cb129 : SUCCESS in 2h 07m 28s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/b0d84f13ef564335b25305ae94f66f98 : SUCCESS in 1h 18m 58s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/13510faddd944beaad9f0e13c1b6885b : SUCCESS in 36m 37s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/36a826ae52a647e28555d87424a02743 : SUCCESS in 38m 01s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/4fd20719fd164f30b36fb786eaab4332 : NODE_FAILURE Node request 300-0021506414 failed in 0s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/4d5f11a3959042d9a10f6a730cf910a9 : SUCCESS in 33m 36s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/18faf164a0564525b1d8a91a636bd7ad : TIMED_OUT in 2h 32m 56s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/162122b7567f43918f6479b92b184566 : SUCCESS in 2h 52m 11s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/4fc841fa29b64ca2a0e8baeb161ae96e : TIMED_OUT in 2h 33m 57s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/695f4d643cf94c3985c980fa03d9bc79 : NODE_FAILURE Node request 300-0021506419 failed in 0s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"bed62c7751ff561f1ce9c52320a10c3e715996ba","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-21 14:33:19.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"b580cfcf8c33088727d379bbe406a15e5f809aa7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-21 16:56:35.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0789581274104eac89d80aaa535aacc7\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d82bda8463e8482287864cde0c3207a1 : SUCCESS in 6m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/591780fd002243348572efe32491111d : SUCCESS in 4m 20s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a3f53de2fdfa4bd6bf414c91c4dd2304 : SUCCESS in 5m 37s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d96ab5dfddfa49ccbdba715db19e264b : SUCCESS in 5m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/462ab33451e646bca4f81bd47b5d2987 : SUCCESS in 12m 55s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/073370adb2b5426584f04050a5fecf90 : SUCCESS in 4m 49s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/5c33b19768434af594569a7394bb6602 : SUCCESS in 7m 52s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/02e30049b43b4b8fad98e67bce65726c : SUCCESS in 1h 47m 56s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/a72cd2791b914e70b23a0136b117aba4 : SUCCESS in 2h 08m 36s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/3add624d72c74873a91a7c35c0af2bb6 : SUCCESS in 1h 20m 44s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/ffbe50ebfec043df86cff9c4317367a1 : SUCCESS in 37m 33s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/ac03538a02c04e6bbf5a997bd52aa4a7 : SUCCESS in 36m 38s","accounts_in_message":[],"_revision_number":5},{"id":"4a8f27e7f4b639ea51e0d4d6e05eba695c0bc337","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-21 16:56:35.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":5},{"id":"c5751de2e0641a6cf1d9bb5518ec55cf48b1b1e7","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-21 16:57:49.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e751cdb445154fba938ba65b5c157c2e\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e31b237a0bf34edcb0bc9507bd4ac99e : SUCCESS in 54s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"e0bff29efd73568b5ef53b5e2bb5156d57737e8f","revisions":{"aa32c785fda45d048996caf95e7bd0adf2f2bd7c":{"kind":"REWORK","_number":1,"created":"2022-10-14 06:19:17.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/89/861289/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/89/861289/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/89/861289/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/89/861289/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/89/861289/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/89/861289/1"}}},"commit":{"parents":[{"commit":"631cfa8e5d2bcf841e5f5509f4276ceaba466f33","subject":"[sqlalchemy2] Turn on RemovedIn20Warnings","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/631cfa8e5d2bcf841e5f5509f4276ceaba466f33"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2022-10-07 16:03:18.000000000","tz":120},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2022-10-13 17:43:19.000000000","tz":120},"subject":"[sqlalchemy2] move to Session.get()","message":"[sqlalchemy2] move to Session.get()\n\nQuery.get() is a legacy method [0] in sqlalchemy, move to Session.get()\n\n[0] https://docs.sqlalchemy.org/en/14/changelog/migration_20.html#\\\n    orm-query-get-method-moves-to-session\n\nChange-Id: Iea874c9f220c1ae6baa52f24d6f538f747ba087f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/aa32c785fda45d048996caf95e7bd0adf2f2bd7c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/aa32c785fda45d048996caf95e7bd0adf2f2bd7c"}]},"branch":"refs/heads/master"},"b38821475d7bf250fd840da4633b7961d6e6c2c3":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2023-02-27 07:52:40.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/89/861289/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/89/861289/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/89/861289/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/89/861289/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/89/861289/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/89/861289/2"}}},"commit":{"parents":[{"commit":"56f463afd7a0fa379079ea32300127f02b86ef15","subject":"[sqlalchemy2] Turn on RemovedIn20Warnings","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/56f463afd7a0fa379079ea32300127f02b86ef15"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2022-10-07 16:03:18.000000000","tz":120},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-02-24 14:44:49.000000000","tz":-300},"subject":"[sqlalchemy2] move to Session.get()","message":"[sqlalchemy2] move to Session.get()\n\nQuery.get() is a legacy method [0] in sqlalchemy, move to Session.get()\n\n[0] https://docs.sqlalchemy.org/en/14/changelog/migration_20.html#\\\n    orm-query-get-method-moves-to-session\n\nChange-Id: Iea874c9f220c1ae6baa52f24d6f538f747ba087f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b38821475d7bf250fd840da4633b7961d6e6c2c3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b38821475d7bf250fd840da4633b7961d6e6c2c3"}]},"branch":"refs/heads/master"},"18eea046f05d2c173856bbe00a6e383dd214afca":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2023-03-29 08:44:58.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/89/861289/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/89/861289/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/89/861289/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/89/861289/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/89/861289/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/89/861289/3"}}},"commit":{"parents":[{"commit":"5da06bc79ed0b94f0cff9630d9f333be6bcf66b4","subject":"[sqlalchemy2] Turn on RemovedIn20Warnings","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/5da06bc79ed0b94f0cff9630d9f333be6bcf66b4"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2022-10-07 16:03:18.000000000","tz":120},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-03-29 08:12:57.000000000","tz":-240},"subject":"[sqlalchemy2] move to Session.get()","message":"[sqlalchemy2] move to Session.get()\n\nQuery.get() is a legacy method [0] in sqlalchemy, move to Session.get()\n\n[0] https://docs.sqlalchemy.org/en/14/changelog/migration_20.html#\\\n    orm-query-get-method-moves-to-session\n\nChange-Id: Iea874c9f220c1ae6baa52f24d6f538f747ba087f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/18eea046f05d2c173856bbe00a6e383dd214afca"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/18eea046f05d2c173856bbe00a6e383dd214afca"}]},"branch":"refs/heads/master"},"19385c050a0879bb378d28505aadedc27db2f5e9":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2023-04-24 06:50:24.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/89/861289/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/89/861289/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/89/861289/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/89/861289/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/89/861289/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/89/861289/4"}}},"commit":{"parents":[{"commit":"fbb7c5efbed8a811681dc9e086b70d412a7959c5","subject":"[sqlalchemy2] Turn on RemovedIn20Warnings in tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/fbb7c5efbed8a811681dc9e086b70d412a7959c5"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2022-10-07 16:03:18.000000000","tz":120},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-04-24 06:41:53.000000000","tz":-240},"subject":"[sqlalchemy2] move to Session.get()","message":"[sqlalchemy2] move to Session.get()\n\nQuery.get() is a legacy method [0] in sqlalchemy, move to Session.get()\n\n[0] https://docs.sqlalchemy.org/en/14/changelog/migration_20.html#\\\n    orm-query-get-method-moves-to-session\n\nChange-Id: Iea874c9f220c1ae6baa52f24d6f538f747ba087f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/19385c050a0879bb378d28505aadedc27db2f5e9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/19385c050a0879bb378d28505aadedc27db2f5e9"}]},"branch":"refs/heads/master"},"e0bff29efd73568b5ef53b5e2bb5156d57737e8f":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2023-05-03 07:07:19.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/89/861289/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/89/861289/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/89/861289/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/89/861289/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/89/861289/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/89/861289/5"}}},"commit":{"parents":[{"commit":"f21fee9a72460dde701f7b39e89be3d2ad6a967c","subject":"[sqlalchemy2] Turn on RemovedIn20Warnings in tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/f21fee9a72460dde701f7b39e89be3d2ad6a967c"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2022-10-07 16:03:18.000000000","tz":120},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-05-03 07:02:50.000000000","tz":-240},"subject":"[sqlalchemy2] move to Session.get()","message":"[sqlalchemy2] move to Session.get()\n\nQuery.get() is a legacy method [0] in sqlalchemy, move to Session.get()\n\n[0] https://docs.sqlalchemy.org/en/14/changelog/migration_20.html#\\\n    orm-query-get-method-moves-to-session\n\nChange-Id: Iea874c9f220c1ae6baa52f24d6f538f747ba087f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e0bff29efd73568b5ef53b5e2bb5156d57737e8f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e0bff29efd73568b5ef53b5e2bb5156d57737e8f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
