)]}'
{"id":"openstack%2Foctavia~861310","triplet_id":"openstack%2Foctavia~master~I3c3af331e147c8bbff560c42bca4d51e864aac07","project":"openstack/octavia","branch":"master","topic":"sqlalchemy2","attention_set":{},"removed_from_attention_set":{"29244":{"account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"last_update":"2023-06-21 11:24:49.000000000","reason":"\u003cGERRIT_ACCOUNT_29244\u003e replied on the change","reason_account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}}},"hashtags":[],"change_id":"I3c3af331e147c8bbff560c42bca4d51e864aac07","subject":"[sqlalchemy2] use attribute to indicate relationship","status":"MERGED","created":"2022-10-14 06:19:17.000000000","updated":"2023-06-21 17:00:09.000000000","submitted":"2023-06-21 16:58:53.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"861310-sqlalchemy2","meta_rev_id":"5d697a0771a98623838ac5215caf4dceaa3eb531","_number":861310,"virtual_id_number":861310,"owner":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-06-21 16:58:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},{"value":0,"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"recommended":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"all":[{"value":1,"date":"2023-05-03 07:07:19.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-06-15 17:12:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":2,"date":"2023-05-03 07:07:19.000000000","_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},{"value":1,"date":"2023-06-17 05:05:10.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-06-15 17:12:51.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},{"value":0,"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},{"value":0,"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-10-14 06:38:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-10-24 15:30:35.000000000","updated_by":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"reviewer":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"state":"REVIEWER"},{"updated":"2023-04-28 13:02:24.000000000","updated_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"reviewer":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"state":"REVIEWER"},{"updated":"2023-06-15 17:12:51.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2023-06-17 05:05:10.000000000","updated_by":{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},"reviewer":{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},"state":"REVIEWER"}],"messages":[{"id":"ca580ded71501073124afe7f3d2a68559be62e2e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2022-10-14 06:19:17.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"22c6eb806c1190f4970c6db4f59f155050bef4d0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-14 06:38:52.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7ef1284ebb8a4654b43f697e13cfdd9e\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3e7f836c575c4cdf916a4456d61dabfb : SUCCESS in 11m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8e176cf70432401bb37630b9f60c2105 : SUCCESS in 6m 07s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/96c428d1802449589700f6e7644e5e82 : SUCCESS in 4m 44s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6699e26e5f204c3896fdb603c2b4e408 : SUCCESS in 4m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/53c18545939b4ca49398fe20939c70b6 : SUCCESS in 11m 36s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/559ffe58199940a28f49190df6327a19 : SUCCESS in 4m 58s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/77c566f3638c4ee796b03de1d3c1c74b : SUCCESS in 7m 49s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/d5f9b7ca48764dbdace8b3791273a7db : SUCCESS in 8m 28s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/9a31a8804dbb4700a2bfa621ce1900af : SUCCESS in 8m 17s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/a0a7237bff2a438c9f551c36df10d424 : SUCCESS in 8m 58s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"98508797ef93074886f564296bf0053978d3b2a0","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2022-10-24 15:30:35.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"dddc76999cf9007a46f9446ed7d86b937918eaa9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-02-27 07:52:40.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"3084a82237c52c5ef006c93a6c3cd59e41f1aa63","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-27 08:21:29.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/0115e7a2fe3744d1a63bfd89b337c97d\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2c6075236df6496e95f6f033c6daf8d2 : RETRY_LIMIT in 3m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/66db82a593914e298de6a597fdd25dec : SUCCESS in 4m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/965344bf68bf470482ce20895c1aef35 : SUCCESS in 4m 44s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9202445ddc994c89aed97a05f6a2f1c8 : SUCCESS in 4m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8bf82577263e4e58ab33874f54b61cf8 : SUCCESS in 9m 58s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/3fa9a6bea59c4ea6855bad7d0503cad5 : SUCCESS in 4m 34s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/8cfd83987caa4ec29e5f9aa46e02b460 : SUCCESS in 7m 52s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/b38cd8cc64534378800acf7d0aa9d29a : SUCCESS in 6m 31s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/19d5f4e934354eb0abfa4053266cfbb2 : SUCCESS in 6m 49s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/681f4627e7974c96904d62bc643f4b67 : SUCCESS in 8m 53s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"20671577c5f272917737b4308d9aab3e701eee5a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-03-29 08:44:58.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.","accounts_in_message":[],"_revision_number":3},{"id":"429251ae99c5413c29193b13be9c47e04dc1afe6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-29 09:03:38.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b076b8f6399e4f8e9c33ce0adaa4b2f6\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bc44c01c12db491db6901455cf0d5420 : SUCCESS in 10m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/acf73d86a1fe4cdab04b6e343576752f : SUCCESS in 4m 50s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/86555c9db9e14562b0c78c16969b0180 : SUCCESS in 5m 19s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f615a9dabefc4363a3d0e71f11897386 : SUCCESS in 4m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/45b726a8e1374a78a0112901f14596c5 : SUCCESS in 13m 40s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/158359f876b04512971106c1636a4849 : SUCCESS in 5m 21s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/9f7fdb2576ce463d913b5d1c55998159 : SUCCESS in 6m 56s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/55767c6e2bfb44c8a92a140a557ccb9e : SUCCESS in 8m 40s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/a203b68d5a274e4d84aa9d4e67aa0a30 : SUCCESS in 6m 40s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/fb22a9d215fe473fb7e4d0032712d17e : SUCCESS in 9m 04s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"d0bd53f5f4bf6305840a7bdad44ee583f877a557","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-04-24 06:50:24.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.\n\nCopied Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"5f7118c111d70fc17ad9d461c87be8b1cacb1fd5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-24 07:17:34.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a49421eb1e5844bfbd333760561a365b\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7f9c63193fad46179b61834274b5d3a1 : SUCCESS in 8m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/04b4fcd70f924bdead6b1b25452193e3 : SUCCESS in 5m 51s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4631228663814629a4a9d4d28b0b1d09 : SUCCESS in 4m 53s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/176c776cc849400b95e9e8165747c9b3 : SUCCESS in 4m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c9f18b230e404201bc77aa53c421b82d : SUCCESS in 11m 14s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/bdf8dbc701ec4ff7b7f8b36783e3f29d : SUCCESS in 5m 07s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/8c052cc3f3434bbeaa36fb6670c85aff : SUCCESS in 9m 50s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/31875dae3c644ae9aed1454ab9348525 : SUCCESS in 6m 38s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/8f9bae7a251b4824bf1d1d2597017718 : SUCCESS in 8m 00s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/b8b80e4e81c0432789d9e95fc9988435 : SUCCESS in 6m 32s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"aaf5f5c21deabdd5c19cc61fe45c85fbd7db90bc","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2023-04-28 13:02:24.000000000","message":"Patch Set 4: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"2c0c48e96f2cb58cfbad56d73ddf3a819a12c4dd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-05-03 07:07:19.000000000","message":"Uploaded patch set 5: Patch Set 4 was rebased.\n\nCopied Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":5},{"id":"79412920f6a982fe7756b1cd793de9807aff83ec","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-03 07:29:36.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cb12e131846b46d9a20aced370756bde\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/98230f0e87af4e37969537bf41a61a58 : SUCCESS in 9m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6b339c4f92414edaad758a6570045394 : SUCCESS in 5m 15s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f766f32d9d40404ba7624a1c49f04b31 : SUCCESS in 5m 03s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f0c4665ef7014d76bd46ca174630e50a : SUCCESS in 4m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1184994d73654fd392b80fe324c078dd : SUCCESS in 11m 50s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/13d1b13524e4417090472cfc34476075 : SUCCESS in 4m 50s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/9e138f265e714cb489e83b6a448565b0 : SUCCESS in 8m 56s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/ec8cb1ab54d54cbb9f69a76b8cf7610b : SUCCESS in 7m 10s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/4e11a3a12f6249cabb10c27c81646162 : SUCCESS in 7m 15s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/00510a52a3dd4b779c482f56b2d3cf50 : SUCCESS in 7m 09s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"8864daa7e1a521e66944ebe1ec5b770dc892d5a3","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-06-15 17:12:51.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"ec09d592eb329bbdb9b82c0b1042520b19fe92bf","author":{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},"date":"2023-06-17 05:05:10.000000000","message":"Patch Set 5: Code-Review+1","accounts_in_message":[],"_revision_number":5},{"id":"87a56ba3fe4fdba37b5c3c0d71bfd49ea25123b3","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-06-21 11:24:49.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"eaae81a9e44761ace7972c3372f67b21aa8de0b6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-21 12:00:14.000000000","message":"Patch Set 5:\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a79f217720e542e78e5083ce5cd0d65d\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d7f90f11e17d4befb84e3cf92a1c4c6d : SUCCESS in 11m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f64e6c74578743b0b3a2f4130614a2db : SUCCESS in 5m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/89716756fa8c4a3b98eac4a7b548280d : SUCCESS in 5m 02s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8e3ffa673413409283db8f2a68892d60 : SUCCESS in 4m 29s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/504db100b4314118b4492505aafc885d : SUCCESS in 5m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5d612a2ef62e410ab503b23274e2f623 : SUCCESS in 11m 02s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/39ebfdb7877f4ad1bd6253d9eb932eaa : SUCCESS in 5m 23s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/cde539e9b4d0489686afb530093db39d : SUCCESS in 9m 01s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/f08c42c3cc294a98b8f9d089a39a2d4f : SUCCESS in 7m 06s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/f6a443df155442d8a3f53a55d0dbe19e : SUCCESS in 7m 55s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/419cec1ffd7444ac9be149d6d6a59412 : SUCCESS in 9m 00s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"ff226e7c729e0fe85ab3542007adfd9153d36068","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-21 14:33:22.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"511cad315eb343920b27654cb7514c9c0600c597","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-21 16:58:53.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/276a47545b9044c69e792433705f7b73\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7d580963b70948c2bed15e35345b564c : SUCCESS in 5m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d966dd98b57d49b8b057bafb19d84265 : SUCCESS in 4m 49s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/726ed42ff15647149f23b9269e1a691d : SUCCESS in 6m 34s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b0856580cc35428c97501b0ffcbf2639 : SUCCESS in 5m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ca3f777cf6f84968a1164a18b7dc9241 : SUCCESS in 13m 46s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/b6bcfdcec270462abf2ce8c785c2cf91 : SUCCESS in 8m 50s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/d1c328504fc34fb886c5b4504921f288 : SUCCESS in 1h 46m 15s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/dd0db3f3d30942ea9c12b6cf1cd5e6c9 : SUCCESS in 2h 10m 56s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/90ba67501f7441df8063697a79366fc5 : SUCCESS in 1h 11m 49s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/337b7aa5fb19472380c0f4befb1a01f5 : SUCCESS in 40m 57s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/3145c005bea94ec78705af18f0bfbf42 : SUCCESS in 38m 43s","accounts_in_message":[],"_revision_number":5},{"id":"fc187effb2f619548d82ffc38015dcfd85011512","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-21 16:58:53.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":5},{"id":"5d697a0771a98623838ac5215caf4dceaa3eb531","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-21 17:00:09.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/32b22342fc7e4ba690c0621105296564\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/11de0097c8204ad0a8cfd41f5b4cd29e : SUCCESS in 54s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"0c02ba024c5685f3bdf308c7eeba981ead4f93e3","revisions":{"8a90f6bd9ab19173ab5aab7b673b32345b8e5cb7":{"kind":"REWORK","_number":1,"created":"2022-10-14 06:19:17.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/10/861310/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/10/861310/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/10/861310/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/10/861310/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/10/861310/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/10/861310/1"}}},"commit":{"parents":[{"commit":"aa32c785fda45d048996caf95e7bd0adf2f2bd7c","subject":"[sqlalchemy2] move to Session.get()","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/aa32c785fda45d048996caf95e7bd0adf2f2bd7c"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2022-10-07 16:24:49.000000000","tz":120},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2022-10-13 17:44:17.000000000","tz":120},"subject":"[sqlalchemy2] use attribute to indicate relationship","message":"[sqlalchemy2] use attribute to indicate relationship\n\nUsing strings to indicate relationship will be removed in sqlalchemy 2.0\n[0].\n\n[0] https://docs.sqlalchemy.org/en/14/changelog/migration_20.html#\\\n    orm-query-joining-loading-on-relationships-uses-attributes-not-strings\n\nChange-Id: I3c3af331e147c8bbff560c42bca4d51e864aac07\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/8a90f6bd9ab19173ab5aab7b673b32345b8e5cb7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/8a90f6bd9ab19173ab5aab7b673b32345b8e5cb7"}]},"branch":"refs/heads/master"},"9a3aad65e7f358e6feef2658607ce698e92990d3":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2023-02-27 07:52:40.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/10/861310/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/10/861310/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/10/861310/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/10/861310/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/10/861310/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/10/861310/2"}}},"commit":{"parents":[{"commit":"b38821475d7bf250fd840da4633b7961d6e6c2c3","subject":"[sqlalchemy2] move to Session.get()","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b38821475d7bf250fd840da4633b7961d6e6c2c3"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2022-10-07 16:24:49.000000000","tz":120},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-02-24 14:44:49.000000000","tz":-300},"subject":"[sqlalchemy2] use attribute to indicate relationship","message":"[sqlalchemy2] use attribute to indicate relationship\n\nUsing strings to indicate relationship will be removed in sqlalchemy 2.0\n[0].\n\n[0] https://docs.sqlalchemy.org/en/14/changelog/migration_20.html#\\\n    orm-query-joining-loading-on-relationships-uses-attributes-not-strings\n\nChange-Id: I3c3af331e147c8bbff560c42bca4d51e864aac07\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/9a3aad65e7f358e6feef2658607ce698e92990d3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/9a3aad65e7f358e6feef2658607ce698e92990d3"}]},"branch":"refs/heads/master"},"c31561a119d55136d5405301c2f35a56f042c2d0":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2023-03-29 08:44:58.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/10/861310/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/10/861310/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/10/861310/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/10/861310/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/10/861310/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/10/861310/3"}}},"commit":{"parents":[{"commit":"18eea046f05d2c173856bbe00a6e383dd214afca","subject":"[sqlalchemy2] move to Session.get()","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/18eea046f05d2c173856bbe00a6e383dd214afca"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2022-10-07 16:24:49.000000000","tz":120},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-03-29 08:12:57.000000000","tz":-240},"subject":"[sqlalchemy2] use attribute to indicate relationship","message":"[sqlalchemy2] use attribute to indicate relationship\n\nUsing strings to indicate relationship will be removed in sqlalchemy 2.0\n[0].\n\n[0] https://docs.sqlalchemy.org/en/14/changelog/migration_20.html#\\\n    orm-query-joining-loading-on-relationships-uses-attributes-not-strings\n\nChange-Id: I3c3af331e147c8bbff560c42bca4d51e864aac07\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c31561a119d55136d5405301c2f35a56f042c2d0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c31561a119d55136d5405301c2f35a56f042c2d0"}]},"branch":"refs/heads/master"},"942b7b1c6e2be2b087e10293f0bd4c5d11df7220":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2023-04-24 06:50:24.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/10/861310/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/10/861310/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/10/861310/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/10/861310/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/10/861310/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/10/861310/4"}}},"commit":{"parents":[{"commit":"19385c050a0879bb378d28505aadedc27db2f5e9","subject":"[sqlalchemy2] move to Session.get()","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/19385c050a0879bb378d28505aadedc27db2f5e9"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2022-10-07 16:24:49.000000000","tz":120},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-04-24 06:41:53.000000000","tz":-240},"subject":"[sqlalchemy2] use attribute to indicate relationship","message":"[sqlalchemy2] use attribute to indicate relationship\n\nUsing strings to indicate relationship will be removed in sqlalchemy 2.0\n[0].\n\n[0] https://docs.sqlalchemy.org/en/14/changelog/migration_20.html#\\\n    orm-query-joining-loading-on-relationships-uses-attributes-not-strings\n\nChange-Id: I3c3af331e147c8bbff560c42bca4d51e864aac07\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/942b7b1c6e2be2b087e10293f0bd4c5d11df7220"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/942b7b1c6e2be2b087e10293f0bd4c5d11df7220"}]},"branch":"refs/heads/master"},"0c02ba024c5685f3bdf308c7eeba981ead4f93e3":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2023-05-03 07:07:19.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/10/861310/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/10/861310/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/10/861310/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/10/861310/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/10/861310/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/10/861310/5"}}},"commit":{"parents":[{"commit":"e0bff29efd73568b5ef53b5e2bb5156d57737e8f","subject":"[sqlalchemy2] move to Session.get()","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e0bff29efd73568b5ef53b5e2bb5156d57737e8f"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2022-10-07 16:24:49.000000000","tz":120},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-05-03 07:02:50.000000000","tz":-240},"subject":"[sqlalchemy2] use attribute to indicate relationship","message":"[sqlalchemy2] use attribute to indicate relationship\n\nUsing strings to indicate relationship will be removed in sqlalchemy 2.0\n[0].\n\n[0] https://docs.sqlalchemy.org/en/14/changelog/migration_20.html#\\\n    orm-query-joining-loading-on-relationships-uses-attributes-not-strings\n\nChange-Id: I3c3af331e147c8bbff560c42bca4d51e864aac07\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/0c02ba024c5685f3bdf308c7eeba981ead4f93e3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/0c02ba024c5685f3bdf308c7eeba981ead4f93e3"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
