)]}'
{"id":"openstack%2Foctavia~861311","triplet_id":"openstack%2Foctavia~master~I877b65d2204d619abb75510e823356137942fa29","project":"openstack/octavia","branch":"master","topic":"sqlalchemy2","attention_set":{},"removed_from_attention_set":{"29244":{"account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"last_update":"2023-06-21 11:24:58.000000000","reason":"\u003cGERRIT_ACCOUNT_29244\u003e replied on the change","reason_account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}}},"hashtags":[],"change_id":"I877b65d2204d619abb75510e823356137942fa29","subject":"[sqlalchemy2] positional arguments for case","status":"MERGED","created":"2022-10-14 06:19:17.000000000","updated":"2023-06-21 17:01:04.000000000","submitted":"2023-06-21 16:58:55.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"861311-sqlalchemy2","meta_rev_id":"48a1ad2aa425cb5dde697f61d922c135d0886643","_number":861311,"virtual_id_number":861311,"owner":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-06-21 16:58:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},{"value":0,"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":2,"date":"2023-06-15 17:13:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-05-03 07:07:19.000000000","_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},{"value":1,"date":"2023-06-17 05:03:59.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":1,"date":"2023-06-15 17:13:44.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},{"value":0,"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},{"value":0,"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"CC":[{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-10-14 09:08:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-10-24 15:59:31.000000000","updated_by":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"reviewer":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"state":"CC"},{"updated":"2023-04-28 13:17:58.000000000","updated_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"reviewer":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"state":"REVIEWER"},{"updated":"2023-06-15 17:13:44.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2023-06-17 05:03:59.000000000","updated_by":{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},"reviewer":{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},"state":"REVIEWER"}],"messages":[{"id":"d897854ddc3c1e638a3c6269c691694f158bab89","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2022-10-14 06:19:17.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"96da6b024050bf1b71d76f35f641027235ce600a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-14 09:08:20.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6d9ef4a7fa6049a899d3c78c9781fad1\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5615ca0f5383490496fcb4b0db5b29eb : SUCCESS in 8m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f9d9f6809dd046bc83b983864651047e : SUCCESS in 5m 55s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1aa497d454e444149c09296d8877130f : SUCCESS in 4m 48s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/bb1963c3fc794caab6101870382ac260 : SUCCESS in 4m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ec3bf4fb2a1440519ffc09c755d6b276 : SUCCESS in 10m 21s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/3d4e665781cb4e60aa18b6c551a6eb7e : SUCCESS in 4m 52s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/934c1edfc6404e8797662ddfcbd890bb : SUCCESS in 5m 46s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/f75ef208f63f455bbaddd92837748276 : SUCCESS in 4m 57s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/23dc7b40a97f4ec393d4910e74330517 : SUCCESS in 6m 41s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/27c4c231753341ca939189682d89fd69 : SUCCESS in 6m 43s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/241b535660b04a299f992a986bef559f : SUCCESS in 6m 39s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/f4dab635afbb47ca84c3fe3d0fbb7e3f : SUCCESS in 1h 13m 32s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/fc4db951fbb04e98b2e0c9fa459d6ce1 : SUCCESS in 1h 58m 32s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/06398d2593a74b9cbccf9024e4314e33 : SUCCESS in 43m 38s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/cfcbba3d6fd8460f8761a4ed3d73344c : SUCCESS in 1h 04m 11s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/050e2ffd2d65400d81c0a1751a41a2d7 : NODE_FAILURE in 0s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/10bb4954cb03410e92597e26741d7b2f : SUCCESS in 39m 04s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/d2b9d6e912a848d5bf34bb25234cab20 : SUCCESS in 1h 46m 16s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/5042bf8a4a674e12b3462559c8aac842 : SUCCESS in 2h 27m 07s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/f82c0f393fe7406bbfa9d468aa9f9f3f : SUCCESS in 2h 23m 21s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/ed13252f36894926a5a5ae545e5867eb : SUCCESS in 2h 17m 57s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/cf1add0e162c47d5b21d82c08fd3e961 : SUCCESS in 1h 49m 48s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/04c35a7a142b4bd3981a1e96062980d4 : SUCCESS in 2h 12m 04s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"99871df3229de7c38c3b37c6db78c6bc995e0afa","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2022-10-24 15:59:31.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"db61daca4a78500527146ac1b294fc5439d9087f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-02-27 07:52:40.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"ec726fb63a365e371f723abcd817cdda13070d75","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-27 10:38:35.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/e259397ca52e4790bdac2d400edc5a73\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0a0f66d99a274764a57eefce680cb430 : SUCCESS in 10m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b5760b82bcf9461a89f29a4f73935807 : SUCCESS in 5m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f4b63b5b679246108fe7233ec22dd6e1 : SUCCESS in 4m 34s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/245b01d53e7b44dc868d817fc02fa48b : SUCCESS in 4m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/015ba3c63ecf4768a47619b6f7c1e2e7 : RETRY_LIMIT in 1m 53s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/c0a0c5efa1f649259a1b92c8ca95b9eb : SUCCESS in 4m 36s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/7759929a27b8431ea17dafc39b0ce01f : RETRY_LIMIT in 1m 46s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/3cbb5c08d3ec4da0bcf81520eee232c0 : SUCCESS in 4m 47s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/b25d82b272de4000ad14435a1622404a : RETRY_LIMIT in 2m 34s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/08262b74bfad4bc3ac84e843ad282c54 : SUCCESS in 15m 13s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/cf6f7cf35e264fd0a52f6d859454b2a2 : SUCCESS in 7m 43s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/f3400d85962541b9aa5dd7fdd8f5bc27 : FAILURE in 5m 56s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/273302d3bf4a4601ac7b56372eeffbe0 : SUCCESS in 1h 54m 04s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/4cd37a3253d34993bf81acb9429eb8a1 : SUCCESS in 35m 03s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/4889b7b88f524bc388dedeefb253aacb : SUCCESS in 38m 29s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/72420fafa77b4e45b562cf1fbeeef0f8 : SUCCESS in 1h 15m 40s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/a27a4814a4a84dec803da35329845bf2 : SUCCESS in 32m 13s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/46b892cd35f94fb6a728a37fa49c6c67 : SUCCESS in 2h 04m 34s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/c2e13de85ffd4e16af0fe1eee3a9b116 : SUCCESS in 1h 55m 21s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/e1e3cab252f549d4bbee6769ba9f1654 : SUCCESS in 1h 53m 27s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/852567c333014350a9a5bd4a4d21aa0c : SUCCESS in 1h 48m 43s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/67ebf3c14de74662a131121411878584 : SUCCESS in 1h 48m 47s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/56568f449fce4fdf95cf3ca47ea09101 : SUCCESS in 2h 31m 38s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"194635455042101eef95c0f0e8c4564c44ade3c6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-03-29 08:44:58.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.","accounts_in_message":[],"_revision_number":3},{"id":"9c842efed103f97b8159fe632634b55fd6e09e9f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-29 11:36:06.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/1c163a5821954778bc8dda70229900f5\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/22615976741341cea2deba072baeeb64 : SUCCESS in 12m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b78c9b221e1d4fdda61ddf129b50740e : SUCCESS in 5m 56s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/29a47a5b470c4651b15d60ddb89380b9 : SUCCESS in 5m 33s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6748016ef960401d836c6345f2640d12 : SUCCESS in 5m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8b86ee7657414601a69cdc1393fdbbf6 : SUCCESS in 11m 30s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/afb51892699d460aa41ee6240a52aca9 : SUCCESS in 5m 00s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/b23f985ae19d4bbeb1580d2ae4ebd2bb : SUCCESS in 8m 16s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/c78fd8bbce2f4b569ed8be6c67edf0a0 : SUCCESS in 5m 28s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/e38c1bad579d489b9082d81fc83fe493 : SUCCESS in 7m 33s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/eb7715321c1242cc855b414ca4371cbe : SUCCESS in 7m 50s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/3c075846432a4015bce4e3be08b2f538 : SUCCESS in 8m 23s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/469eb2f81857455281c60e70b8a68447 : SUCCESS in 1h 21m 48s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/1c4474e5eaf24f11b4de597edbeae968 : SUCCESS in 2h 15m 37s\n- octavia-v2-dsvm-scenario-jobboard https://zuul.opendev.org/t/openstack/build/4ea8b667e7e0421baa39d91f9d033295 : TIMED_OUT in 2h 33m 17s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/9fb4eecd649a40578cab0b94ff96cbb1 : SUCCESS in 37m 51s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/87715eaea58142d196d93b408bf3e59c : SUCCESS in 45m 29s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/96a31ba271244140b5812cdc8b73602d : SUCCESS in 1h 10m 56s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/600f2a111df54defa264cc43e91a465c : SUCCESS in 39m 07s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/28364fda4cbd42e3a7eb67e2aae1b7ce : SUCCESS in 2h 20m 00s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/0340aa82c505431a82751578c11b0429 : SUCCESS in 2h 36m 33s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/a9a5bfde82af4e449128ee623f1814a5 : TIMED_OUT in 2h 34m 16s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/fc9839e7006b44e29f6925dca96bdd1a : TIMED_OUT in 2h 35m 14s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"6a8e78088c16aa9204c636ac6ee4c8defd476714","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-04-24 06:50:24.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"891fe099ebeb9b734d47e5bc0bad8dfe04e29e14","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-24 09:36:41.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/42f80dc2d482425da0c942d66de4d3a4\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9216fcf0527d4901a4e1d88954139055 : SUCCESS in 13m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fdb05eee9c8049bfaea7bf3dc88c8aa2 : SUCCESS in 5m 26s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/73bcd32ad3094c02bff43d141b6aea09 : SUCCESS in 5m 20s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2d9b2bdadcaf48829efb642ed4da5a33 : SUCCESS in 4m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c046feda3fba440f80a7f83f69ed6d39 : SUCCESS in 11m 56s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/9534803b5fa94372ac0e589749c4bf03 : SUCCESS in 5m 50s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/45d3a271adde432d8ba82d0dd5dbc776 : SUCCESS in 7m 08s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/e79e5bde9b5a4b8395de9dcbe2d76920 : SUCCESS in 4m 30s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/55772c5876514c0f8e2d4c21eebbc168 : SUCCESS in 6m 55s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/75fd03033c54491c8ae05fd691891696 : SUCCESS in 8m 10s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/fc403813a5bc464f8c8168f31b0c7915 : SUCCESS in 7m 45s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/deedd3d1f9a445fd93a7c279b50d58ed : SUCCESS in 1h 30m 30s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/73e57e840f1e447fb52f8a32fda798ce : FAILURE in 11m 55s\n- octavia-v2-dsvm-scenario-jobboard https://zuul.opendev.org/t/openstack/build/cd6946f11deb4bdebc9a89c737b160b3 : FAILURE in 11m 54s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/2be26fd8690b4f7082c582c90d6cedbd : FAILURE in 13m 46s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/be6afb264bbd4dba8a6639a5ffee90fa : FAILURE in 31m 53s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/af39454d2ed74b7bbb31347ad59a95d6 : SUCCESS in 57m 13s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/29b08a782937426c88917e13f88ba20c : FAILURE in 13m 07s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/9c9c6483825b409cb1b0fea513e6863a : FAILURE in 12m 28s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/a1df9386e3be4c3580d1698285eb9d18 : FAILURE in 18m 14s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/5eb8032aea124a7484c387f7ac4d90c9 : FAILURE in 11m 50s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/698f223f2ba64b738acec10de081c852 : SUCCESS in 2h 30m 30s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"6ccd7edbc84ef40924d14c4c39caec38fe87e878","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2023-04-28 13:17:58.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"834b838e7fb03a8dfbe9687a6a93f6f12a8e39c2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-05-03 07:07:19.000000000","message":"Uploaded patch set 5: Patch Set 4 was rebased.\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"eea045d21950803a750ffd43e11d032429304a38","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-03 10:08:52.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/f0f1c6f14cd84057a8ea04f68cc7113b\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/be0ec9ca415440d7a329cf9501affdc2 : SUCCESS in 12m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4ae76bed00e3473e99867dcbe33436e9 : SUCCESS in 5m 19s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/10f05c34dee048418800b2bf94695931 : SUCCESS in 4m 30s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d9f1a3766cfd4b1b850d8229223ce4da : SUCCESS in 4m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6ca02bc168c04a3e80c146f881082ddc : SUCCESS in 12m 52s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/8fac74f2b0be48eea182dfa127772231 : SUCCESS in 5m 41s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/654ed75a4cca4e98aa10d70d779424ab : SUCCESS in 9m 05s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/57494e9c9be84778972fc39c3055d9ab : SUCCESS in 4m 41s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/602600de1d334d1786ad681a8d428572 : SUCCESS in 6m 49s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/a35d2dc144424684b4c6d2d3369f033f : SUCCESS in 9m 45s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/08b7e01f6d314eeb87ee2d59f80937a3 : SUCCESS in 7m 00s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/9c62b4778e1044d094f26d8cc543c59a : SUCCESS in 1h 44m 40s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/9ce21458741f4b40bf1b75edcb5f0a04 : SUCCESS in 2h 19m 39s\n- octavia-v2-dsvm-scenario-jobboard https://zuul.opendev.org/t/openstack/build/4995ee0da03f41ff8e42e21fe9fbc4b6 : SUCCESS in 2h 35m 58s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/b75e90fd608045cb95c591b2d5dfae07 : NODE_FAILURE Node request 300-0021120528 failed in 0s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/8b63164b24be4feabd3560bebd54974f : SUCCESS in 45m 27s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/2e65a30e93c942bd9c187b8d75278839 : NODE_FAILURE Node request 300-0021120530 failed in 0s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/4bc082cfe5e9446680194527f3b9d0e4 : SUCCESS in 41m 16s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/f30e25760f1f48ed9cb2f1200c8db5df : NODE_FAILURE Node request 300-0021120532 failed in 0s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/0196d954f2284d0392f52dc472aed784 : SUCCESS in 2h 47m 33s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/f245371d2504423d9922a6d4c9ddd224 : NODE_FAILURE Node request 300-0021120534 failed in 0s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/655addd0e7624f50b16554ac3ea038c2 : NODE_FAILURE Node request 300-0021120535 failed in 0s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"fd510b50512e48b3c53035bfb18c2040c3774dc5","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-06-15 17:13:44.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"c0b54e4b647003cf83f9b863b0f81dfebc6f8648","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-15 20:18:20.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/e33ebb539bda47dcb605e7ea46addeaf\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/841753da77da4a1f91496c40e927916b : SUCCESS in 10m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/67ee77f3791b47888784864cfc17882d : SUCCESS in 5m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/61d4492cbe1b434e949083b12edea8a3 : SUCCESS in 4m 29s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5b91daf7800440248bff9a9170dc1445 : SUCCESS in 4m 31s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ca3531cc5f084ecdb1e8b3f86a635c78 : SUCCESS in 23m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/23c16a85427044b78c7b8bd487b4eed6 : SUCCESS in 10m 13s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/437dd095ce19480d8af90d63b391b1da : SUCCESS in 5m 22s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/5aab7cdb1db5452993d8da2a94b5d00e : SUCCESS in 7m 16s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/b78c6a8cb44841efaf4a3ea481d04d42 : SUCCESS in 4m 34s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/4a7e2b121d2b4113be69269a4a80f9d3 : SUCCESS in 6m 45s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/3da375e718b64b89945a1fedf8cf0e31 : SUCCESS in 8m 38s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/7333712447dd4983a16d5f9c9a00e6e8 : SUCCESS in 8m 06s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/2db0de5f0ec649508262b370db8b1a22 : SUCCESS in 1h 34m 53s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/2a5a2e2ba7574f4083ce0ea4011020e1 : SUCCESS in 2h 00m 49s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/ef98cd44223c40fe989599342fdef448 : SUCCESS in 1h 12m 13s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/aa1e353383f14ccaab287f90e93ac882 : FAILURE in 35m 27s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/8ca8af5cf0c0404fa254fec84457d47d : SUCCESS in 1h 10m 12s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/73f1aebc89544212ae8aad8d46232260 : SUCCESS in 41m 41s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/54a9191690b24943aac7e5595fdfdd7c : SUCCESS in 46m 44s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/611dbaa4c5724d9fa23ce259af2fc730 : FAILURE in 52m 37s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/4886ab317dc044b38e9a5c6199ba03df : SUCCESS in 47m 05s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/904f562c49984a93ad5a2bc6fe02e1f2 : TIMED_OUT in 2h 32m 53s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/228266336d3243989f3d2cc823c9de40 : TIMED_OUT in 3h 03m 08s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/a2db4a571623460d994173eb88666cf2 : TIMED_OUT in 2h 32m 39s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/705fad960d954bc6a42298dead286eee : FAILURE in 1h 38m 52s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"646b801fefa771195423758a9bd7226ea514f7a3","author":{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},"date":"2023-06-17 05:03:59.000000000","message":"Patch Set 5: Code-Review+1","accounts_in_message":[],"_revision_number":5},{"id":"d53dcd47111409b31fcf626be774078792303753","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-06-21 11:24:58.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"bf514642384506e6a965ba2eb8d73595a7c649dd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-21 14:47:20.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/195230c939694e1585254a059c1c01b4\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c2e8b6d697cb410686b9ae10dbc572d9 : SUCCESS in 10m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f87f349600ec43a099990d145f4de3f0 : SUCCESS in 5m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7cec083b8cfb49b7ab3c75f7d98a8f32 : SUCCESS in 4m 56s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/713d0b6365884b1b8c2d04512b2f3400 : SUCCESS in 5m 36s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2b69dea458cc4a3eb6e9190f2c3c520a : SUCCESS in 5m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/500e8f0b8d8a43bb8f4d8fe99e5d55d7 : SUCCESS in 13m 35s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/adb6a2d2edca45748ff6d6e2848d741f : SUCCESS in 6m 06s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/fc906350965c4e0fad3ebe2258517fa0 : SUCCESS in 8m 04s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/d47d0ab8225942ca83e492e9ae434448 : SUCCESS in 5m 06s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/daa349a4f8814a6baf36e0dfc378f102 : SUCCESS in 7m 09s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/074b22a80f7c4c15a6ee032acd2959ee : SUCCESS in 7m 47s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/4c38dffe2ecb498cb4bbc72eec244395 : SUCCESS in 8m 21s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/d0db10cebd944e7bb365c9b0af3c3aab : SUCCESS in 1h 43m 56s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/a43234dca2544ff3bc6ed4b376085923 : SUCCESS in 2h 01m 38s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/e836e725c5774d188ce726d6284320ea : SUCCESS in 1h 13m 26s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/81431e14e55441ed872872632b0104ba : SUCCESS in 2h 06m 06s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/e71f674e7e824b49982b3bb17d08c48d : SUCCESS in 1h 20m 29s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/c1e4e714b73a41e28e8ed893cdb9e4af : SUCCESS in 38m 08s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/6490079ad30f4615a6a8814b942cbe00 : SUCCESS in 35m 51s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/e90be79e3dc9458d98420ebf4e24f1af : NODE_FAILURE Node request 300-0021506439 failed in 0s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/86346be63a41451daffd437be07399f3 : SUCCESS in 30m 30s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/540730565c124a45a10dd7f5923a529f : TIMED_OUT in 2h 32m 52s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/217d6b7bef4b482bab6ebae2c29542df : SUCCESS in 2h 44m 27s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/6790fe74120a48e48329654c5d372935 : TIMED_OUT in 2h 33m 26s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/c7cac4cef2554382b19d2a5f8593ec76 : NODE_FAILURE Node request 300-0021506444 failed in 0s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"c660e2bd9a242fbefdeda7fd3c0c0b11ba2e0de0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-21 14:48:03.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"f91da976142eb4653e6b6201a6878a4544536a74","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-21 16:58:55.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2fc285e419ef4319a4986d6bcbaca99c\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/519da2c305ff417dacfe8b34c93a1d4d : SUCCESS in 5m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/344c88f5ebe54e99b23b7a51a54da691 : SUCCESS in 5m 12s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a09ca39a1a2b49f7b1a6498ee0d77585 : SUCCESS in 5m 25s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/dae85c92e5ee4f1daeb405928f5b0897 : SUCCESS in 5m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aa69f232c760436cb445a62d46063a5a : SUCCESS in 11m 57s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/b043f5faf43049639303518479f38dc4 : SUCCESS in 5m 20s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/e295b2966e5d4692bebf081ea5debf92 : SUCCESS in 9m 04s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/fedc31a87f804209bd5ff9b9b5f516b3 : SUCCESS in 1h 45m 37s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/8c8cee6b9b8641ac8f065c51216cc1c9 : SUCCESS in 1h 58m 46s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/71814752a9ec46c4a413d8a533ea50af : SUCCESS in 1h 15m 17s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/7dc0739257574190928940a39ba3f4af : SUCCESS in 37m 32s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/307fc88493214067bf3d7c0317136d03 : SUCCESS in 35m 56s","accounts_in_message":[],"_revision_number":5},{"id":"fb23ba5672a42ea1f821ec728dfe5369b809d602","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-21 16:58:55.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":5},{"id":"48a1ad2aa425cb5dde697f61d922c135d0886643","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-21 17:01:04.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/50a630db240e4c5faddee686d3e83ee7\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7b99c6c2fd5f4286a9c86baa4ce9e80e : SUCCESS in 48s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"c412058fbb4f8bcf02db396a1343116fc8061578","revisions":{"f534ca550f997dab5cc10575c0e9d82e71dfcc9a":{"kind":"REWORK","_number":1,"created":"2022-10-14 06:19:17.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/11/861311/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/11/861311/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/11/861311/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/11/861311/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/11/861311/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/11/861311/1"}}},"commit":{"parents":[{"commit":"8a90f6bd9ab19173ab5aab7b673b32345b8e5cb7","subject":"[sqlalchemy2] use attribute to indicate relationship","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/8a90f6bd9ab19173ab5aab7b673b32345b8e5cb7"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2022-10-07 16:40:27.000000000","tz":120},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2022-10-13 17:45:05.000000000","tz":120},"subject":"[sqlalchemy2] positional arguments for case","message":"[sqlalchemy2] positional arguments for case\n\nSome functions/classes now take positional arguments instead of tuples\n[0].\n\n[0] https://docs.sqlalchemy.org/en/14/changelog/migration_20.html#\\\n    select-no-longer-accepts-varied-constructor-arguments-columns-are-passed-positionally\n\nChange-Id: I877b65d2204d619abb75510e823356137942fa29\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/f534ca550f997dab5cc10575c0e9d82e71dfcc9a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/f534ca550f997dab5cc10575c0e9d82e71dfcc9a"}]},"branch":"refs/heads/master"},"64ba26176e5a21531c0ca4cc08abf7fc4a29b51d":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2023-02-27 07:52:40.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/11/861311/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/11/861311/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/11/861311/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/11/861311/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/11/861311/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/11/861311/2"}}},"commit":{"parents":[{"commit":"9a3aad65e7f358e6feef2658607ce698e92990d3","subject":"[sqlalchemy2] use attribute to indicate relationship","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/9a3aad65e7f358e6feef2658607ce698e92990d3"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2022-10-07 16:40:27.000000000","tz":120},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-02-24 14:44:49.000000000","tz":-300},"subject":"[sqlalchemy2] positional arguments for case","message":"[sqlalchemy2] positional arguments for case\n\nSome functions/classes now take positional arguments instead of tuples\n[0].\n\n[0] https://docs.sqlalchemy.org/en/14/changelog/migration_20.html#\\\n    select-no-longer-accepts-varied-constructor-arguments-columns-are-passed-positionally\n\nChange-Id: I877b65d2204d619abb75510e823356137942fa29\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/64ba26176e5a21531c0ca4cc08abf7fc4a29b51d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/64ba26176e5a21531c0ca4cc08abf7fc4a29b51d"}]},"branch":"refs/heads/master"},"a057ffec7c8a21f17a059378c83a6e446d6cdf77":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2023-03-29 08:44:58.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/11/861311/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/11/861311/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/11/861311/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/11/861311/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/11/861311/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/11/861311/3"}}},"commit":{"parents":[{"commit":"c31561a119d55136d5405301c2f35a56f042c2d0","subject":"[sqlalchemy2] use attribute to indicate relationship","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c31561a119d55136d5405301c2f35a56f042c2d0"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2022-10-07 16:40:27.000000000","tz":120},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-03-29 08:12:57.000000000","tz":-240},"subject":"[sqlalchemy2] positional arguments for case","message":"[sqlalchemy2] positional arguments for case\n\nSome functions/classes now take positional arguments instead of tuples\n[0].\n\n[0] https://docs.sqlalchemy.org/en/14/changelog/migration_20.html#\\\n    select-no-longer-accepts-varied-constructor-arguments-columns-are-passed-positionally\n\nChange-Id: I877b65d2204d619abb75510e823356137942fa29\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a057ffec7c8a21f17a059378c83a6e446d6cdf77"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a057ffec7c8a21f17a059378c83a6e446d6cdf77"}]},"branch":"refs/heads/master"},"c34caef01a7ab7e8884ab247e7926896e98857da":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2023-04-24 06:50:24.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/11/861311/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/11/861311/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/11/861311/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/11/861311/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/11/861311/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/11/861311/4"}}},"commit":{"parents":[{"commit":"942b7b1c6e2be2b087e10293f0bd4c5d11df7220","subject":"[sqlalchemy2] use attribute to indicate relationship","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/942b7b1c6e2be2b087e10293f0bd4c5d11df7220"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2022-10-07 16:40:27.000000000","tz":120},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-04-24 06:41:53.000000000","tz":-240},"subject":"[sqlalchemy2] positional arguments for case","message":"[sqlalchemy2] positional arguments for case\n\nSome functions/classes now take positional arguments instead of tuples\n[0].\n\n[0] https://docs.sqlalchemy.org/en/14/changelog/migration_20.html#\\\n    select-no-longer-accepts-varied-constructor-arguments-columns-are-passed-positionally\n\nChange-Id: I877b65d2204d619abb75510e823356137942fa29\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c34caef01a7ab7e8884ab247e7926896e98857da"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c34caef01a7ab7e8884ab247e7926896e98857da"}]},"branch":"refs/heads/master"},"c412058fbb4f8bcf02db396a1343116fc8061578":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2023-05-03 07:07:19.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/11/861311/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/11/861311/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/11/861311/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/11/861311/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/11/861311/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/11/861311/5"}}},"commit":{"parents":[{"commit":"0c02ba024c5685f3bdf308c7eeba981ead4f93e3","subject":"[sqlalchemy2] use attribute to indicate relationship","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/0c02ba024c5685f3bdf308c7eeba981ead4f93e3"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2022-10-07 16:40:27.000000000","tz":120},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-05-03 07:02:50.000000000","tz":-240},"subject":"[sqlalchemy2] positional arguments for case","message":"[sqlalchemy2] positional arguments for case\n\nSome functions/classes now take positional arguments instead of tuples\n[0].\n\n[0] https://docs.sqlalchemy.org/en/14/changelog/migration_20.html#\\\n    select-no-longer-accepts-varied-constructor-arguments-columns-are-passed-positionally\n\nChange-Id: I877b65d2204d619abb75510e823356137942fa29\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c412058fbb4f8bcf02db396a1343116fc8061578"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c412058fbb4f8bcf02db396a1343116fc8061578"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
