)]}'
{"id":"openstack%2Foctavia~861312","triplet_id":"openstack%2Foctavia~master~I9c3d148b1c4bff284d105e799846f5ce70c55046","project":"openstack/octavia","branch":"master","topic":"sqlalchemy2","attention_set":{},"removed_from_attention_set":{"29244":{"account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"last_update":"2023-06-21 11:25:06.000000000","reason":"\u003cGERRIT_ACCOUNT_29244\u003e replied on the change","reason_account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}}},"hashtags":[],"change_id":"I9c3d148b1c4bff284d105e799846f5ce70c55046","subject":"[sqlalchemy2] Use subquery to select from a query","status":"MERGED","created":"2022-10-14 06:19:17.000000000","updated":"2023-06-21 17:20:39.000000000","submitted":"2023-06-21 17:19:29.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"861312-sqlalchemy2","meta_rev_id":"90df08978986d25bb234660fb153f3645515ff72","_number":861312,"virtual_id_number":861312,"owner":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-06-21 17:19:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},{"value":0,"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"recommended":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"all":[{"value":1,"date":"2023-05-03 07:07:19.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-06-15 17:14:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":2,"date":"2023-05-03 07:07:19.000000000","_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},{"value":1,"date":"2023-06-17 05:05:02.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-06-15 17:14:28.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},{"value":0,"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},{"value":0,"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-10-14 09:09:54.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-10-24 16:07:53.000000000","updated_by":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"reviewer":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"state":"REVIEWER"},{"updated":"2023-04-28 14:12:48.000000000","updated_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"reviewer":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"state":"REVIEWER"},{"updated":"2023-06-15 17:14:28.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2023-06-17 05:05:02.000000000","updated_by":{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},"reviewer":{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},"state":"REVIEWER"}],"messages":[{"id":"d9892dd1890787f10304f1bfe2d3b5d6887108b6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2022-10-14 06:19:17.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"22ccb5ddfaed39ad15fc5f70a423b469755abf3c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-14 09:09:54.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/15e996c72bf842b3878172e1294184f4\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3aaa7c72a5f54f378e4f2089f70830c0 : SUCCESS in 8m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fb1c953837604b118f7289ec3b0c499d : SUCCESS in 6m 23s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b628853e3c2e4138ad7ce020ce47dddc : SUCCESS in 4m 41s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c2d53c73129d4bba9f4bef36770bc5c8 : SUCCESS in 5m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8a34bdd477b04fa7bb233c1841f1cebf : SUCCESS in 12m 38s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/925f7f22c60d4c0aa41cc583e6487640 : SUCCESS in 4m 51s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/8b2d24ebaf194cd592b66137e0e6cf55 : SUCCESS in 7m 59s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/5b6d515c2d6a4a9b8b67c3b7dd3a4474 : SUCCESS in 5m 33s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/9e4eff98eb9e4f2580bfa97d91f59799 : SUCCESS in 7m 12s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/ebd56ed9d32543209b306b2cd2cfd45d : SUCCESS in 6m 13s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/53ea1d1305634894a6d4973dacf95b8f : SUCCESS in 8m 06s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/4bc061b46c7743c284904b6f4c9516fd : SUCCESS in 1h 41m 14s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/2de54e74ffb24528947548340934e826 : SUCCESS in 1h 35m 37s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/e5ff05a46ea849c1b6fe205cacae1aad : SUCCESS in 43m 08s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/6dce21a139a04e8bae6f1c65e451d785 : SUCCESS in 53m 15s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/bffb0d93f5364d8b94f1a157d6bd30ca : SUCCESS in 1h 11m 26s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/017928f07a4b44c8868fc1dc0bb5e7a9 : SUCCESS in 37m 57s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/7a8d45dc2007400e8866bc0fe58b37b1 : SUCCESS in 1h 33m 14s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/de4e850490c04676bda56cfed4f7d52a : SUCCESS in 1h 51m 00s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/4c2a88155e6f4396b1d22e83f05a393d : SUCCESS in 2h 00m 30s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/7bd82ef7e54b4725ae4d5f6ac84bb1f2 : SUCCESS in 1h 44m 27s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/65a596a6f75948f4a7bfe7bff0ffb800 : SUCCESS in 2h 01m 48s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/621550d08a8d46e39088136eebccaa12 : SUCCESS in 2h 25m 13s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"3435423c833048eac9af6c7d758140eaf829c28d","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2022-10-24 16:07:53.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c1515418bd62e286668ac5579fd185cd95cb5a3e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-02-27 07:52:40.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"5bf40ed68f70bab975887ed930718bab4d79c2d0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-27 11:07:31.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/c480fc038bfa497f99d75ffdcb4b7bbd\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/010e882559b743c2a7ab0ebaa9e042b6 : SUCCESS in 9m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c3c2211fdc2944b880c3a5617d5d2b5c : SUCCESS in 4m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7b4bef715b704710b29f864f5ef32afd : SUCCESS in 4m 31s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4bdf823e122941ee9bf5098af78e16f2 : SUCCESS in 5m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c306cac9412d448497d9c1746b849e14 : SUCCESS in 10m 59s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/50072694f0c84b949bf67d1ee5aac65e : SUCCESS in 4m 59s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/82857ef346bb47599453b157b4a113c8 : SUCCESS in 7m 37s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/453fb24474e74ea587e359a42ec6ba58 : RETRY_LIMIT in 3m 47s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/f5a882a84ad343d3a4b7fd974e967dae : SUCCESS in 7m 34s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/1f8e214259004817b5258eb37abe3e59 : SUCCESS in 7m 32s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/ae243bf1c15b4e17a1190d97a56febfa : SUCCESS in 7m 26s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/1ac6f88ef95b4550833ebe7f733dd5b7 : SUCCESS in 1h 51m 33s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/e61b56fdd3b74bbb8357ab699ad0a6bc : SUCCESS in 1h 51m 09s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/bd1f2980d4da4b9bb82e6bb43fd04e40 : SUCCESS in 34m 19s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/f8a0389443f447988feb5140709c8ed6 : SUCCESS in 43m 03s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/2f2ef6c86cbe4f45b63bb80047cc184d : SUCCESS in 1h 02m 06s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/86efb66d240647f89b8084b4ce73f222 : SUCCESS in 35m 01s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/6d3d86e022be46218c643bde11d61067 : SUCCESS in 2h 22m 20s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/df86080e8dc04f0ab2855592f04e5634 : SUCCESS in 2h 58m 18s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/956aae72f36c460d8eacd36fbb75f460 : SUCCESS in 1h 28m 26s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/d24b093ec1fa4d78b47e50bdff792f35 : SUCCESS in 2h 31m 26s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/8fe283c1b78e4781917191a6aa6cd5ca : SUCCESS in 2h 31m 22s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/18a88e49d3cf49d08125042a85f48f13 : TIMED_OUT in 2h 33m 28s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"f6067f0ccbf4b5238477f4041b1149486397f833","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-03-29 08:44:58.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.","accounts_in_message":[],"_revision_number":3},{"id":"72b33b4a1a53ff665c6fc9d3822a6d682e64813c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-29 12:00:07.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/c2c8e1dd5acd41909a380d78229201be\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a19862277b784db0a2778bac75a5e512 : SUCCESS in 10m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9762957e6393407eae690fa244ab571a : SUCCESS in 5m 39s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/672cf71df21746d29bd4b2e1c280fce1 : SUCCESS in 8m 56s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ba631f47fdc241b3b44fe07604069b47 : SUCCESS in 5m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c0ee67aed72f44d0a6390b86ddecde45 : SUCCESS in 13m 07s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/32bbd6aae4d44489bd1713ef47f26d03 : SUCCESS in 5m 07s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/4f523fcb68524ad184867de619504f85 : SUCCESS in 8m 08s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/b9758a5e72a54b428f73e277d3520f64 : SUCCESS in 4m 43s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/c2576104d8184802adc55396561ecd6d : SUCCESS in 8m 08s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/ec61bfb13b6e4b1b9e5e6e589a48c53f : SUCCESS in 7m 43s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/c93bb4e6167548cdbcd9929fde04f5d4 : SUCCESS in 11m 51s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/55f3676b04694e2ea5a720539199c18c : SUCCESS in 1h 36m 00s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/cd2360020a3044cbb50cdfa61633b95d : TIMED_OUT in 2h 33m 07s\n- octavia-v2-dsvm-scenario-jobboard https://zuul.opendev.org/t/openstack/build/c4e88dd10fb347a48df79a65ecd70493 : TIMED_OUT in 2h 32m 56s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/faa498af4c204e8a9759d947b19bed16 : SUCCESS in 39m 58s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/6f64b0bed7c94c1b89a9be7d6d1410e9 : SUCCESS in 45m 35s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/9293de3364614b36bc8e5d67a0d825e0 : SUCCESS in 1h 02m 42s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/e91b2df62a944634b14824a7b78ca87d : SUCCESS in 37m 04s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/927af2ee9b7841d89e00867b7552e128 : SUCCESS in 2h 21m 48s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/6a3bf221090a4f1b8687bb725c640766 : SUCCESS in 3h 00m 24s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/b80237669e324b47a0b1791e566fb62d : SUCCESS in 2h 20m 55s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/a43c2c4785824a0fabf4d299fa7f59fe : SUCCESS in 2h 30m 58s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"ed277df34b1a7d8875af3d2186698df4b7370b42","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-04-24 06:50:24.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.\n\nCopied Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"e99c1f90f57a5ede26db1407cf19fe8d474390bd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-24 09:21:17.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/fc6980c05a6249ae8d95c3d3de8ce610\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/43007b25f90b420c94b8565412d3140f : SUCCESS in 13m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/db6ddfa9318f4fd08d4b6bdfddd2bb12 : SUCCESS in 4m 32s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c3d26f4d078f493f97de63e376b0b401 : SUCCESS in 5m 10s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f927e7fd88d14aa3871471dc0556255f : SUCCESS in 4m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/65cde70933134a63ba5fab9ed26c45fd : SUCCESS in 10m 54s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/48ba24d0a050471ea4ffff75a2363d58 : SUCCESS in 5m 57s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/d243bb1adb704702af13758238b92bb6 : SUCCESS in 6m 26s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/a63f02e4fbc64826a1fcdb2ef840d759 : SUCCESS in 4m 53s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/e6904e0eba52470bb3ede082338a57c9 : SUCCESS in 7m 36s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/3ea7728cb27048ed84c246ed032b67df : SUCCESS in 12m 26s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/69d496854c074ec4836ae1c126c02c75 : SUCCESS in 11m 29s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/23fe26b46bd44f5eb93935d76e05f115 : SUCCESS in 1h 24m 40s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/78c08540ab7343e0bbc1a8057382bc8a : FAILURE in 21m 44s\n- octavia-v2-dsvm-scenario-jobboard https://zuul.opendev.org/t/openstack/build/49d427840be449baba286e6d746ace2a : FAILURE in 10m 43s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/2cc5cc5803474d6395fbb00985007229 : FAILURE in 13m 29s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/7ea6b19c1ba7434a9a9ef55d1b975463 : FAILURE in 32m 58s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/8b6581107f3b4938baaf545d0dbb9c14 : SUCCESS in 1h 00m 35s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/d2a6dad34346462898dfdef177246ad5 : FAILURE in 11m 16s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/bedb181272264124a49ed964442b1f3e : FAILURE in 13m 49s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/661289df987d44658e8e543fc4df5eda : FAILURE in 14m 58s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/3ee6456ef8ce4e9db6d46c2059148739 : FAILURE in 9m 33s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/7233c087d25f43668cb518152a02c342 : SUCCESS in 2h 22m 24s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"dd570646914b32c8265fbab1de4256833480d2ff","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2023-04-28 14:12:48.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"4e8f1d453bb010a91f305463a2f1e871338d56fb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-05-03 07:07:19.000000000","message":"Uploaded patch set 5: Patch Set 4 was rebased.\n\nCopied Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"2bf7d20feac97e27804f30937f77fe3bbaaced98","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-03 09:57:26.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9a3ca51c6d8a4c5d987593e6cf17e012\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3149e77b0c034935a7aaec335bfa713f : SUCCESS in 11m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ff977be5f2f34376923bcf6030cec255 : SUCCESS in 5m 40s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/bf32f286bffc4f52bf349bde4fee3f04 : SUCCESS in 4m 42s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d27fbd5fbb1849d68a93270fd7e943d7 : SUCCESS in 4m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/73bb5e9d0f3a45fe845dc7a3d2ef20c3 : SUCCESS in 11m 28s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/d57adb303cb64a8380eb99b1eb99460b : SUCCESS in 5m 00s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/f4389b3a443b4107905f1f8764c7a091 : SUCCESS in 6m 59s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/525bdced857b493e874716087a1545ee : SUCCESS in 4m 44s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/5bfb0c539a074eb9848fe03ebfd4482f : SUCCESS in 7m 28s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/8f50f320b15e4042830036f9313b8634 : SUCCESS in 7m 26s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/65fc212e21d348e0ac8c103191263d43 : SUCCESS in 7m 01s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/17bdf5403fa545d590eac1e6a38d78cd : SUCCESS in 1h 43m 42s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/d34b26e04a98406baed19f48b5949f07 : SUCCESS in 2h 33m 34s\n- octavia-v2-dsvm-scenario-jobboard https://zuul.opendev.org/t/openstack/build/cbe7c21c3ef04565b43960e594f7f2d8 : SUCCESS in 2h 26m 31s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/721c99b832b84b09af20e0e4915899f9 : SUCCESS in 43m 51s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/640123134b40426c8ecb5205b3177f5c : SUCCESS in 46m 16s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/6b6ac244dddd48d18c96dd78b8b95996 : SUCCESS in 1h 09m 45s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/f1300033dd8b4f4694a9a9da704dde07 : SUCCESS in 41m 32s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/a3e19aef61e9421da90e6d6b53403035 : SUCCESS in 2h 29m 26s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/a0280e126de84e1a9c2129a20c65c632 : SUCCESS in 2h 24m 08s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/89101e7414024d5ea7884a527330612b : SUCCESS in 2h 12m 50s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/9b5fb5feba4549e1862ab31097838e63 : NODE_FAILURE Node request 300-0021120557 failed in 0s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"3eb68d29e5cc66adc43312f217fbdb42eb25ae0e","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-06-15 17:14:28.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"aade1eb31401ca58bc087d34624734f9ec5d6d22","author":{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},"date":"2023-06-17 05:05:02.000000000","message":"Patch Set 5: Code-Review+1","accounts_in_message":[],"_revision_number":5},{"id":"3fc7e7a6f3aed041f32375a2d8e5733b07e10e3e","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-06-21 11:25:06.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"74503c10451ea568ba479948455aed2b7d1b8004","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-21 14:48:06.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"81dd4903520d56b434a6e5b058b5d9cc9466e09e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-21 15:07:33.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/51fd83ac3517431fb98361f63cdc3f7b\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7537f57afe6d4cfc8772b6a346bcd799 : SUCCESS in 9m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/76dfcd732d9a41ac8111dbf8a01866c1 : SUCCESS in 4m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f1497446a8064607b78569a0847a78f5 : SUCCESS in 4m 47s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/778ac9531f124d459f2bb93f12bbbc61 : SUCCESS in 5m 06s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5337a680432e4a20972a07edfe93bdf9 : SUCCESS in 6m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/64fc8df5ae824bd596e2afc299d85e64 : SUCCESS in 14m 16s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/fc14f0ea5cf94a4ba45efed4845d4ba3 : SUCCESS in 5m 23s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/7ff1ddd25d1f4ac3a0f5b89baae06f2a : SUCCESS in 6m 35s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/1ba3a929deb84f18aac88d7409682a3e : SUCCESS in 4m 59s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/7273efeab275480e98dfb5e4846d8bbf : SUCCESS in 7m 53s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/1bf40a2145384e97aed96c95a0900cc4 : SUCCESS in 9m 14s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/9485d318369342b9933a1932dfc23a70 : SUCCESS in 7m 08s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/1fbefb2ab67341fb8975ada742bdb4e5 : SUCCESS in 1h 37m 50s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/128947d038e944b2b4959d38a7f94cea : SUCCESS in 2h 04m 23s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/04f0d116139348c8834eeaaf897db289 : SUCCESS in 1h 24m 30s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/18d2ac5e0b0e4de6a14ce82a927adbed : SUCCESS in 2h 11m 36s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/15c735eafe6746b683e6ce453e3f45db : SUCCESS in 1h 20m 19s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/038d29e2f2c14a519f8c60e9cc4b0f8c : SUCCESS in 41m 19s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/a3dc8990ef034e7aa5eca3aaf1d563bf : SUCCESS in 37m 14s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/64ae3f000d194cfebffc9912e3baa877 : NODE_FAILURE Node request 300-0021506464 failed in 0s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/817580921b8f4795bf7e722dcd68daea : SUCCESS in 29m 49s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/f37c9d5fb01042268b928a6b8a795342 : TIMED_OUT in 2h 33m 32s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/409d43abe4454952be9b8e697ccfee81 : SUCCESS in 2h 35m 12s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/b72d3ea4b17e4612924062a7bd3ecbb5 : TIMED_OUT in 2h 33m 34s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/bc0f5d7b6c4143f8ba06b8b772b1d09d : NODE_FAILURE Node request 300-0021506469 failed in 0s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"3be131977a74db5811576fcbdb83d80b8d4520c5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-21 17:19:29.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/67fc3215cb764c8db179d7af881b4289\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0612e7906baf4baf95d0a9a1f30ac644 : SUCCESS in 5m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d960d34883e143548129b81894c403cd : SUCCESS in 5m 22s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/afe2993042ca469e9ce1a808a7cafb3c : SUCCESS in 5m 34s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/fceff7dc3fe44fcbaa4813bcd1b8c9ba : SUCCESS in 5m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6180b0470e25423faf924f9f8aafad8b : SUCCESS in 13m 19s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/e253423ebbb84e4f9150434c5b3fc46d : SUCCESS in 6m 39s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/26469ae0741c4b0bb78e56a19b19d0ca : SUCCESS in 7m 23s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/aaf254706e8b4866bd3fef2af1c5630b : SUCCESS in 2h 18m 16s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/75d8804a0006425cab690b611b3dd7ff : SUCCESS in 2h 04m 36s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/c8eab9d92a1a450580164e0faff86fb2 : SUCCESS in 2h 11m 51s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/b65eb14ca76045a4b900bb2eaee1ca0e : SUCCESS in 36m 29s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/6e77577c55564699a6c5d8c2d227c030 : SUCCESS in 37m 23s","accounts_in_message":[],"_revision_number":5},{"id":"deb845f863c496e0e3b3060b7b34002d2fd8fd73","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-21 17:19:29.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":5},{"id":"90df08978986d25bb234660fb153f3645515ff72","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-21 17:20:39.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/417af15cd5ef4d46be7be7171df2d997\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7bd9adbea23a4c148b24fa818f982d38 : SUCCESS in 49s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"49d066cc3a4d8821789736e291d151922876d012","revisions":{"4fb70341f95869c31b92bf63f1ceb219773da552":{"kind":"REWORK","_number":1,"created":"2022-10-14 06:19:17.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/12/861312/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/12/861312/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/861312/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/861312/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/861312/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/12/861312/1"}}},"commit":{"parents":[{"commit":"f534ca550f997dab5cc10575c0e9d82e71dfcc9a","subject":"[sqlalchemy2] positional arguments for case","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/f534ca550f997dab5cc10575c0e9d82e71dfcc9a"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2022-10-07 17:29:30.000000000","tz":120},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2022-10-13 17:46:19.000000000","tz":120},"subject":"[sqlalchemy2] Use subquery to select from a query","message":"[sqlalchemy2] Use subquery to select from a query\n\nQuery.select_from will be removed in sqlalchemy 2.0 [0]\n\n[0] https://docs.sqlalchemy.org/en/14/changelog/migration_20.html#\\\n    selecting-from-the-query-itself-as-a-subquery-e-g-from-self\n\nChange-Id: I9c3d148b1c4bff284d105e799846f5ce70c55046\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/4fb70341f95869c31b92bf63f1ceb219773da552"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/4fb70341f95869c31b92bf63f1ceb219773da552"}]},"branch":"refs/heads/master"},"c1035bb70f9f79ebc81a2ed2d4e88e7dc92ede92":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2023-02-27 07:52:40.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/12/861312/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/12/861312/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/861312/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/861312/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/861312/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/12/861312/2"}}},"commit":{"parents":[{"commit":"64ba26176e5a21531c0ca4cc08abf7fc4a29b51d","subject":"[sqlalchemy2] positional arguments for case","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/64ba26176e5a21531c0ca4cc08abf7fc4a29b51d"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2022-10-07 17:29:30.000000000","tz":120},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-02-24 14:44:49.000000000","tz":-300},"subject":"[sqlalchemy2] Use subquery to select from a query","message":"[sqlalchemy2] Use subquery to select from a query\n\nQuery.select_from will be removed in sqlalchemy 2.0 [0]\n\n[0] https://docs.sqlalchemy.org/en/14/changelog/migration_20.html#\\\n    selecting-from-the-query-itself-as-a-subquery-e-g-from-self\n\nChange-Id: I9c3d148b1c4bff284d105e799846f5ce70c55046\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c1035bb70f9f79ebc81a2ed2d4e88e7dc92ede92"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c1035bb70f9f79ebc81a2ed2d4e88e7dc92ede92"}]},"branch":"refs/heads/master"},"671f49f1f82b757814fcf8f4e917aa34bcac3ed4":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2023-03-29 08:44:58.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/12/861312/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/12/861312/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/861312/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/861312/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/861312/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/12/861312/3"}}},"commit":{"parents":[{"commit":"a057ffec7c8a21f17a059378c83a6e446d6cdf77","subject":"[sqlalchemy2] positional arguments for case","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a057ffec7c8a21f17a059378c83a6e446d6cdf77"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2022-10-07 17:29:30.000000000","tz":120},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-03-29 08:12:57.000000000","tz":-240},"subject":"[sqlalchemy2] Use subquery to select from a query","message":"[sqlalchemy2] Use subquery to select from a query\n\nQuery.select_from will be removed in sqlalchemy 2.0 [0]\n\n[0] https://docs.sqlalchemy.org/en/14/changelog/migration_20.html#\\\n    selecting-from-the-query-itself-as-a-subquery-e-g-from-self\n\nChange-Id: I9c3d148b1c4bff284d105e799846f5ce70c55046\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/671f49f1f82b757814fcf8f4e917aa34bcac3ed4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/671f49f1f82b757814fcf8f4e917aa34bcac3ed4"}]},"branch":"refs/heads/master"},"05f437507a93396dd9d7d17ff27fe64c64c43324":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2023-04-24 06:50:24.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/12/861312/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/12/861312/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/861312/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/861312/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/861312/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/12/861312/4"}}},"commit":{"parents":[{"commit":"c34caef01a7ab7e8884ab247e7926896e98857da","subject":"[sqlalchemy2] positional arguments for case","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c34caef01a7ab7e8884ab247e7926896e98857da"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2022-10-07 17:29:30.000000000","tz":120},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-04-24 06:41:53.000000000","tz":-240},"subject":"[sqlalchemy2] Use subquery to select from a query","message":"[sqlalchemy2] Use subquery to select from a query\n\nQuery.select_from will be removed in sqlalchemy 2.0 [0]\n\n[0] https://docs.sqlalchemy.org/en/14/changelog/migration_20.html#\\\n    selecting-from-the-query-itself-as-a-subquery-e-g-from-self\n\nChange-Id: I9c3d148b1c4bff284d105e799846f5ce70c55046\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/05f437507a93396dd9d7d17ff27fe64c64c43324"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/05f437507a93396dd9d7d17ff27fe64c64c43324"}]},"branch":"refs/heads/master"},"49d066cc3a4d8821789736e291d151922876d012":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2023-05-03 07:07:19.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/12/861312/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/12/861312/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/861312/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/861312/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/861312/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/12/861312/5"}}},"commit":{"parents":[{"commit":"c412058fbb4f8bcf02db396a1343116fc8061578","subject":"[sqlalchemy2] positional arguments for case","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c412058fbb4f8bcf02db396a1343116fc8061578"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2022-10-07 17:29:30.000000000","tz":120},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-05-03 07:02:50.000000000","tz":-240},"subject":"[sqlalchemy2] Use subquery to select from a query","message":"[sqlalchemy2] Use subquery to select from a query\n\nQuery.select_from will be removed in sqlalchemy 2.0 [0]\n\n[0] https://docs.sqlalchemy.org/en/14/changelog/migration_20.html#\\\n    selecting-from-the-query-itself-as-a-subquery-e-g-from-self\n\nChange-Id: I9c3d148b1c4bff284d105e799846f5ce70c55046\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/49d066cc3a4d8821789736e291d151922876d012"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/49d066cc3a4d8821789736e291d151922876d012"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
