)]}'
{"id":"openstack%2Foctavia~862131","triplet_id":"openstack%2Foctavia~master~Ib98d50b7b1c790a180efab1d6626f6db4b3998e1","project":"openstack/octavia","branch":"master","topic":"di-create-jammy","attention_set":{},"removed_from_attention_set":{"31664":{"account":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"last_update":"2023-02-13 13:02:28.000000000","reason":"Change was submitted"},"34429":{"account":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"last_update":"2023-02-13 08:28:55.000000000","reason":"\u003cGERRIT_ACCOUNT_34429\u003e replied on the change","reason_account":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}}},"hashtags":[],"change_id":"Ib98d50b7b1c790a180efab1d6626f6db4b3998e1","subject":"Update diskimage-create.sh to set jammy as default Ubuntu version","status":"MERGED","created":"2022-10-20 12:24:38.000000000","updated":"2023-02-13 13:04:46.000000000","submitted":"2023-02-13 13:02:28.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":6,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"862131-di-create-jammy","meta_rev_id":"dfb49c3bf4545270fb996952c645eae3c7270542","_number":862131,"virtual_id_number":862131,"owner":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-02-13 13:02:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"recommended":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"all":[{"value":1,"date":"2022-12-13 17:54:10.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":2,"date":"2023-02-09 17:35:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":2,"date":"2023-02-09 17:45:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":1,"date":"2023-02-09 17:45:33.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-10-20 16:54:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-12-13 17:54:10.000000000","updated_by":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"reviewer":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"state":"REVIEWER"},{"updated":"2023-02-09 17:35:03.000000000","updated_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"REVIEWER"},{"updated":"2023-02-09 17:45:33.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"}],"messages":[{"id":"4f6799c21ccecb1198de51137a627e1ee1a49f78","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-10-20 12:24:38.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7b10b1e3a681c1a584eb6bddb9356e4eb87b9f38","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-10-20 12:30:34.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"6b32f3d012652618d78cf043bc4dd4dd36cafcf8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-10-20 14:17:03.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"64bf4d51aa7a191dfe8e1a55126106b17b47de0a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-10-20 14:19:21.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"9ff994c125a567ef6184b5a544fa14a3b024f321","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-20 16:54:57.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/e1e8bcfb4639478481b88a244d90a39d\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9488bf49b6c647278bf9a5ef2bc08240 : SUCCESS in 9m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a3e2e9617ccf42cd9e8063734fd217c9 : SUCCESS in 8m 35s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9902cb53a6b444d9bd95d1bceceda189 : SUCCESS in 5m 46s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ae9bf6196ddf4eda986ea78a544cba37 : SUCCESS in 4m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a70f58d86d654de7bfa87f98b1f168a0 : SUCCESS in 12m 41s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/51630a6e080448479877e659ecbeacf9 : SUCCESS in 4m 51s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/c59e12bd4f144fa788b63c476cb9a946 : SUCCESS in 5m 54s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/afa6ef123fd54b958875003dffaacee7 : SUCCESS in 10m 04s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/9b60a57d0294497a977cba218d9657c8 : SUCCESS in 4m 46s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/59388772a50a4e0c82758f16f635fb70 : SUCCESS in 9m 03s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/be58eeca26594d8db4faa0728ea5ba9a : SUCCESS in 8m 37s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/5c138fa38c404454b04f6039d3873dde : SUCCESS in 8m 04s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/30ba64e3bad9410db6c16d304e76a85b : SUCCESS in 1h 45m 04s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/38fe2353ef664e85ab50c66ba84e0e14 : SUCCESS in 1h 44m 06s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/066c63e29b2e46bc8776d50dedfcb02b : FAILURE in 33m 58s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/19be8042974d473582ecd9bd8f3d85ef : SUCCESS in 49m 45s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/0118b4b932fa43e4960790fe15364394 : SUCCESS in 1h 03m 18s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/0183f09d524747a78a4206908ce64a5d : SUCCESS in 34m 07s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/51706f1d4f0b4a6e9757851a273dad5b : SUCCESS in 1h 28m 24s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/7513be5f958c40e2968771757ea4c83c : SUCCESS in 2h 32m 17s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/c7481323fe8244598789e9a5c5087aeb : SUCCESS in 2h 21m 25s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/5ea4fdb4d0f143f9801cecb4dd6839a4 : SUCCESS in 1h 51m 58s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/9a1d31c5c0054aa7ba81f3b9a07ab9fa : SUCCESS in 2h 21m 59s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/1c0dc4228d814eb18a82845d020354ff : SUCCESS in 2h 34m 25s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"dea2e8af8ec0774d7b606422c69add5391a4eb1a","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-10-21 08:12:51.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"db5d2e0e011e7b450da11e86a1c101d90a2c75af","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-21 10:43:27.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/16978f5969dd4195a0c804d0ad53ad28\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4552cdf3d19d41939a1fce00aa5b2ce3 : SUCCESS in 8m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9d23274f316f4b9aa4b2c55f7ee640bc : SUCCESS in 6m 06s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0d0ee384b7754097ae783e18ee69510c : SUCCESS in 4m 33s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/59107a28ab93451f9da707647ab9a192 : SUCCESS in 4m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/676b394334e848299e5a105d684e7e77 : SUCCESS in 11m 22s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/04ccfc59c1084a3290273b088872e656 : SUCCESS in 5m 02s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/6d19d9e757494eebb21266df1dd217c8 : SUCCESS in 5m 08s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/2e67fb99372541a0b30cffb257d65e32 : SUCCESS in 7m 22s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/e67d35b223ec4536b924472468d51f70 : SUCCESS in 4m 34s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/47e94e98865645bd8b0e5d29cae46f92 : SUCCESS in 7m 39s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/63a4b932efad4a4bbea1bbc6add257fa : SUCCESS in 7m 17s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/7f24bac8ee724c8781c79c3c1d602153 : SUCCESS in 7m 00s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/9d43b5e6191546fc8c26afef08691699 : SUCCESS in 1h 21m 51s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/b0a49eec50a84735992d1d1ad4a0bf94 : SUCCESS in 1h 54m 42s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/270c438959b648b6b59972816e46af93 : SUCCESS in 38m 47s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/2ec7b2e3c39e42e0bbccde13906e3d1e : SUCCESS in 48m 29s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/4cbba02b080c4f3d9fb3e236e67944cb : SUCCESS in 1h 12m 44s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/e757d4c57ab448f1959c06be34073bf1 : SUCCESS in 36m 39s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/b79e181f58b14498b7c1359336be82a4 : SUCCESS in 1h 44m 35s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/2585d77ab84540faac4a2a457661f570 : SUCCESS in 2h 07m 39s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/0866c042ea6340938debc591f18b7a23 : SUCCESS in 1h 31m 01s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/1fad2d28eeab4922a9c546a4c2726aee : SUCCESS in 1h 43m 45s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/b2c65831741d44328fddd22083e0683e : SUCCESS in 1h 35m 06s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/4ded149b4edc489bae20b73ab0950641 : SUCCESS in 2h 24m 23s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"b40582c05b5b17775435b593623c3e6f56fbf23c","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2022-12-13 17:54:10.000000000","message":"Patch Set 4: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"116278a7214351f23c2a61cdf454376445eea676","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-12-13 18:07:27.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"e453f339bb0dec95ca6ee1b9908b54f4056e136b","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-12-27 10:33:00.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"0b823bdc8bcfc247d06b753eb1aa99ba4bd2b304","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-02-09 17:35:03.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"aad1b73f1abe4a985b2200a036c0015990619ab6","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-02-09 17:45:33.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"742f321da70fffff1d148beb6d996f6501725940","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-09 17:46:25.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"db245fbf047886331e29cb5f43a0d62dfc5b3d47","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-09 18:09:40.000000000","message":"Patch Set 4: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/66da9bece1a84d34ad9e8eaeb7555bc6\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e6e05eadb0d74fb4b24f6a9852e4def2 : SUCCESS in 5m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6d4627c15b9449f99804d556771e54c0 : POST_FAILURE in 10m 26s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/597687dc351c49168d759ad56de832c7 : SUCCESS in 8m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0406a23a1b9f44bc8f990c874292931f : CANCELED\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/55195452754646fabe9d9d4ead969483 : CANCELED\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/cc4e01b6f679437a94075a8f29ff190e : SUCCESS in 4m 37s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/a5d0678b42c64098b9a060f76258a865 : CANCELED\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/f0ddfcbd2ac94e6d8df9def06e1e16d4 : CANCELED\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/63d2a06bd95b4b74a1b30b6ecaf02224 : CANCELED\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/10741756386d40dd954ddeebc9a5b29d : CANCELED\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/b620c570a59240d7b49e60e8982ffccd : CANCELED","accounts_in_message":[],"_revision_number":4},{"id":"80448ca949db0e0ca65836716bcc14a9b1dd0fe4","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2023-02-13 08:28:55.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"d5b1998291c67cc6738f1a3f28c7c231ae45a31d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-13 11:04:13.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5e72df5a643945e0b21ef6eccfbff457\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/777420d9651d453899252531efe1978a : SUCCESS in 10m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b5bc766c055b4c44a4feb0f193f8b4cb : SUCCESS in 5m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/474dccbef60447cfb9c3669ca8730ebf : SUCCESS in 5m 05s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f90f8f08e030462792d31980e817f713 : SUCCESS in 5m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fbd2816e60444ddfbf30c33c2ee137bc : SUCCESS in 15m 27s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f2485a913d0249ac81bcfb512855a311 : SUCCESS in 4m 24s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/a186cda4955c46c28fbeb1b46f0ce888 : SUCCESS in 7m 03s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/eaffae83d9bd47f598cfd41f6fec220d : SUCCESS in 7m 27s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/c79becc891014baabdb2eafd0d9e4e49 : SUCCESS in 4m 10s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/d2696eb944ff44edacb38b51f07d127c : SUCCESS in 7m 01s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/f4abf480785f4c9c97ab9c62f2aee67c : SUCCESS in 6m 34s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/08416f856e9e4db6a637400f361f4126 : SUCCESS in 6m 29s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/742bbfd0940c4336a4af1b780be32114 : SUCCESS in 1h 10m 49s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/c33823454333421f9e1f1fa5f2efc9e9 : SUCCESS in 2h 06m 08s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/965b0ed6982b4f69ad3681010f4ed99d : SUCCESS in 39m 02s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/510e1cf5e34a418e8f0f009555ba09fa : SUCCESS in 42m 18s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/33342ef864014d33adc1d060aa20fce4 : SUCCESS in 1h 05m 16s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/8a54b661c37a4a34878b11c78718a101 : SUCCESS in 34m 17s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/6b73fc5592d24f64adcb90ab36938b45 : TIMED_OUT in 2h 32m 57s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/0513f66bdfd745d9a6dea6fd96155ccf : SUCCESS in 2h 27m 08s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/71bdc9fcb19d4222a53ae2c309ee049e : SUCCESS in 2h 17m 11s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/0aa61ed4ce8b4cf8b33203584a3c4187 : SUCCESS in 2h 12m 48s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/1dbee9bb1b5e48c8961d6d2328c82bcd : SUCCESS in 2h 07m 18s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/1fea1e4968994abc924f0d7583f2b923 : SUCCESS in 2h 27m 13s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"a8a1e79ea50a859ca53463c62a1cabede15b040d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-13 11:05:01.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"0e30fb5cb174cb3ab6c6fcfda9963fcd7f13d7b1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-13 13:02:28.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c7a4e3487c8c4920ad13589f1c70fe4e\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/265475dbbe564ab9b125c0e75a735fd9 : SUCCESS in 5m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6b7e386ebbc0403295745f427c552d42 : SUCCESS in 4m 35s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/878bb7ee85db49b1a50ae34002e56844 : SUCCESS in 4m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c798561528cb49a3b6ebe958461d8a67 : SUCCESS in 12m 31s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fee69c607e654867bd786abe42d1d5c2 : SUCCESS in 4m 50s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/45bef04d513b4b4ea3b72c9e1ed46d54 : SUCCESS in 4m 06s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/d9d57fb7a9a642f6876865eacc8bd87f : SUCCESS in 7m 48s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/1ea8adcc0c8648038e5aad8c6f0bc694 : SUCCESS in 1h 34m 18s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/a687856f021b4f0b938e895839a9ae05 : SUCCESS in 1h 42m 39s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/3b5bac1a1cf84657ac433d9e0e7cfbe3 : SUCCESS in 42m 49s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/315f7cb9ec3849089964115b40c5efb3 : SUCCESS in 40m 10s","accounts_in_message":[],"_revision_number":4},{"id":"0082c98e4839057cc88a90c3e71436acaaef3a81","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-13 13:02:28.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"dfb49c3bf4545270fb996952c645eae3c7270542","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-13 13:04:46.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d5e28d647537427ea9a82f118374ea5d\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2bbb0e98e98e41ec856f6ccedd26af00 : SUCCESS in 59s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fd08dc57b10445d79e8b859c49fdf0a1 : SUCCESS in 56s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"37ddc6fba4c5d216a0088f7e8a8c0baf01f84b64","revisions":{"2613550729b490e04db535bbe4505a91a0bf5b7c":{"kind":"REWORK","_number":1,"created":"2022-10-20 12:24:38.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/31/862131/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/31/862131/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/31/862131/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/31/862131/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/31/862131/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/31/862131/1"}}},"commit":{"parents":[{"commit":"344d6d76ce4481f30fe295114551243d61e7eea8","subject":"Fix prometheus-proxy service name in Red Hat-based distros","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/344d6d76ce4481f30fe295114551243d61e7eea8"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-10-20 12:24:28.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-10-20 12:24:28.000000000","tz":120},"subject":"Update diskimage-create.sh to set jammy as default Ubuntu version","message":"Update diskimage-create.sh to set jammy as default Ubuntu version\n\nChange-Id: Ib98d50b7b1c790a180efab1d6626f6db4b3998e1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/2613550729b490e04db535bbe4505a91a0bf5b7c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/2613550729b490e04db535bbe4505a91a0bf5b7c"}]},"branch":"refs/heads/master"},"16bfb9825a517b0d65b17437cdd557170c77a027":{"kind":"REWORK","_number":2,"created":"2022-10-20 12:30:34.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/31/862131/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/31/862131/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/31/862131/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/31/862131/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/31/862131/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/31/862131/2"}}},"commit":{"parents":[{"commit":"344d6d76ce4481f30fe295114551243d61e7eea8","subject":"Fix prometheus-proxy service name in Red Hat-based distros","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/344d6d76ce4481f30fe295114551243d61e7eea8"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-10-20 12:24:28.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-10-20 12:30:25.000000000","tz":120},"subject":"Update diskimage-create.sh to set jammy as default Ubuntu version","message":"Update diskimage-create.sh to set jammy as default Ubuntu version\n\nChange-Id: Ib98d50b7b1c790a180efab1d6626f6db4b3998e1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/16bfb9825a517b0d65b17437cdd557170c77a027"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/16bfb9825a517b0d65b17437cdd557170c77a027"}]},"branch":"refs/heads/master"},"c5435104a698353bb60b7143b9abe38c7e5d9b97":{"kind":"REWORK","_number":3,"created":"2022-10-20 14:17:03.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/31/862131/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/31/862131/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/31/862131/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/31/862131/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/31/862131/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/31/862131/3"}}},"commit":{"parents":[{"commit":"344d6d76ce4481f30fe295114551243d61e7eea8","subject":"Fix prometheus-proxy service name in Red Hat-based distros","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/344d6d76ce4481f30fe295114551243d61e7eea8"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-10-20 12:24:28.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-10-20 14:16:55.000000000","tz":120},"subject":"Update diskimage-create.sh to set jammy as default Ubuntu version","message":"Update diskimage-create.sh to set jammy as default Ubuntu version\n\nChange-Id: Ib98d50b7b1c790a180efab1d6626f6db4b3998e1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c5435104a698353bb60b7143b9abe38c7e5d9b97"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c5435104a698353bb60b7143b9abe38c7e5d9b97"}]},"branch":"refs/heads/master"},"37ddc6fba4c5d216a0088f7e8a8c0baf01f84b64":{"kind":"REWORK","_number":4,"created":"2022-10-20 14:19:21.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/31/862131/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/31/862131/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/31/862131/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/31/862131/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/31/862131/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/31/862131/4"}}},"commit":{"parents":[{"commit":"344d6d76ce4481f30fe295114551243d61e7eea8","subject":"Fix prometheus-proxy service name in Red Hat-based distros","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/344d6d76ce4481f30fe295114551243d61e7eea8"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-10-20 12:24:28.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-10-20 14:19:09.000000000","tz":120},"subject":"Update diskimage-create.sh to set jammy as default Ubuntu version","message":"Update diskimage-create.sh to set jammy as default Ubuntu version\n\nChange-Id: Ib98d50b7b1c790a180efab1d6626f6db4b3998e1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/37ddc6fba4c5d216a0088f7e8a8c0baf01f84b64"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/37ddc6fba4c5d216a0088f7e8a8c0baf01f84b64"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[]}
