)]}'
{"id":"openstack%2Foctavia~862230","triplet_id":"openstack%2Foctavia~stable%2Fzed~Ie739cd582103d1aab3021c56654e395b5919358f","project":"openstack/octavia","branch":"stable/zed","attention_set":{},"removed_from_attention_set":{"32962":{"account":{"_account_id":32962,"name":"Jakob Meng","email":"code@jakobmeng.de","username":"jakobmeng@web.de"},"last_update":"2022-10-24 18:35:39.000000000","reason":"\u003cGERRIT_ACCOUNT_32962\u003e replied on the change","reason_account":{"_account_id":32962,"name":"Jakob Meng","email":"code@jakobmeng.de","username":"jakobmeng@web.de"}}},"hashtags":[],"change_id":"Ie739cd582103d1aab3021c56654e395b5919358f","subject":"Fix prometheus-proxy service name in Red Hat-based distros","status":"MERGED","created":"2022-10-21 07:23:49.000000000","updated":"2022-10-24 22:45:58.000000000","submitted":"2022-10-24 22:44:11.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"862230","cherry_pick_of_change":861145,"cherry_pick_of_patch_set":1,"meta_rev_id":"7d55d6ca741bfb0901d63994a3321fbb11f54523","_number":862230,"virtual_id_number":862230,"owner":{"_account_id":32962,"name":"Jakob Meng","email":"code@jakobmeng.de","username":"jakobmeng@web.de"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-10-24 22:44:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":2,"date":"2022-10-24 14:51:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-10-24 13:44:23.000000000","_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":1,"date":"2022-10-24 14:51:31.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"CC":[{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-10-21 07:23:49.000000000","updated_by":{"_account_id":32962,"name":"Jakob Meng","email":"code@jakobmeng.de","username":"jakobmeng@web.de"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"CC"},{"updated":"2022-10-21 09:47:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-10-24 13:44:23.000000000","updated_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"reviewer":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"state":"REVIEWER"},{"updated":"2022-10-24 14:51:31.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"}],"messages":[{"id":"1d2ae4ba90e30d156c78ea5b5eedb34f9d8829bb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32962,"name":"Jakob Meng","email":"code@jakobmeng.de","username":"jakobmeng@web.de"},"date":"2022-10-21 07:23:49.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"25e8e203a61c43806c2a501822597662361b10f6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-21 09:47:47.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3b8232f9ab1747bc825563034fb47183\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e9748806a7b84b7d9e095e3085f9c7d9 : SUCCESS in 9m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5afe87a3002241caab2b99525d9bfee1 : SUCCESS in 6m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2659b64db41e4beca6371e91c8aad162 : SUCCESS in 4m 17s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5b38f81d4d3d45f3950fc006a0d94149 : SUCCESS in 4m 43s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/cecbdcd4571341e5a70260b66c0eb888 : SUCCESS in 5m 24s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bed5754b8dae46889345449bff1630f5 : SUCCESS in 12m 53s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/075785be37f04162a7399764ad04aa83 : SUCCESS in 4m 48s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/87012e02d557491c988a5d9590969ad8 : SUCCESS in 7m 02s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/30be936a883f4ddcb8fbffdd5901ddc2 : SUCCESS in 5m 01s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/80857422ac9f4a9ba56b20ac906df1c8 : SUCCESS in 6m 57s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/8d9bd5c6d7b24028a5680075819416f3 : SUCCESS in 8m 44s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/215c841152dd44dc99d5901d4b13bf7f : SUCCESS in 8m 24s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/23c015cd578c49838a544f41a9d6ad89 : SUCCESS in 1h 21m 51s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/fc45613a27dc48fb960018237f256fdb : SUCCESS in 1h 39m 31s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/98283a2bf28f49f9a672197360b8684b : SUCCESS in 42m 41s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/6c266c7650414f90ae460dc2204e0625 : SUCCESS in 57m 27s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/c648bf34b8414ed2a3b03216ef64dc80 : SUCCESS in 1h 10m 58s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/a7a06158820541188e6dea4260c9caf2 : SUCCESS in 40m 08s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/806cae6dd41b4cfca8e11eb63c6db0ef : SUCCESS in 1h 47m 37s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/cfc0cdd97e6048ffbddf1cdb53c129cf : SUCCESS in 2h 21m 24s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/efa6beb580064646920a9884675beddb : SUCCESS in 1h 58m 46s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/bedc19e43f5641298c62340975a0f186 : SUCCESS in 1h 40m 58s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/6eb9ebd7cc66403c8b4a14142d2bc258 : SUCCESS in 1h 58m 06s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/4f6613d6ad2448a881d9bca0d2e068b3 : SUCCESS in 1h 50m 35s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"09581c760bea904affa6f90a2b1efa9670ad98d4","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-10-24 13:44:23.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"d6eb709f9c7cff1e9385e3e82218f52c1fed410e","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2022-10-24 14:51:31.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c66d36ec96560f2753fa1d5fd6494b944b28f142","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-24 14:52:11.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"be7e6f4e57eca9c801165897ce02723a34674fac","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-24 15:02:17.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/03f38e0d04c3468e9a8cd57359dd34ab\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d290473247b0471db6f4191139f9c9b4 : CANCELED\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/95517d95a8a14aaea72a4f946b4c6064 : SUCCESS in 4m 35s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/87bb0596ae264963ac0777a772818243 : SUCCESS in 5m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/64c918d87e104a0eb076a2bb6641c9cc : CANCELED\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/2d93e54c9dc5467abc05fe5fa15bcbbd : SUCCESS in 4m 48s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/3ce460874cda403fab517b25d062d3e5 : CANCELED\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/6a00f08f956b46f086ea640efb77725d : CANCELED\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/3df39ab0b3ed47d29cac02a3e4c4f15e : FAILURE in 9m 07s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/c464df9747e74f5fb8bd4305772effe2 : CANCELED\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/e61f308317994c0787256c3ebf1df525 : CANCELED","accounts_in_message":[],"_revision_number":1},{"id":"1135cae0ba2894bba702934a5797e90e2dedd531","author":{"_account_id":32962,"name":"Jakob Meng","email":"code@jakobmeng.de","username":"jakobmeng@web.de"},"date":"2022-10-24 18:35:39.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"4a832fcf0b855cdcad6bb5ad3bc8cfa3dad1f2b8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-24 20:43:59.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9cb315f3534543f1a73a72082a4b46b4\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2aebcfe15cd14b5286730407d6c4cf2e : SUCCESS in 9m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/332f32cdb4044d809405ccbce134852c : SUCCESS in 5m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/65eed02ac6b54c52815b39e79ce26007 : SUCCESS in 4m 26s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/dcadfc40126a4eb4aeeedb779438b6b6 : SUCCESS in 5m 25s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7c07ea6657fe43e2a2b49eb27b868ee1 : SUCCESS in 4m 31s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ccd75102af65408d8aba9d6fce20d188 : SUCCESS in 11m 24s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/af318fe87f3d407cae5d26132dc859a5 : SUCCESS in 4m 57s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/5c8c2d9c32e5475b93253d23ed9b5542 : SUCCESS in 6m 45s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/49116da3b76a48c39cced2f8cd1406d3 : SUCCESS in 5m 08s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/3868525ef96f460ab609dc400257056a : SUCCESS in 6m 14s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/4c5488087453425784aa054cb0a03d21 : SUCCESS in 6m 59s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/339a22a6fbbc48cda19fcf04d59117cb : SUCCESS in 6m 25s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/396af07f453446248930d3a56f6e1bbf : SUCCESS in 1h 11m 57s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/f474db54d5c8429088fc71124affaae2 : SUCCESS in 1h 34m 51s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/a84b62ef91624b6ab8b667e7044fe22e : SUCCESS in 40m 51s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/83c5a7018aa74a1b9474a2843622a25b : SUCCESS in 52m 21s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/bcbadc576fc048aa86b01d0ac1df9811 : SUCCESS in 1h 11m 33s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/83c7f708aa78465982d745224e796210 : SUCCESS in 31m 16s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/96e6fff275e146768578a5ab5bdfafdb : SUCCESS in 1h 35m 06s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/c1b8a1129591403ba5677b57469a8ea1 : SUCCESS in 2h 06m 55s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/8a0eb113471e4a0fbb42dd1a14c77128 : SUCCESS in 1h 31m 06s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/fc73d3a2a53b4a8eada626dea3f05f37 : SUCCESS in 1h 45m 16s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/9b6c4df5850b452b9809c3eae66fe4b0 : SUCCESS in 1h 46m 46s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/5d3a1c87cb804852b71dc75ae29203db : SUCCESS in 1h 43m 04s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ea07317fc45ba7286f733772af90eff03b4d00d3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-24 20:44:42.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"90fc5d56e0923f99266a3de5d0e54e2946b96a49","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-24 22:44:11.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/61a9a106ead8443b85c36f2edbbb4078\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0f9e51815ffb45cea198be007201056e : SUCCESS in 6m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bd8bb6b4c41542fdbf118cbb8473a6c9 : SUCCESS in 4m 25s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/62070e6ae11c4ef09b148c7a59f2580e : SUCCESS in 4m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d7160e2259f44668b96a3043468e8855 : SUCCESS in 13m 17s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/3ad25f7fb8a648aab637cd1f486d4b18 : SUCCESS in 4m 44s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/7cad6182263349c78080e63cf960f6c6 : SUCCESS in 7m 02s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/71fc27d2a51148d496118abdb419fb2e : SUCCESS in 1h 43m 06s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/c9ea7f117d1a4d769ccdaa271d705efd : SUCCESS in 1h 58m 42s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/209aa7aba28745bc854261c04fb02acc : SUCCESS in 34m 00s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/70b852185f584979be95be8868cc929d : SUCCESS in 1h 03m 40s","accounts_in_message":[],"_revision_number":1},{"id":"a37cc262ec8bebd0106eabe6fdb808cbf668f7ff","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-24 22:44:11.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"7d55d6ca741bfb0901d63994a3321fbb11f54523","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-24 22:45:58.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9781706d603c4508ba3bf770e0c0c5f8\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f49f9fcde58c445881713ad54c017c1e : SUCCESS in 1m 28s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"48fad7c2b8608cbd174e0b3d7953ad15fb2bf942","revisions":{"48fad7c2b8608cbd174e0b3d7953ad15fb2bf942":{"kind":"REWORK","_number":1,"created":"2022-10-21 07:23:49.000000000","uploader":{"_account_id":32962,"name":"Jakob Meng","email":"code@jakobmeng.de","username":"jakobmeng@web.de"},"ref":"refs/changes/30/862230/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/30/862230/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/30/862230/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/30/862230/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/30/862230/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/30/862230/1"}}},"commit":{"parents":[{"commit":"4f2a019db769ed3fd95e3cd2c0fa69458cf77187","subject":"Update TOX_CONSTRAINTS_FILE for stable/zed","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/4f2a019db769ed3fd95e3cd2c0fa69458cf77187"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2022-10-13 07:46:11.000000000","tz":120},"committer":{"name":"Jakob Meng","email":"code@jakobmeng.de","date":"2022-10-21 07:23:49.000000000","tz":0},"subject":"Fix prometheus-proxy service name in Red Hat-based distros","message":"Fix prometheus-proxy service name in Red Hat-based distros\n\nThe prometheus-proxy service is called octavia-prometheus-proxy in RDO\u0027s\nRPMs [0].\nAdd a svc-map for the service name to fix the amphora-agent element for\nthese distros.\n\n[0] https://review.rdoproject.org/r/c/openstack/octavia-distgit/+/44152\n\nPartial-Bug: 1992668\n\nChange-Id: Ie739cd582103d1aab3021c56654e395b5919358f\n(cherry picked from commit 344d6d76ce4481f30fe295114551243d61e7eea8)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/48fad7c2b8608cbd174e0b3d7953ad15fb2bf942"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/48fad7c2b8608cbd174e0b3d7953ad15fb2bf942"}]},"branch":"refs/heads/stable/zed"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[]}
