)]}'
{"id":"openstack%2Foctavia~862231","triplet_id":"openstack%2Foctavia~stable%2Fyoga~Ie739cd582103d1aab3021c56654e395b5919358f","project":"openstack/octavia","branch":"stable/yoga","attention_set":{},"removed_from_attention_set":{"32962":{"account":{"_account_id":32962,"name":"Jakob Meng","email":"code@jakobmeng.de","username":"jakobmeng@web.de"},"last_update":"2022-10-25 08:17:21.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ie739cd582103d1aab3021c56654e395b5919358f","subject":"Fix prometheus-proxy service name in Red Hat-based distros","status":"MERGED","created":"2022-10-21 07:24:04.000000000","updated":"2022-10-25 08:19:11.000000000","submitted":"2022-10-25 08:17:21.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"862231","cherry_pick_of_change":861145,"cherry_pick_of_patch_set":1,"meta_rev_id":"62aca20399f62c40c52ff1d2b144d7ea5d256efa","_number":862231,"virtual_id_number":862231,"owner":{"_account_id":32962,"name":"Jakob Meng","email":"code@jakobmeng.de","username":"jakobmeng@web.de"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-10-25 08:17:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-10-24 20:33:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":2,"date":"2022-10-25 06:26:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":2,"date":"2022-10-24 20:33:11.000000000","_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":1,"date":"2022-10-25 06:26:21.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-10-21 07:24:04.000000000","updated_by":{"_account_id":32962,"name":"Jakob Meng","email":"code@jakobmeng.de","username":"jakobmeng@web.de"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"CC"},{"updated":"2022-10-21 09:10:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-10-24 13:44:51.000000000","updated_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"reviewer":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"state":"REVIEWER"},{"updated":"2022-10-24 14:52:02.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2022-10-25 06:26:21.000000000","updated_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"REVIEWER"}],"messages":[{"id":"c2a971db43d20a1047a640ad92d7d4d8272ad5dd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32962,"name":"Jakob Meng","email":"code@jakobmeng.de","username":"jakobmeng@web.de"},"date":"2022-10-21 07:24:04.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"da4509e50f6e89a7747602fcc993db9d7b89b87b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-21 09:10:00.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/a38463771cb14332bed245f92b3227b3\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4dc6a8243232481ba76b2d2368198b85 : SUCCESS in 9m 07s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/33f5529b5b1b4fbdadd5089d40168c07 : SUCCESS in 7m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4af86f203ace41949340d50a20afc237 : SUCCESS in 5m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3a9cfb27924242669d8cc21514055e27 : SUCCESS in 3m 58s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/18309dee8b5844ad9f91babdb5f1a96a : SUCCESS in 4m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fa32329a9abc475199c6293e80f18d4f : SUCCESS in 10m 27s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/01760c9d89b14e4b8eddd9e590a482c1 : FAILURE in 4m 45s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/376b203ef314482ab5ddd5eb38bb6867 : FAILURE in 5m 11s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/164a7b2a1598480e9aff558fcdb2a386 : SUCCESS in 4m 09s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/c5f0314115a04be1af7a72ad77937c15 : SUCCESS in 7m 12s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/9d1aa858b14543ccad872cef82a5d1b1 : SUCCESS in 1h 00m 57s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/fe8c698a35ba4515a8577115a21e8641 : SUCCESS in 1h 28m 52s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/68beb186346f4029858f25b088e64d3d : SUCCESS in 35m 22s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/fd25d64d41ac4b56ad94389e1fac2631 : SUCCESS in 47m 01s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/b6746081dfd94b3196e8e27ba0a340b3 : SUCCESS in 1h 04m 29s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/7468dd3fb9ba43179dc0ddf818f0b133 : SUCCESS in 32m 42s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/d1cdff3957114bbe854682457ba6b1f4 : SUCCESS in 1h 41m 18s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/57b18988f0be4c48baa17c997f82361f : SUCCESS in 1h 43m 49s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/5ae35069abfa467492ddbead87a4cea4 : SUCCESS in 1h 39m 29s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/661dfa981c084cd9add3660df62acc67 : SUCCESS in 1h 27m 17s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/b41446979c1d45979cf0d6861b1bb78b : SUCCESS in 1h 43m 03s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8-stream https://zuul.opendev.org/t/openstack/build/9620e7ac0d554a50b61ce74010ef8f74 : FAILURE in 32m 22s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"7c71c810864cd482a6136a4267ff1260b8f7d61f","author":{"_account_id":32962,"name":"Jakob Meng","email":"code@jakobmeng.de","username":"jakobmeng@web.de"},"date":"2022-10-21 10:02:11.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"8325b8132aaae5f5ed9dfbcaf4543b2f3bd137bf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-21 11:56:04.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/69e4c928bb6b40e4b7e62a8b997aec90\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/34c635f10f5d4a5a9283e56cc636af28 : SUCCESS in 9m 13s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/481e5f553b624549a4c08b496b385804 : SUCCESS in 7m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c8e48520190946708b28a6acecee84c5 : SUCCESS in 5m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e95dfba5cc6b49adb06b159876f9e3dd : SUCCESS in 4m 56s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b4d1a39fce4a4caeb64204835a1c4d8e : SUCCESS in 5m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9eecaa85dc504b909ce2c8fe6aaac761 : SUCCESS in 12m 22s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/6f064eb7d47d4d3aaa47f81155ce4a8e : FAILURE in 4m 35s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/2140d3a81d174940b054a66f7a7701b7 : FAILURE in 4m 34s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/7a5f1ce15aa64e90a3f0316147b546b4 : SUCCESS in 4m 37s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/8a3372049b604d368e5e721e31fce677 : SUCCESS in 8m 52s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/2b2bac7b69264782a1f138ddf5ab00b7 : SUCCESS in 1h 21m 55s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/318c4e5b2e274fee899a07eb9f77659f : SUCCESS in 1h 36m 20s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/bb3d06df94b1491f8c576608fa7bdde4 : SUCCESS in 40m 24s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/c0926862185547a391acafda37ba8407 : SUCCESS in 48m 28s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/f22eb2a63d0b49f5b58d3bd39f377063 : SUCCESS in 1h 13m 41s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/32bc741e48b44f01a4b35afed59e63e1 : SUCCESS in 35m 28s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/f68c5f3e5d9047ec97aca73c82908596 : SUCCESS in 1h 27m 58s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/185a7ad643884480b310463043a21e69 : SUCCESS in 1h 52m 07s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/945dbc67564348e18957bd69d48906c1 : SUCCESS in 1h 47m 55s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/6496dcbdd4ac4fcf803aa7b9c885a006 : SUCCESS in 1h 41m 57s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/b80d973d8ac2413fbda9cdfd57714d57 : SUCCESS in 1h 40m 59s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8-stream https://zuul.opendev.org/t/openstack/build/63ad91838d5c4568a551eda96c8dbca7 : FAILURE in 34m 30s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"1837c5c608c0e5dda5af7b66186091214c081bf6","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-10-24 13:44:51.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"dcae5b675db205e92d2cabdb2c9c73e6c9cbd58f","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2022-10-24 14:52:02.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"734c149038329ba2fec3c1a66f37d5b0f71d8c8d","author":{"_account_id":32962,"name":"Jakob Meng","email":"code@jakobmeng.de","username":"jakobmeng@web.de"},"date":"2022-10-24 18:37:13.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"247c962e4f825e883a4ce0dae6b0337d2e23e48a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-24 20:26:36.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/92fdd1f458ac4b8f9313ca03e009f38d\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a6019ac72cc844df993ad8885fdb2933 : SUCCESS in 8m 21s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b3716ee1b0f04b4481ea1aff271fc7fd : SUCCESS in 9m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a59d074d69b5492b913221c6a7fe3e65 : SUCCESS in 6m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/841432fcf08f4eff98d07ad0529a1b64 : SUCCESS in 4m 34s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f0fa1551bdae4c43aaf201de90ce1a45 : SUCCESS in 4m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7d55eb8a09124711a8c5e740c38fdb83 : SUCCESS in 12m 21s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/0d384ad017f1454692de12595ede2fa8 : FAILURE in 5m 07s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/908739c97436447babd172f8af3c0e98 : FAILURE in 6m 08s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/f407a76990ff4cf290b167996c23163b : SUCCESS in 4m 53s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/15bf30559272459a97d0d110f78102ba : SUCCESS in 7m 49s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/e93eac6683cd411089de5735a47da55e : SUCCESS in 1h 25m 24s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/5613cf251db643b28088cdf6c343cb8d : SUCCESS in 1h 34m 54s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/d7c79afaa29f4bbea936ba983b36f056 : SUCCESS in 37m 23s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/14a68fe1c29f4b54b30a53024d5afb25 : SUCCESS in 45m 57s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/1b11c09427d74172a79bfe4c98276472 : SUCCESS in 53m 11s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/9d4c43cef9064029a0d4a4b44ace3d75 : SUCCESS in 32m 01s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/ab7a84d35ddb4b57819e518506db69e7 : SUCCESS in 1h 28m 50s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/5482f0ab886c482c9c2dad986ab9a1f3 : SUCCESS in 1h 47m 09s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/82f23268929f4ed2af613d1904016405 : SUCCESS in 1h 24m 58s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/1b255f481b534a47a91f5e01836443c1 : SUCCESS in 1h 31m 57s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/f60b3e219ee6459ba2cc2900ef62e82d : SUCCESS in 1h 43m 56s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8-stream https://zuul.opendev.org/t/openstack/build/d3440225737a472eb975773e183610a2 : FAILURE in 35m 22s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"c323292a38c59e4d58a17a74dda43b145ca16d24","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2022-10-24 20:33:11.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"927d302b8f8883418f18de78b1556562c94a5aad","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-24 23:02:43.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/947d45ec33f2442fa8c7bad11220d7af\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/80c597225f35457ead0d307f2fdf29c5 : SUCCESS in 7m 04s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/284ca70396e647c088e8a7c718261f90 : SUCCESS in 9m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/12b971a6f3654609830f37b4d553d404 : SUCCESS in 5m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8837d6812ef945248adadae390e9aca8 : SUCCESS in 7m 35s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0311c51fc706429198e14870e7c4438b : SUCCESS in 5m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/065b8d53a8b24f3393ec7348cab96722 : SUCCESS in 14m 44s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/563a0583223349a981d33d141bda1cf0 : SUCCESS in 5m 38s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/ac95f1dd17e346798c3a80497e1ef68c : SUCCESS in 6m 19s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/9cdf8fc015244d6083c9ef92e7dcb4d0 : SUCCESS in 5m 35s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/4b745e169d0e46c0987a30ac4f1f6565 : SUCCESS in 9m 12s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/3176b76787ed461d8676f0c84092a154 : SUCCESS in 1h 46m 53s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/3d763cf546a24804b158843932cb5852 : SUCCESS in 2h 01m 47s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/4f6e21ece6a94a3f8fe77cd0a78f976b : SUCCESS in 35m 03s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/d3861f11634f426ba96f49c2aff91e98 : SUCCESS in 43m 53s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/a83a500cd71d41b6a3101d3dd1c03b18 : SUCCESS in 1h 07m 17s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/554955246d1d4f0a8bda8669f7e10a4b : SUCCESS in 34m 45s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/d29a9ff22d054685a6cfa90d5a8a8ced : SUCCESS in 1h 27m 00s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/2837568bca394359b5f31aaeeaab12e2 : SUCCESS in 1h 57m 51s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/1039529364e24f5a80b14b06905c2c99 : SUCCESS in 1h 50m 19s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/ca695c1077754090b5fa216298d48786 : SUCCESS in 2h 00m 05s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/d49d70f8a97e4a3bb2c7c4b31fa50db9 : SUCCESS in 2h 07m 45s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8-stream https://zuul.opendev.org/t/openstack/build/ce0a255436f2411d923cd1219a25e72e : FAILURE in 33m 58s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"8c4da80590ef2aa0f1cb00c7d7514c85949ae4a8","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2022-10-25 06:26:21.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"b273f05dbafd8d136976d3b77e29bd68d722a271","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-25 06:27:00.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"7026a934a632c6dca0bc2a9d854c4b61e92c1ee3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-25 08:17:21.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/126307f7e70d4069b5bc999d7db5a53b\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c062a05f910c4802a448ca90f95cd5f7 : SUCCESS in 6m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0e5d6969051b4e1aa0335e7911fa13be : SUCCESS in 4m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9fa9d754f5334777a57a9345835c4286 : SUCCESS in 4m 16s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/641a4a4320a94b7ea60a08d26ddfce3c : SUCCESS in 5m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/33fd8836557c40a5a71c41fb05805d48 : SUCCESS in 8m 53s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/2dbc51d681234b769e705e6f39c8ecd6 : SUCCESS in 4m 37s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/d057892c16f04295b4a47ce4ebe5f657 : SUCCESS in 6m 50s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/4abf4cd719fc44e0982d676707d902a4 : SUCCESS in 1h 49m 46s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/cff15902ddad4d22af7f587ea67e11dd : SUCCESS in 1h 40m 48s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/3a8af388977e45dbb7989494765c2a9e : SUCCESS in 37m 32s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/cf6f3496b434454fbc7df551fcd71940 : SUCCESS in 46m 15s","accounts_in_message":[],"_revision_number":2},{"id":"685d681e1cffba253355a88e899c112351109e34","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-25 08:17:21.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"62aca20399f62c40c52ff1d2b144d7ea5d256efa","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-25 08:19:11.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/46f6b99f5eb44c59980e53f31d818284\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ef345ffd1d4f4682b452fc0ddc7bc9a7 : SUCCESS in 1m 33s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"5efaf36096a8659813c5fe035717d721d494b52d","revisions":{"25cab926d596729a31ae1bf954918d2ca2214732":{"kind":"REWORK","_number":1,"created":"2022-10-21 07:24:04.000000000","uploader":{"_account_id":32962,"name":"Jakob Meng","email":"code@jakobmeng.de","username":"jakobmeng@web.de"},"ref":"refs/changes/31/862231/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/31/862231/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/31/862231/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/31/862231/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/31/862231/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/31/862231/1"}}},"commit":{"parents":[{"commit":"bb4cde2febd6be746ff4874be21b3021da2b4e48","subject":"Merge \"Fix PortNotFound exception when updating a LB after a failover\" into stable/yoga","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/bb4cde2febd6be746ff4874be21b3021da2b4e48"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2022-10-13 07:46:11.000000000","tz":120},"committer":{"name":"Jakob Meng","email":"code@jakobmeng.de","date":"2022-10-21 07:24:04.000000000","tz":0},"subject":"Fix prometheus-proxy service name in Red Hat-based distros","message":"Fix prometheus-proxy service name in Red Hat-based distros\n\nThe prometheus-proxy service is called octavia-prometheus-proxy in RDO\u0027s\nRPMs [0].\nAdd a svc-map for the service name to fix the amphora-agent element for\nthese distros.\n\n[0] https://review.rdoproject.org/r/c/openstack/octavia-distgit/+/44152\n\nPartial-Bug: 1992668\n\nChange-Id: Ie739cd582103d1aab3021c56654e395b5919358f\n(cherry picked from commit 344d6d76ce4481f30fe295114551243d61e7eea8)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/25cab926d596729a31ae1bf954918d2ca2214732"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/25cab926d596729a31ae1bf954918d2ca2214732"}]},"branch":"refs/heads/stable/yoga"},"5efaf36096a8659813c5fe035717d721d494b52d":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2022-10-24 20:33:11.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/31/862231/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/31/862231/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/31/862231/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/31/862231/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/31/862231/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/31/862231/2"}}},"commit":{"parents":[{"commit":"0d8a06bd2a961228e9138ab6d8216bd55205d1d5","subject":"Use python3.9 for tips jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/0d8a06bd2a961228e9138ab6d8216bd55205d1d5"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2022-10-13 07:46:11.000000000","tz":120},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2022-10-24 20:33:11.000000000","tz":0},"subject":"Fix prometheus-proxy service name in Red Hat-based distros","message":"Fix prometheus-proxy service name in Red Hat-based distros\n\nThe prometheus-proxy service is called octavia-prometheus-proxy in RDO\u0027s\nRPMs [0].\nAdd a svc-map for the service name to fix the amphora-agent element for\nthese distros.\n\n[0] https://review.rdoproject.org/r/c/openstack/octavia-distgit/+/44152\n\nPartial-Bug: 1992668\n\nChange-Id: Ie739cd582103d1aab3021c56654e395b5919358f\n(cherry picked from commit 344d6d76ce4481f30fe295114551243d61e7eea8)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/5efaf36096a8659813c5fe035717d721d494b52d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/5efaf36096a8659813c5fe035717d721d494b52d"}]},"branch":"refs/heads/stable/yoga","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[]}
