)]}'
{"id":"openstack%2Foctavia~862599","triplet_id":"openstack%2Foctavia~master~Ia0c1d4107161fff8d1f3071d57860bfd265f596f","project":"openstack/octavia","branch":"master","topic":"remove-split-listeners","attention_set":{},"removed_from_attention_set":{"31664":{"account":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"last_update":"2023-01-12 12:22:15.000000000","reason":"\u003cGERRIT_ACCOUNT_31664\u003e replied on the change","reason_account":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"}},"29244":{"account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"last_update":"2023-01-13 09:15:58.000000000","reason":"\u003cGERRIT_ACCOUNT_29244\u003e replied on the change","reason_account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}},"11628":{"account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"last_update":"2023-01-12 15:39:23.000000000","reason":"\u003cGERRIT_ACCOUNT_11628\u003e replied on the change","reason_account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},"34429":{"account":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"last_update":"2023-02-04 08:39:25.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ia0c1d4107161fff8d1f3071d57860bfd265f596f","subject":"Remove support for split listener configuration","status":"MERGED","created":"2022-10-25 11:41:17.000000000","updated":"2023-02-04 08:40:37.000000000","submitted":"2023-02-04 08:39:25.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":34,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"862599-remove-split-listeners","meta_rev_id":"2a80371fedf29a8113a092215cb0d1a33e4f1d93","_number":862599,"virtual_id_number":862599,"owner":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-02-04 08:39:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"recommended":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"all":[{"value":1,"date":"2023-02-02 16:44:03.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-02-02 16:44:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":2,"date":"2023-02-03 17:29:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"all":[{"value":0,"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":1,"date":"2023-02-04 06:34:56.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"rejected":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"all":[{"value":0,"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":-2,"date":"2023-02-02 16:44:03.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-10-25 14:54:33.000000000","updated_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"REVIEWER"},{"updated":"2022-10-25 17:11:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-10-26 14:16:37.000000000","updated_by":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"reviewer":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"state":"REVIEWER"},{"updated":"2023-01-11 20:55:22.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"}],"messages":[{"id":"ec071dc733a008bfcc35cdddedaf1fe3a39b01b2","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-10-25 11:41:17.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"75ca0f0c8e33464b844f79ff6c3c22c3e4ad3aa7","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-10-25 14:37:52.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"dd4faadd370397b3e504f05cac79c220a776dd44","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-10-25 14:39:40.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"52521fc4d192be78ef99ba52ff6f4a7b58a78e6d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-10-25 14:57:57.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"8fdc0df337b5b6c815bb26f6acbcc7773005e60d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-25 17:11:38.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b239ac672e5d40beb9eb25b3f5093d94\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5971c21608b74b27a6f06e6139e8e30f : SUCCESS in 11m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/498743bc6f60481ba849009619ffed4d : SUCCESS in 6m 30s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/022219734552493eb46df4bcc9d4431a : SUCCESS in 4m 42s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a7b5bd0ff4da4af4a03d838d8d3f3b55 : SUCCESS in 5m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/59de7aff11f54bd2820455a62017b2cb : SUCCESS in 12m 21s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/058cd91cadd94a8dbe43c324e909d825 : SUCCESS in 4m 29s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/fac7d08c8e744e53b97693d1d2f273e9 : SUCCESS in 5m 05s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/710aeec4dfff4dd9b24677d40c216665 : SUCCESS in 7m 39s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/e187be04b12c4d378d21cd495e515349 : SUCCESS in 5m 21s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/a33ee7c1898b485883faf650c1edcf89 : SUCCESS in 7m 18s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/f0b646f324844043a6cbd4f3b3a612b1 : SUCCESS in 7m 02s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/4d7d046172fe466a8a5bdfbbad2881a8 : SUCCESS in 7m 48s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/ae350d2f6ed44c63857cb33c44c3f57d : SUCCESS in 1h 33m 25s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/f2a4239a75a049bf97ba5403ca49c7ae : SUCCESS in 1h 30m 38s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/7cf4e435a308430b8a44d76a9ca79503 : SUCCESS in 32m 54s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/f0700ad52928480f9ab2bef907561377 : SUCCESS in 48m 53s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/6b3c56c140484fd2bca6ed0ad87e03a5 : SUCCESS in 1h 18m 34s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/c9c64575e1e3441d9f580590569adb33 : SUCCESS in 33m 51s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/cc0d8b7a309f4995a04465bc74048117 : SUCCESS in 1h 52m 07s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/b23554d5055345bab9ea335366fe4c76 : SUCCESS in 2h 11m 51s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/a94549505c1a4b79ae83abb857216a1e : SUCCESS in 1h 46m 35s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/99c1de62222e435b9da921f1d015b364 : SUCCESS in 1h 42m 14s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/051157cf1a184b04b847bd660e6491e0 : SUCCESS in 1h 32m 39s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/3757493414a84e688c342ac15b7073eb : SUCCESS in 1h 55m 35s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"28bda456a812ffcaaba50d7c151c4d15a1346f89","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2022-10-27 06:56:24.000000000","message":"Patch Set 4: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"d2829626af2532764bbd442fb155a8784a81deb0","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2022-10-27 07:00:51.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"8ac0a8fab12a8abb6e189bd586dc7f8c8ebb72d9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-10-27 09:22:56.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"e1e4f67a339033fd35f7b9ec43847535510086b7","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-10-27 09:23:50.000000000","message":"Patch Set 5:\n\n(3 comments)","accounts_in_message":[],"_revision_number":5},{"id":"1a000b595452ef77bede344c31c50d781053b015","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2022-10-27 10:47:42.000000000","message":"Patch Set 5: Code-Review-1\n\n(4 comments)","accounts_in_message":[],"_revision_number":5},{"id":"7ae63204e6e75c815d6f3a18e749e6923dd98ef7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-27 11:49:56.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5794688b93184449b4f6548600edd9a6\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f4decfa3e9d34632a7929d61ce6a1d8e : SUCCESS in 9m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7b161d719ca04636b7be19e12dc5c444 : SUCCESS in 6m 33s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4eea9ab2483c4e66a204f101c022d048 : SUCCESS in 4m 40s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4345f744f50b41b4a5179d453feb85a8 : SUCCESS in 4m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/79df0dabc63d4dc6a53c9403b11f3689 : SUCCESS in 11m 23s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f8ddaa3c1a404d4c92992823ea8b6c70 : SUCCESS in 5m 08s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/7f0ccdf8fba1493cb811d94ad4f4e83a : SUCCESS in 5m 01s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/4c3c141ebeb045008bdcbd1fd42c0044 : SUCCESS in 8m 39s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/c6b74b486a214a82bebefa37da5eb39b : SUCCESS in 4m 19s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/c460492348914792a6c4ce95e609d0dc : SUCCESS in 7m 34s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/9802cf2a56014bb8a8432eb7482a24b4 : SUCCESS in 7m 21s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/409c5679c9c744cdb04e3431594e16aa : SUCCESS in 7m 29s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/f320ae1236c24a76bbd78ed2aa2d0e25 : SUCCESS in 1h 27m 37s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/c6914df973e84b94946ffbfc9255a4b2 : SUCCESS in 2h 24m 04s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/aec42b9d11364892ae2b5dbb35a873fd : SUCCESS in 40m 05s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/584355db2a974bbf8cea064ab11ff05d : SUCCESS in 53m 30s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/7c7b29a93dc64482982496cb21fbc5e8 : SUCCESS in 1h 14m 31s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/ae1fdb4d7409427c8780fde81e92efa3 : SUCCESS in 39m 14s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/7d42e5a8a0644d6a9e5fb7db47164c54 : SUCCESS in 1h 36m 29s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/431d8a28d9014efa95dbe0e5916b5d81 : SUCCESS in 2h 24m 21s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/3aa83d3ba9ac445cb78732bcde8fac6c : SUCCESS in 1h 58m 27s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/8aea80729ec4470994a0d0025737c5b7 : SUCCESS in 1h 47m 20s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/2ba788fa757c4b538463885a5ad012ee : SUCCESS in 1h 32m 00s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/3f93931c17ed4b4696e3eae156520dce : SUCCESS in 2h 06m 45s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"50a8536b5c5bf709cb1566b66f50e87c5a14489f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-10-27 12:47:59.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"b1c5aca5ba94179b1604ac274d5790fa0ba968c4","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-10-27 12:49:18.000000000","message":"Patch Set 6:\n\n(4 comments)","accounts_in_message":[],"_revision_number":6},{"id":"489a43895b9a0fc827e5de01f77f50d8fc0a3cb2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-10-27 12:59:44.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"4ac6bdae2fbc2cdcaddf1766e123d627289f8c65","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-27 15:07:34.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e2c6bd3c5c8341f2a2afbf913adfc4be\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cfc5f9ccdd824136ab0264533a5943b0 : SUCCESS in 9m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/674ac08bed63495194b714b80e18e027 : SUCCESS in 7m 12s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b0dad470dbd442999ba525f68b6f0623 : SUCCESS in 5m 22s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c06110f921d4477db57c12e1ce76ae7f : SUCCESS in 5m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7edd7ff4b64c4884b6e97dd5786240f4 : SUCCESS in 10m 45s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cd7d1418f9404389b6477cd1c6eac484 : SUCCESS in 4m 37s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/ab3de0f2b148420c88bde049c2e58f18 : SUCCESS in 4m 45s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/d71ba0335ca749ad88871bccefd02579 : SUCCESS in 7m 42s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/7ef248c6d8d4490abfc0d8c05361aadf : SUCCESS in 4m 01s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/9422dfbc819142a6830f3f3767458ec3 : SUCCESS in 6m 17s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/7382808ceb9f44b3bdda72e3f0980e12 : SUCCESS in 7m 21s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/90a7730e71a341239cab397f47221296 : SUCCESS in 9m 24s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/a9ba6d83d95e460697351ffa5f4dc96e : SUCCESS in 1h 45m 04s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/c52eaa0a8bcb49ad918d27a3b46aeee8 : SUCCESS in 1h 34m 51s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/d4d704108f8f4c12983d024dc2065272 : SUCCESS in 37m 15s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/61bafb1e2cf547cf9835b4d8e38e126a : SUCCESS in 48m 41s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/f2bb2bbb77e8452e9bee2a6bb4d3f1c9 : SUCCESS in 1h 14m 04s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/5ff688ce0f0044b994ca2cf0a0a0bdb0 : SUCCESS in 35m 45s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/a0a7ff211e464f8a93d878e24df124b0 : SUCCESS in 1h 27m 16s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/4f732478869b4980828c1e374f70d902 : SUCCESS in 2h 06m 18s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/333f1660281e4c899285167c0917c8e3 : SUCCESS in 1h 52m 27s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/85030d798f8640149f05f3d215e15882 : SUCCESS in 1h 46m 31s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/07a36adea72b423cbb4e87087c693e07 : SUCCESS in 1h 45m 42s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/94a46f26dcb8494882caa85df0035854 : FAILURE in 27m 14s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"4b7af647e1446d1d52f965e29e94fb8ee1efd369","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2022-10-27 16:04:08.000000000","message":"Patch Set 7:\n\n(3 comments)","accounts_in_message":[],"_revision_number":7},{"id":"510723ff8a58e50eba1b352138b211d901487d30","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-12-28 09:21:34.000000000","message":"Patch Set 8: Patch Set 7 was rebased","accounts_in_message":[],"_revision_number":8},{"id":"212ee7bc75b4f043cc65cfaf5ac65fb8ccc6dbac","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-28 12:19:01.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/f445638195294dbc8b09096dbdeeb1e5\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/dd628cce485b47ac92062b396facc7c6 : FAILURE in 3m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9cb5e1bd48e64cfabcb3c10eb6c69e26 : FAILURE in 3m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f1034fc41e534f2eb7def30c3485be71 : FAILURE in 3m 29s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/da3492c4be3347a3badc88c28cb6af2a : FAILURE in 3m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/25c79ad7b52d4366b7de1ced0fbf305e : FAILURE in 4m 05s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/923e693dc2144e75a9799e5ac59555ab : SUCCESS in 5m 33s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/9e870187cf8e46bc85ddd6427a134539 : FAILURE in 5m 13s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/f71799d6d9b84ed08d250b865fe8c331 : FAILURE in 5m 30s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/79b9d05932954c3ea4d7e88dda606eb7 : SUCCESS in 4m 56s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/a6ed2fe918e94ae997928a3a2f81765d : FAILURE in 5m 13s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/cb35db9c69b943b2861e3083149d341b : FAILURE in 5m 03s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/95e2a903bc0b4ab3b73b047b4cd76d2c : FAILURE in 5m 46s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/a50b7f70e1234a3fab5844bac8e70cda : SUCCESS in 1h 15m 26s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/8d08ae95d35a4932b8a355e0a1cee453 : SUCCESS in 2h 24m 56s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/c71e758e49d7409bb08ddbc418763c2c : SUCCESS in 41m 06s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/2a2db80deab242338ab15d95aba34d7b : SUCCESS in 1h 05m 26s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/810df79085e141699afe8bd757e6d9ab : FAILURE in 1h 24m 01s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/996690972f4d41a493bd260e1234ab64 : SUCCESS in 30m 59s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/2532e82ed44d4b098b2578aade787084 : SUCCESS in 2h 22m 42s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/afd66622c09c4396b39d13167a2f85fe : SUCCESS in 2h 37m 38s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/4ba420d723ba44218f8d8cfafbfba1e6 : SUCCESS in 2h 31m 01s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/977d81dffc374458b2645968c37583e3 : TIMED_OUT in 2h 34m 34s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/8fbcbec563074f9cabe31afc2251c4a9 : SUCCESS in 2h 20m 57s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/e49aab0040754e36b93c66f1584abb9b : SUCCESS in 2h 24m 54s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"3dd0df770a010d50b0ace3adb7405ece7b20ac3c","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2023-01-11 16:06:50.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"da1de4d57ea4aeaa66af714cb0cac676c00859b4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-11 18:25:00.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/821fc433f5e54863ab464e5b90a4c5a9\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f568082a293e4550acd4ac87658d3a4e : SUCCESS in 11m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f74a00a8db9a4b029f54eee736e8f11e : SUCCESS in 6m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/431262adb62e471e8f480e090cb48e4d : SUCCESS in 5m 26s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2eeb640c078d4859a07fa7a246395929 : SUCCESS in 4m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/76b4626c69b747f983cfcb9fd15503a0 : SUCCESS in 14m 05s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/424eaa2a368e4baeb54f97b39715f6c3 : SUCCESS in 5m 04s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/11ee30d3ab56425c83b146d085ef5645 : SUCCESS in 5m 30s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/570d256e6fc246d88c7ddbc5ac4dec66 : SUCCESS in 7m 54s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/ea0ee6c0c552428c8c6bca40ce2e3209 : SUCCESS in 4m 50s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/56e66e409f994bb7ba43eacd12314cb4 : SUCCESS in 7m 52s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/62a89a0a393444daaf7a039a62620fc9 : SUCCESS in 7m 48s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/7dd4d437aaf14de094e9a72069b1bcdf : SUCCESS in 7m 32s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/45c8555d3e8a45dea6d5bbffef643668 : SUCCESS in 1h 34m 53s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/6700beeb21ff424690a8b5c98f35ab7d : SUCCESS in 1h 49m 56s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/616abf11eedf47e2886f7dd4410edb12 : SUCCESS in 33m 00s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/b22415d9ef5f45ff85d3bf2ab2ce9bb8 : SUCCESS in 48m 42s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/b5728756ff6e4e99b5e531d0c371d3b6 : SUCCESS in 1h 14m 02s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/54d4049b60fd4bfe8e6271674675df18 : SUCCESS in 30m 04s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/7fc7482c7e2e4af19a2d5b3e4640af4e : SUCCESS in 1h 54m 59s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/d89054a720c34e59afe2d07bf64e1823 : SUCCESS in 2h 13m 46s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/c880a878856043aa8518ea3a158292be : SUCCESS in 2h 07m 17s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/ead3d7212ec54963958ba0326f35697b : SUCCESS in 2h 02m 35s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/fb7fbb3d609c48e1ac1f35a7418748b0 : SUCCESS in 1h 44m 24s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/30692cf2a4ff4cfbb55289c558624acb : SUCCESS in 2h 10m 21s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"bd0dcd0d63d3d7b60a29f35308a1b9e8447363b3","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-01-11 20:55:22.000000000","message":"Patch Set 8: Code-Review-1\n\n(4 comments)","accounts_in_message":[],"_revision_number":8},{"id":"5ad49dd70fe7e58af769a2f56b2c74ff35ddcd76","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2023-01-12 10:21:47.000000000","message":"Patch Set 9: Patch Set 8 was rebased","accounts_in_message":[],"_revision_number":9},{"id":"c7f40a11a20a86e7d5689241604b8ec9b35a1adb","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2023-01-12 10:24:32.000000000","message":"Patch Set 9: Backport-Candidate-2\n\n(3 comments)","accounts_in_message":[],"_revision_number":9},{"id":"5bf9910e6ca3cb569f9b395804b87e365c49132d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2023-01-12 10:24:51.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"9b4925640bdd498552d1ec19811ec78d63d30a58","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2023-01-12 12:22:15.000000000","message":"Patch Set 10: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"08910ff6a209d11c789c8bf0f0e9ce4d41f6ab30","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-12 12:37:30.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8fcdedc251f445f196a2bb0380e6549d\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e4d65bf4bbcf464b9ac439766f3bc5c8 : SUCCESS in 9m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8c43a6629e3f4fe28ab56263d97c78c8 : SUCCESS in 5m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/66362657cb9046edbec0ddf60717a955 : SUCCESS in 4m 53s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6824d006db5d4eab9678e27da8be90d1 : SUCCESS in 4m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7d4a578153b745728d67d99edfc7f495 : SUCCESS in 11m 31s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f0916ee44dbd455298841485435dba4d : SUCCESS in 4m 21s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/41cc56a96c7b4a78a77ec68f352067a7 : SUCCESS in 5m 01s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/4b7adf1c2f68456b93d3fd39e8754207 : SUCCESS in 7m 56s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/64a379b21e1746d5bd32a0d6e399de29 : SUCCESS in 4m 53s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/376d3575e0a8444dbda76cb33e657b4b : SUCCESS in 7m 41s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/3b475d282cc640a0a5ec614695bd4d3d : SUCCESS in 6m 37s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/2311f3b4901a4cb989b552ffaf81ed7b : SUCCESS in 7m 50s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/a627033ce4cc4e8f9edde21a30509506 : SUCCESS in 1h 47m 41s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/930c5b90becf4840955eb3e5ba6adaae : SUCCESS in 1h 50m 33s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/ff0fdee97492417e84a22803732d8c14 : SUCCESS in 35m 41s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/1460266e32da480fb99bc7e9c49827aa : SUCCESS in 52m 54s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/d849cdcea28249b2bb158047fa8e0167 : SUCCESS in 1h 16m 10s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/3ad8e409ae1842269ef6db2e4eb1dbc7 : SUCCESS in 32m 19s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/060b9da138444f2c9b914610da4afee4 : SUCCESS in 1h 44m 14s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/d27e114085524c3abc0879e5fb355a72 : SUCCESS in 1h 56m 14s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/494f4e621ca145cea3cf980e1f894dc4 : SUCCESS in 1h 48m 55s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/bddad79b02ca40609d8cd817f722851d : SUCCESS in 1h 41m 18s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/97b78b0aecc54d19b7bc45add5bcddca : SUCCESS in 1h 56m 40s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/2107ebd5dd234115a6342a0bf41d84d8 : SUCCESS in 2h 10m 02s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"38a428e336d35e396cb75103a50e2afd2cae8ed8","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-01-12 15:39:23.000000000","message":"Patch Set 10: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"4d668f6d456dbe6ce95b8243960dce3954344f12","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-01-13 07:57:57.000000000","message":"Patch Set 10: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"d5d567bcd8f39e8fa8916364a13888387a56823f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2023-01-13 08:05:37.000000000","message":"Patch Set 11: Patch Set 10 was rebased","accounts_in_message":[],"_revision_number":11},{"id":"013a0a4d94160fd3df1d328b4be86039c2ec759b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2023-01-13 08:10:09.000000000","message":"Uploaded patch set 12.","accounts_in_message":[],"_revision_number":12},{"id":"a1195ae7adfee2d965ce5010a8baf8d46a8fa239","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2023-01-13 08:15:41.000000000","message":"Patch Set 12: Backport-Candidate-2\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"33f9d32f7f3fbae49c81596b93b84b64760e6653","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-01-13 09:15:58.000000000","message":"Patch Set 12: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"4132413c62aaa373cb4ea061ceaa7e57ffdd177a","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2023-01-13 10:52:16.000000000","message":"Patch Set 12:\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"82815cf362e1fea9be41a3c1962b384d8b635c09","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-13 11:03:18.000000000","message":"Patch Set 12: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/f69efd76fdcf4723b790cb1094ea299b\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/790dce7473ee48509e2690bf7ccd0e60 : SUCCESS in 7m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d0b4c345e896435daf494547f4f11ed9 : SUCCESS in 4m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3fdb69dbf95d41dfa6a876b58ab142b4 : SUCCESS in 4m 40s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/38cc71a68c264581af0e5cef0c6c565d : SUCCESS in 5m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dd036dd6dc8c4d52b7190c28e5d29c14 : SUCCESS in 10m 05s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/25bf8edb883445a9867f00bf3af118ac : SUCCESS in 5m 00s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/bcdaec0bd7c641a59ff7642fa5f7acc6 : SUCCESS in 5m 06s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/f3ebaf4699ee48c786beadfe10244e9a : SUCCESS in 8m 00s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/e10e64220e0f4e8eab4d1916ab13125a : SUCCESS in 4m 32s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/8053062208c4497aa9a62a0db7c0d189 : SUCCESS in 7m 14s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/ed63a9bc16274e0aab0346c7bf9bdc25 : SUCCESS in 8m 19s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/8a1b5f6acb1a4c43bbaaef87b066b466 : SUCCESS in 5m 42s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/247ac5b413664436ae022b2ad5826433 : SUCCESS in 1h 35m 39s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/5b6b7aea93cd492f925e8c1b1f565d1c : SUCCESS in 2h 20m 36s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/927e9db586b8480c9fa1493ffddc9359 : SUCCESS in 35m 03s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/7d85c639212a479ba3dfebd29c1b2f1c : FAILURE in 2h 48m 49s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/183bc067a6dc4998ab4c74d78ff1469e : SUCCESS in 1h 21m 04s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/160c790f2d2f44868ff3dc9545c475e8 : SUCCESS in 42m 54s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/e3eab29244634108bd74fc399b066e28 : SUCCESS in 1h 59m 41s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/85d63c77d36743df8e8475dc4ef4f33e : FAILURE in 2h 36m 00s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/2297edfa84b64e688fb10ec6b45efd1f : SUCCESS in 2h 01m 04s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/830b3d1af74b48b29651a16486c9e84e : SUCCESS in 2h 27m 32s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/757f8e07939f4a3c8dfdad5b3e9af3b6 : SUCCESS in 2h 14m 28s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/aca2b62182f0493b86b8205b5ff3abee : SUCCESS in 2h 10m 58s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"431003b9bd1b18c4056affd3ef477b1b033250c7","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2023-01-13 11:22:43.000000000","message":"Patch Set 12:\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"5539d52c4837e64fdf5529483aa1ed96cc84a4c8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-13 14:03:11.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1271369d0ccb4cf1a9e3e0be5a6f57b0\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0bbfc858528f483dae25c80f8dcfc952 : SUCCESS in 10m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ed9695e75aa74105a35ea8c91a153a0f : SUCCESS in 5m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/de3c1d290c964a949ecd1d9e275cb604 : SUCCESS in 4m 19s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7b43c7506c084114bb1434183e02727a : SUCCESS in 4m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/63a86a79c5d94f7d9b9e5d12d5ab8ee9 : SUCCESS in 10m 31s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5cedcc16bff8444ea85bdf7caa57183a : SUCCESS in 5m 46s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/041944fca56247428b0b4fb9bbde27ef : SUCCESS in 5m 00s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/b3456d09016944f59d267d306dfd8146 : SUCCESS in 8m 37s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/7b254bae383c486d8c839a1736d89e2c : SUCCESS in 4m 41s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/acadf11088ad45d990d89d1a5818bedf : SUCCESS in 6m 50s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/05a27d539b284b46949395bd5e2fcc88 : SUCCESS in 7m 15s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/99b2fe77fbd24c6ca842f661c87acc9f : SUCCESS in 6m 28s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/49e631ef70434df7bbfbb51d5c9dc425 : SUCCESS in 1h 21m 25s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/8b3d6ae9441e4987affb61a8e18bab73 : SUCCESS in 1h 51m 47s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/68667ea027b24e7cbc5f2bf461bf7243 : SUCCESS in 44m 21s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/921ef381bca94746995fa14330b760b6 : SUCCESS in 1h 32m 29s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/44007d97f2e342cf9a056a3e7eceee74 : SUCCESS in 1h 23m 06s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/bf46056acef94b149bb20289bbde86d6 : SUCCESS in 37m 55s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/0f50206e0d094edab0a144a7a26ee0c3 : SUCCESS in 1h 56m 47s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/be485c6aef9345348e7f5437ed9cc047 : FAILURE in 2h 38m 52s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/c5c33fb6af344a1aaf53270ca85d2557 : SUCCESS in 2h 32m 48s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/20bb9d7d97c146629a9f612b86352763 : SUCCESS in 2h 07m 57s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/896034da7b7a4ec090a6911b915d2e19 : SUCCESS in 2h 04m 31s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/868b2366be954d51a2b72e02879da78a : SUCCESS in 2h 34m 21s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"bdf4d3c7b17931ffbd091defa22fad1033e7625e","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2023-01-16 10:14:07.000000000","message":"Patch Set 12: Code-Review+1","accounts_in_message":[],"_revision_number":12},{"id":"e9bd2e786732ea984c5c773cbacd2dd117daa131","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-02-02 16:16:45.000000000","message":"Patch Set 12: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"4c5d04341efd6bfa8ad57ba31590bba018f8d764","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-02 16:17:32.000000000","message":"Patch Set 12: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":12},{"id":"56839794ba5e597f2e83ce923d8c410c6b138b87","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-02 16:23:29.000000000","message":"Patch Set 12: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/9297133680ce4679a7edd1b66d0ad2fc\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/91638ed2aed446cf8365991c2477efb2 : FAILURE in 5m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ee5c2692c6e84911a611e3334d0c250e : CANCELED\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ad1f438d4598498395d59396e6628f88 : SUCCESS in 4m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2cead7e617d24f3ca26d6d24afed6216 : CANCELED\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a4472dca39684ac790b097c205d7c57b : SUCCESS in 5m 02s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/11d96aba35a84e668a6425d41bbae04c : CANCELED\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/40bfc75ecdd440f48b2268a2040a9e64 : CANCELED\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/adfd4cfaa20b4d6ebc871693899de4c2 : CANCELED\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/e70577fff9ef43498bb5ea313c5aa73f : CANCELED\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/7cff3471257a4d2dbbf24714c3506aa6 : CANCELED\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/a93de0612bc74e82b41b6958e47cf81f : CANCELED\n\nWarning:\n  Comments left for invalid file octavia/amphorae/backends/agent/api_server/amphora_info.py\n  Comments left for invalid file octavia/amphorae/backends/agent/api_server/keepalived.py\n  Comments left for invalid file octavia/amphorae/backends/agent/api_server/keepalivedlvs.py\n  Comments left for invalid file octavia/amphorae/backends/agent/api_server/loadbalancer.py\n  Comments left for invalid file octavia/amphorae/backends/agent/api_server/osutils.py\n  Comments left for invalid file octavia/amphorae/backends/agent/api_server/plug.py\n  Comments left for invalid file octavia/amphorae/backends/agent/api_server/server.py\n  Comments left for invalid file octavia/amphorae/backends/agent/api_server/util.py\n  Comments left for invalid file octavia/amphorae/backends/utils/haproxy_query.py\n  Comments left for invalid file octavia/amphorae/drivers/health/heartbeat_udp.py\n  Comments left for invalid file octavia/api/common/types.py\n  Comments left for invalid file octavia/api/drivers/amphora_driver/v1/driver.py\n  Comments left for invalid file octavia/api/v2/controllers/health_monitor.py\n  Comments left for invalid file octavia/api/v2/controllers/listener.py\n  Comments left for invalid file octavia/common/validate.py\n  Comments left for invalid file octavia/controller/worker/v1/flows/load_balancer_flows.py\n  Comments left for invalid file octavia/controller/worker/v1/tasks/database_tasks.py\n  Comments left for invalid file octavia/controller/worker/v2/flows/load_balancer_flows.py\n  Comments left for invalid file octavia/controller/worker/v2/tasks/database_tasks.py\n  Comments left for invalid file octavia/distributor/drivers/noop_driver/driver.py","accounts_in_message":[],"_revision_number":12},{"id":"51273b81cb4b0629f43654f03ddcc69e0845a103","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2023-02-02 16:44:03.000000000","message":"Patch Set 13: Patch Set 12 was rebased","accounts_in_message":[],"_revision_number":13},{"id":"e1149b8dccf10e49df6c51013ed64989fb52e85c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-02 19:22:12.000000000","message":"Patch Set 13: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/84de120ea1f549d19a8642ede1392793\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/71b70d2c8ae44d8fae6b3a5a27ce15b7 : SUCCESS in 11m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a4e2ce7374044f9fa798d168129759f6 : FAILURE in 5m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5dff8fae3f2f451bacc2d0ebb410edbf : SUCCESS in 5m 08s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d5319f467c3c4804b6502a2ae7cd895c : SUCCESS in 5m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fb61bd04112e49dca3fa93aea5231064 : SUCCESS in 12m 27s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/aa3346a8685d487b8f18f1cf5609b57b : SUCCESS in 4m 25s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/a3970aab190e4a06a5e13043ed0e7862 : SUCCESS in 5m 28s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/b5fbadafbf1343809411a9d3ccb984b7 : SUCCESS in 8m 38s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/49ff239b24fb4df09ab1e29fc329e44b : SUCCESS in 5m 39s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/3dbe8019b6124764be483d75e9d42f8b : SUCCESS in 7m 12s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/56441de202c1463ca07583de475e8175 : SUCCESS in 7m 41s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/4ddf3d6ffbc7400b81c19f65eabe6651 : SUCCESS in 10m 59s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/05b91742f7744766b5ae5e14e88d7918 : SUCCESS in 1h 19m 05s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/711a0907434942dc94d6ab1643dab56f : TIMED_OUT in 2h 33m 47s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/6fcc152b76994ad3bc0a624141902021 : SUCCESS in 45m 58s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/1d5bfeed522f4845b5658ed4e1987c3f : SUCCESS in 53m 04s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/0290c1e815824abebcbe7bc9436aa878 : SUCCESS in 1h 15m 31s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/9779cce6aa32409390bca116e0090ebb : SUCCESS in 32m 55s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/9cbe527facac4315946db46e4d95a3f2 : SUCCESS in 2h 29m 09s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/d5466e8df6a84f2e9ee64f4f7322470a : SUCCESS in 2h 20m 26s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/e63471d3205f4fb7b0307308c27d7a09 : TIMED_OUT in 2h 34m 20s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/d271680e3850441ba011c80eb69bbec0 : SUCCESS in 2h 29m 25s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/4be7758012c44e17ac8e825cb4e4fc1f : SUCCESS in 1h 48m 16s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/8e1ca7ff751e4f869f7ee27014e2d031 : SUCCESS in 2h 19m 24s (non-voting)\n\nWarning:\n  Comments left for invalid file octavia/amphorae/backends/agent/api_server/amphora_info.py\n  Comments left for invalid file octavia/amphorae/backends/agent/api_server/keepalived.py\n  Comments left for invalid file octavia/amphorae/backends/agent/api_server/keepalivedlvs.py\n  Comments left for invalid file octavia/amphorae/backends/agent/api_server/loadbalancer.py\n  Comments left for invalid file octavia/amphorae/backends/agent/api_server/osutils.py\n  Comments left for invalid file octavia/amphorae/backends/agent/api_server/plug.py\n  Comments left for invalid file octavia/amphorae/backends/agent/api_server/server.py\n  Comments left for invalid file octavia/amphorae/backends/agent/api_server/util.py\n  Comments left for invalid file octavia/amphorae/backends/utils/haproxy_query.py\n  Comments left for invalid file octavia/amphorae/drivers/health/heartbeat_udp.py\n  Comments left for invalid file octavia/api/common/types.py\n  Comments left for invalid file octavia/api/drivers/amphora_driver/v1/driver.py\n  Comments left for invalid file octavia/api/v2/controllers/health_monitor.py\n  Comments left for invalid file octavia/api/v2/controllers/listener.py\n  Comments left for invalid file octavia/common/validate.py\n  Comments left for invalid file octavia/controller/worker/v1/flows/load_balancer_flows.py\n  Comments left for invalid file octavia/controller/worker/v1/tasks/database_tasks.py\n  Comments left for invalid file octavia/controller/worker/v2/flows/load_balancer_flows.py\n  Comments left for invalid file octavia/controller/worker/v2/tasks/database_tasks.py\n  Comments left for invalid file octavia/distributor/drivers/noop_driver/driver.py","accounts_in_message":[],"_revision_number":13},{"id":"2a0df57f1ee4ab11d57525abdd88a0eee7b9bdf4","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-02-03 17:29:17.000000000","message":"Patch Set 13: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":13},{"id":"ec7ef3492fccfde8604f78ac229a843fcfd69fe8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-03 19:54:34.000000000","message":"Patch Set 13: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d2bf2116342f41feb92f5601470fe835\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d18d12ade8d94a65bc1bbb29a108aa2e : SUCCESS in 13m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/340d4ed6b9e24da6ab17f73e635634ac : SUCCESS in 6m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d4b9abf6729549b5b49c6c5a34ff69e7 : SUCCESS in 4m 16s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/cd4284eaabaa4a8f8cf150b94228418f : SUCCESS in 5m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4287d5636d35413dae4e85946cf8484b : SUCCESS in 11m 06s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/eefc1240317243958c69c02252905b56 : SUCCESS in 4m 22s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/9a6c71e2a7ed47fe8f35717904de5e2d : SUCCESS in 5m 10s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/4e22ec8cc5e34b8bbf296cf25569a1ec : SUCCESS in 7m 20s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/e08dccd622694715b79d5fcaf259e06d : SUCCESS in 6m 18s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/dce6b755ac2a4741b70c61927c805e41 : SUCCESS in 6m 38s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/5cdc449df9b94439a94f796a99ff8048 : SUCCESS in 7m 20s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/978b54faeb6846f68f184b0d2f641d2a : SUCCESS in 6m 57s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/1a00a2f7deee472b8debec695dfe3144 : SUCCESS in 1h 11m 12s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/8055ca2ae3714dca9b62730234a68ea1 : SUCCESS in 1h 30m 16s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/f0b6757607e04aadac40d7891391942b : SUCCESS in 32m 49s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/a6aff3fae5cc417fb14f60810e80e3c8 : SUCCESS in 1h 02m 21s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/7f034156dc9f46858b656b03892343df : SUCCESS in 1h 09m 32s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/8a213ac0aed54355a81776870e80796f : SUCCESS in 33m 21s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/3911e203e3f8470b81461a3e01f9cb2c : SUCCESS in 1h 42m 34s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/e50cc37152ca4b038c597825771ac6cb : SUCCESS in 2h 18m 04s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/187f8dd1a56042a1bd0c5bda5e2f1afa : SUCCESS in 2h 14m 03s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/bfff18b0cd8741fd82c7d322fb5a4f81 : SUCCESS in 1h 33m 50s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/a00c326a23b946e7905fce01807ef6ef : SUCCESS in 2h 09m 28s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/7c51d2880c3e4f83b0618ab965ece439 : SUCCESS in 2h 15m 53s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"cd70d4bf728af2d1f6909b241d3311ba82362cbc","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-02-04 06:34:56.000000000","message":"Patch Set 13: Workflow+1","accounts_in_message":[],"_revision_number":13},{"id":"c80b033d8cdc6cc0caa82eae4936229366ef0cae","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-04 06:35:44.000000000","message":"Patch Set 13: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":13},{"id":"33121ef2f1b17c8f87c004afb186c6bb0395691f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-04 08:39:24.000000000","message":"Patch Set 13: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d4fcb11a0589425689ad2850156351f7\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/28bbd90cfc044a48bcf7a61dff0e8177 : SUCCESS in 4m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/53216236f49f454f9311635700db95ff : SUCCESS in 4m 18s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6f1afb2546764042ac0222d45b733b0b : SUCCESS in 4m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/adb98b68626b432b899ddf03cca28387 : SUCCESS in 10m 24s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/308936414a544cfa91af46cdd981e46e : SUCCESS in 4m 22s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/f7b3a94f738946e5aae0937615f1c9a9 : SUCCESS in 5m 03s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/0d775b8f1ee14781957b399b4480b1be : SUCCESS in 6m 28s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/a2e0cac2c18e40339dd4a0369cb1d895 : SUCCESS in 58m 45s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/ea369c2fd3d94391beb183f25dfe60d9 : SUCCESS in 2h 02m 36s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/3e4e499332a54a869534581f4607e39b : SUCCESS in 45m 17s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/a22c8e8c6a96454b87c6f1f39c503aee : SUCCESS in 53m 15s","accounts_in_message":[],"_revision_number":13},{"id":"83078f2183ada6cc57da9138bcc877a9b7f6985d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-04 08:39:25.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":13},{"id":"2a80371fedf29a8113a092215cb0d1a33e4f1d93","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-04 08:40:37.000000000","message":"Patch Set 13:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/eab925f920ed4abc879e9cd02e7e4feb\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/86e193aace8647fea7083dfc344c358f : SUCCESS in 52s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/00816f6ac958458d9651670e03725751 : SUCCESS in 44s","accounts_in_message":[],"_revision_number":13}],"current_revision_number":13,"current_revision":"7a2df883a09d2f9877c640a77914c2f2b62f86e7","revisions":{"b190c825d8958594d096a029d87b24853d1bd166":{"kind":"REWORK","_number":1,"created":"2022-10-25 11:41:17.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/99/862599/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/99/862599/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/99/862599/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/99/862599/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/99/862599/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/99/862599/1"}}},"commit":{"parents":[{"commit":"344d6d76ce4481f30fe295114551243d61e7eea8","subject":"Fix prometheus-proxy service name in Red Hat-based distros","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/344d6d76ce4481f30fe295114551243d61e7eea8"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-10-25 11:39:36.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-10-25 11:39:36.000000000","tz":120},"subject":"Remove support for split listener configuration","message":"Remove support for split listener configuration\n\nRemove support for Amphora API 0.5 or earlier.\nDuring the PTG for the Antelope cycle it has been decided\nto remove split listener configuration.[1]\n\n[1]: https://etherpad.opendev.org/p/antelope-ptg-octavia\n\nChange-Id: Ia0c1d4107161fff8d1f3071d57860bfd265f596f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b190c825d8958594d096a029d87b24853d1bd166"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b190c825d8958594d096a029d87b24853d1bd166"}]},"branch":"refs/heads/master"},"c6bce0c5512a66927bf7d25c8bde33540cf4b10f":{"kind":"REWORK","_number":2,"created":"2022-10-25 14:37:52.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/99/862599/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/99/862599/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/99/862599/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/99/862599/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/99/862599/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/99/862599/2"}}},"commit":{"parents":[{"commit":"344d6d76ce4481f30fe295114551243d61e7eea8","subject":"Fix prometheus-proxy service name in Red Hat-based distros","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/344d6d76ce4481f30fe295114551243d61e7eea8"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-10-25 11:39:36.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-10-25 14:35:46.000000000","tz":120},"subject":"Remove support for split listener configuration","message":"Remove support for split listener configuration\n\nRemove support for Amphora API 0.5 or earlier.\nDuring the PTG for the Antelope cycle it has been decided\nto remove split listener configuration.[1]\n\n[1]: https://etherpad.opendev.org/p/antelope-ptg-octavia\n\nChange-Id: Ia0c1d4107161fff8d1f3071d57860bfd265f596f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c6bce0c5512a66927bf7d25c8bde33540cf4b10f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c6bce0c5512a66927bf7d25c8bde33540cf4b10f"}]},"branch":"refs/heads/master"},"de48bce5cf3eaa56f2e7643179243eb69ed0c567":{"kind":"REWORK","_number":3,"created":"2022-10-25 14:39:40.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/99/862599/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/99/862599/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/99/862599/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/99/862599/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/99/862599/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/99/862599/3"}}},"commit":{"parents":[{"commit":"344d6d76ce4481f30fe295114551243d61e7eea8","subject":"Fix prometheus-proxy service name in Red Hat-based distros","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/344d6d76ce4481f30fe295114551243d61e7eea8"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-10-25 11:39:36.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-10-25 14:39:31.000000000","tz":120},"subject":"Remove support for split listener configuration","message":"Remove support for split listener configuration\n\nRemove support for Amphora API 0.5 or earlier.\nDuring the PTG for the Antelope cycle it has been decided\nto remove split listener configuration.[1]\n\n[1]: https://etherpad.opendev.org/p/antelope-ptg-octavia\n\nChange-Id: Ia0c1d4107161fff8d1f3071d57860bfd265f596f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/de48bce5cf3eaa56f2e7643179243eb69ed0c567"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/de48bce5cf3eaa56f2e7643179243eb69ed0c567"}]},"branch":"refs/heads/master"},"558ccdc0d48dfde0710abef16a0258a326c3bcce":{"kind":"REWORK","_number":4,"created":"2022-10-25 14:57:57.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/99/862599/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/99/862599/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/99/862599/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/99/862599/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/99/862599/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/99/862599/4"}}},"commit":{"parents":[{"commit":"344d6d76ce4481f30fe295114551243d61e7eea8","subject":"Fix prometheus-proxy service name in Red Hat-based distros","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/344d6d76ce4481f30fe295114551243d61e7eea8"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-10-25 11:39:36.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-10-25 14:57:49.000000000","tz":120},"subject":"Remove support for split listener configuration","message":"Remove support for split listener configuration\n\nRemove support for Amphora API 0.5 or earlier.\nDuring the PTG for the Antelope cycle it has been decided\nto remove split listener configuration.[1]\n\n[1]: https://etherpad.opendev.org/p/antelope-ptg-octavia\n\nChange-Id: Ia0c1d4107161fff8d1f3071d57860bfd265f596f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/558ccdc0d48dfde0710abef16a0258a326c3bcce"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/558ccdc0d48dfde0710abef16a0258a326c3bcce"}]},"branch":"refs/heads/master"},"b45f58ac25367b4d115b02a14b2526ba96a861da":{"kind":"REWORK","_number":5,"created":"2022-10-27 09:22:56.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/99/862599/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/99/862599/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/99/862599/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/99/862599/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/99/862599/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/99/862599/5"}}},"commit":{"parents":[{"commit":"344d6d76ce4481f30fe295114551243d61e7eea8","subject":"Fix prometheus-proxy service name in Red Hat-based distros","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/344d6d76ce4481f30fe295114551243d61e7eea8"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-10-25 11:39:36.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-10-27 09:22:47.000000000","tz":120},"subject":"Remove support for split listener configuration","message":"Remove support for split listener configuration\n\nRemove support for Amphora API 0.5 or earlier.\nDuring the PTG for the Antelope cycle it has been decided\nto remove split listener configuration.[1]\n\n[1]: https://etherpad.opendev.org/p/antelope-ptg-octavia\n\nChange-Id: Ia0c1d4107161fff8d1f3071d57860bfd265f596f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b45f58ac25367b4d115b02a14b2526ba96a861da"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b45f58ac25367b4d115b02a14b2526ba96a861da"}]},"branch":"refs/heads/master"},"f6c4045d58985941f85911c8f95c0e777254ba3f":{"kind":"REWORK","_number":6,"created":"2022-10-27 12:47:59.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/99/862599/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/99/862599/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/99/862599/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/99/862599/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/99/862599/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/99/862599/6"}}},"commit":{"parents":[{"commit":"344d6d76ce4481f30fe295114551243d61e7eea8","subject":"Fix prometheus-proxy service name in Red Hat-based distros","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/344d6d76ce4481f30fe295114551243d61e7eea8"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-10-25 11:39:36.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-10-27 12:47:49.000000000","tz":120},"subject":"Remove support for split listener configuration","message":"Remove support for split listener configuration\n\nRemove support for Amphora API 0.5 or earlier.\nDuring the PTG for the Antelope cycle it has been decided\nto remove split listener configuration.[1]\n\n[1]: https://etherpad.opendev.org/p/antelope-ptg-octavia\n\nChange-Id: Ia0c1d4107161fff8d1f3071d57860bfd265f596f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/f6c4045d58985941f85911c8f95c0e777254ba3f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/f6c4045d58985941f85911c8f95c0e777254ba3f"}]},"branch":"refs/heads/master"},"99443ad6b304ef4f611b2221cdeb89c0763fbde0":{"kind":"REWORK","_number":7,"created":"2022-10-27 12:59:44.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/99/862599/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/99/862599/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/99/862599/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/99/862599/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/99/862599/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/99/862599/7"}}},"commit":{"parents":[{"commit":"344d6d76ce4481f30fe295114551243d61e7eea8","subject":"Fix prometheus-proxy service name in Red Hat-based distros","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/344d6d76ce4481f30fe295114551243d61e7eea8"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-10-25 11:39:36.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-10-27 12:58:33.000000000","tz":120},"subject":"Remove support for split listener configuration","message":"Remove support for split listener configuration\n\nRemove support for Amphora API 0.5 or earlier.\nDuring the PTG for the Antelope cycle it has been decided\nto remove split listener configuration.[1]\n\n[1]: https://etherpad.opendev.org/p/antelope-ptg-octavia\n\nChange-Id: Ia0c1d4107161fff8d1f3071d57860bfd265f596f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/99443ad6b304ef4f611b2221cdeb89c0763fbde0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/99443ad6b304ef4f611b2221cdeb89c0763fbde0"}]},"branch":"refs/heads/master"},"6826e8b7481a9b83ec928bc481990c5a0dcd9176":{"kind":"TRIVIAL_REBASE","_number":8,"created":"2022-12-28 09:21:34.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/99/862599/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/99/862599/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/99/862599/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/99/862599/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/99/862599/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/99/862599/8"}}},"commit":{"parents":[{"commit":"ac3618b140b1b8ba02f2263abf910c17b3f08aff","subject":"Merge \"Fix typo in docs section title\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/ac3618b140b1b8ba02f2263abf910c17b3f08aff"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-10-25 11:39:36.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-12-28 09:21:34.000000000","tz":0},"subject":"Remove support for split listener configuration","message":"Remove support for split listener configuration\n\nRemove support for Amphora API 0.5 or earlier.\nDuring the PTG for the Antelope cycle it has been decided\nto remove split listener configuration.[1]\n\n[1]: https://etherpad.opendev.org/p/antelope-ptg-octavia\n\nChange-Id: Ia0c1d4107161fff8d1f3071d57860bfd265f596f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/6826e8b7481a9b83ec928bc481990c5a0dcd9176"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/6826e8b7481a9b83ec928bc481990c5a0dcd9176"}]},"branch":"refs/heads/master","description":"Rebase"},"2c408c99a8b87fc4aeaaab8b1ab2d2ca19138c8b":{"kind":"TRIVIAL_REBASE","_number":9,"created":"2023-01-12 10:21:47.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/99/862599/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/99/862599/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/99/862599/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/99/862599/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/99/862599/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/99/862599/9"}}},"commit":{"parents":[{"commit":"250b54aa15cb6e5d028ad83f1fa55607445476db","subject":"Merge \"Add cpu-pinning element\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/250b54aa15cb6e5d028ad83f1fa55607445476db"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-10-25 11:39:36.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2023-01-12 10:21:47.000000000","tz":0},"subject":"Remove support for split listener configuration","message":"Remove support for split listener configuration\n\nRemove support for Amphora API 0.5 or earlier.\nDuring the PTG for the Antelope cycle it has been decided\nto remove split listener configuration.[1]\n\n[1]: https://etherpad.opendev.org/p/antelope-ptg-octavia\n\nChange-Id: Ia0c1d4107161fff8d1f3071d57860bfd265f596f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/2c408c99a8b87fc4aeaaab8b1ab2d2ca19138c8b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/2c408c99a8b87fc4aeaaab8b1ab2d2ca19138c8b"}]},"branch":"refs/heads/master","description":"Rebase"},"1820a8a2c796d3aab58fb3ad31b16df5497e17c7":{"kind":"REWORK","_number":10,"created":"2023-01-12 10:24:51.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/99/862599/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/99/862599/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/99/862599/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/99/862599/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/99/862599/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/99/862599/10"}}},"commit":{"parents":[{"commit":"250b54aa15cb6e5d028ad83f1fa55607445476db","subject":"Merge \"Add cpu-pinning element\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/250b54aa15cb6e5d028ad83f1fa55607445476db"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-10-25 11:39:36.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2023-01-12 10:24:07.000000000","tz":60},"subject":"Remove support for split listener configuration","message":"Remove support for split listener configuration\n\nRemove support for Amphora API 0.5 or earlier.\nDuring the PTG for the Antelope cycle it has been decided\nto remove split listener configuration.[1]\n\n[1]: https://etherpad.opendev.org/p/antelope-ptg-octavia\n\nChange-Id: Ia0c1d4107161fff8d1f3071d57860bfd265f596f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/1820a8a2c796d3aab58fb3ad31b16df5497e17c7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/1820a8a2c796d3aab58fb3ad31b16df5497e17c7"}]},"branch":"refs/heads/master"},"d748c39fcf663f70f0686eec9172efd7062e3a4f":{"kind":"TRIVIAL_REBASE","_number":11,"created":"2023-01-13 08:05:37.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/99/862599/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/99/862599/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/99/862599/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/99/862599/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/99/862599/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/99/862599/11"}}},"commit":{"parents":[{"commit":"eea6ac62fd29616ffc6a8c6b5c8d1e9ca0efb734","subject":"Merge \"Optimize HAProxy SSL cache size\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/eea6ac62fd29616ffc6a8c6b5c8d1e9ca0efb734"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-10-25 11:39:36.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2023-01-13 08:05:37.000000000","tz":0},"subject":"Remove support for split listener configuration","message":"Remove support for split listener configuration\n\nRemove support for Amphora API 0.5 or earlier.\nDuring the PTG for the Antelope cycle it has been decided\nto remove split listener configuration.[1]\n\n[1]: https://etherpad.opendev.org/p/antelope-ptg-octavia\n\nChange-Id: Ia0c1d4107161fff8d1f3071d57860bfd265f596f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d748c39fcf663f70f0686eec9172efd7062e3a4f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d748c39fcf663f70f0686eec9172efd7062e3a4f"}]},"branch":"refs/heads/master","description":"Rebase"},"9a6eb74648b79d55c4e94629d106642dd18943d5":{"kind":"REWORK","_number":12,"created":"2023-01-13 08:10:09.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/99/862599/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/99/862599/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/99/862599/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/99/862599/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/99/862599/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/99/862599/12"}}},"commit":{"parents":[{"commit":"eea6ac62fd29616ffc6a8c6b5c8d1e9ca0efb734","subject":"Merge \"Optimize HAProxy SSL cache size\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/eea6ac62fd29616ffc6a8c6b5c8d1e9ca0efb734"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-10-25 11:39:36.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2023-01-13 08:06:28.000000000","tz":60},"subject":"Remove support for split listener configuration","message":"Remove support for split listener configuration\n\nRemove support for Amphora API 0.5 or earlier.\nDuring the PTG for the Antelope cycle it has been decided\nto remove split listener configuration.[1]\n\n[1]: https://etherpad.opendev.org/p/antelope-ptg-octavia\n\nChange-Id: Ia0c1d4107161fff8d1f3071d57860bfd265f596f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/9a6eb74648b79d55c4e94629d106642dd18943d5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/9a6eb74648b79d55c4e94629d106642dd18943d5"}]},"branch":"refs/heads/master"},"7a2df883a09d2f9877c640a77914c2f2b62f86e7":{"kind":"TRIVIAL_REBASE","_number":13,"created":"2023-02-02 16:44:03.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/99/862599/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/99/862599/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/99/862599/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/99/862599/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/99/862599/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/99/862599/13"}}},"commit":{"parents":[{"commit":"b687bd0a4cee6fa302d97f1c1895420fcd7ff69f","subject":"Merge \"Filter out details from taskflow logs with v2+jobboard\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b687bd0a4cee6fa302d97f1c1895420fcd7ff69f"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-10-25 11:39:36.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2023-02-02 16:44:03.000000000","tz":0},"subject":"Remove support for split listener configuration","message":"Remove support for split listener configuration\n\nRemove support for Amphora API 0.5 or earlier.\nDuring the PTG for the Antelope cycle it has been decided\nto remove split listener configuration.[1]\n\n[1]: https://etherpad.opendev.org/p/antelope-ptg-octavia\n\nChange-Id: Ia0c1d4107161fff8d1f3071d57860bfd265f596f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/7a2df883a09d2f9877c640a77914c2f2b62f86e7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/7a2df883a09d2f9877c640a77914c2f2b62f86e7"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}}]}],"submit_requirements":[]}
