)]}'
{"id":"openstack%2Foctavia~862600","triplet_id":"openstack%2Foctavia~stable%2Fwallaby~Id77ce36f59b71d309f153e5c1d44059f162ee440","project":"openstack/octavia","branch":"stable/wallaby","attention_set":{},"removed_from_attention_set":{"29244":{"account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"last_update":"2022-11-16 19:30:17.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Id77ce36f59b71d309f153e5c1d44059f162ee440","subject":"Fix barbican client with application credentials/trusts","status":"MERGED","created":"2022-10-25 11:50:18.000000000","updated":"2022-11-16 19:31:27.000000000","submitted":"2022-11-16 19:30:17.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"862600","meta_rev_id":"bedc285696ca6e7438126e60c6b8e4d9379f3054","_number":862600,"virtual_id_number":862600,"owner":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-11-16 19:30:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"recommended":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":1,"date":"2022-10-25 12:51:14.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":2,"date":"2022-11-16 17:33:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-11-10 13:21:09.000000000","_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":1,"date":"2022-11-16 17:33:09.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-10-25 12:51:14.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2022-10-25 14:02:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-11-10 13:21:09.000000000","updated_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"reviewer":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"state":"REVIEWER"},{"updated":"2022-11-16 17:33:09.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"}],"messages":[{"id":"eb44748fc90240fec28162dd203a2a6f860af4bb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2022-10-25 11:50:18.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2d56c64b0e6b66b689344e3db918f6002cbd5b1d","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2022-10-25 12:51:14.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"c751f3e7605fa1390db0da9eb87e9dd9c2d6d248","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-25 14:02:52.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/50d13d0745b24cef86c8dc7ae0b03b41\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4cb66c36d18a42399bf6565c3fe6c487 : SUCCESS in 9m 07s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f92a68b68fd349d1a8a4a846ecb7f2f3 : SUCCESS in 6m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2268db8c24db46feb65c64bfae56200c : SUCCESS in 4m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/15d53673276e48ebace8c4c9f13d0665 : SUCCESS in 6m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/df72bc0b82c242b8a442e03af2b48fe6 : SUCCESS in 5m 15s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/19dd6f675d374acd9d94204cd159b70e : SUCCESS in 5m 39s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4d9129dd028e4ccb9f40458f010fd096 : SUCCESS in 12m 45s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3e694c323c094b53bc2942efe7d0537a : SUCCESS in 4m 54s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/2c01a8d1c3dd4c70b6268da5437cc42d : SUCCESS in 5m 05s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/40aab34f95474da5870a5b4e1789bdc5 : SUCCESS in 6m 42s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/4941da5468714c8b977a5b41f2c4d1a5 : SUCCESS in 4m 27s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/aeecad1d1b7e4f899c9748b1c826c2a9 : SUCCESS in 6m 13s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/10aa7b99fada47e9948a9ebecf61c5f1 : SUCCESS in 56m 08s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/36158b4c720f4132b9e1029924420d60 : SUCCESS in 1h 33m 52s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/032dd393eb89492b8e6f66129f5b808e : SUCCESS in 34m 13s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/7399111b959a44e89afb8a8aeb2e73e5 : SUCCESS in 34m 02s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/ef0ece15f8cd4b13b678da424b370105 : SUCCESS in 35m 39s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/55cbf227f0004239a234dd92b85421e1 : RETRY_LIMIT in 9m 07s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/05daa39474aa44dda947635d0b218625 : SUCCESS in 1h 31m 29s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/5f8ecec0e6114ceba4d958b253343955 : FAILURE in 29m 09s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/e2acf0daee284f64863a51b1046f7e45 : SUCCESS in 1h 35m 27s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/32937557ddbf4b8b911bebacdc2716ba : SUCCESS in 1h 44m 55s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/09f9456c98e742c09123b70daca29055 : SUCCESS in 2h 03m 46s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/2696c328da684cca9d1aced09505099e : NODE_FAILURE in 0s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"82d336ed25f9a1812867c29e6ed818e50534f713","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-11-10 13:21:09.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"e8ff4013c9b90123b1a11c17f5598f78638a1da3","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2022-11-16 17:33:09.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"87c51530e51c41bad080f00e0912e039d5096306","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-16 17:33:50.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"692c43c6027ebc2b24176b3b2e6f168e553bc679","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-16 19:30:17.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/537c1ebe64424f73ab2359adb7879368\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f37df089dace4630923e30b6a70fda77 : SUCCESS in 7m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e45e50baca744c13bb43d36ed2998665 : SUCCESS in 5m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6084dd54809d47e3a4a9afabaf463611 : SUCCESS in 5m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d81b7349cd04493cb3225b9f90f2e285 : SUCCESS in 5m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a6dbcd14f04b4057b89569d3ca786086 : SUCCESS in 10m 02s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f781c670e2fe4735a12fcecaf80fc81b : SUCCESS in 5m 52s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/607442e791bd4e99b21c55a266fe6cfb : SUCCESS in 4m 35s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/5a11501c68014462b2df79f1070f9828 : SUCCESS in 8m 03s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/0f67e4b24acb46fc952d14f630555836 : SUCCESS in 1h 16m 25s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/1a59a8534b124c19958e29112d2c2d22 : SUCCESS in 1h 48m 58s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/d2e0e3c4b70a4da2b1dcccccac5b1431 : SUCCESS in 43m 30s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/d0c1bed5fe3947cc93d3e1df83f3464d : SUCCESS in 39m 40s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/93485c810a53474dbc20ff24f659eeb0 : SUCCESS in 35m 52s","accounts_in_message":[],"_revision_number":1},{"id":"32f0bfaa7bef79ba86bccb755aaf9a00e4d1ad1b","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-16 19:30:17.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"bedc285696ca6e7438126e60c6b8e4d9379f3054","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-16 19:31:27.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c9a3ab9cb17a46e99972836560086b5b\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eeac40a8220c4308881d38bd5a529616 : SUCCESS in 51s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/de26d239e2074692b5756fa12e860457 : SUCCESS in 39s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"b20bb2eb23d236421cca5a621b25ada0da3297f6","revisions":{"b20bb2eb23d236421cca5a621b25ada0da3297f6":{"kind":"REWORK","_number":1,"created":"2022-10-25 11:50:18.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/00/862600/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/00/862600/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/00/862600/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/00/862600/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/00/862600/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/00/862600/1"}}},"commit":{"parents":[{"commit":"401d4b9545622b38809f42ec46e29b33fa06ba92","subject":"Merge \"Fix PortNotFound exception when updating a LB after a failover\" into stable/wallaby","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/401d4b9545622b38809f42ec46e29b33fa06ba92"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2022-08-25 08:33:21.000000000","tz":120},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2022-10-25 11:50:07.000000000","tz":120},"subject":"Fix barbican client with application credentials/trusts","message":"Fix barbican client with application credentials/trusts\n\nIt seems that keystoneauth1.identity.generic.token doesn\u0027t handle\nproperly the application credential/trust tokens passed by the context\nof the requests. When using app credentials, Octavia failed to retrieve\nthe certificates from barbican.\nSwitching to keystoneauth1.token_endpoint fixes the issue, the\nauth tokens are correctly passed to the barbican client.\n\nStory: 2007619\nTask: 39737\n\nChange-Id: Id77ce36f59b71d309f153e5c1d44059f162ee440\n(cherry picked from commit ce7f27e3b7ef6a94501ce975fb0e9dadcffb822b)\n(cherry picked from commit 851510f4c74658b128790da1aab410e245355a90)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b20bb2eb23d236421cca5a621b25ada0da3297f6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b20bb2eb23d236421cca5a621b25ada0da3297f6"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[]}
