)]}'
{"id":"openstack%2Foctavia~862671","triplet_id":"openstack%2Foctavia~master~I6ca990b7080e35021e8d8dac922e9546eac3da2f","project":"openstack/octavia","branch":"master","attention_set":{"38562":{"account":{"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"},"last_update":"2026-03-16 18:56:33.000000000","reason":"\u003cGERRIT_ACCOUNT_37598\u003e replied on the change","reason_account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}},"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2026-03-16 18:56:24.000000000","reason":"Reviewer was added"},"11628":{"account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"last_update":"2023-02-07 07:21:49.000000000","reason":"\u003cGERRIT_ACCOUNT_29244\u003e replied on the change","reason_account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}},"37598":{"account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"last_update":"2026-03-17 07:30:45.000000000","reason":"\u003cGERRIT_ACCOUNT_29244\u003e replied on the change","reason_account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}}},"removed_from_attention_set":{"29244":{"account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"last_update":"2026-03-17 07:30:45.000000000","reason":"\u003cGERRIT_ACCOUNT_29244\u003e replied on the change","reason_account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}}},"hashtags":[],"change_id":"I6ca990b7080e35021e8d8dac922e9546eac3da2f","subject":"Add zookeeper backend for jobboard in devstack","status":"NEW","created":"2022-10-26 08:02:11.000000000","updated":"2026-03-17 07:30:45.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":14,"unresolved_comment_count":6,"has_review_started":true,"meta_rev_id":"fd940636770980ab62d82e6fe2d1e8d51c1aa694","_number":862671,"virtual_id_number":862671,"owner":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"tag":"autogenerated:zuul:check","value":1,"date":"2026-03-16 18:53:06.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"rejected":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"all":[{"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":-1,"date":"2026-03-17 07:30:45.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-10-26 10:36:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-02-06 19:04:16.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2024-09-26 06:46:46.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"CC"},{"updated":"2026-03-16 18:56:24.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"CC"},{"updated":"2026-03-16 18:56:24.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2026-03-16 18:56:33.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"},"state":"REVIEWER"}],"messages":[{"id":"4989547e543b31397f9ddbc67164800f7bcc0bf4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2022-10-26 08:02:11.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e551a34653d9ea40d782b23037a6875f03a20e02","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-26 10:36:19.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2af6d04840754d69b125b48462d23f2f\n\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/bbed134f1e0042ef812cc2cab571ca7a : SUCCESS in 2h 00m 41s\n- octavia-v2-dsvm-scenario-amphora-v2-zookeeper https://zuul.opendev.org/t/openstack/build/f97ae997398c4fa9803cd5832af63cf3 : FAILURE in 2h 20m 03s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/b9e8ad05660143d2b1d73fed19cb0f72 : SUCCESS in 2h 27m 18s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"dbc09450c1619da5c17f273ba1a04d58c2bab617","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2022-10-26 13:12:50.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"aa1b52b3d28bace797979e492618f88972c8e6da","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-26 15:03:25.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/11470a07e90f40258133e76a66b267ef\n\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/ec11051ab3824c4d9469831e807afa01 : SUCCESS in 1h 40m 22s\n- octavia-v2-dsvm-scenario-amphora-v2-zookeeper https://zuul.opendev.org/t/openstack/build/337e6772f32b4131b9c183b9294ea259 : FAILURE in 48m 43s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/a45656e906ee4a5aa5ad53a1f155f913 : SUCCESS in 1h 30m 36s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"c1c8adea2d53c0400b3cfe9c3df5ac52d413c9f1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2022-10-26 16:45:28.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"9be3aa53e6af7d70f488c519c6bfa3ac83427947","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-26 17:45:00.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bf99f3835e6f4d84b864a2f4fad51a08\n\n- octavia-v2-dsvm-scenario-amphora-v2-zookeeper https://zuul.opendev.org/t/openstack/build/10f594d6915247d693d30c361c8c29d8 : FAILURE in 57m 31s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"cc1583850c56aacbeb076ed9b5d01221b490b82e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2022-10-26 17:55:28.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"37d3bf69bab3296956b84925a7ad230291e5acf0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-26 18:32:59.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8a59afc766ff419b84e51e4d2a381d39\n\n- octavia-v2-dsvm-scenario-amphora-v2-zookeeper https://zuul.opendev.org/t/openstack/build/c06cace77e6845c2a24b0ab50d5a1f77 : SUCCESS in 35m 52s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"75b30742397f0fc880b8cb9d43d2266a8ca8db12","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2022-10-27 06:02:57.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"335c2e2cf8e5f642fe45ab5defb6b2772718a19b","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2022-10-27 07:17:44.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"0074394f58ef842ec57db1783ef37a8b89702ecb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-27 08:32:34.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/be7ac3a8c7bd424ab343e8345ebdb711\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/683386689bcf4225bd6c9b275fc82f86 : SUCCESS in 2m 13s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9bb3396699d2464eb054511b5282ace0 : SUCCESS in 9m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9aa627b4cef14201bfdc3c4d1e8f1cda : SUCCESS in 6m 20s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/716de2986dbd4a72bb1dc30d98acabfb : SUCCESS in 4m 59s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/63dddabf2d86451c8526e6e254eb9457 : SUCCESS in 4m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cc5c100e8b884f929382ea3739a77b8b : SUCCESS in 10m 54s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/9aa787919129475c813977f5b475192c : SUCCESS in 5m 00s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/17ee6d89b2d44a818c99d9cceb8ff1dc : SUCCESS in 6m 34s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/46b4629713504216846ca205465935c2 : FAILURE in 4m 18s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/8ac6bc22e5d744b3b52f939e69b70558 : SUCCESS in 6m 33s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/ed56b3b6c489430181fc793adaf81c71 : SUCCESS in 7m 15s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/1b218696697c4d11b5c6f89030aca630 : SUCCESS in 6m 46s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/043a0bc3b6ca447789b2cd86e0490d87 : SUCCESS in 1h 23m 28s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/a7a97a2962ac45a9805f94e0845ac6cb : SUCCESS in 1h 47m 53s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/f60423ec5edd44e59b06a06a2beeb097 : SUCCESS in 45m 31s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/9444dc0fc9734ad89221e437f4174176 : FAILURE in 40m 15s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/f62c8785339b4d5383bb615d9351772e : SUCCESS in 1h 08m 45s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/b5f3fcb875ff4fd8bbe81ddf979561ed : SUCCESS in 41m 26s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/9fee222f026542f9a2a1de64c0d85849 : SUCCESS in 2h 05m 13s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/335e6c985de74f1aabda5c6c4b12cb5e : SUCCESS in 2h 26m 52s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/65a20e3e1b794349a47ee5215927733e : SUCCESS in 1h 45m 34s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-zookeeper https://zuul.opendev.org/t/openstack/build/cc879c6d8478461ba78eed5dd5e0c5f6 : SUCCESS in 1h 59m 36s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/56974188c18f4d328cf7093320aaa400 : SUCCESS in 1h 51m 02s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/f0c6b6ae047343aebc3d752c6d8391dc : SUCCESS in 2h 18m 39s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"7e1b01565b29aa8efecb2feac928fcf0223f363e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2022-10-27 12:20:06.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"4b6983b99d21cb05ca1874f6414222517db7eb75","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-27 14:58:53.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/d5c549963d344dcaa35d5ef58af9e654\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/484232bf99194cc3a430e67a7f44d24e : SUCCESS in 8m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c316ae7c9fc74bffbc01be342a4ed98f : SUCCESS in 5m 46s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f4b37984bfc14222acd2301cc283741f : SUCCESS in 6m 50s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/35c13e2497fb43089e944569a6989633 : SUCCESS in 4m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/20292cf353764f168a7529c6ac87ce79 : SUCCESS in 9m 57s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/058ac106ce8c41ef8c904084bc11383e : SUCCESS in 6m 18s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/aebbe005b8894a8a986c2b7f5f692375 : SUCCESS in 6m 40s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/0c1984674b1146d6842de5708d53af58 : SUCCESS in 5m 12s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/9a62b146e256405586a52c4ec1289c8b : SUCCESS in 8m 08s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/7771332a10044010bff78e78b69196f9 : SUCCESS in 6m 54s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/8beed724b1b74a08b9fe9bd83dad2379 : SUCCESS in 6m 06s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/27152c75f54e48808bbaa0767473d970 : SUCCESS in 1h 24m 53s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/af049957007149699dd607b1434b5639 : TIMED_OUT in 2h 35m 06s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/494d0fac3b43429685dca9b1b7035810 : SUCCESS in 38m 16s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/892bef4e4cd945528c44e2e814eae476 : FAILURE in 45m 19s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/fe72132581334285ac65c8d821ff078d : SUCCESS in 1h 05m 16s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/4fa7126df0d747ebb084c8f068394a51 : SUCCESS in 35m 48s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/5f692e1f3cac4471be6134b7da54e3da : FAILURE in 21m 10s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/acb0b0e540c74528ae669ca39ea2501a : SUCCESS in 2h 29m 37s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/6e1a01a859c34d2ab988191b18f976c4 : SUCCESS in 1h 46m 09s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-zookeeper https://zuul.opendev.org/t/openstack/build/43ba491ae09242a2b9cf1f4a957ac655 : FAILURE in 2h 03m 08s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/e4860cf5c1bf42ac9724854c58e9759e : SUCCESS in 2h 28m 26s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/43a7fad5c8f849feaa98c8ee3e5162a3 : SUCCESS in 2h 14m 19s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"37f916c16838b4a5c8722ba80febbeeeca05d4ae","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2022-10-27 15:14:31.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"cc30a544380c9e14ce032dbc1e3060e75838a607","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-27 17:30:25.000000000","message":"Patch Set 6:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/764f6ce4086a404585ceda8e96b79f74\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7de83f42f93f48a59bb1ae0146212fbc : SUCCESS in 16m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2d992a11a39449d9baf4471c1547df5f : SUCCESS in 7m 44s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f1070924d8a34bceabbf3d7b8bb438c0 : SUCCESS in 5m 18s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/cb21abbc2a0044acb5c4efb823e42ea2 : SUCCESS in 4m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d796182f3bb841319f872dbdf4cdfd7c : SUCCESS in 13m 52s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/fe778e85e7074d4f824447b3092414a2 : SUCCESS in 4m 58s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/8d6edd7a72834026ae5a4dd0527da012 : SUCCESS in 8m 25s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/89e27120083841bda51bd06baf95c963 : SUCCESS in 4m 55s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/022ba0ca898946ea88cd98bd8f168ff0 : SUCCESS in 7m 00s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/b7db7881f9144d23b0e87d3d79473bfc : SUCCESS in 8m 31s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/75d551746f5c4ae79020e68be88c7063 : SUCCESS in 7m 08s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/e669267f398c4e12808b9bd82ea7ae9b : SUCCESS in 1h 49m 30s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/d1ca8e6ae03745fa8499f310e0305244 : FAILURE in 27m 18s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/7ae12aeff84f41ae99acc09a5fe82c6e : SUCCESS in 37m 40s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/7108f613327a479fa9e3ff8bc3455a03 : FAILURE in 34m 04s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/5737c6fd727245d6b152936568a7acfb : SUCCESS in 1h 14m 17s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/756122d343b14ab795c6f6a6203b1426 : SUCCESS in 30m 30s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/3f55079245504a7ea3c40310758ac759 : FAILURE in 29m 04s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/bfa6826c5b254ee18d327b2bf7d7a1a8 : SUCCESS in 2h 13m 09s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/bf7de0bd37c447f593a10f03b7f15d9d : SUCCESS in 1h 45m 03s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-zookeeper https://zuul.opendev.org/t/openstack/build/145efc28e1ea4070852792eafaf0c0a4 : SUCCESS in 1h 44m 44s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/d1b361d41a9548d0afa786a66770f670 : SUCCESS in 1h 46m 54s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/ae89b2de3c6c4b5894dcbfa0045bdaa8 : SUCCESS in 2h 06m 30s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"e9063d01b80dfcfa8338ac13cca7fcf7d9eefcbb","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2022-10-31 09:44:46.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"5da47d3a1108e24329d5ffa71f42b176d90c623c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-31 12:59:40.000000000","message":"Patch Set 6:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/0dfbc26651364b6b825e92efc84db940\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/db49e8a9ad5c4642a4093a612542a73f : SUCCESS in 11m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ff839a3ce4ab4393b38543f4932b8136 : SUCCESS in 6m 53s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a23b07a2686b46c1a4ce76f03067c361 : SUCCESS in 4m 18s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/737fb0da19234ebdacb063b064c44d8f : SUCCESS in 5m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/87e291ef1a8a4cd2aaa2426a46779ac7 : SUCCESS in 11m 32s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/b807d57355f145489043160864589e56 : SUCCESS in 6m 03s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/134408dd98e0496a9e713a4fd2202e58 : SUCCESS in 8m 29s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/0f9c26a706594be9a59cfe4fdffd31e3 : SUCCESS in 4m 43s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/a8034118316b4c6baadcb9cca757ec30 : SUCCESS in 7m 38s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/d41b8234b9084cf19e5a47f0bdde3ecd : SUCCESS in 6m 30s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/6bd5af305bb340d89891ebf70dbd08a6 : SUCCESS in 9m 30s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/9be867d1d2434600947460a908bb7869 : SUCCESS in 1h 19m 26s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/c045361415374f25a87a52aa6daf33db : SUCCESS in 2h 15m 23s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/3921611695ab452daf09769c7985a91e : SUCCESS in 43m 40s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/08b940d871ec4a928d0b2fe20924c1a5 : FAILURE in 45m 50s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/5774b2330d5d453dbba96974bc904dff : SUCCESS in 1h 17m 12s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/79af64da169d429bb3cfea589dabc00b : SUCCESS in 37m 17s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/2724e691efbb43c19fd35baa0767a2d9 : SUCCESS in 2h 09m 48s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/8dc5927e4c3d4c7aac5a238332873e4b : SUCCESS in 2h 51m 44s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/55cafbd5b096414dadc7c1cc33e73d18 : SUCCESS in 2h 17m 51s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-zookeeper https://zuul.opendev.org/t/openstack/build/89be56597ea54c3fa58c9442e115f362 : SUCCESS in 2h 10m 39s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/3563f6f1afc247c7a422cbf2b79df1d5 : SUCCESS in 2h 34m 16s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/fc3c8ce1ec444bc1b42ed358869d0096 : SUCCESS in 2h 35m 25s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"6596b7e73f957527654acd40c4b6d1b2908a4b54","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2022-11-30 07:41:32.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"0e6f6f9b96a3ed596deb2726b683f16153d32784","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-30 10:21:48.000000000","message":"Patch Set 6:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/5757e17199314542a41500de20594b7e\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a6cbfd82142c42f69c96a47af4486f22 : SUCCESS in 9m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bfc6e594d9794eaf810a6e128468d823 : SUCCESS in 5m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8188c5925a544107868f7d90bd62c997 : SUCCESS in 4m 17s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ab6a563533934473868ceb8460bf2c2d : SUCCESS in 5m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/475574a7434349249b37a8aac7cbb394 : SUCCESS in 12m 10s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/bc2b7d9c8845489d9adbc15891314353 : SUCCESS in 5m 04s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/c5a3a46945304052891a3e46c88378d3 : SUCCESS in 9m 12s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/2ab903b23370443d95fffc20ecc7df37 : SUCCESS in 5m 09s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/24da8c126d304b33bcd7acc6047de0aa : SUCCESS in 7m 21s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/6e81df5b97194fa0b6c235dec5f7eec2 : SUCCESS in 7m 36s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/8461c9fd7e754c558b20c24d5573868d : SUCCESS in 7m 58s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/0acd980811fc441e9892c351412b3b82 : SUCCESS in 1h 20m 04s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/760a8a7f49404fa881ac8d86c23197b0 : SUCCESS in 2h 32m 36s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/05a09750b53842c9a4329d96ad005228 : SUCCESS in 45m 34s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/fbd9f9af61c24067961074ca37416eb6 : FAILURE in 51m 13s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/7ae9f077fe644bbda5ecf15a5c88eef8 : SUCCESS in 1h 27m 12s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/2f6bb55520ce480bbe8c2a01d02cf69e : SUCCESS in 42m 21s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/14a17288c4634a0281c1574f20cfa1e9 : TIMED_OUT in 2h 34m 08s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/cba5bb2bef4541e789650f9ed616a750 : SUCCESS in 2h 37m 02s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/f0177feb26874a41a51876728be67f77 : SUCCESS in 2h 00m 48s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-zookeeper https://zuul.opendev.org/t/openstack/build/87b03e2f738a4957a9238ffd4fd4c283 : FAILURE in 1h 59m 02s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/bd78593963184b0e89d25a88e417f2c5 : SUCCESS in 2h 22m 44s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/45a152ddbdbe4aaabf72e93b9b9b10b3 : TIMED_OUT in 2h 33m 44s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"152a58bc6ed22d630c5301bd21fb2c82802c3247","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-01-12 11:19:16.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"51d2678a0efb31eaff0da5fa492bc6a0de068bc3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-12 12:16:54.000000000","message":"Patch Set 7: Verified-1\n\n(2 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/723289f096cb4d66804ee85beec551ed\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e27d8e8895bb49d9afab71f16fe029a4 : SUCCESS in 8m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6951703bdbce4862a6b597d401f00ee9 : FAILURE in 5m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f08ce344b4f24c2a93f0f621d46e97e1 : SUCCESS in 4m 25s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d8f7d6fded3a4118bfb8e5857465132e : SUCCESS in 5m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/27cdd9227c414623ba8530ea9edc738f : SUCCESS in 11m 52s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/fa7e89e791fe4a378e9b80e416220602 : SUCCESS in 4m 56s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/45d2c28030e4479a9de6c2359147adad : SUCCESS in 8m 53s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/409f880497b84489a83165781337ad9f : SUCCESS in 4m 40s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/f333f7c122ff4968a6aea2c79fb9fa5b : SUCCESS in 8m 16s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/b2f00392c4704c34819238758c5d429b : SUCCESS in 8m 21s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/e3ccc10fd818467dbb36a62737e6584e : SUCCESS in 6m 52s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/7da8db3016e84a1aab477e77d8fa44e8 : FAILURE in 8m 04s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/0fc0e40ed70f4678975993ec3a903c69 : FAILURE in 10m 20s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/038ded651da74bc6837f0a45b537618e : FAILURE in 7m 49s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/7f670ee94d1f4f69afa07f2dd2f472ff : FAILURE in 49m 03s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/f480f909029742d28a9000cc5e4b9cf4 : FAILURE in 10m 59s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/cb3287dc4c5c44fea6b15ec5cfa2e5fe : FAILURE in 9m 40s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/cc831a635004431884428a6c6d1c92fe : FAILURE in 8m 02s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/8e01e2f24ba141deaa1d97c87dd0e333 : FAILURE in 16m 57s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/ad520e7ce3dd42309b8dcf9c70019274 : FAILURE in 8m 56s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-zookeeper https://zuul.opendev.org/t/openstack/build/dc278517481f4ac6ad1f0544be23559a : FAILURE in 10m 19s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/2d909b7c33b14f8982d7bbc54affc75a : FAILURE in 9m 20s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/57ab395d056144148c28acf9ed28a0f3 : FAILURE in 12m 58s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"b4a143fe5be14183fdd379e54c12f382219c3318","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-01-12 12:25:10.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"94dd30a32cf139cf1c250abcad75be547f65f2bc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-12 15:19:34.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/35e7bf61f3204b31a33d0541d29e5ae4\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/842d411cbe724afea576ac330868db29 : SUCCESS in 10m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/83cb874c4f054b048786374c2828b2e1 : SUCCESS in 5m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bb2023e667034e4d97c1f130cc93b170 : SUCCESS in 4m 59s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5e9f53f7d6594f9ea0b590d45aeb7049 : SUCCESS in 5m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ab9ea733d67741abb0bcbb7d1b79c8c1 : SUCCESS in 10m 43s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/4be6d70feb814458a34113ed32687aec : SUCCESS in 5m 38s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/4ca113fed4ac45e298cee06de3850777 : SUCCESS in 8m 11s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/c97f5d0713584b56a61064cd16dab277 : SUCCESS in 5m 40s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/3b820d5e32af4f98aeacb2e4325f68ce : SUCCESS in 7m 31s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/3403421e660349d4a6f2d3781ec21fa8 : SUCCESS in 8m 03s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/0b838716bc05476e94a347894725a7c4 : SUCCESS in 7m 17s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/0aef3e3b520149ac91a9a8999ea3d8a4 : SUCCESS in 1h 25m 45s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/84f6243a09954d5381dbbc2239b776fd : SUCCESS in 1h 45m 07s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/591e7bb10524447db0bcd8d09fbf2738 : SUCCESS in 33m 19s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/b02a9f4f762144f388f7420ccffd8901 : SUCCESS in 1h 04m 13s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/8c85fa597dcf4c2eb85e71f412130495 : SUCCESS in 1h 23m 18s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/29d1ae59c11646f4adcea3aa6053562c : SUCCESS in 30m 03s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/9dce63f94312425cb0f22468ac1bf1cb : SUCCESS in 1h 53m 30s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/f9653ebc661749af9d9559fd76862a74 : SUCCESS in 2h 50m 34s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/5ea7354f1caa409a990180c05811c2e1 : SUCCESS in 2h 17m 38s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-zookeeper https://zuul.opendev.org/t/openstack/build/ed43f90f0da647dfbee034c4e3409a56 : TIMED_OUT in 2h 33m 06s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/fa429874ba5849a7a9b46bbebe8a6d86 : SUCCESS in 1h 42m 35s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/b1009442302041178e7d4e2278be936b : TIMED_OUT in 2h 35m 14s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"e0e92b40a61cc2b6d411d8803cbead757bd4ce37","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-02-06 19:04:16.000000000","message":"Patch Set 8: Code-Review-1\n\n(5 comments)","accounts_in_message":[],"_revision_number":8},{"id":"471ad84366bb192b4fa29abc61f13af7e4a29b97","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-02-07 07:21:49.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"43e639836da41a32f23e998f75761f8eed732367","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-04-12 16:36:31.000000000","message":"Patch Set 8: Workflow-1","accounts_in_message":[],"_revision_number":8},{"id":"29cbda47da9779141661f66bf8c51fc3a5f375fb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-10-16 09:43:37.000000000","message":"Uploaded patch set 9.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n* Workflow-1\n","accounts_in_message":[],"_revision_number":9},{"id":"591a9faca05e81b944489a1f557316be25bf81db","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-16 12:24:50.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/46e6695356f84d5d8c7b7b86f809b607\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2489e1d9fb294db3b18af4385f335951 : SUCCESS in 11m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/239035107ba24440a88f7050a1285a7e : SUCCESS in 5m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f4aee88add0242fda7d6ba054f2d95fe : SUCCESS in 5m 25s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d1bd80e5829a4b45b0b0f892ada5d770 : SUCCESS in 4m 49s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/79a81d25e40044b494c098c80f23c21d : SUCCESS in 4m 53s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/44e970e51a784a25a5dfff378425a4ff : SUCCESS in 5m 59s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/16820a34e58e4ee58327201e252b0151 : SUCCESS in 12m 26s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/728c038a1bcc46dc8b8366061a31297b : SUCCESS in 5m 41s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/6a6c1a5b77ce47368272b48bafb5ea8a : SUCCESS in 9m 00s\n- octavia-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/e0d7b56f65084ba7b4bfce6f04f2f118 : SUCCESS in 11m 18s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/01ce1f529b54407ea623ea860c950e8a : SUCCESS in 4m 39s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/f5a3311a3d1d4d429241263f79d0b63b : SUCCESS in 7m 40s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/b7b482b8bc7548168ec2002de405881c : SUCCESS in 8m 47s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/1f4835ba1c2a437ead2e7556f69fdb67 : SUCCESS in 7m 24s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/7dbf1ab988d640c68bb67f6e337de373 : SUCCESS in 1h 14m 44s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/486a42afb2104d0fa97b88c30e8bafc9 : SUCCESS in 1h 34m 29s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/269519c96a694e96bcc25cf5134f3afb : SUCCESS in 1h 06m 25s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/8c6de4bc4b0543baa4ce7e0bbdaa2d10 : SUCCESS in 1h 59m 00s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/5a211029fe364fd692708981b47dd425 : SUCCESS in 1h 14m 22s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/ba5a6673313e493288c1063a01c58147 : SUCCESS in 37m 34s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/931ac5e48b8e4bbca8c61c3d37bc6857 : SUCCESS in 46m 01s\n- octavia-grenade-skip-level https://zuul.opendev.org/t/openstack/build/ed02bd8030f2442aa3b126014fbdeb9f : SUCCESS in 38m 33s (non-voting)\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/3059318ed4cb46479d0e05d62dae15fc : SUCCESS in 1h 08m 25s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/0394fa1bc3d74c70a18b95f84f1acd1b : SUCCESS in 35m 14s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/cf2946c94e4b45be890f3c3b12dfc616 : SUCCESS in 1h 52m 23s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/fa1bd22cfe604335b4e28ef1aa5fe3ca : SUCCESS in 2h 38m 24s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/85434f28c6124b7c8a44b7436af9c633 : SUCCESS in 1h 54m 36s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/995a6dac788641a480b436c7ca125b1e : TIMED_OUT in 2h 34m 51s (non-voting)\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard-zookeeper https://zuul.opendev.org/t/openstack/build/f60a9e267e81467fa014760217884e3b : FAILURE in 58m 35s (non-voting)\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard-zookeeper https://zuul.opendev.org/t/openstack/build/d4dcd739ce114ba1b779e8d5956f3baf : FAILURE in 2h 18m 28s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"9ce69cec30bf9c0f9a6bf93c85bba55490b248a9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2024-09-25 16:44:38.000000000","message":"Uploaded patch set 10.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":10},{"id":"6e8d4d74bca85045e7c5c680fa489caae0cc1878","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-25 19:32:53.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/420466209e804bd5b34b854f75a80b30\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7ea3265415a9465bb834404fc526cd8c : SUCCESS in 11m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3251d2a31e3749339575c34d1b0f7774 : SUCCESS in 4m 39s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/caf082e9d5f14edea0726bd4a0289983 : SUCCESS in 4m 11s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/6840d7bb78ff46c6a459ac55f6e79af5 : SUCCESS in 4m 51s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/f4dff0d7fed54015b850abea5bb0f4aa : SUCCESS in 7m 01s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aaf79f29b85d46af8d8da3296fb4328b : SUCCESS in 9m 59s\n- octavia-tox-py311-tips https://zuul.opendev.org/t/openstack/build/ff6a71b85fcb45039c72c403ee4734a4 : SUCCESS in 4m 50s\n- octavia-tox-functional-py311-tips https://zuul.opendev.org/t/openstack/build/5c57007acb274682a768237d7a6dd544 : SUCCESS in 6m 17s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/000f57e875cd4c8e8938b080fdfa850a : SUCCESS in 3m 58s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/186b32e5f73a452fae81b0428dcc0199 : SUCCESS in 8m 17s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/0e1034ec7404450fba9eababf7fe2546 : SUCCESS in 8m 45s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/a1f3b3e69b234c57baac6c803ed3a607 : SUCCESS in 6m 13s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/18d31096b4114a679684b9328b94084a : SUCCESS in 1h 30m 21s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/03197580d4b44a8996440c9959cca326 : SUCCESS in 1h 38m 44s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/7f352f3b7191416cb4b72b2749f78cad : SUCCESS in 1h 13m 28s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/a97638ea571c48a9b0263ff9f29e2ee9 : SUCCESS in 2h 19m 30s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/acc18e21a36546a98e403f0a8f5a2309 : SUCCESS in 1h 04m 59s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/383ce463c7e34f91a132d4d266015dd7 : SUCCESS in 31m 03s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/1558036eff5b46ae9eed69686514d45e : SUCCESS in 36m 34s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/4402ab508f58485ba011a5dfd7fc5c11 : SUCCESS in 45m 07s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/d562ab70c6d047f5ae724379097b6fd9 : SUCCESS in 34m 02s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/e72ee72f9c0d4ed6b5e8b1d1228f3325 : TIMED_OUT in 2h 32m 59s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/664c0014360a4c838a6c482fdbb23ca6 : SUCCESS in 2h 46m 40s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/6124f397eeff479bbc29204fb821708f : SUCCESS in 2h 18m 20s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/b01d018549144b1aa2517e6f8b517fac : TIMED_OUT in 2h 35m 29s (non-voting)\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard-zookeeper https://zuul.opendev.org/t/openstack/build/fd538ecab1604cce9e8ab78f22fc1d3f : FAILURE in 53m 20s (non-voting)\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard-zookeeper https://zuul.opendev.org/t/openstack/build/b3f945eae96646e997481d17c6662ec6 : FAILURE in 1h 47m 41s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"9f04968da4c96297c5963eba20c80e21c64e3136","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2024-09-26 06:00:58.000000000","message":"Uploaded patch set 11.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":11},{"id":"be606103d94c08dc4ff4a5794dc6afd0ee09b43e","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2024-09-26 06:01:56.000000000","message":"Patch Set 10:\n\n(2 comments)","accounts_in_message":[],"_revision_number":10},{"id":"409409ff08d3c4343cdb568d154862717fce7469","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2024-09-26 06:46:46.000000000","message":"Patch Set 11:\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"cffdc5ab3b40a0c722ae3513e1c6c38e12817e04","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-26 09:10:59.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/11ad156e735c4f23b56e8b4fd38d39be\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/eff3d8f28e4b4a08955ddb4c754d9eb7 : SUCCESS in 10m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d573137bf63c408abbbf0e8bd30d9438 : SUCCESS in 3m 31s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7b7d8804a733440f91053cfe2f28172c : SUCCESS in 4m 31s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/278ea5545e3041daa5803d17d805296d : SUCCESS in 4m 05s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/de1673d7f0514f459011a7430508e1fe : SUCCESS in 4m 43s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e3eb0e02ab434e4abb54231b081eff53 : SUCCESS in 9m 50s\n- octavia-tox-py311-tips https://zuul.opendev.org/t/openstack/build/a1700b5fd3484def9160c637a72e662e : SUCCESS in 5m 39s\n- octavia-tox-functional-py311-tips https://zuul.opendev.org/t/openstack/build/b23a5e6bb4a9492e91e8ba8ec7e4849f : SUCCESS in 5m 25s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/0fbc252eba70445da9d30f4cfde1e4c9 : SUCCESS in 4m 21s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/4494d579b803400b83d26a2d4edbc684 : SUCCESS in 7m 23s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/e2982d47f4244684b72584b993bff057 : SUCCESS in 7m 20s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/6dd23ea170814ca6a514e463eb78e0ac : SUCCESS in 6m 43s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/d5c5a10442d94218a8a4353f6d48a024 : SUCCESS in 48m 57s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/a7302debccaf46d08fc45aa0fa041399 : SUCCESS in 2h 06m 39s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/f827265253fb4073a0c085774fb712fa : SUCCESS in 53m 18s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/ade431ae576d4fcfbcb6e77a93ce9aa5 : SUCCESS in 2h 09m 49s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/16d6d1abacc54098a9f2a9d4ca0c9ea6 : SUCCESS in 49m 30s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/c7c98b166c424b408a6b3ff6fcb550df : SUCCESS in 42m 11s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/a08d6f15d0a64507990240166223b0aa : SUCCESS in 34m 54s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/83f9153a04664d90a335612481707ad8 : SUCCESS in 1h 01m 19s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/39f553e0fea0471f91a4c8930d78b756 : SUCCESS in 29m 13s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/b4dbbc1727ef48529c398bb33174e034 : TIMED_OUT in 2h 32m 58s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/babbd86a6af045eba3bfde2b2c786e6a : SUCCESS in 3h 02m 33s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/d10e3222dffd46108676ee8471cd439c : SUCCESS in 2h 21m 45s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/8d3f5f9103bc43de944016745b01a1d7 : SUCCESS in 2h 24m 00s (non-voting)\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard-zookeeper https://zuul.opendev.org/t/openstack/build/cd9595de7a204de791f789d20cac8583 : SUCCESS in 1h 39m 22s (non-voting)\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard-zookeeper https://zuul.opendev.org/t/openstack/build/22c7dedb681b4700b072d515219c2a98 : SUCCESS in 1h 21m 07s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"7e81c3e36ae67e3ea96947945fc4b4bd1cd86798","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2024-09-26 09:12:09.000000000","message":"Uploaded patch set 12.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":12},{"id":"1fb3adbdb020924197ab5462c550f12afa1a9157","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-26 09:12:32.000000000","message":"Patch Set 12: Verified-1\n\n(1 comment)\n\nZuul encountered a syntax error while parsing its\nconfiguration in the repo openstack/octavia on branch master.  The\nproblem was:\n\n  Job octavia-v2-dsvm-scenario-traffic-ops-jobboard-etcd not defined\n\nThe problem appears in the the \"opendev.org/openstack/octavia\" project stanza:\n\n  project:\n      templates:\n        - check-requirements\n        - periodic-stable-jobs-neutron\n        - openstack-cover-jobs\n        - openstack-python3-jobs\n        - publish-openstack-docs-pti\n        - release-notes-jobs-python3\n        - octavia-tox-tips\n  ...\n\n  in \"openstack/octavia/zuul.d/projects.yaml@master\", line 3","accounts_in_message":[],"_revision_number":12},{"id":"886323958efcb2eab1704df49dfd607b0ba6ef55","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2024-09-26 09:14:58.000000000","message":"Uploaded patch set 13.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":13},{"id":"25fd7674af32b6ba06216ce288e4b84b6775d1d1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-26 12:23:37.000000000","message":"Patch Set 13: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/9f8733a5b1964cc7aa277e10647a1d69\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fd7feb5babd8493b8f0014faf8ce0caf : POST_FAILURE in 10m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/481ba72b058640499a41b541b2a4aa5d : SUCCESS in 7m 16s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8565c5628b524cce8b1c5da03f449e7b : SUCCESS in 7m 54s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/298d9cb512ee424587791905a845beea : SUCCESS in 6m 20s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/6e53fa7d13e443b583a22a2155587287 : SUCCESS in 5m 16s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a5625021c50e4a49906fecfddc4bc98a : SUCCESS in 13m 55s\n- octavia-tox-py311-tips https://zuul.opendev.org/t/openstack/build/95d53be298ae4f01bbc86ada9a47dedb : SUCCESS in 4m 37s\n- octavia-tox-functional-py311-tips https://zuul.opendev.org/t/openstack/build/946c2c8b5e8f41268826fcaea39c32c4 : SUCCESS in 6m 25s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/17ff6c4f755a4fa489693ef30ee5f17e : SUCCESS in 4m 03s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/386534a11347464b971f5e264622e5b8 : SUCCESS in 6m 54s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/e2c7cb70c46544bfb2a54c37242fac64 : SUCCESS in 8m 54s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/529bf4601c924fe4a8f748fa732dc33f : SUCCESS in 7m 11s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/406f0a83d3e146cf8d555db2ae998714 : SUCCESS in 1h 28m 57s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/850e348613db409dab2f717abcd1b493 : SUCCESS in 1h 43m 37s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/ed545d80f91545c78c88ab04489b4520 : SUCCESS in 1h 00m 07s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/5e99f3b3e8ce4454acc99486d89e3b95 : SUCCESS in 2h 16m 10s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/c0bab023f3944af58b14d9301f32897c : SUCCESS in 56m 50s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/1a05f702b2b646b19ab48b47dd94aaec : SUCCESS in 45m 28s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/28888295177c405493869ae0b14ac254 : SUCCESS in 39m 49s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/13f8124dca07428285963efb7b40b62f : SUCCESS in 57m 59s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/995fce50b7164724a8b55fb2184b1068 : SUCCESS in 31m 09s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/e99fa011bff641328c4b76c556e90fb5 : TIMED_OUT in 2h 33m 47s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/23f9dce426114f648c761d5aa6b7faf8 : TIMED_OUT in 3h 03m 55s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/2d347a10869e47a29948f69df13f31b9 : TIMED_OUT in 2h 32m 37s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/02d8e83fac084e0cb94e9d26b3b16b05 : TIMED_OUT in 2h 34m 41s (non-voting)\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard-zookeeper https://zuul.opendev.org/t/openstack/build/3772651064914db59093eae57a5b8fa0 : SUCCESS in 1h 51m 57s (non-voting)\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard-zookeeper https://zuul.opendev.org/t/openstack/build/84931bff4e424385b575f6755c5748c0 : SUCCESS in 1h 08m 55s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"1d39f65f5030f3e3f7c943024d7f4133f0cf98d8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2024-10-01 09:34:50.000000000","message":"Patch Set 14: Patch Set 13 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":14},{"id":"19b1c84a0a6d87cb229fd2c9d3189f5804c10f96","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-01 12:42:05.000000000","message":"Patch Set 14: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/726ed5304ed2475a8e5a112e17f1c852\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/32d2de3bf86144de9bb29b4ae8114c1a : SUCCESS in 11m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1398fb640f484236b57c74f4a8d135ca : SUCCESS in 6m 32s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c4ed36b4fecd406494f9ad5e481dbdad : SUCCESS in 4m 04s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/ed2f00b0f78745a0be0c8d3cd40b3ad2 : SUCCESS in 4m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5c7e28c4819648e296f201c881f62e48 : SUCCESS in 9m 58s\n- octavia-tox-py311-tips https://zuul.opendev.org/t/openstack/build/9275bb0835cb4c21aae82b0a33711e6a : SUCCESS in 4m 55s\n- octavia-tox-functional-py311-tips https://zuul.opendev.org/t/openstack/build/3beee28bec6245c88e47eb6a7717a632 : SUCCESS in 6m 17s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/685ddb2c08b9492095cc998ee703501b : SUCCESS in 3m 01s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/b9e1b9968aa94bd7b7f15882c6f18013 : SUCCESS in 6m 33s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/8e434dee8af04fb9a562fcbceae01c85 : SUCCESS in 5m 36s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/6fad4239c8ec4d0ba8fa7da52ad5ce44 : SUCCESS in 7m 11s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/c22f161ba92e41529e9e79bcba26715d : SUCCESS in 47m 31s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/bd04da22e6c242759f439f1206ec75e6 : SUCCESS in 1h 34m 20s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/89ea2b941eaa4fb6bb25febfbd03261c : SUCCESS in 49m 07s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/d54823d2f17f414ebc261c357157189a : SUCCESS in 1h 38m 27s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/b78975e78c004e298b66ebc92d4f6567 : SUCCESS in 1h 27m 20s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/b70299526a024a4a85306a93e7c6596b : SUCCESS in 53m 50s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/b48cd0515327430e957ec4915883b173 : FAILURE in 25m 32s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/15052eca2bd64945a7783430adb1a50e : SUCCESS in 1h 04m 36s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/d400d2d3d61647e896d0278731e6b257 : SUCCESS in 40m 50s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/ce85d4c2bcda4784959570c77ba1fc15 : TIMED_OUT in 2h 34m 01s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/41460adc4bcc4e70ae66918a538fb3fa : TIMED_OUT in 3h 04m 58s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/b9554c06c7304856b8b9441ec3143c21 : SUCCESS in 2h 21m 08s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/018772df7c6641e4b6143c8efb32cf58 : TIMED_OUT in 2h 33m 58s (non-voting)\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard-zookeeper https://zuul.opendev.org/t/openstack/build/ad967999a33a4e9db389cb55af7ff044 : SUCCESS in 2h 23m 00s (non-voting)\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard-zookeeper https://zuul.opendev.org/t/openstack/build/fedfec3a157344b283e0dc042c57d155 : SUCCESS in 1h 04m 20s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"c89aa17eea879e5e02e7f239a8dddaefc4f47fc8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-03-16 15:45:41.000000000","message":"Patch Set 15: Commit message was updated.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":15},{"id":"bd93bd57d3cc5037729c3055c2a11f54c5038b8a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-03-16 15:46:06.000000000","message":"Patch Set 16: Commit message was updated.","accounts_in_message":[],"_revision_number":16},{"id":"2b02146995e03ba2c2d211bcd9887a3b76ab5b8c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-03-16 15:46:10.000000000","message":"Patch Set 17: Patch Set 16 was rebased","accounts_in_message":[],"_revision_number":17},{"id":"9b916ba6d2e3428af2f38962a7f0763e052a4829","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-16 18:53:06.000000000","message":"Patch Set 17: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4bac7fdee0b34581821d88d317131f3b\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/949bd498d8f049fd906130cb88fb7741 : SUCCESS in 15m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8c747ae87ca0419daefedfc9d9e30167 : SUCCESS in 6m 15s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/44859c5243e9408aa6a57f890511f015 : SUCCESS in 4m 27s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/30c5edc3285042158dc1aefe30a9e3ac : SUCCESS in 6m 28s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/89e46d2167334af6b586c26ff5e1d083 : SUCCESS in 6m 55s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ca3c378a482e4eaa9b371952696fc384 : SUCCESS in 5m 54s\n- octavia-tox-py313-tips https://zuul.opendev.org/t/openstack/build/7f34fcfaacb54bf9b8ff0bf34b498213 : SUCCESS in 3m 32s\n- octavia-tox-functional-py313-tips https://zuul.opendev.org/t/openstack/build/35fd28e35cc6453c8ff0a750e019aef2 : SUCCESS in 10m 06s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/d463965f03fc414abb69a60a2a0c77b8 : SUCCESS in 2m 46s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/203b880cdeee4c748ef624ea64f27159 : SUCCESS in 7m 42s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/a5e449162d964eb1a3ade94e87f8c264 : SUCCESS in 5m 20s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/1d47ccd5eefd43c6a8aae4303f45ded2 : SUCCESS in 1h 05m 10s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/be3a86fd72a64e0d9f5992d688db6eb6 : SUCCESS in 1h 41m 47s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/7172880a85d24a5795d6f75f72719f06 : SUCCESS in 1h 11m 14s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/e83876b0fa0a444c843f537b25e61f43 : SUCCESS in 1h 46m 50s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/c7f07ce04d0d4126af04c61e07d6cb74 : SUCCESS in 1h 31m 14s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/1fea0c742c9645a18262ca530fab4c15 : SUCCESS in 40m 46s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/b68f2e2397c44f5da624f8f76e1475ed : SUCCESS in 45m 50s\n- octavia-grenade-skip-level https://zuul.opendev.org/t/openstack/build/9daf19cb29244fef8d465768b86d39ba : SUCCESS in 45m 07s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/2cf081d6735e4703a844c61da15c6c0b : FAILURE in 36m 45s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/fc7a367ede754adca7185cc46185469a : FAILURE in 51m 47s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/9301d3e9624841528109110900a9e82a : TIMED_OUT in 3h 05m 02s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/b62f09360ae944d19d96cee00d2c5026 : SUCCESS in 1h 38m 29s (non-voting)\n- octavia-v2-dsvm-scenario-centos-10-stream-traffic-ops https://zuul.opendev.org/t/openstack/build/e19a12e4dc314d9495678463a7bb23db : FAILURE in 10m 04s (non-voting)\n- octavia-v2-dsvm-scenario-centos-10-stream-non-traffic-ops https://zuul.opendev.org/t/openstack/build/2731d07d1b0541469553f759fc3e7ee5 : FAILURE in 9m 12s (non-voting)\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard-zookeeper https://zuul.opendev.org/t/openstack/build/3b79cce8171a4041be564b706fd99add : SUCCESS in 1h 48m 54s (non-voting)\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard-zookeeper https://zuul.opendev.org/t/openstack/build/b4ec6327fb5440908f852f40b6a681fc : SUCCESS in 1h 25m 09s (non-voting)","accounts_in_message":[],"_revision_number":17},{"id":"fd940636770980ab62d82e6fe2d1e8d51c1aa694","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2026-03-17 07:30:45.000000000","message":"Patch Set 17: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":17}],"current_revision_number":17,"current_revision":"5d54a8b83a2d4fb95cfaf64196c8f08f63a21ddc","revisions":{"bc68450f90bdc32c5cebb9079a4701ee1add5aaf":{"kind":"REWORK","_number":1,"created":"2022-10-26 08:02:11.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/71/862671/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/71/862671/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/71/862671/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/71/862671/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/71/862671/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/71/862671/1"}}},"commit":{"parents":[{"commit":"344d6d76ce4481f30fe295114551243d61e7eea8","subject":"Fix prometheus-proxy service name in Red Hat-based distros","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/344d6d76ce4481f30fe295114551243d61e7eea8"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2022-10-26 07:29:44.000000000","tz":120},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2022-10-26 08:00:40.000000000","tz":120},"subject":"WIP Add zookeeper backend for jobboard in devstack","message":"WIP Add zookeeper backend for jobboard in devstack\n\nChange-Id: I6ca990b7080e35021e8d8dac922e9546eac3da2f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/bc68450f90bdc32c5cebb9079a4701ee1add5aaf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/bc68450f90bdc32c5cebb9079a4701ee1add5aaf"}]},"branch":"refs/heads/master"},"35515bc93bc88a80714e5b84ad03499c6de959b3":{"kind":"REWORK","_number":2,"created":"2022-10-26 13:12:50.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/71/862671/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/71/862671/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/71/862671/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/71/862671/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/71/862671/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/71/862671/2"}}},"commit":{"parents":[{"commit":"344d6d76ce4481f30fe295114551243d61e7eea8","subject":"Fix prometheus-proxy service name in Red Hat-based distros","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/344d6d76ce4481f30fe295114551243d61e7eea8"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2022-10-26 07:29:44.000000000","tz":120},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2022-10-26 13:12:44.000000000","tz":120},"subject":"WIP Add zookeeper backend for jobboard in devstack","message":"WIP Add zookeeper backend for jobboard in devstack\n\nChange-Id: I6ca990b7080e35021e8d8dac922e9546eac3da2f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/35515bc93bc88a80714e5b84ad03499c6de959b3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/35515bc93bc88a80714e5b84ad03499c6de959b3"}]},"branch":"refs/heads/master"},"d21855b3c42bb342ff92e9d2fa7f5caf20adb9c0":{"kind":"REWORK","_number":3,"created":"2022-10-26 16:45:28.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/71/862671/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/71/862671/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/71/862671/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/71/862671/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/71/862671/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/71/862671/3"}}},"commit":{"parents":[{"commit":"344d6d76ce4481f30fe295114551243d61e7eea8","subject":"Fix prometheus-proxy service name in Red Hat-based distros","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/344d6d76ce4481f30fe295114551243d61e7eea8"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2022-10-26 07:29:44.000000000","tz":120},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2022-10-26 16:31:00.000000000","tz":120},"subject":"WIP Add zookeeper backend for jobboard in devstack","message":"WIP Add zookeeper backend for jobboard in devstack\n\nChange-Id: I6ca990b7080e35021e8d8dac922e9546eac3da2f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d21855b3c42bb342ff92e9d2fa7f5caf20adb9c0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d21855b3c42bb342ff92e9d2fa7f5caf20adb9c0"}]},"branch":"refs/heads/master"},"8c448cf601606c9fe055aa7bcffa017473d75645":{"kind":"REWORK","_number":4,"created":"2022-10-26 17:55:28.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/71/862671/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/71/862671/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/71/862671/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/71/862671/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/71/862671/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/71/862671/4"}}},"commit":{"parents":[{"commit":"344d6d76ce4481f30fe295114551243d61e7eea8","subject":"Fix prometheus-proxy service name in Red Hat-based distros","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/344d6d76ce4481f30fe295114551243d61e7eea8"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2022-10-26 07:29:44.000000000","tz":120},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2022-10-26 17:55:21.000000000","tz":120},"subject":"WIP Add zookeeper backend for jobboard in devstack","message":"WIP Add zookeeper backend for jobboard in devstack\n\nChange-Id: I6ca990b7080e35021e8d8dac922e9546eac3da2f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/8c448cf601606c9fe055aa7bcffa017473d75645"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/8c448cf601606c9fe055aa7bcffa017473d75645"}]},"branch":"refs/heads/master"},"97f8a4ec54d9f9bb8ebe893e17c4723f9108e354":{"kind":"REWORK","_number":5,"created":"2022-10-27 06:02:57.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/71/862671/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/71/862671/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/71/862671/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/71/862671/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/71/862671/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/71/862671/5"}}},"commit":{"parents":[{"commit":"344d6d76ce4481f30fe295114551243d61e7eea8","subject":"Fix prometheus-proxy service name in Red Hat-based distros","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/344d6d76ce4481f30fe295114551243d61e7eea8"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2022-10-26 07:29:44.000000000","tz":120},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2022-10-27 05:36:48.000000000","tz":120},"subject":"Add zookeeper backend for jobboard in devstack","message":"Add zookeeper backend for jobboard in devstack\n\nZookeeper backend is enabled if OCTAVIA_JOBBOARD_BACKEND is set to\n\"zookeeper\", it also requires enabling the \"octavia-zookeeper\" service\nin devstack.\n\nThis commit also adds a octavia-v2-dsvm-scenario-amphora-v2-zookeeper\njob to the check pipeline, it replaces the\noctavia-v2-dsvm-scenario-amphora-v2 job which was a duplicate of the\noctavia-v2-dsvm-scenario job.\n\nChange-Id: I6ca990b7080e35021e8d8dac922e9546eac3da2f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/97f8a4ec54d9f9bb8ebe893e17c4723f9108e354"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/97f8a4ec54d9f9bb8ebe893e17c4723f9108e354"}]},"branch":"refs/heads/master"},"db60109ab6ba6b33ce4833e9da4d43221021d982":{"kind":"REWORK","_number":6,"created":"2022-10-27 12:20:06.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/71/862671/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/71/862671/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/71/862671/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/71/862671/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/71/862671/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/71/862671/6"}}},"commit":{"parents":[{"commit":"344d6d76ce4481f30fe295114551243d61e7eea8","subject":"Fix prometheus-proxy service name in Red Hat-based distros","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/344d6d76ce4481f30fe295114551243d61e7eea8"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2022-10-26 07:29:44.000000000","tz":120},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2022-10-27 12:19:47.000000000","tz":120},"subject":"Add zookeeper backend for jobboard in devstack","message":"Add zookeeper backend for jobboard in devstack\n\nZookeeper backend is enabled if OCTAVIA_JOBBOARD_BACKEND is set to\n\"zookeeper\", it also requires enabling the \"octavia-zookeeper\" service\nin devstack.\n\nThis commit also adds a octavia-v2-dsvm-scenario-amphora-v2-zookeeper\njob to the check pipeline, it replaces the\noctavia-v2-dsvm-scenario-amphora-v2 job which was a duplicate of the\noctavia-v2-dsvm-scenario job.\n\nChange-Id: I6ca990b7080e35021e8d8dac922e9546eac3da2f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/db60109ab6ba6b33ce4833e9da4d43221021d982"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/db60109ab6ba6b33ce4833e9da4d43221021d982"}]},"branch":"refs/heads/master"},"288c464e6cb954fe50a1c579e162e295bc875bed":{"kind":"REWORK","_number":7,"created":"2023-01-12 11:19:16.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/71/862671/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/71/862671/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/71/862671/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/71/862671/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/71/862671/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/71/862671/7"}}},"commit":{"parents":[{"commit":"344d6d76ce4481f30fe295114551243d61e7eea8","subject":"Fix prometheus-proxy service name in Red Hat-based distros","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/344d6d76ce4481f30fe295114551243d61e7eea8"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2022-10-26 07:29:44.000000000","tz":120},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-01-12 11:19:08.000000000","tz":60},"subject":"Add zookeeper backend for jobboard in devstack","message":"Add zookeeper backend for jobboard in devstack\n\nZookeeper backend is enabled if OCTAVIA_JOBBOARD_BACKEND is set to\n\"zookeeper\", it also requires enabling the \"octavia-zookeeper\" service\nin devstack.\n\nThis commit also adds a octavia-v2-dsvm-scenario-amphora-v2-zookeeper\njob to the check pipeline, it replaces the\noctavia-v2-dsvm-scenario-amphora-v2 job which was a duplicate of the\noctavia-v2-dsvm-scenario job.\n\nChange-Id: I6ca990b7080e35021e8d8dac922e9546eac3da2f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/288c464e6cb954fe50a1c579e162e295bc875bed"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/288c464e6cb954fe50a1c579e162e295bc875bed"}]},"branch":"refs/heads/master"},"11e6ea1c0d64cc4d8ca8588551b621dbba3fddb3":{"kind":"REWORK","_number":8,"created":"2023-01-12 12:25:10.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/71/862671/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/71/862671/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/71/862671/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/71/862671/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/71/862671/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/71/862671/8"}}},"commit":{"parents":[{"commit":"344d6d76ce4481f30fe295114551243d61e7eea8","subject":"Fix prometheus-proxy service name in Red Hat-based distros","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/344d6d76ce4481f30fe295114551243d61e7eea8"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2022-10-26 07:29:44.000000000","tz":120},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-01-12 12:23:44.000000000","tz":60},"subject":"Add zookeeper backend for jobboard in devstack","message":"Add zookeeper backend for jobboard in devstack\n\nZookeeper backend is enabled if OCTAVIA_JOBBOARD_BACKEND is set to\n\"zookeeper\", it also requires enabling the \"octavia-zookeeper\" service\nin devstack.\n\nThis commit also adds a octavia-v2-dsvm-scenario-amphora-v2-zookeeper\njob to the check pipeline, it replaces the\noctavia-v2-dsvm-scenario-amphora-v2 job which was a duplicate of the\noctavia-v2-dsvm-scenario job.\n\nChange-Id: I6ca990b7080e35021e8d8dac922e9546eac3da2f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/11e6ea1c0d64cc4d8ca8588551b621dbba3fddb3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/11e6ea1c0d64cc4d8ca8588551b621dbba3fddb3"}]},"branch":"refs/heads/master"},"be8a61191ab5213cb0a8f1b9eefe61fba71ff798":{"kind":"REWORK","_number":9,"created":"2023-10-16 09:43:37.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/71/862671/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/71/862671/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/71/862671/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/71/862671/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/71/862671/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/71/862671/9"}}},"commit":{"parents":[{"commit":"860d2f6aadfba9cef2c695d072dd48e29a7e85c8","subject":"Merge \"Fix incorrect masquerade rules in multivip LBs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/860d2f6aadfba9cef2c695d072dd48e29a7e85c8"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2022-10-26 07:29:44.000000000","tz":120},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-10-16 09:43:11.000000000","tz":120},"subject":"Add zookeeper backend for jobboard in devstack","message":"Add zookeeper backend for jobboard in devstack\n\nZookeeper backend is enabled if OCTAVIA_JOBBOARD_BACKEND is set to\n\"zookeeper\", it also requires enabling the \"octavia-zookeeper\" service\nin devstack.\n\nThis commit also adds 2 jobs to the check pipeline, they replace the\noctavia-v2-dsvm-scenario-amphora-v2 job which was a duplicate of the\noctavia-v2-dsvm-scenario job.\n\nChange-Id: I6ca990b7080e35021e8d8dac922e9546eac3da2f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/be8a61191ab5213cb0a8f1b9eefe61fba71ff798"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/be8a61191ab5213cb0a8f1b9eefe61fba71ff798"}]},"branch":"refs/heads/master"},"e2cf996059c284acc84ed26d1b5648290921952a":{"kind":"REWORK","_number":10,"created":"2024-09-25 16:44:38.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/71/862671/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/71/862671/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/71/862671/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/71/862671/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/71/862671/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/71/862671/10"}}},"commit":{"parents":[{"commit":"3c6f10f189eecb0020120819c89ec658085f4953","subject":"Fix pep8 with pylint 3.3.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/3c6f10f189eecb0020120819c89ec658085f4953"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2022-10-26 07:29:44.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2024-09-25 16:44:29.000000000","tz":120},"subject":"Add zookeeper backend for jobboard in devstack","message":"Add zookeeper backend for jobboard in devstack\n\nZookeeper backend is enabled if OCTAVIA_JOBBOARD_BACKEND is set to\n\"zookeeper\", it also requires enabling the \"octavia-zookeeper\" service\nin devstack.\n\nThis commit also adds 2 jobs to the check pipeline, they replace the\noctavia-v2-dsvm-scenario-amphora-v2 job which was a duplicate of the\noctavia-v2-dsvm-scenario job.\n\nChange-Id: I6ca990b7080e35021e8d8dac922e9546eac3da2f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e2cf996059c284acc84ed26d1b5648290921952a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e2cf996059c284acc84ed26d1b5648290921952a"}]},"branch":"refs/heads/master"},"0dbbdcdec7a4ac301c4047d8621c3bf07ea5dfa0":{"kind":"REWORK","_number":11,"created":"2024-09-26 06:00:58.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/71/862671/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/71/862671/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/71/862671/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/71/862671/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/71/862671/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/71/862671/11"}}},"commit":{"parents":[{"commit":"3c6f10f189eecb0020120819c89ec658085f4953","subject":"Fix pep8 with pylint 3.3.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/3c6f10f189eecb0020120819c89ec658085f4953"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2022-10-26 07:29:44.000000000","tz":120},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2024-09-26 06:00:42.000000000","tz":120},"subject":"Add zookeeper backend for jobboard in devstack","message":"Add zookeeper backend for jobboard in devstack\n\nZookeeper backend is enabled if OCTAVIA_JOBBOARD_BACKEND is set to\n\"zookeeper\", it also requires enabling the \"octavia-zookeeper\" service\nin devstack.\n\nThis commit also adds 2 jobs to the check pipeline, they replace the\noctavia-v2-dsvm-scenario-amphora-v2 job which was a duplicate of the\noctavia-v2-dsvm-scenario job.\n\nChange-Id: I6ca990b7080e35021e8d8dac922e9546eac3da2f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/0dbbdcdec7a4ac301c4047d8621c3bf07ea5dfa0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/0dbbdcdec7a4ac301c4047d8621c3bf07ea5dfa0"}]},"branch":"refs/heads/master"},"9e84ab56d7fa70e583eabf4e3659e1217e4a4aeb":{"kind":"REWORK","_number":12,"created":"2024-09-26 09:12:09.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/71/862671/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/71/862671/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/71/862671/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/71/862671/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/71/862671/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/71/862671/12"}}},"commit":{"parents":[{"commit":"3c6f10f189eecb0020120819c89ec658085f4953","subject":"Fix pep8 with pylint 3.3.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/3c6f10f189eecb0020120819c89ec658085f4953"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2022-10-26 07:29:44.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2024-09-26 09:12:00.000000000","tz":120},"subject":"[DNM] Test zookeeper backend for 2023.1","message":"[DNM] Test zookeeper backend for 2023.1\n\nCherry-Picked from https://review.opendev.org/c/openstack/octavia/+/862671\nfor checking on Zookeeper connection issues on 2023.1\n\nChange-Id: I6ca990b7080e35021e8d8dac922e9546eac3da2f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/9e84ab56d7fa70e583eabf4e3659e1217e4a4aeb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/9e84ab56d7fa70e583eabf4e3659e1217e4a4aeb"}]},"branch":"refs/heads/master"},"2cce2e44efa1cbc56954a180a6463e403ba55450":{"kind":"REWORK","_number":13,"created":"2024-09-26 09:14:58.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/71/862671/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/71/862671/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/71/862671/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/71/862671/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/71/862671/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/71/862671/13"}}},"commit":{"parents":[{"commit":"3c6f10f189eecb0020120819c89ec658085f4953","subject":"Fix pep8 with pylint 3.3.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/3c6f10f189eecb0020120819c89ec658085f4953"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2022-10-26 07:29:44.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2024-09-26 09:14:50.000000000","tz":120},"subject":"Add zookeeper backend for jobboard in devstack","message":"Add zookeeper backend for jobboard in devstack\n\nZookeeper backend is enabled if OCTAVIA_JOBBOARD_BACKEND is set to\n\"zookeeper\", it also requires enabling the \"octavia-zookeeper\" service\nin devstack.\n\nThis commit also adds 2 jobs to the check pipeline, they replace the\noctavia-v2-dsvm-scenario-amphora-v2 job which was a duplicate of the\noctavia-v2-dsvm-scenario job.\n\nChange-Id: I6ca990b7080e35021e8d8dac922e9546eac3da2f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/2cce2e44efa1cbc56954a180a6463e403ba55450"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/2cce2e44efa1cbc56954a180a6463e403ba55450"}]},"branch":"refs/heads/master"},"cdd9aafcd89eb60867a364499c265c39fa6ef764":{"kind":"TRIVIAL_REBASE","_number":14,"created":"2024-10-01 09:34:50.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/71/862671/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/71/862671/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/71/862671/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/71/862671/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/71/862671/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/71/862671/14"}}},"commit":{"parents":[{"commit":"b65d730316341b21c296583e2cc32090e300a16a","subject":"Merge \"Fixed retry functions with tenacity 8.5.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b65d730316341b21c296583e2cc32090e300a16a"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2022-10-26 07:29:44.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2024-10-01 09:34:50.000000000","tz":0},"subject":"Add zookeeper backend for jobboard in devstack","message":"Add zookeeper backend for jobboard in devstack\n\nZookeeper backend is enabled if OCTAVIA_JOBBOARD_BACKEND is set to\n\"zookeeper\", it also requires enabling the \"octavia-zookeeper\" service\nin devstack.\n\nThis commit also adds 2 jobs to the check pipeline, they replace the\noctavia-v2-dsvm-scenario-amphora-v2 job which was a duplicate of the\noctavia-v2-dsvm-scenario job.\n\nChange-Id: I6ca990b7080e35021e8d8dac922e9546eac3da2f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/cdd9aafcd89eb60867a364499c265c39fa6ef764"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/cdd9aafcd89eb60867a364499c265c39fa6ef764"}]},"branch":"refs/heads/master","description":"Rebase"},"4709d9065de63950c04d391569fa48ca4c0d193a":{"kind":"NO_CODE_CHANGE","_number":15,"created":"2026-03-16 15:45:41.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/71/862671/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/71/862671/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/71/862671/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/71/862671/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/71/862671/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/71/862671/15"}}},"commit":{"parents":[{"commit":"b65d730316341b21c296583e2cc32090e300a16a","subject":"Merge \"Fixed retry functions with tenacity 8.5.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b65d730316341b21c296583e2cc32090e300a16a"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2022-10-26 07:29:44.000000000","tz":120},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-03-16 15:45:41.000000000","tz":0},"subject":"Add zookeeper backend for jobboard in devstack","message":"Add zookeeper backend for jobboard in devstack\n\nZookeeper backend is enabled if OCTAVIA_JOBBOARD_BACKEND is set to\n\"zookeeper\", it also requires enabling the \"octavia-zookeeper\" service\nin devstack.\n\nThis commit also adds 2 jobs to the check pipeline, they replace the\noctavia-v2-dsvm-scenario-amphora-v2 job which was a duplicate of the\noctavia-v2-dsvm-scenario job.\n\nChange-Id: I6ca990b7080e35021e8d8dac922e9546eac3da2f\nSigned-off-by: Gregory Thiemonge \u003cgthiemon@redhat.com\u003e\nSigned-off-by: Dmitriy Rabotyagov \u003cnoonedeadpunk@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/4709d9065de63950c04d391569fa48ca4c0d193a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/4709d9065de63950c04d391569fa48ca4c0d193a"}]},"branch":"refs/heads/master","description":"Edit commit message"},"301c4d5b5467dc6566a927c87abe4c1f1a3a3311":{"kind":"NO_CODE_CHANGE","_number":16,"created":"2026-03-16 15:46:06.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/71/862671/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/71/862671/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/71/862671/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/71/862671/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/71/862671/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/71/862671/16"}}},"commit":{"parents":[{"commit":"b65d730316341b21c296583e2cc32090e300a16a","subject":"Merge \"Fixed retry functions with tenacity 8.5.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b65d730316341b21c296583e2cc32090e300a16a"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2022-10-26 07:29:44.000000000","tz":120},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-03-16 15:46:06.000000000","tz":0},"subject":"Add zookeeper backend for jobboard in devstack","message":"Add zookeeper backend for jobboard in devstack\n\nZookeeper backend is enabled if OCTAVIA_JOBBOARD_BACKEND is set to\n\"zookeeper\", it also requires enabling the \"octavia-zookeeper\" service\nin devstack.\n\nThis commit also adds 2 jobs to the check pipeline, they replace the\noctavia-v2-dsvm-scenario-amphora-v2 job which was a duplicate of the\noctavia-v2-dsvm-scenario job.\n\nChange-Id: I6ca990b7080e35021e8d8dac922e9546eac3da2f\nCo-authored-by: Dmitriy Rabotyagov \u003cnoonedeadpunk@gmail.com\u003e\nSigned-off-by: Gregory Thiemonge \u003cgthiemon@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/301c4d5b5467dc6566a927c87abe4c1f1a3a3311"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/301c4d5b5467dc6566a927c87abe4c1f1a3a3311"}]},"branch":"refs/heads/master","description":"Edit commit message"},"5d54a8b83a2d4fb95cfaf64196c8f08f63a21ddc":{"kind":"TRIVIAL_REBASE","_number":17,"created":"2026-03-16 15:46:10.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/71/862671/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/71/862671/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/71/862671/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/71/862671/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/71/862671/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/71/862671/17"}}},"commit":{"parents":[{"commit":"79cf1f787fa9c03adba8575dc94324b97963f84e","subject":"Update master for stable/2026.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/79cf1f787fa9c03adba8575dc94324b97963f84e"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2022-10-26 07:29:44.000000000","tz":120},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-03-16 15:46:10.000000000","tz":0},"subject":"Add zookeeper backend for jobboard in devstack","message":"Add zookeeper backend for jobboard in devstack\n\nZookeeper backend is enabled if OCTAVIA_JOBBOARD_BACKEND is set to\n\"zookeeper\", it also requires enabling the \"octavia-zookeeper\" service\nin devstack.\n\nThis commit also adds 2 jobs to the check pipeline, they replace the\noctavia-v2-dsvm-scenario-amphora-v2 job which was a duplicate of the\noctavia-v2-dsvm-scenario job.\n\nChange-Id: I6ca990b7080e35021e8d8dac922e9546eac3da2f\nCo-authored-by: Dmitriy Rabotyagov \u003cnoonedeadpunk@gmail.com\u003e\nSigned-off-by: Gregory Thiemonge \u003cgthiemon@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/5d54a8b83a2d4fb95cfaf64196c8f08f63a21ddc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/5d54a8b83a2d4fb95cfaf64196c8f08f63a21ddc"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":["label:Workflow\u003dMIN"],"failing_atoms":["label:Workflow\u003dMAX"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
