)]}'
{"id":"openstack%2Foctavia~863544","triplet_id":"openstack%2Foctavia~master~I304e06d1d187070b260347926f3b516072827807","project":"openstack/octavia","branch":"master","topic":"update-vip-default-arg","attention_set":{},"removed_from_attention_set":{"34429":{"account":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"last_update":"2026-07-31 08:08:13.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I304e06d1d187070b260347926f3b516072827807","subject":"Align signature of AbstractNetworkDriver.update_vip method","status":"MERGED","created":"2022-11-03 17:15:09.000000000","updated":"2026-07-31 08:09:24.000000000","submitted":"2026-07-31 08:08:13.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"863544-update-vip-default-arg","meta_rev_id":"68f2626207b3c9ece739f70be81e470f2b758dcc","_number":863544,"virtual_id_number":863544,"owner":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":0,"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-07-31 08:08:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"recommended":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"all":[{"value":1,"date":"2022-12-13 13:36:33.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":1,"date":"2023-02-27 02:45:02.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":2,"date":"2023-02-15 08:20:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"all":[{"value":0,"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":0,"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":1,"date":"2026-07-31 06:38:25.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":0,"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-11-03 19:22:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-12-13 13:36:33.000000000","updated_by":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"reviewer":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"state":"REVIEWER"},{"updated":"2023-02-15 08:20:03.000000000","updated_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"REVIEWER"},{"updated":"2023-02-27 02:45:02.000000000","updated_by":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"reviewer":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"state":"REVIEWER"}],"messages":[{"id":"a4611634d59dcf9afb17c67821476d1b576e523a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-11-03 17:15:09.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4155e8ca901b56fe1778f0fd3e6fab2686f08e5c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-03 19:22:29.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/42759afe4b5d4c11bb9a3ee17159a8cb\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e2f054ffd2584f55a45c0539a8dc8905 : SUCCESS in 7m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/188b240277bf4800b0679b3acd2cc9af : SUCCESS in 5m 41s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/827dc935288d48018cef8ea249a7ffc8 : SUCCESS in 3m 59s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/aca5fc9020e44ef1b997c3a3d90f22dc : SUCCESS in 4m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d9495172f4264e3cb2b5a6f5172e852c : SUCCESS in 8m 54s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/b67fb03042d246c9814086471da133cd : SUCCESS in 4m 45s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/7f2568e387c54734acaf28b3f9ae875e : SUCCESS in 6m 03s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/30a4f58a5efe403d884bfae77dc361b3 : SUCCESS in 4m 05s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/0a6eb7f3c419479ba9f41c642dcedef0 : SUCCESS in 7m 00s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/0bf150b17deb45f780b8bbb0929cbd6e : SUCCESS in 7m 09s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/43fa18d8ea054f9994f1222a7cfcd679 : SUCCESS in 5m 46s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/5c19b6a31c1d4475b047e342047fb1da : SUCCESS in 1h 34m 16s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/70faaf2f0e8c425cbaa4ee3022cabbec : SUCCESS in 1h 39m 46s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/edd0f16d92f540c3bafb5ba54ccfc828 : SUCCESS in 34m 50s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/1c2fcbfca9be421ab174b0d3ff090fda : SUCCESS in 47m 47s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/448ed39ab7b24caaa9a3d7a0f1415bb7 : SUCCESS in 1h 10m 03s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/9b4121fa4a7744ceb1557023a5c97576 : SUCCESS in 29m 40s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/6a3cc8f026e64fc29e7857fb7dc5a7d0 : SUCCESS in 1h 41m 15s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/53d35160eaee449fa3258037cd90e746 : SUCCESS in 1h 40m 33s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/1f00c76b304b4b8dac7dda4b82826811 : SUCCESS in 1h 48m 03s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/4ec96d0b2e854483a4fcb226683534e9 : SUCCESS in 1h 54m 17s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/6e092f6dd15147fb90ecd535bb38211e : SUCCESS in 1h 44m 33s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/e3623e3a201e4d9fa60bfaed68752c36 : SUCCESS in 2h 04m 55s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"c8e4b02c729c7c45d346242ea53bb240a1c03dee","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2022-12-13 13:36:33.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"046d6bbb959c53788b991d4b283feb9a528a2767","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-02-15 08:20:03.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"261bd17847d5d9b8706a2a8c8e8445faa18f9ff2","author":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"date":"2023-02-27 02:45:02.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"6071d9c34c87cbc0a7123aade9148101d6ffdcad","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2026-07-31 06:38:25.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"2852ce431131c5d254737f36ab52992a53f08319","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-31 06:38:59.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"f4fccc09c12000b7f2198152751d1053676b9c62","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-31 08:08:13.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ef0159db0bad4d29a5fe7a74a3840980\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7f009a9bbc1543dba160dc2ed828fe4e : SUCCESS in 3m 40s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/9896ef4849ac4f9ead214131f729ae98 : SUCCESS in 4m 40s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/0c51508ce6d04bd082672035bff29e3b : SUCCESS in 3m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b113317b7e9842cfaae0e0778e25230e : SUCCESS in 12m 23s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/673012bf52474ef4947e4503b80edeef : SUCCESS in 8m 31s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/1070116027ef44659d1fbf2bb871adc6 : SUCCESS in 3m 40s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/ac7bed1cfd764695a8bac90be25609c4 : SUCCESS in 6m 09s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/aeb071dedc7d445196e86fb051231b82 : SUCCESS in 1h 11m 57s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/56521b41aa114139b06f9892a59cb2cb : SUCCESS in 1h 28m 34s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/8a233ebe64854e77b2a5a500c18d2bba : SUCCESS in 1h 00m 18s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/7418143f279448beacf34fb492e34c40 : SUCCESS in 35m 46s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/b26174797f454f1ab5802c2e6b7fbf57 : SUCCESS in 48m 21s\n- octavia-grenade-skip-level https://zuul.opendev.org/t/openstack/build/f94523c090034def8a6b073d843d7e58 : SUCCESS in 42m 49s","accounts_in_message":[],"_revision_number":1},{"id":"561fbe985d17bad5f5bc72bc54f4844d63db1fdb","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-31 08:08:13.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"68f2626207b3c9ece739f70be81e470f2b758dcc","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-31 08:09:24.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2ae414cbad584c77bfd95fd4fd766119\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e7197c5320b54db28352b5cad46da650 : SUCCESS in 54s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a054a91555735a7ef9a77d36aff2d37c74fa9204","revisions":{"a054a91555735a7ef9a77d36aff2d37c74fa9204":{"kind":"REWORK","_number":1,"created":"2022-11-03 17:15:09.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/44/863544/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/44/863544/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/44/863544/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/44/863544/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/44/863544/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/44/863544/1"}}},"commit":{"parents":[{"commit":"344d6d76ce4481f30fe295114551243d61e7eea8","subject":"Fix prometheus-proxy service name in Red Hat-based distros","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/344d6d76ce4481f30fe295114551243d61e7eea8"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-11-03 17:10:14.000000000","tz":60},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-11-03 17:10:14.000000000","tz":60},"subject":"Align signature of AbstractNetworkDriver.update_vip method","message":"Align signature of AbstractNetworkDriver.update_vip method\n\nThe update_vip() method gets called in network_tasks.py without setting\nthe for_delete argument. That works because the actual network drivers\nimplement that function using a default for the for_delete argument.\nThis patch updates the signature of the respective base class in case a\ndriver does not overwrite this method.\n\nChange-Id: I304e06d1d187070b260347926f3b516072827807\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a054a91555735a7ef9a77d36aff2d37c74fa9204"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a054a91555735a7ef9a77d36aff2d37c74fa9204"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
