)]}'
{"id":"openstack%2Foctavia~864896","triplet_id":"openstack%2Foctavia~stable%2Fussuri~I1e3551b1a7a09d2a38106eba21fb5536114883f7","project":"openstack/octavia","branch":"stable/ussuri","topic":"ussuri-test","attention_set":{},"removed_from_attention_set":{"34429":{"account":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"last_update":"2022-12-12 08:22:58.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I1e3551b1a7a09d2a38106eba21fb5536114883f7","subject":"DNM Test potential fix for failing barbican test","status":"ABANDONED","created":"2022-11-17 13:23:10.000000000","updated":"2022-12-12 08:22:58.000000000","total_comment_count":2,"unresolved_comment_count":0,"work_in_progress":true,"has_review_started":false,"meta_rev_id":"b0276a438541139d610b6c0613f9c54a507c524e","_number":864896,"virtual_id_number":864896,"owner":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2022-11-17 23:29:32.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}]},"pending_reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}]},"reviewer_updates":[{"updated":"2022-11-17 15:37:04.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-11-18 10:53:20.000000000","updated_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"CC"}],"messages":[{"id":"e15f3828c20496a2d070ba792ffd3e98c34e4ef0","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-11-17 13:23:10.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e4e314dfbd0db95a029da8fd1df53cc3c230185b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-17 15:37:04.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/a0414aa758ab443e8c9525ff19a0152f\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ee8fb94afac64374b9ec2ab539286582 : SUCCESS in 9m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/03534810c4a5478296cc42931148554d : FAILURE in 4m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/df3d3ee5688e41b99799d04bba9934cb : SUCCESS in 4m 52s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/2ea6079da2ae48799b4d457ff2b52a06 : SUCCESS in 5m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/da058700f6c1425981a0ba2c67fa7cc5 : SUCCESS in 12m 12s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/1210d5f4932c49d9a8407e5bdaf45744 : SUCCESS in 5m 11s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/333f2140503648c8b9dc70113f3403ca : SUCCESS in 8m 18s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/b077963586634056bb62a2873523b470 : SUCCESS in 6m 34s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/1640a43dc7374a468f9e7c49e60b57c5 : SUCCESS in 50m 58s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/39938898dd9f44f29bd717a617e1d37e : SUCCESS in 1h 44m 28s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/a143c7f69e1c44eabfb764799a2e9219 : FAILURE in 38m 21s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/f7f8674579224342aea07042586cd72e : SUCCESS in 33m 43s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/dec9e658f46f4915a1612d3c1cb0e663 : SUCCESS in 44m 20s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/098aefa988084476b3686617c9f8d334 : SUCCESS in 1h 32m 37s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/b1a77996ccd340f18c2371709f502ab0 : RETRY_LIMIT in 5m 43s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/033d183693704f308f6f66c958d89153 : SUCCESS in 2h 02m 27s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"0b59ba132562edea3292f3965ab067d1c6743bbf","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-11-17 16:30:27.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"0c15d9665845966f2b9a41a70522e0dfe967ec77","tag":"autogenerated:gerrit:abandon","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-11-17 17:15:53.000000000","message":"Abandoned\n\ntest succeeded","accounts_in_message":[],"_revision_number":1},{"id":"3510dc59008aaec0fed4930313ad72df53447ca7","tag":"autogenerated:gerrit:restore","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-11-17 17:18:43.000000000","message":"Restored","accounts_in_message":[],"_revision_number":1},{"id":"cafe3556c6b821199fe4312a44740c82b0fb4b2b","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-11-17 17:30:56.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"c04b9bf53f4eb7f430fbdf17e8b1f78128e5a278","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-17 19:05:45.000000000","message":"Patch Set 2: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/82dcd3a1e3f24b81b565bd150b2fd474\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3da7bf39702e4a48b6927fc7da040aab : SUCCESS in 9m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e8ed244376814b92b9b71b8cede71673 : FAILURE in 5m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/833524feab014dc6befa51b071450ffc : SUCCESS in 5m 49s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/3561e61f451644ceb9b4aa5943a08c58 : SUCCESS in 4m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/50f4de033fbd4cccbef65283b6c6819b : SUCCESS in 11m 48s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/caa1b003bb7747a290f8314fad4440f4 : SUCCESS in 4m 34s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/fd5e2977969d4508a792096d40ac392a : SUCCESS in 7m 17s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/8f1a8ba433334955b395465175312164 : SUCCESS in 9m 55s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/3fcc3b45244343408cf8ea026b7ba440 : SUCCESS in 43m 21s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/01f4d214c9144aa692ebed3accdb2158 : SUCCESS in 1h 32m 38s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/10f38b2df71745ae9fb57619019f0403 : SUCCESS in 34m 15s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/a38a5f7ca8dd413e820bb15f75cb345f : SUCCESS in 31m 03s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/ecdf70606ea14757943620b99b6a0bc4 : SUCCESS in 32m 22s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/c3f73567183c49a9be56a19fdf4e6ccb : SUCCESS in 1h 33m 14s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/0bec07e2061e450195ead138ea01b449 : RETRY_LIMIT in 5m 39s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/48ccd5444f0340c283db9c05d2aae76d : SUCCESS in 1h 31m 50s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"f71af0f60881cb49346f8ee0468a2a2fa9500f48","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-11-17 21:02:53.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"90989ebb24c67191379a8ea68b8a6252ee5d70c4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-17 23:29:32.000000000","message":"Patch Set 3: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/8e31475892d145fc80e315d7c7487ee9\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3f1b8e1b97a64cad9f174caf5237cd66 : SUCCESS in 10m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bd56327250a440e0b48bfd3d7107e8c6 : FAILURE in 4m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2098f2ca157142618dceee9326e64f8f : SUCCESS in 5m 18s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/34f9b93a3b85435081111caf3af1b138 : SUCCESS in 5m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b11e3fb9a5864219bc1a8f399c7a616a : SUCCESS in 10m 51s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/e078767213614d6ea5416a7ed8e51007 : SUCCESS in 6m 12s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/c69600d573124c9cbd282c50e4998b50 : SUCCESS in 8m 10s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/780ca379102f4f67a6e25e743b7a04f7 : SUCCESS in 7m 51s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/1afa072dbf46431aae00ac4bf3b5da7e : SUCCESS in 1h 08m 02s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/297c119c1c084bb49087068c7cdd53f9 : FAILURE in 1h 50m 38s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/6010331f3cab49b1a0974bb39db432bd : SUCCESS in 47m 13s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/ac137e389ed4433792d87d9fad0d21fd : SUCCESS in 37m 54s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/2e09f29eb0934d1d89af020026a27b9c : SUCCESS in 43m 09s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/8d3db368e30a4664ac110022d7053c25 : SUCCESS in 2h 05m 11s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/e39a6951e6d24cc1a2de1816c683a0ae : RETRY_LIMIT in 5m 28s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/b4f2a2882b084c46a1004041b581efe3 : SUCCESS in 2h 19m 13s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"70b156179220c8d7a582ccfbf723185009eb85ec","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2022-11-18 10:53:20.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"e3b00a12b653456ffe84cef73e08a29051fb62de","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-18 12:39:44.000000000","message":"Patch Set 3:\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/e3bfe7f6cae84284af5f7d6fa87efc76\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fde16005fc9741938f80175bd9238f00 : SUCCESS in 7m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/05ece875da234c07b0d4b6062855ebe7 : FAILURE in 5m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9d8d0025f5da434d97962505a654fcb1 : SUCCESS in 5m 49s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/79193d93d2c640e2960ff4d5a843c22c : SUCCESS in 5m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/312c311593b34806a0fde9f65d547c3f : SUCCESS in 10m 24s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/65bf74e052b64b5b8e24d6ac7e5b800a : SUCCESS in 7m 18s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/62f797d74a0747ee9ea81c7746877286 : SUCCESS in 7m 49s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/32eabc95e4c142f9b3f267d323856ee6 : SUCCESS in 8m 20s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/b9af25724d254487a321a0a4698f689d : SUCCESS in 54m 18s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/8a5c3830da394fcfb721440edad376d7 : SUCCESS in 1h 36m 23s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/f38ce9d03c8f4ab0a4b5434cc5321223 : SUCCESS in 43m 07s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/2adad7dd1ab44dadad1186b637c4ee49 : SUCCESS in 42m 34s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/9f00cf92d3624da09ac21e896c04682a : SUCCESS in 43m 22s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/3ed1731d281d4cdbb77cb544b27d67cd : SUCCESS in 1h 43m 24s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/d85825aa3bfe42aa8a64f2f7269ea3ed : RETRY_LIMIT in 4m 57s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/94822d28fdfc46a38e5b21bf4daac3a6 : SUCCESS in 1h 30m 49s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"b0276a438541139d610b6c0613f9c54a507c524e","tag":"autogenerated:gerrit:abandon","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-12-12 08:22:58.000000000","message":"Abandoned\n\nfix has been merged.","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"a1de3529dace24520d4824dcf73c1974fde45f3d","revisions":{"af7a95b01b3d32698ebac2a79ebc982fea3d894f":{"kind":"REWORK","_number":1,"created":"2022-11-17 13:23:10.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/96/864896/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/96/864896/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/96/864896/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/96/864896/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/96/864896/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/96/864896/1"}}},"commit":{"parents":[{"commit":"e79cca355208535204b8e9f50f543ccda53e3aed","subject":"Merge \"Fix PortNotFound exception when updating a LB after a failover\" into stable/ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e79cca355208535204b8e9f50f543ccda53e3aed"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-11-17 13:22:09.000000000","tz":60},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-11-17 13:22:09.000000000","tz":60},"subject":"Test potential fix for failing barbican test","message":"Test potential fix for failing barbican test\n\nDepends-on: I3da638eb97939186b923a5854f2ea260b61953e8\nChange-Id: I1e3551b1a7a09d2a38106eba21fb5536114883f7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/af7a95b01b3d32698ebac2a79ebc982fea3d894f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/af7a95b01b3d32698ebac2a79ebc982fea3d894f"}]},"branch":"refs/heads/stable/ussuri"},"6f73d7f6e5f6bb9ea510fce3fc55cf1b566c3d83":{"kind":"REWORK","_number":2,"created":"2022-11-17 17:30:56.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/96/864896/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/96/864896/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/96/864896/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/96/864896/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/96/864896/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/96/864896/2"}}},"commit":{"parents":[{"commit":"e79cca355208535204b8e9f50f543ccda53e3aed","subject":"Merge \"Fix PortNotFound exception when updating a LB after a failover\" into stable/ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e79cca355208535204b8e9f50f543ccda53e3aed"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-11-17 13:22:09.000000000","tz":60},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-11-17 17:30:41.000000000","tz":60},"subject":"Test potential fix for failing barbican test","message":"Test potential fix for failing barbican test\n\nDepends-on: I3da638eb97939186b923a5854f2ea260b61953e8\nChange-Id: I1e3551b1a7a09d2a38106eba21fb5536114883f7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/6f73d7f6e5f6bb9ea510fce3fc55cf1b566c3d83"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/6f73d7f6e5f6bb9ea510fce3fc55cf1b566c3d83"}]},"branch":"refs/heads/stable/ussuri"},"a1de3529dace24520d4824dcf73c1974fde45f3d":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2022-11-17 21:02:53.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/96/864896/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/96/864896/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/96/864896/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/96/864896/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/96/864896/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/96/864896/3"}}},"commit":{"parents":[{"commit":"e79cca355208535204b8e9f50f543ccda53e3aed","subject":"Merge \"Fix PortNotFound exception when updating a LB after a failover\" into stable/ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e79cca355208535204b8e9f50f543ccda53e3aed"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-11-17 13:22:09.000000000","tz":60},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-11-17 21:02:53.000000000","tz":0},"subject":"DNM Test potential fix for failing barbican test","message":"DNM Test potential fix for failing barbican test\n\nDepends-on: I3da638eb97939186b923a5854f2ea260b61953e8\nChange-Id: I1e3551b1a7a09d2a38106eba21fb5536114883f7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a1de3529dace24520d4824dcf73c1974fde45f3d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a1de3529dace24520d4824dcf73c1974fde45f3d"}]},"branch":"refs/heads/stable/ussuri","description":"Edit commit message"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
