)]}'
{"id":"openstack%2Foctavia~866327","triplet_id":"openstack%2Foctavia~master~I686cfdef78de927fa4bc1921c15e8d5853fd2ef9","project":"openstack/octavia","branch":"master","topic":"remove-neutronclient","attention_set":{},"removed_from_attention_set":{"31664":{"account":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"last_update":"2023-05-03 18:39:18.000000000","reason":"Change was submitted"},"11628":{"account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"last_update":"2023-05-03 16:10:02.000000000","reason":"\u003cGERRIT_ACCOUNT_11628\u003e replied on the change","reason_account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},"29244":{"account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"last_update":"2023-04-26 06:47:58.000000000","reason":"\u003cGERRIT_ACCOUNT_29244\u003e replied on the change","reason_account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}},"34429":{"account":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"last_update":"2023-05-03 18:39:18.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I686cfdef78de927fa4bc1921c15e8d5853fd2ef9","subject":"Remove python-neutronclient","status":"MERGED","created":"2022-12-01 11:56:28.000000000","updated":"2023-05-03 18:40:41.000000000","submitted":"2023-05-03 18:39:18.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":72,"unresolved_comment_count":2,"has_review_started":true,"submission_id":"866327-remove-neutronclient","meta_rev_id":"d3a927f8f4904def8ad1e743e1ddd093e582dbb5","_number":866327,"virtual_id_number":866327,"owner":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":0,"_account_id":34451,"name":"Fernando Royo","email":"froyo@redhat.com","username":"froyo"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-05-03 18:39:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"recommended":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"all":[{"value":0,"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":1,"date":"2023-04-26 14:57:11.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":34451,"name":"Fernando Royo","email":"froyo@redhat.com","username":"froyo"},{"value":1,"date":"2023-04-20 14:56:25.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-05-03 16:10:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":2,"date":"2023-04-26 06:47:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":0,"_account_id":34451,"name":"Fernando Royo","email":"froyo@redhat.com","username":"froyo"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-05-03 16:10:02.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":0,"_account_id":34451,"name":"Fernando Royo","email":"froyo@redhat.com","username":"froyo"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},{"_account_id":34451,"name":"Fernando Royo","email":"froyo@redhat.com","username":"froyo"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-12-01 13:18:44.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-12-02 00:50:48.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2023-01-10 18:22:32.000000000","updated_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"CC"},{"updated":"2023-01-13 17:06:02.000000000","updated_by":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"reviewer":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"state":"REVIEWER"},{"updated":"2023-04-20 14:56:25.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2023-04-26 06:47:58.000000000","updated_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"REVIEWER"},{"updated":"2023-04-26 14:57:11.000000000","updated_by":{"_account_id":34451,"name":"Fernando Royo","email":"froyo@redhat.com","username":"froyo"},"reviewer":{"_account_id":34451,"name":"Fernando Royo","email":"froyo@redhat.com","username":"froyo"},"state":"REVIEWER"}],"messages":[{"id":"fabff524416f6a9ed54bbdd25463c843c9ed5e8a","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-12-01 11:56:28.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9ef453e3310be05ab5c694fd434a716f0cb75343","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-01 13:18:44.000000000","message":"Patch Set 1: Verified-1\n\n(46 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/47aa9b045d4e40dd8a58e98277aedd70\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/e1209a02e3bc448fa503197c97b84e46 : SUCCESS in 3m 09s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c68127f79ac84f66bd2e8cb7c4bf4af3 : FAILURE in 4m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8fbb7b1152b04e668b9407a1c9d658da : FAILURE in 4m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1f079aaadf684fef907b0c685138fe3e : FAILURE in 4m 32s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ebb4acb0eded4400854bcd2b42a3ecd2 : FAILURE in 5m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0404034c14d6472683b3419f8de97976 : FAILURE in 8m 29s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/dae1f4f9e5a348b6b63fb601f5aaafa6 : FAILURE in 5m 17s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/df49dbe9fbc5464e85e6709444b084ca : SUCCESS in 8m 28s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/e86aa24b1b064cf6bc6328fd93850a11 : SUCCESS in 4m 42s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/d36a2c3253a343eb9d55c478d1325364 : SUCCESS in 8m 30s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/d8863ebe442b421aa3d75108482f9fe4 : SUCCESS in 9m 14s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/f43b3b9bbea14db8912eb399c3e9a533 : SUCCESS in 7m 27s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/64f811f66c4245da8cc3d3595597c9e9 : SUCCESS in 1h 20m 06s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/40ccff032df740b0946634c122e742d3 : FAILURE in 35m 04s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/9ccdc98879c844b78d12fa5dcabd5c25 : FAILURE in 37m 41s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/651a2ccb79794a5e864b7c7f2b85e242 : FAILURE in 58m 40s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/f6f0813e7d974345a57ff1fe27aff8f5 : FAILURE in 1h 08m 25s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/e0aae3a1b5b447ca9fced00c42ec3586 : FAILURE in 34m 12s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/50563943296749a7b3ac65e4447648d5 : FAILURE in 38m 17s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/a4e6edf07b3c41af9d6c220eb0cdcb32 : FAILURE in 49m 54s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/3c978245cfa445b68ccf7bbf42a771f4 : FAILURE in 37m 16s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/35ccb794f6f84ead9d1c770dc784c4e9 : FAILURE in 37m 16s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/8698956c0e444f3d838ee1ee693c5eb5 : FAILURE in 31m 04s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/5dce4be2093949d9bea7699be111bf1d : FAILURE in 37m 38s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"28c3acb8882f92accd2d567e8cfa40d27e091c0a","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-12-01 16:10:21.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"2f9a5cb9088f0f8d63d5e37d77f0056014d5c1d3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-01 18:14:09.000000000","message":"Patch Set 2: Verified-1\n\n(50 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/ce748f09a47b4e3695498fa1a661632e\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/5afef2c3501644af9a82f92f2e232cb7 : SUCCESS in 2m 22s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4f7c4f1e1cf343d9b8b395d90d0c5093 : FAILURE in 4m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d050a9f94f5649eca0daff9c18882fba : FAILURE in 4m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/64271f8d61a54507b049ca3c0275db55 : FAILURE in 3m 56s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4ff1662127fd44d4b08f47cf5d73eb77 : FAILURE in 4m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f661767b4bd64d51869644ad3c683bf0 : FAILURE in 6m 58s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/0f0c0265c72d4fe6a8ea59d82f997bf3 : FAILURE in 4m 31s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/0daafcb4c40e431d94ee88e8ae1ff723 : SUCCESS in 7m 49s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/2f71b4bd1c424fb4bddc4980cedae920 : SUCCESS in 5m 07s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/d8c5b171a82a4bf681c5256be0b550dc : SUCCESS in 7m 16s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/21fbe3de515443ccb72103c865a54c11 : SUCCESS in 8m 27s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/76926791d03e4363862fb1fff35d7875 : SUCCESS in 7m 09s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/3c92e791d15349d1a51b9931654984dc : SUCCESS in 1h 14m 23s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/669bd32328db4d6e9a4dc3513ef3e29e : FAILURE in 48m 11s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/4f8f85740b4f469b88aab250ff9772ee : FAILURE in 35m 34s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/9ebed231a75043ca85cf8eebdc9fb30c : FAILURE in 2h 01m 14s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/a8aa8413fbf64e54a429e9b9776905e3 : FAILURE in 1h 00m 10s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/d45fc83f5f0a44e9a96b2d40a22f1d29 : FAILURE in 29m 24s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/ebe393cac24241978476c4bdbdf29187 : FAILURE in 32m 26s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/47e07a4882df42f9bd07299fe83a3ff6 : FAILURE in 1h 56m 56s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/af7b4c1dfdc54c448bdeb1d45f10ab84 : FAILURE in 37m 24s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/b5b589601ea34c42b5d7a25b3113ea59 : FAILURE in 37m 20s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/847b50d257344bc69bece911357a9528 : FAILURE in 37m 39s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/7c2131e9791f470fb6cce2b949202be2 : FAILURE in 46m 55s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"f6edf1ceab45f5a5773224ab949914bf91a52fd7","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2022-12-02 00:50:48.000000000","message":"Patch Set 2: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"8de101365549b8c99b8e1c459b5ddaacc4ae58f4","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-12-05 15:35:07.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"656b58edaaf37fb2870611ea01d1f2bfc8ca33c4","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-12-05 15:36:40.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"0c22241de099f8098ce748270a2dc08d1aa661eb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-05 16:53:01.000000000","message":"Patch Set 3: Verified-1\n\n(58 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/830af1b63d034b2eb4c2f6571d9addce\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/68f189a5edbb420e95933f477c5c29c3 : SUCCESS in 2m 09s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4b61610aa22f457ebe214e03ee7fe1a9 : FAILURE in 3m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8bfa71917d594233ba2ed3549c17c321 : FAILURE in 4m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/eb4e41ce8f5349ef9ac017e92a6ebae4 : FAILURE in 3m 48s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c8591c4b40494988a881203efd7b9c9f : FAILURE in 3m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b2a7fe4613c042249a314bc97039e6cf : FAILURE in 6m 11s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/95db1ea408e44e18a342840f8f650d97 : FAILURE in 4m 19s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/e36af6cda8b041dd90eee2b5f1706b7d : SUCCESS in 7m 37s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/ab71ff05db6e4da282e82fd20398fdd2 : SUCCESS in 4m 36s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/7672d2bfb6f1492ca5e7d3c8edc3d12d : SUCCESS in 6m 44s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/6b97a8d64a81444c95a3074addc76133 : SUCCESS in 8m 52s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/282115fb825c476c8422ca4e3ac82082 : SUCCESS in 7m 24s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/93c4a0a01b07445198cff3277456cb3e : SUCCESS in 1h 16m 14s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/5aca9e5ec5a34e4c84332d180f2dc5a3 : FAILURE in 28m 37s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/8f5f3e6a02974a059865ed2283c9bf4e : FAILURE in 29m 56s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/5d73ed3ff53f4244ada5636d5a21af44 : FAILURE in 42m 22s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/fc48a84855b0438f88d06051b6068989 : FAILURE in 52m 57s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/73ac0484d20a47f6b55c59f824e879f0 : FAILURE in 27m 26s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/72914b71f16a441587cc76467f814e89 : FAILURE in 29m 56s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/7f884ed684d2454eb80613f52770318d : FAILURE in 39m 34s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/4bcc4902705d49eb9701eb8a5e69661c : FAILURE in 28m 10s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/120362e219f94180bc4573a0f3a15c78 : FAILURE in 27m 46s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/1e294de221064f9b80f96d0025cc5db7 : FAILURE in 27m 10s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/e1f268a4201642c8bf70ef24ca15561e : FAILURE in 38m 09s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"93c16b28118ad73e4f74355552abec7c92d76da6","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-12-07 12:27:00.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"8be2d0d6bccaece0c0164b59f767d84709c128c4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-07 13:57:03.000000000","message":"Patch Set 4: Verified-1\n\n(57 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/404914ed934a4a7087bf507692e4f5f0\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/cc1370b56f254c1d99153afce60c2566 : SUCCESS in 2m 15s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/341392ec70a54674bfd49272fe21681a : FAILURE in 4m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/45c5d98d2089476595efcca4380cde61 : FAILURE in 4m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b93107ec3c0d49a09bbbb96da97b31e9 : FAILURE in 4m 08s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8f2d7ac3de0343e6a4350c54d30a5823 : FAILURE in 4m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f2af6260fcb745fb9a12b2d58ed507de : FAILURE in 6m 31s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/84aa9e2dba5548daa7187429f4721faf : FAILURE in 4m 19s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/353806f8944b4f8e9b13a235120091e2 : SUCCESS in 7m 10s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/d3fa38c29384459db26e7a8dff6a1529 : SUCCESS in 4m 50s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/bc03f2f850474de2929d32b2e07833da : SUCCESS in 6m 57s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/4f279f37f3a24d2981a9826a891c972a : SUCCESS in 6m 37s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/e4be99777ebd43f5bec4df498564636d : SUCCESS in 8m 21s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/a23307db62324b0bb19816e01cd04ff0 : SUCCESS in 1h 28m 09s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/272515de1933412fb0fb0029ca8aff41 : FAILURE in 30m 12s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/c00f329e291a4ac2877b217703a56c1f : FAILURE in 31m 15s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/5710da020da7418c8570db65fd38f870 : FAILURE in 41m 29s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/86f0bd644279473d8844e2923845caa8 : FAILURE in 51m 14s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/706c723838b94416b8d9d0f6b0f5ee7c : FAILURE in 28m 25s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/a85dde57e5b04026b5698c824576f116 : FAILURE in 28m 17s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/ced9d76122444963bafae657c59a93f3 : FAILURE in 40m 47s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/bd761e20c79544b1a1911c955799080b : FAILURE in 28m 36s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/4ba9670e11614bdd9c95ba1e272ffa87 : FAILURE in 27m 43s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/8898fb8f86d94d4a99e6f20924c08c52 : FAILURE in 30m 35s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/f0a088cc318b4faf89aa953e69c00d2c : FAILURE in 39m 55s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"e0c1f8dcf7183f31ddba007d06dece1d10d0dba1","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-12-12 10:07:46.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"a4ce3c5540fbd1fde950120149f65c6b36af060f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-12 11:46:10.000000000","message":"Patch Set 5: Verified-1\n\n(58 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/40e68e73a5ec4174a782644b80789fa6\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/9788f9f4352b4e07a43beb9a691c8cb2 : SUCCESS in 2m 01s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1ecbeb89d2564e14be984e63e85ac2c7 : FAILURE in 4m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/85531969990d4222a6a0072f8a511b2e : FAILURE in 4m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/932bf2f4f5e74eefa4b782351da27a34 : FAILURE in 4m 04s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3a8abf4be44b49d6960d1c9d51677b7c : FAILURE in 4m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/045e0094cdc541b881c0cf361494bc74 : FAILURE in 6m 09s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/43db0f43fa17415f8a463156b160f8b7 : FAILURE in 4m 17s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/fa4bb2900ad64b7980cc6d68c7cf8770 : SUCCESS in 8m 48s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/9c0a5e6b9b2e43bfac7ac0cd351fa363 : SUCCESS in 4m 55s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/67b0ed1d5dc34477bc1ac6c77896ed33 : SUCCESS in 8m 24s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/fa5b03e741c547419418986c3371d3ed : SUCCESS in 7m 16s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/ca3a26ccc61b4090b6bb0901b2cd46d9 : SUCCESS in 7m 41s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/c711ccf5376c4d9e956d30f71e6141b4 : SUCCESS in 1h 30m 38s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/a3fd67dbf3744183bf18a98e7445e5ff : FAILURE in 27m 26s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/85095148db334037853e2d2aadf35957 : FAILURE in 28m 04s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/d3ce35ddb73c42a3ada4b49ca0948b92 : FAILURE in 42m 16s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/b5682693766345a5b6f2bb00371d327c : FAILURE in 57m 51s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/6bb667a399104880ac3715a829b7626a : FAILURE in 26m 39s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/a04badffdb3344cda10afee96a8b6b8a : FAILURE in 28m 43s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/135d0706bbd44e90b120c7ba5aa4f1b2 : FAILURE in 39m 37s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/b8b63ecb409c408887207eb83abfa32e : FAILURE in 34m 31s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/e5cf0302d17749a9b62a61fcb0ef25dc : FAILURE in 28m 41s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/7706d8509363470db4465aaefd2b2d3a : FAILURE in 38m 15s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/0cad32f43d80448db4a97e836d3d058d : FAILURE in 45m 28s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"ad26995b5d0540cf69e64ccc815c0b0976f1b0e8","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-12-12 18:04:58.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"496281ddf185975abc0d13558a2fcdbeff0e05fa","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-12-12 18:06:01.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"71ee24e7418f447e30ad3ff5db1ec355ec067e0e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-12 19:30:27.000000000","message":"Patch Set 6: Verified-1\n\n(48 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/91e6c603e8e544a8b5d4b3b9519ad43c\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/1688b8b83699414db9042187108aff9f : SUCCESS in 2m 02s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5feca40a5e964c6bbcf6e0dfd39e5cb5 : FAILURE in 3m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bd2b4987c37349c8875df378c3447316 : FAILURE in 3m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/49908b4aa73941faa8187afff8b8722f : FAILURE in 3m 59s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6586d2a3ea324cfb8e048dc28bb420b1 : FAILURE in 3m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fd7ad1644b6a43a9a1f8f47c70ea5193 : FAILURE in 4m 58s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/a9cf658e452040cf91b92f723f3d31d1 : FAILURE in 3m 55s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/9c9ef7488d494ed4979c04ba584d3d18 : FAILURE in 4m 10s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/332df32da21b4512914dae1368aba705 : FAILURE in 3m 48s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/abd849751df6433ab33120b1be85796d : FAILURE in 3m 56s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/c93b3e539dc946eb9e4fad518f111306 : FAILURE in 3m 59s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/f686476c878c4623a74be90c5605b5f6 : FAILURE in 3m 34s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/16bfc4f2707f42aa851dfe1739e46547 : SUCCESS in 1h 20m 39s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/523c01c900da41b39ff9a72faa8f11b0 : FAILURE in 28m 15s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/ac7f379eec46401fa6dbbdc693d3152d : FAILURE in 31m 28s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/b955272fd8e04d319cbc76e5f118b5f7 : FAILURE in 47m 23s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/961359c3d65a485cbcbe4a23d2eba36c : FAILURE in 52m 50s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/582c2dac7f414a6092e82cff0e3f105c : FAILURE in 27m 20s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/b016448fc80c47e585aba9f522b51250 : FAILURE in 29m 43s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/ea4c0a9841744d4192b6ca984e609a67 : FAILURE in 39m 34s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/a35b477c466243d89d601c4d5bb60ff2 : FAILURE in 34m 42s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/6615439dd6b24591be6b79ef8630b82d : FAILURE in 28m 00s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/5018e0a9b36346e786e754639786f9ef : FAILURE in 15m 07s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/ec7bfd995b214a409f6445d3df511536 : FAILURE in 41m 02s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"2cf79f2109f3bf067b318e51292027c9985fe804","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-12-15 11:01:29.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"d42bd1a54949a1efacac87cc3e09a6b9d6972b15","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-15 12:19:03.000000000","message":"Patch Set 7: Verified-1\n\n(45 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/efdc0c9e97954478af22cb6a4b43f8d5\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/aad647f692d241979a2357f2240ada64 : SUCCESS in 2m 49s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c5ee8f0639ed435b8ee13c7bd6a8e308 : FAILURE in 5m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/56d2c90387884f6da8e8aff3667e5864 : FAILURE in 4m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ab020bacd5a940129088204750288c9a : FAILURE in 4m 42s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/64dcb354f7364feca042d51b4b20b315 : FAILURE in 4m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/18fc986908424721a1c3649bff0e91f9 : FAILURE in 4m 55s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/1611272b41d1400f89171e6595193709 : FAILURE in 4m 37s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/861a6bb2cf6f4840a8f5858b9fff4118 : FAILURE in 3m 59s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/ef4a82003b7f4d78b6eac7e33ea10521 : SUCCESS in 4m 03s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/6038ad607fac4d1cab4f79abe5c2a0b3 : FAILURE in 3m 43s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/54619027944d40118484e8f4248fe7b5 : FAILURE in 3m 51s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/70fe8be0fdd34ab1a6f5715181d474b1 : FAILURE in 4m 04s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/a1996023c68241bfb45921e0c0032f88 : SUCCESS in 1h 14m 32s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/5b933db4ab754b238444e14fa2c25c03 : FAILURE in 28m 55s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/3085e28cbb1a4cc2bc40a29b5257f26e : FAILURE in 31m 35s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/76305aadcf674a1ab94be4ed9c8168fa : FAILURE in 43m 07s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/ece0b466df1e4b02ae096e2f11467f65 : FAILURE in 53m 15s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/ee42d52e0a93449f8afb44b41f323f34 : FAILURE in 28m 33s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/20e053d7119848eab22f07f57da37d1c : FAILURE in 31m 54s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/d8eedb6444e440f7bec138025ee7af01 : FAILURE in 37m 13s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/2350730a61064f16b05f069eec3de8c5 : FAILURE in 30m 00s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/c41834c9859347ec96eed3ddebbfedfb : FAILURE in 31m 07s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/1ee9ea60f62d4e36853072a996039c11 : FAILURE in 30m 51s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/614254311bef4c498b6d2781f12e246b : FAILURE in 38m 20s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"145e3a26bee82999b4e78563c0907b2355af4f27","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-12-15 15:41:50.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"7a692a5ecab87c6c9f89b9e0725df8f4abeb299a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-15 17:31:07.000000000","message":"Patch Set 8: Verified-1\n\n(40 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/9819d640adb842e4abf8683600ddbb42\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/91b9643f1624407483dbc443b6b9e0bf : SUCCESS in 2m 11s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/33f4960bc5864ea690fbbc997259eff0 : FAILURE in 4m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/059da797987f4b2ca9efdfabb8cf2539 : FAILURE in 4m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/90add33264d341e5a3dd6638dd741b6e : FAILURE in 4m 16s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9cdd3fa331ae4d008f50fe857b6776da : FAILURE in 3m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/66e26ce37bbe4447a8e52df8b84cdf87 : FAILURE in 5m 12s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/7f31fc5c7ebd4967a72e5af1034865f4 : FAILURE in 4m 41s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/207bd7360e8e4a8b863ec3d5e4afe0a5 : FAILURE in 4m 01s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/7374a1cd633846048566d0a8f7df51ee : SUCCESS in 4m 25s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/dc5cfd7179f2416a9a81038e81cecf5a : FAILURE in 3m 55s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/9c6b254f07764aa5ae3e9c30b33d54f5 : FAILURE in 3m 44s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/fde69ce9f1984375afcc63a3519ce491 : FAILURE in 3m 36s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/042a5df7cf2243b1830a672fad89c951 : SUCCESS in 1h 42m 03s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/9f7f6000c40b4ce2848637da122982c1 : FAILURE in 28m 19s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/5bbfd51444664f11950de6114253b2d9 : FAILURE in 28m 24s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/fd48eed97b294a9ea19bace2f922731c : FAILURE in 43m 22s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/7ad8b26cb61049d6b9b59fcb6d0f2e5b : FAILURE in 58m 51s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/98ff521c70f3490cb17ac237c4ff8b83 : FAILURE in 26m 44s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/3b52c5e892c444abaf43b7a59ce83597 : FAILURE in 31m 07s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/efccf86491934badb0c9d6d605992c09 : FAILURE in 38m 31s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/a8f6b20d547b4f92957bba7632c6b0a6 : FAILURE in 28m 31s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/4bd7a6e2483f4afc8ae049819125a446 : FAILURE in 26m 44s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/d94b65b1014545dc87aca0a3763a9894 : FAILURE in 32m 14s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/25144f4f6f16415b90cdcf38e7d635cb : FAILURE in 39m 55s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"a9d65568a20c5739f9026ae21c5ae851affe6701","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-12-16 12:32:32.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"7b390960dc29718e61c2b092e798229218e6d8fc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-16 13:36:46.000000000","message":"Patch Set 9: Verified-1\n\n(40 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/7f4b09a197b6412b995847a4fc328407\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/67d02bfdb10748beb5a379ce61b57a01 : SUCCESS in 1m 55s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/637f82a2faac4067b885e38e4ea9c12b : FAILURE in 3m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/43443d2bfd6747b48ee47b05d49c1845 : FAILURE in 3m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1dab07132ac64c2fbd36732a7c3ea8ed : FAILURE in 3m 36s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7add0f4ab85d47e48456244bd27db086 : FAILURE in 3m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/16d1daacac7e496a8186b4db12a2bcce : FAILURE in 5m 44s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/a106b88454af4fa3833a3d57a77dc0b2 : FAILURE in 3m 56s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/7779dc5fd6c84c78afc03d271e3902a1 : FAILURE in 4m 09s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/04ced4f86910448e80960c13d6190a27 : SUCCESS in 4m 00s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/2a977605aaf546e49a7b3bb0107c48cc : FAILURE in 3m 20s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/abb96165720e416480fd87eaba631ff8 : FAILURE in 4m 05s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/89d4f2f8c9ea45fb92b17d4abe4ea73b : FAILURE in 3m 46s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/34d6731151d9458f8b7333ac433d8a39 : SUCCESS in 58m 50s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/bcc9215d45e34887aa786b82ff4aeb8d : FAILURE in 26m 22s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/2496b482896240c3bb85cdce5fed6316 : FAILURE in 27m 31s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/a4d694063a564e33a4c8d8e6b3e35436 : FAILURE in 47m 41s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/6def860cc2c140be9645c3769f848ac8 : FAILURE in 51m 35s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/d3f71bb80af640e0836fe3221a55a16f : FAILURE in 32m 47s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/57d90db0034247f381b95acf97a96639 : FAILURE in 19m 22s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/f3167632e60f4aa4bfc1521bc10b69dd : FAILURE in 38m 08s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/d5aed74b2fe249cbba0bdf76d09f26a4 : FAILURE in 28m 43s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/abdc27cd031d492c80e43db94a0e580c : FAILURE in 28m 51s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/cc08309bc5344dbab5fc99bf1df9336a : FAILURE in 29m 16s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/f74cb06408ab4b74af3e9382a1fcfcb9 : FAILURE in 33m 41s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"e2e87f1f24dcddb7f5b39c1375d25be974a0f5b5","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-12-19 15:50:43.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"1f2c519ba7ba9d170488c0697a54503f2cc45047","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-19 18:28:34.000000000","message":"Patch Set 10: Verified-1\n\n(37 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/9e329f353d134edb8e9b2d6568004da4\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/c417c6c4639c4e6fbe8bc99efc592440 : SUCCESS in 2m 10s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3495137600ba44f9914eb9cb5947af09 : FAILURE in 4m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/39833f9b6095474681cde5c49b565e01 : FAILURE in 4m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2f843b0268384df6aaed668a39e5e644 : FAILURE in 4m 24s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0d03a858ec654884ab89f85dffd3de1f : FAILURE in 5m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5b270f636e8646dfbfa67bda9ca28090 : SUCCESS in 12m 43s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/404c9d1b24b54f3b8b7b1514f8b86c1a : FAILURE in 5m 09s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/6a0bf78f54454664a85dd8a93d15c6d8 : SUCCESS in 9m 40s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/e3838b5e93bf42168786c5029d0fe035 : FAILURE in 4m 17s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/4b69c67235f544b19beb9044587528e5 : SUCCESS in 8m 20s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/674a2070039841098324a9455520a872 : SUCCESS in 7m 07s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/10ec0dd1375540beb2d9da8eff88e313 : SUCCESS in 8m 38s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/518b253c38d54d39af57fa19670e28b3 : SUCCESS in 1h 14m 53s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/0c50a0b5765746c1b5e605b7a5c3d9ef : SUCCESS in 1h 47m 50s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/422067e61f3142399843149ee3cef1f4 : SUCCESS in 38m 42s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/74e7613fa02c40cebf57aa756107b015 : FAILURE in 44m 06s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/d4777977aa3f44afbe7e1a34a9929c0b : FAILURE in 1h 18m 12s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/f0220442244a4bbf997d7d987104fea1 : SUCCESS in 35m 47s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/092ad63e9abc403b9d2532c7e1a2a565 : SUCCESS in 2h 08m 10s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/74869a486efe4f288475f0abc7d357c5 : SUCCESS in 2h 21m 22s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/5e7b5eda201c4fdbb655875da916682f : SUCCESS in 1h 50m 04s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/de2bd62bee044a6687ac5d5deabb24c9 : SUCCESS in 2h 33m 02s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/9b82167083eb4edc8f479ff56d1fb71d : SUCCESS in 2h 13m 29s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/1af1f1a202c14efeac9196863fcc7cb5 : SUCCESS in 2h 24m 25s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"ee85457cb813251ca7b4ccb5859fa2ae63b7ea7a","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-12-20 15:59:18.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"b224eaa0c34c62dbf01dd84296af93547e7a3fd9","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-12-20 16:09:25.000000000","message":"Uploaded patch set 12: Patch Set 11 was rebased.","accounts_in_message":[],"_revision_number":12},{"id":"4c101e510a9f1af95d22dddab45a58ff303e0f2e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-20 17:19:56.000000000","message":"Patch Set 12: Verified-1\n\n(26 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/dd269b503f744091b56ca57cc9128cbe\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/e224fc80db27429eb13a631113697650 : SUCCESS in 2m 55s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/40f29b4db36a4c2cb8b27580806a177c : FAILURE in 5m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b001d39fb1d74b8fa92d4e6b0338662b : FAILURE in 5m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/96afaabe09214c16ad6b74108dbf870e : FAILURE in 5m 35s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/eca77e13892b4d25aef33c657cce479b : FAILURE in 6m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8df9aa71dc6f49b5b91152f0f5e70460 : SUCCESS in 12m 44s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/bc75d22ab3d348b7a1c86fb65a48f331 : FAILURE in 5m 09s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/0985842ac116494086a17e4d8e346909 : SUCCESS in 7m 47s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/9b35247302a54b898c2e64f1ff5e51aa : FAILURE in 5m 03s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/ddb43b886440407593fe1e0a71e66522 : SUCCESS in 8m 05s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/db6fca89851143d8bc0049e977c49084 : SUCCESS in 7m 14s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/32ea591da3ce4512a255bb7b68258348 : SUCCESS in 9m 01s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/8c2dae2a2924452d84796f83ad6b279d : SUCCESS in 1h 09m 31s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/c225420f2ef4438aa2b7a3d8aad2a33f : FAILURE in 36m 01s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/c9d4f64d0e1f4eadbdb0c8ab186672ba : FAILURE in 33m 17s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/cb91175ae2364682bc6e8253ad4f0700 : FAILURE in 51m 20s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/67ff12cc4578425b872a1a2f7eefd75d : FAILURE in 54m 10s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/33b415c97ccf446184690d013b850ff3 : FAILURE in 31m 01s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/5ba87191461b4489a377af5ddd807bad : FAILURE in 18m 49s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/3590dd67b40847e3a6c6c4c035d9770e : FAILURE in 42m 36s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/c521ab526ff14e809470259cdb1c607e : FAILURE in 32m 22s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/b11aef5f32e94f548a845362cccc796f : FAILURE in 32m 45s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/af1cf97c99984614b02f85e77d0665c4 : FAILURE in 33m 50s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/058bf55dd3374e138e426b0bed5b632e : FAILURE in 38m 22s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"559dde3791247368afa14dfdad3d7697180e2e23","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-12-21 11:30:58.000000000","message":"Uploaded patch set 13.","accounts_in_message":[],"_revision_number":13},{"id":"0b7c6b841c3455c652369e79bf6ae0d8739fa2c3","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-12-21 12:29:48.000000000","message":"Uploaded patch set 14.","accounts_in_message":[],"_revision_number":14},{"id":"ea1e7859b5dfa5ae85a7f0577004bbea69971d1d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-21 13:57:09.000000000","message":"Patch Set 14: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/39f58d182eb54cc4a95b7cb5a03b0f96\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/840e65cadf4f4cb59d5f427ba1f72f52 : SUCCESS in 2m 11s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3781a58dffe2490aad9b4ccfe9c22915 : SUCCESS in 10m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a9111d4e186b4d72a7d82045400f7002 : FAILURE in 5m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/33836db976ea4d95a5086be0a161eec5 : SUCCESS in 5m 07s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5c4d1c1f930544e594e002aee3b32161 : SUCCESS in 4m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/abedf7eb45ab447fa366b66b0361de31 : SUCCESS in 11m 15s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/7f930562180741d5be6fa719df416852 : SUCCESS in 5m 23s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/be440e30d9ff493286fd8c2fa5ed2b98 : SUCCESS in 7m 50s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/b1431967d69e41ed8ff510a6fea468c8 : SUCCESS in 5m 32s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/1ce648fdbf60472983c7d96d1f1efc00 : SUCCESS in 8m 14s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/f259b7b1be084f63a29a055d1d397ea5 : SUCCESS in 8m 19s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/c1010ee07bdb4c92bd218c230f6b26c8 : SUCCESS in 8m 34s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/73a695c3fe8e4f9d99d6ac4666218bae : SUCCESS in 1h 21m 45s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/2c37015ecfcb4657a6392c4eb962b2c6 : FAILURE in 29m 03s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/4a9e61b4436c4ace9b05a1b5c41a1530 : FAILURE in 28m 27s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/3ed56329341d40e48f04d42cb7c74215 : FAILURE in 42m 11s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/d3bf9d2013884de690ce1862d8a45d92 : FAILURE in 1h 03m 16s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/83a0470d13a247ca88a778c55a56ccc3 : FAILURE in 26m 44s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/42dbcb151d66450d9506456fd927eea2 : FAILURE in 29m 09s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/e59533d2656a4156acb1b38e663419de : FAILURE in 39m 26s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/601e18bcd3594d6bb598504d52c068f3 : FAILURE in 27m 29s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/b1c336f7aa9c49fb9afadb3cd446c784 : FAILURE in 28m 42s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/0b8f75bb4652437b89f8bb5b2a12f979 : FAILURE in 26m 33s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/b27ce1b64f7d4aad845355be19a16ace : FAILURE in 36m 10s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"2db4930514624aafabdf06ca469a350702cc7ae2","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-12-29 10:00:56.000000000","message":"Uploaded patch set 15: Patch Set 14 was rebased.","accounts_in_message":[],"_revision_number":15},{"id":"969838fb340fce6900902f6bf188cc320682eddb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-29 11:19:38.000000000","message":"Patch Set 15: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/76337e6719c844309153b70e38244fa3\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/93b14d9d2e40431dbbb66182733a5d19 : SUCCESS in 2m 27s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/458f75fdf3db4786aa479651d78b1d51 : FAILURE in 5m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/02c02183d44447c8ad39e5b00ea11a42 : FAILURE in 4m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d2371b8f0b4b40f6b5a559cd8072b3c4 : FAILURE in 4m 28s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4acf1aab250b4cf2a813f38560ed1b8a : FAILURE in 3m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a1a64fe74b4b4363a89402ccea19bfe1 : FAILURE in 4m 54s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/18505448562648108a42ae1d3b184ed2 : FAILURE in 5m 11s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/caad310da0aa410fb76dd47fdbe601c6 : FAILURE in 5m 46s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/3c61f4e74be644ec997c2e9aafa1d310 : SUCCESS in 3m 44s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/55be9b206f014e67b824e4d7de964f28 : FAILURE in 4m 42s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/cf90fff075914843802ea404bcc1db57 : FAILURE in 4m 50s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/839a445f3c9147c7aa01a30225b44196 : FAILURE in 5m 16s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/aefaa96060594035826c32788bc304a4 : SUCCESS in 1h 07m 47s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/d312fd81df624d94a90bc73580b01622 : FAILURE in 37m 34s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/95009e7b4ab6400eab694cf07fddbbb9 : FAILURE in 28m 18s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/3061a4a0b8864c7dbd2617bb11028b28 : FAILURE in 59m 12s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/32a3659e9d3c435698eeae48d968ae82 : FAILURE in 52m 38s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/171b7ea0bb25417faf63b2c0ed1d98cd : FAILURE in 33m 02s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/28f8fe191bb94bca8b69a67489fa6687 : FAILURE in 33m 13s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/3b65f37e0cbf48f79f82d9975aee5b44 : FAILURE in 51m 07s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/81d619bf80e645b1b06c8454bbc48f93 : FAILURE in 29m 04s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/f4bc314242dc4bfeb71f00f7f17a1570 : FAILURE in 29m 18s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/2695ba20a0cf410186544a13d1263b5a : FAILURE in 34m 35s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/7e382d5ae43c46bcbc3e36466e5607ee : FAILURE in 33m 35s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"986408e0901826f9e75493dbbdce608767456e96","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2023-01-04 11:47:52.000000000","message":"Patch Set 16: Commit message was updated.","accounts_in_message":[],"_revision_number":16},{"id":"578ac0a9233e4e9138939360699924ebde49b639","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2023-01-04 11:49:40.000000000","message":"Uploaded patch set 17: Commit message was updated.","accounts_in_message":[],"_revision_number":17},{"id":"65c63b369325fee9eb0217deb562667dc8f0f47d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-04 13:34:34.000000000","message":"Patch Set 17: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/45eab4b1f2964da59f47e7a168c9a6d7\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/1a0ae2b237ca4314979b787c15f8b812 : SUCCESS in 2m 16s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/06b961cdfffa4cc889821832e3294083 : SUCCESS in 12m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/deaea54bb4ab4d11b39473a481df7811 : FAILURE in 5m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3c7461eb57ac47478c3622711f34507a : SUCCESS in 4m 35s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/abedf082ebe24eda9adca73eec5b800e : SUCCESS in 5m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b8a4fb69abbb45ab94721d2a692ec980 : SUCCESS in 11m 01s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/85c3fe1c66474eaa9ea836a66ccb5f8e : SUCCESS in 5m 16s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/22f71214838247a7a3e3cf789eb547ac : SUCCESS in 6m 29s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/6189ec547bdf49d3984e5d2964caca12 : SUCCESS in 4m 01s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/42ef801ac2834afbb046439d4040e880 : SUCCESS in 8m 22s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/ff5abb40b5e449a5b0bd992cc91c61b8 : SUCCESS in 7m 36s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/33086f0154a14b1ca51058541671138c : SUCCESS in 8m 17s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/8bb2861b8c354821b82ef56fb841ca7a : SUCCESS in 1h 39m 33s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/af442d41d5f84f1bad07a96a08b55e2d : FAILURE in 33m 00s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/1aaf1f9a10c14414ab5b9454d287e3c4 : FAILURE in 36m 42s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/f65634a03d4f4a67909654f32e85ba3d : FAILURE in 52m 53s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/2c3a74c6f79540f994bf6ecd39629130 : FAILURE in 1h 06m 01s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/75f2e90c397f462c91ff1f581d85615b : FAILURE in 30m 55s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/8a8ca06f960a40a7a0fa208e45aae09f : FAILURE in 32m 08s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/c0ee4a1ce1334265862c6f6458159e70 : FAILURE in 43m 25s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/3ea8df535b8c4104a2f586db4b070f74 : FAILURE in 37m 38s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/ea3d791ac3cf4e5a8d870ee549cb8932 : FAILURE in 33m 20s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/ebeb992e4c1745bab1d6a59f4671d914 : FAILURE in 32m 25s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/6198a5e08e3946d08c79dc629e40f38a : FAILURE in 43m 08s (non-voting)","accounts_in_message":[],"_revision_number":17},{"id":"6119762d7e7b0e94d6e0d10a5087590d70ff020a","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2023-01-05 11:07:27.000000000","message":"Uploaded patch set 18.","accounts_in_message":[],"_revision_number":18},{"id":"9e659f867c8e66aca6141c62f9b7b89b7850f0a5","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2023-01-05 11:08:31.000000000","message":"Patch Set 18:\n\n(2 comments)","accounts_in_message":[],"_revision_number":18},{"id":"895798bb9906d646527c371a223130823940a234","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-01-05 13:30:35.000000000","message":"Uploaded patch set 19.","accounts_in_message":[],"_revision_number":19},{"id":"d0c65b626b91575f84261a2055bbb43ae0f29287","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-01-05 14:25:08.000000000","message":"Uploaded patch set 20.","accounts_in_message":[],"_revision_number":20},{"id":"a1fa0c14702cf67425970d832952eb6b23d8a099","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-05 16:51:53.000000000","message":"Patch Set 20: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/65b3b6a92c8d4c6ba9693619293a0a68\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/f8e263e5cec54bd1946cc8b3a6445e64 : SUCCESS in 2m 10s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/477099952b9a4b868d7bcbc4d06f6b00 : SUCCESS in 11m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f9009163421f40a6b815a888a4d28d9e : SUCCESS in 5m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/574077d7ccad4e0a906f80f95375f1af : SUCCESS in 4m 07s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0c36232815ac485cbfc1332f51f89237 : SUCCESS in 4m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4a2b1227067848478869efc593dcfb2c : SUCCESS in 11m 58s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/74d707edb85b43f49ffb419fda296a87 : SUCCESS in 5m 00s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/14546db13707446691ef304ca3be01b5 : SUCCESS in 7m 53s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/3bf5101337014fb6a2c768d2abf1da38 : SUCCESS in 4m 08s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/f46f5d7f107f4caa8537aa70b99798fa : SUCCESS in 7m 47s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/e3a50fc8b58e4f15a66d8a326abd101e : SUCCESS in 7m 33s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/6b6b1d48231244198d5b8b032f5ba9eb : SUCCESS in 7m 31s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/c0342f187dfe481bbaeb76f9447b0afc : SUCCESS in 1h 21m 16s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/9377cad911c0421dbacf784539b45c81 : SUCCESS in 1h 55m 17s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/3350963de49e4632a8eb194ab6a636d9 : SUCCESS in 34m 52s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/5328a029bcf04d1aaf98e9e245dedced : SUCCESS in 1h 03m 57s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/cf20c195279f4601b1f4f25d3b25af14 : SUCCESS in 1h 14m 58s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/fe6508585dce402b8de8fbf23dabac1e : SUCCESS in 30m 10s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/1f472214fb724529855d60e1bcdc31bf : SUCCESS in 1h 48m 49s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/5afb190128d9439cb97e5ec86f414440 : SUCCESS in 2h 13m 04s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/c711b014df2f4ff986b5ce7410253af4 : SUCCESS in 2h 13m 59s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/dcb8b4e087fe4f43bb454c9f48018c01 : SUCCESS in 1h 49m 20s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/97da3d318edc407899bf7481c53195ca : SUCCESS in 1h 41m 05s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/ea42d44fdb2242248335e1b2a11ddee6 : SUCCESS in 2h 24m 03s (non-voting)","accounts_in_message":[],"_revision_number":20},{"id":"50a1e1e66af722a2a7bb00ee1e6a22ad694ab522","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2023-01-10 17:46:37.000000000","message":"Patch Set 20: Workflow-1\n\n(1 comment)\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":20},{"id":"7d275d71bdff7c820c5cfa6ab4e26b4fb779b003","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-01-10 18:22:32.000000000","message":"Patch Set 20:\n\n(1 comment)","accounts_in_message":[],"_revision_number":20},{"id":"3c22a155355d885a8eab391d7eceb6390fdfeb0d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2023-01-11 01:15:55.000000000","message":"Uploaded patch set 21.","accounts_in_message":[],"_revision_number":21},{"id":"4541f46ae08350bbc5a7fbcf28e2843adad524ab","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-11 03:51:40.000000000","message":"Patch Set 21: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/57eeac8622cd443baf55ee537888cd58\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/14d23c10758b4acab3cf01f1d5e06ce4 : SUCCESS in 2m 04s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bcf804362a5e4bf2892e5c0d05a4f7cd : SUCCESS in 10m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/72b7df176b154ef38a47fda7fbd3691a : SUCCESS in 5m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1454b49228f94d559a506a869d1748b4 : SUCCESS in 4m 27s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5a70ef4adc9e4779ae06d78403194edd : SUCCESS in 4m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/14dcac3d31dd460881db2a6caa9c5f22 : SUCCESS in 10m 50s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5ae744bc90e6428a830c15f4e21cf183 : SUCCESS in 4m 39s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/750d7de8fb8f4badacdc2e8acf019f92 : SUCCESS in 4m 43s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/aa6a724690fc4ad4a2d134354591a85b : FAILURE in 2m 08s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/308663b0b2644bf4b48d2541e329b5a6 : SUCCESS in 4m 28s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/43bb2b558bea436caa3525ec731da736 : FAILURE in 2m 28s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/217b303b3bff42d6b132f678a2e13974 : FAILURE in 1m 58s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/1fd875a34d8841feb13ca5e08a76e798 : FAILURE in 2m 11s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/a48087e7e863450fa399cf72d3f55cd3 : SUCCESS in 1h 44m 43s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/13ac5605ab21426ab2a0c3f578227cdc : FAILURE in 19m 07s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/7ec44d3967574e5fa6209893ca0cdbe5 : SUCCESS in 42m 27s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/c19d4bf10fe04ea8816666d0f853be5e : SUCCESS in 1h 07m 01s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/5783b6c477d245e09564bf3327d989ac : SUCCESS in 1h 08m 53s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/80eba8b03ffe4573a02778b53414ebc6 : SUCCESS in 34m 30s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/43c3ff74ee2e44ca919ccdf45f4fa927 : SUCCESS in 2h 02m 50s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/112c7f355a984d4cbb4b0a4e4c687d43 : SUCCESS in 2h 14m 09s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/15a7df8abc654b10aeebf089e2cbff0e : SUCCESS in 2h 10m 44s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/ab88e188f6ef4b8e8feb64b041a6b374 : SUCCESS in 1h 58m 11s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/1c78b339c955448d9ac9312927418b5c : SUCCESS in 2h 07m 35s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/39434c7e6d5a490baba165d4cd1634ee : SUCCESS in 2h 33m 45s (non-voting)","accounts_in_message":[],"_revision_number":21},{"id":"a51f1358ab56b0e4788007fdb54bf96b0a336e43","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2023-01-11 10:00:56.000000000","message":"Patch Set 21:\n\n(1 comment)","accounts_in_message":[],"_revision_number":21},{"id":"454f64a90ee815737cbda642abc3dfe718e626f0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2023-01-11 10:16:59.000000000","message":"Uploaded patch set 22.","accounts_in_message":[],"_revision_number":22},{"id":"8eab619a997c246b9fec4e4dade08626097d3a01","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2023-01-11 11:26:52.000000000","message":"Patch Set 22: Backport-Candidate-2\n\n(1 comment)","accounts_in_message":[],"_revision_number":22},{"id":"389a5922478b0fec1377434ebb6f56da4f1a337a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-11 12:43:41.000000000","message":"Patch Set 22: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/444455560e744a8397fc12c84a6c479e\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/fd61f41e0c9a45fea6a6f9a307a91512 : SUCCESS in 2m 50s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/53d8b58230de430c9e1cf7f4916157c2 : SUCCESS in 12m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4f05f2e626724046b7b3c0850c2516ae : SUCCESS in 5m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7bc5db13b66b4609a83ea7485d0d0b51 : SUCCESS in 7m 04s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/cd854422bf674bd39be15bdc7b7d4eb8 : SUCCESS in 6m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9f20e9d89d304ea69013676d78c9b22f : SUCCESS in 12m 29s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b260b562af46466898db248c3de9ce33 : SUCCESS in 6m 01s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/63bc5aa555ac49caa1b318e1b95efb06 : SUCCESS in 5m 57s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/2bb7095bdfb44dc99a5bf7b3d8255c12 : SUCCESS in 9m 33s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/85dfe117e98642578ba037d23054dd0c : SUCCESS in 5m 06s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/5ab7435c569148919fd19237c1c3626b : SUCCESS in 6m 04s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/b566ad3d6437477e8fe3d450be8ecbc8 : SUCCESS in 6m 38s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/b58cb583917e4dc6850b92ec06b6dc7d : SUCCESS in 6m 38s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/af137ec2d9314643b70549ee3e9adea7 : SUCCESS in 1h 45m 50s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/f4605a85313140b59fbe8949e2546b2c : SUCCESS in 1h 47m 05s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/af0f5b9a8822475d8eb768c4b54c73d4 : SUCCESS in 35m 30s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/a09ca2e2a48f4777ae8ca9563d47faff : SUCCESS in 51m 24s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/ed3da3b163b84378867ac0e91007c5bc : SUCCESS in 1h 06m 37s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/a3d8c05ddbd24b43b2cb2f6c57725713 : SUCCESS in 30m 17s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/fbd0dae564a24a5a9b766ff1212c9e55 : SUCCESS in 1h 44m 04s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/6fe773bf886b498188e2479bdb05e2dc : SUCCESS in 2h 12m 30s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/8873aa7c70bd411ca853d9cc17389c0f : SUCCESS in 2h 02m 01s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/39b2dda49c3f4eb6bbad6de835610889 : SUCCESS in 1h 44m 04s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/8940d0beac4a428dbdc928761a6aebab : SUCCESS in 1h 42m 17s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/2de6a2d5cd69409e8fb4b0d2082f64b9 : SUCCESS in 2h 24m 06s (non-voting)","accounts_in_message":[],"_revision_number":22},{"id":"e697f11a13765a69688461df4a59842f5a465ca0","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-01-13 13:34:52.000000000","message":"Patch Set 22:\n\n(3 comments)","accounts_in_message":[],"_revision_number":22},{"id":"0df5a85c8912d59f81026b5359f741bd755275e3","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2023-01-13 17:06:02.000000000","message":"Patch Set 22: Code-Review-1\n\n(11 comments)","accounts_in_message":[],"_revision_number":22},{"id":"5449cca18e0399e80d603a115b4601d9573b07b6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2023-01-19 15:29:42.000000000","message":"Uploaded patch set 23.","accounts_in_message":[],"_revision_number":23},{"id":"b2669c8ef4c8759f7476e6450043c9dba832da68","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2023-01-19 16:08:10.000000000","message":"Uploaded patch set 24.","accounts_in_message":[],"_revision_number":24},{"id":"5062fa728139de499607a2d95ffd064ea0d25d77","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2023-01-19 16:14:16.000000000","message":"Patch Set 24:\n\n(11 comments)","accounts_in_message":[],"_revision_number":24},{"id":"0b04eec2774f25ce8912e57b9b8d0ecaad1d80ea","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-01-19 16:17:32.000000000","message":"Patch Set 24:\n\n(1 comment)","accounts_in_message":[],"_revision_number":24},{"id":"0f023f853d344d3cc64b6843e5c9b96e36fb4ce9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-19 18:52:15.000000000","message":"Patch Set 24: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/6bc048d6e8b9421392d5bf8562761862\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/c3460c1cae6c45fdbfae3232aa07d10d : SUCCESS in 2m 10s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0d0197daeec643cd9b23f950ccbb6f6f : FAILURE in 5m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/637abbd8d90645f2984a197489e8ce65 : SUCCESS in 5m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c567d7b8c9a24b0b9d58f7a27d78a245 : FAILURE in 4m 17s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/840a7d1aaf7741d6bdcbe9d90cb88be6 : FAILURE in 5m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/11c3430415c649a6bdee65902df5ff33 : SUCCESS in 13m 12s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/589d8fccff5b4347ba174a5c8d63d08f : SUCCESS in 5m 38s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/8d21c4f622b7456d925526a5d00510fb : FAILURE in 5m 56s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/a3256d09252d4c61bc11fc10b11772ab : SUCCESS in 8m 54s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/c69dba06aea249cf92fd31eb0f74c102 : SUCCESS in 5m 36s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/efaaee6e69ec4c41a5029ff5faa9405f : SUCCESS in 7m 52s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/507d7ce5ed914b289004fc918ab2fcff : SUCCESS in 7m 31s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/0618321acba04c8ea945fca385a3f2a6 : SUCCESS in 7m 49s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/0458048dad194627af92ecb2120aa819 : SUCCESS in 1h 25m 09s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/e3bf22a355094dbb833c6c8abe15bcac : SUCCESS in 1h 38m 49s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/695eb41583e041a1816d746b100a28e1 : SUCCESS in 34m 13s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/96e77b60baa24ff28a79e5fff75147b1 : FAILURE in 42m 53s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/70ac8a25635a4652a47894eba11d8f96 : SUCCESS in 1h 03m 06s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/a62cb98610fd43089a42b826b6851543 : SUCCESS in 30m 50s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/faa8ea0cdaac4ce8af1ec4d0376239b7 : SUCCESS in 1h 51m 33s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/461c9a6a94d54f8b8772bc60813791b4 : SUCCESS in 1h 55m 46s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/289b14242b7848668d30278ea69bd488 : SUCCESS in 2h 03m 38s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/2e808ffb5bb6460a94f2a0781e69af1b : SUCCESS in 1h 33m 12s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/7bf025c6136848f5bc03b4d1fae8cc27 : SUCCESS in 1h 59m 19s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/0cd3e370a28b4f44b50133820e9bd01e : SUCCESS in 2h 34m 59s (non-voting)","accounts_in_message":[],"_revision_number":24},{"id":"c95613764c0faec59d3074db3f3719bcd4686f71","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2023-01-23 08:49:09.000000000","message":"Uploaded patch set 25.","accounts_in_message":[],"_revision_number":25},{"id":"24cadd973fec54a965d645a6681ec467c52eac37","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-23 11:24:15.000000000","message":"Patch Set 25: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ac2eae505bdb4e3b8bdd7cd8579db685\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/20dcbc7e66b347a4a5a2854bd8468cb5 : SUCCESS in 2m 48s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/416397277bc849b0ab8ac17140d15668 : SUCCESS in 10m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/683632aa91b141c096b7e9a0b56ed812 : SUCCESS in 5m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4a26734cab0649dcaedee2e6e9273d7b : SUCCESS in 4m 27s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c5fbcb194e9a4506915c014906b80eb9 : SUCCESS in 4m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ab8adc82b5f54a739cd2e78446248983 : SUCCESS in 12m 16s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2996dd6db4a043b0aa492a7435d73fe2 : SUCCESS in 5m 07s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/fcf3c4d538d34e14ae46a364a57947c2 : SUCCESS in 5m 08s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/746bffdde4d54468a25ba2a7e88389de : SUCCESS in 12m 10s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/2a1508b223004c129a877a138e60d433 : SUCCESS in 4m 49s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/aebd4fbcf87b4592980d691e0b57355c : SUCCESS in 7m 03s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/870e610e89474329a6d69e17593647ff : SUCCESS in 8m 26s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/8872b23ab1044b5b8b02a45d99cedba1 : SUCCESS in 8m 23s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/8a110278800146c1a6713e219a1eeb4a : SUCCESS in 1h 20m 34s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/c7e1bdd4a4674ceda17e309d5b45f859 : SUCCESS in 2h 26m 49s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/4c5cfe8ab6f64e87ac357dd551d45202 : SUCCESS in 42m 41s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/9f466089fd4a4c8abd39ccb4053f64de : SUCCESS in 1h 11m 10s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/d01d3a4d419a4b248d8ff6e9ab581594 : SUCCESS in 1h 17m 20s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/ae43e37f96134231b2a66cf9a621b658 : SUCCESS in 32m 30s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/40457dfa65774ebda7dcd2321edc3c15 : SUCCESS in 1h 44m 58s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/e552b67406c046a1873908bfeeea1939 : SUCCESS in 2h 17m 04s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/8118c68b13fa4242be3dd61b7af6f3e8 : SUCCESS in 2h 21m 34s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/55a8975fdab6410c9a980f7a7d5ae178 : SUCCESS in 1h 54m 04s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/25b115827f964bccb25fa60b9f564234 : SUCCESS in 2h 06m 22s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/f51b848e5d184fb5b95dfa74b90abf2a : SUCCESS in 2h 29m 16s (non-voting)","accounts_in_message":[],"_revision_number":25},{"id":"b0f6aa7b21d747a7faa27e29fc386c7de46f6f88","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2023-01-23 12:27:28.000000000","message":"Uploaded patch set 26: Commit message was updated.","accounts_in_message":[],"_revision_number":26},{"id":"df5ada9d332a40e43a87467b01ff54bdba6aa4df","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-23 15:10:00.000000000","message":"Patch Set 26: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9a57d5bcf278485bbd67b37418e72c02\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/62ed66b43c1a4b988132cab1957af8cb : SUCCESS in 1m 47s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8ba665d0c41f4af191274fc10ab5491e : SUCCESS in 9m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9d29a79e93f347168454cd80153a43c4 : SUCCESS in 4m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8554d4ae2ad449f2a2c36f7e8940410e : SUCCESS in 4m 18s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/fd6624eddf324daa94eb2ff93d4acfc2 : SUCCESS in 4m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/84242d7a03b243d8b8ac0f258a825e71 : SUCCESS in 11m 28s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6bb6dcad9f244b54aa2d9369fc00cdb4 : SUCCESS in 4m 09s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/e7930e2f3c284678b488b5784aa3d7cb : SUCCESS in 5m 41s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/a46d83c1ad1b49cf9c487e47edf8aa5f : SUCCESS in 6m 52s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/78118c78d84e4576b6ce65560e9a37ed : SUCCESS in 5m 48s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/7562de8562e84eec99ea340eff1ace47 : SUCCESS in 7m 18s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/21cad68fa8ef4f8782e0d824409e35c1 : SUCCESS in 7m 58s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/604c2c4f03ad48e18b4946ab60632cba : SUCCESS in 8m 48s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/6ac9b7d4fd624c8aa5b4b3890b3900d4 : SUCCESS in 1h 14m 46s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/385538cec57d40189269266c5986b02b : SUCCESS in 2h 28m 22s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/5d5f4ab489214fddb2ee00a46739b55c : SUCCESS in 34m 41s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/eb09fb17fa3a4b8ba07c26293aa473ec : SUCCESS in 49m 24s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/4c1b26b707a04bac8783e0619b250cc4 : SUCCESS in 1h 17m 09s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/49cc125f23594f439ef7d62dc07720ef : SUCCESS in 31m 41s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/7c94963eb6e646fe92360ac5100f70d1 : SUCCESS in 2h 20m 43s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/ef2567497495498d9b77493f8e705f5a : SUCCESS in 2h 06m 02s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/03929a1861f24045b7344dd6f85bd14d : SUCCESS in 2h 01m 55s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/513200bd580c411d8a0ab40e8abc38f5 : SUCCESS in 2h 00m 41s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/faba3cf84b144c85b11f5f0f6921ec1d : SUCCESS in 2h 05m 16s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/7e6ffe4911624ec9878253628629688a : SUCCESS in 2h 33m 52s (non-voting)","accounts_in_message":[],"_revision_number":26},{"id":"2b969951931272ae444a9025cddaf90b4450f175","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2023-01-24 08:17:30.000000000","message":"Uploaded patch set 27.","accounts_in_message":[],"_revision_number":27},{"id":"0ec4a56ee090e408913c8dea43bbe1a8d9bfb895","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2023-01-24 10:12:08.000000000","message":"Patch Set 27: Backport-Candidate-2\n\n(2 comments)","accounts_in_message":[],"_revision_number":27},{"id":"9141e778ddd31bb39c31b879023458bb6d1bfa46","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-24 10:53:41.000000000","message":"Patch Set 27: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/df321bf8c9804bebafd9ba49e4c29da0\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/1ec0b3d2ccce4351b91bdf2de3292ad9 : SUCCESS in 1m 58s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/582a65e553324691bac541622c157848 : SUCCESS in 8m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/80f8e5b2a5b3473397493bd12a1755f1 : SUCCESS in 4m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/98c69f090c0146538900e21caf796a3b : SUCCESS in 4m 33s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/1f72f0497c494964b89f10590b1b06d7 : SUCCESS in 4m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/30c80a58be12417e85f051fd10bba8ca : SUCCESS in 11m 05s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0837ae5e3e5e428c898d3e31db733966 : SUCCESS in 4m 19s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/9157ad6603b148de9edc5b1e5160a547 : SUCCESS in 4m 42s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/75a68ad7b694477ca84e1114c7e00386 : SUCCESS in 8m 40s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/d06c5c4e36c74bdb95a4c4a02a02f93f : SUCCESS in 4m 28s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/ddc86a102c374c5ab64fdeb74b94cbf0 : SUCCESS in 7m 03s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/41d3382835b94a2dbc27b751cde8e0dc : SUCCESS in 7m 27s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/3c7f795870254d5a92332ad3770ef972 : SUCCESS in 7m 04s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/8b09b75e4cd64dfea8ba0b9cebff0f47 : SUCCESS in 1h 33m 11s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/9b5bc6e655644fafacad809dbae9b512 : SUCCESS in 1h 29m 37s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/e52d1dfb23014f4ea05da4905ab55242 : SUCCESS in 34m 12s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/b1bd273915244f12a89636bcdb418c76 : SUCCESS in 52m 06s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/609607ae6fc941fe9ba94e7134016235 : SUCCESS in 1h 19m 00s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/01690a11ae1f4109ab6b1eac1439eea6 : SUCCESS in 32m 07s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/f3aeb76318f0492c98b96e721384ffa6 : SUCCESS in 1h 49m 07s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/87fca119106845dfa522e586fa930c08 : SUCCESS in 2h 13m 40s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/599a5e24ba8a434891e417d3d0993977 : SUCCESS in 2h 02m 49s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/9772fce0c37e4f6faad9766e13ad43a6 : SUCCESS in 1h 46m 30s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/aa414c2276c540ce9b75c7e7958b251f : SUCCESS in 1h 51m 50s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/a56840950181473887a2a75e24eaebac : SUCCESS in 2h 33m 49s (non-voting)","accounts_in_message":[],"_revision_number":27},{"id":"cf9492f02c3bf53995f1784de7fcd7af9898d544","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2023-01-24 12:29:10.000000000","message":"Uploaded patch set 28.","accounts_in_message":[],"_revision_number":28},{"id":"ec2b9b51982bd00cba13466cb293d234dd6dcd33","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2023-01-24 12:37:03.000000000","message":"Uploaded patch set 29.","accounts_in_message":[],"_revision_number":29},{"id":"f53b5a66428ade3933e657c9e912510f06658819","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-24 15:14:13.000000000","message":"Patch Set 29: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/468caa7ff91545db9ce26aabd12b9aca\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/79486fcb46ea4a48af6c529c5885b634 : SUCCESS in 2m 10s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/489f071db2a94971b41c11ee4ece02a3 : FAILURE in 4m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fc468c3db68841608f4fdf5a7c625a01 : SUCCESS in 5m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ef6acfb3d3f4421596c37d35dcea1f9a : FAILURE in 4m 46s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/965b605b60314fcda968d812a6e70265 : FAILURE in 4m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6e47b3120df24ecb8ded04f85e92f77c : SUCCESS in 13m 13s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0722c7cf42cb4c7f8cb4cd22f62e92eb : SUCCESS in 4m 04s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/9de72aa1adf047c4b39502bacb6aea14 : FAILURE in 5m 24s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/78138cc5a13642cc8be7d3e0c6e8fd91 : SUCCESS in 7m 06s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/ed6b92fa13cd4581a43772c94d9f0c50 : SUCCESS in 4m 05s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/2fbd79e77c884c1b9a219a429e4b3fcb : SUCCESS in 9m 31s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/7737adc9a8b749ceb09564e532e39605 : SUCCESS in 7m 05s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/437fe4484d724e0683c1130d80f34374 : SUCCESS in 7m 19s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/e1fb403ee5f140e3b738a12c17be0b4b : SUCCESS in 1h 34m 42s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/a6804196348d4cc59302c48a49fcd8db : SUCCESS in 1h 51m 05s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/0f7545134d794369b8350fe74f971315 : SUCCESS in 36m 42s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/e19dd31b11f84c09bbf9e99dcebb2f31 : SUCCESS in 50m 12s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/85effe52d14b4f7f9e6b67580279c78e : SUCCESS in 1h 12m 15s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/fe506be475b84efc9da5ddb05cb4a1de : SUCCESS in 33m 25s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/d6fbc924fdb84d11a7e6cdb37944b4eb : SUCCESS in 1h 49m 28s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/91221c7a0ed0411d955e589636545e75 : SUCCESS in 2h 15m 15s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/901ede5a6da446a6adc298bd05437484 : SUCCESS in 2h 14m 02s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/bab5ae8389624f5bbaec354d9094c0f0 : SUCCESS in 1h 34m 55s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/b0dd5a1f170d4d2b876d088f5b16ea22 : SUCCESS in 1h 55m 03s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/14c1064abc664e80a21ef54d8c6f8568 : TIMED_OUT in 2h 34m 56s (non-voting)","accounts_in_message":[],"_revision_number":29},{"id":"58ef8c95e86d586c3d21026e59df6a6b3909c856","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2023-01-24 16:11:46.000000000","message":"Patch Set 29: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":29},{"id":"35fd2e6ec7ddb2635fbee2b3a20bc539c9514850","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2023-01-24 16:15:14.000000000","message":"Patch Set 29:\n\n(1 comment)","accounts_in_message":[],"_revision_number":29},{"id":"031b5446df96f9b584874dc204fe139eba706ebf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2023-01-24 16:43:46.000000000","message":"Uploaded patch set 30.","accounts_in_message":[],"_revision_number":30},{"id":"d4c7d81eb71f0879eb1aba2d551c005de8159211","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-24 19:32:13.000000000","message":"Patch Set 30: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/44947991c3894edb9af34136598d0240\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/4ea04b6950104e3084e00856c31fa057 : SUCCESS in 2m 01s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b419c280d8c743758d23fad5fbbf560f : SUCCESS in 10m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1948cf4461404ccfad42e3c3dda7df7f : SUCCESS in 4m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c4aa62b89bdd48ac87ec59e6d2cf5d61 : SUCCESS in 4m 31s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d70e4420929542028e300e0b6657ee5d : SUCCESS in 4m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/580118459c7b44fb8a386f5eb9c92fe5 : SUCCESS in 11m 23s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/82541572604d459f8fb9b66775cdbccb : SUCCESS in 5m 00s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/e8419dddada14861890fbf0b1ec3fc82 : SUCCESS in 4m 33s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/4bd2a65065c84bb6ba0e2b063b5ce894 : SUCCESS in 7m 50s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/fb7db021f735449082f17ae35c1f3396 : SUCCESS in 4m 53s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/bcca77ebbb9e4a4095d47abf42b18ad8 : SUCCESS in 7m 33s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/f73ad2322694404a90c4e419601301fa : SUCCESS in 7m 39s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/d8d68328210e49b4b825d30f793b4423 : SUCCESS in 8m 13s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/5dfe4a86ff754ef18861ffdb553285ce : SUCCESS in 1h 26m 15s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/0e941512136a4457b16a161be35c1f8c : SUCCESS in 1h 41m 12s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/6081d03e99904268a09217d21b9fd53e : SUCCESS in 35m 49s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/200f163cc9564fa8a0a4b5126a9188f0 : SUCCESS in 48m 46s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/6106cfeddf8f41f18e3ac563d65696d8 : SUCCESS in 1h 21m 57s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/7ab3ab5ea0ba4b91ac51e75f9e452753 : SUCCESS in 31m 22s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/39db93cfa1094ab1a715917b5a860a4d : SUCCESS in 2h 03m 34s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/b810f28ba42d4917b2a22f9edb682968 : SUCCESS in 1h 57m 44s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/f87dcee8c7354831ab47700d84a9bf75 : SUCCESS in 2h 06m 33s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/41e701cef5e046938bc6f344f06da827 : SUCCESS in 1h 45m 20s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/08f5e045566945d290d5ba1b74b5358e : SUCCESS in 1h 38m 33s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/29ff378db36d4d3abaac98420afb00ce : TIMED_OUT in 2h 34m 40s (non-voting)","accounts_in_message":[],"_revision_number":30},{"id":"9c1b434a90893b341b7a8277b76b71931be72a9d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2023-02-06 11:32:04.000000000","message":"Uploaded patch set 31.","accounts_in_message":[],"_revision_number":31},{"id":"5810d4ce481d90d509c1aab1c9e882cbed49c433","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-06 14:06:48.000000000","message":"Patch Set 31: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/0f39a01c50934082a5e547aef2dfdc16\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/e10b9968acd44291a9e48a0749945979 : SUCCESS in 1m 53s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1ae7bd8d40594a189d551d6db3b587a8 : SUCCESS in 9m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/802f24f1a9d04da2891516e28bdb5377 : SUCCESS in 5m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e7107f83f8d44ab9beeead9c7484c35a : SUCCESS in 7m 22s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/26ece0a657084e5db019903645a9887c : SUCCESS in 4m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5e16a274926b4e22a4d641ba3ecf0fb0 : SUCCESS in 11m 55s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9b0bd1f0ebff478bb3d84d29567195ff : SUCCESS in 5m 11s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/422cd9e09cd944e38deaf70d9ecfe80b : SUCCESS in 4m 42s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/544f487988a34b2d8efde0f3d788d99a : SUCCESS in 7m 46s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/b5ede3df67dc4794924a56c8d4fedbed : SUCCESS in 5m 01s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/50940325a2a042ef81233bb9b7894d65 : SUCCESS in 7m 28s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/2989595538534073ab435f2faae53cd4 : SUCCESS in 8m 46s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/4be2c8652ea2483aa712aaece359bd83 : SUCCESS in 6m 34s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/55baf2827dca4e57ae74f49839c1f354 : SUCCESS in 1h 21m 55s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/2fbedf1a89384e33a5807b905a5f5171 : SUCCESS in 2h 11m 42s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/14968c4cc3f34e00a774e8f0300101c3 : SUCCESS in 40m 45s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/950b802d41f04812a335d43e818ac8be : FAILURE in 39m 32s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/6de7cee086b7484a91f52f02f155fb20 : SUCCESS in 1h 14m 53s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/985df43ca5dd4c6d93f61f690a80ae7c : SUCCESS in 37m 57s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/2ea57b8d5a024ea8b05f1b3eca90e404 : SUCCESS in 2h 20m 56s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/00998aaf444a434f8049ef4e6a688f84 : FAILURE in 40m 11s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/98272aa3ad4d40abbf1bca13da9650f0 : SUCCESS in 2h 10m 13s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/c7288cc3c3484a2ba143b3078d54e1aa : SUCCESS in 1h 57m 59s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/829681c3a8d64e5aaccee811eac57d4d : SUCCESS in 2h 07m 05s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/f240337ed73d46f3937038ae4ce0b68c : SUCCESS in 2h 27m 54s (non-voting)","accounts_in_message":[],"_revision_number":31},{"id":"ba2e38b404ba468120ec5d4956ef43faebffb9b4","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2023-02-06 14:27:23.000000000","message":"Patch Set 31:\n\n(1 comment)","accounts_in_message":[],"_revision_number":31},{"id":"42ea631686f191cad270315d1359b032c42c8ef3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-06 17:24:40.000000000","message":"Patch Set 31:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/2c8ca041aa7141bba7491c0d8f3b3657\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/a81ba064b83042d1a95aa31cb329428f : SUCCESS in 2m 20s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/25cbfcef72924d31b168eafa22a9edad : SUCCESS in 11m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/988b15fd24f7404eb2a582750d381cce : SUCCESS in 4m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/426e95c4d9b3463db771e2e05ca7efa2 : SUCCESS in 4m 58s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/11b5eb4117d444249ff4887ca7eb002e : SUCCESS in 5m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8842c4f795094c0b86f4ae712d75361c : SUCCESS in 10m 36s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/90e16f7dba6544369415512b64fce95c : SUCCESS in 5m 03s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/f4255b29eaf14693a153eef8dd2b2f7b : SUCCESS in 6m 08s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/2e2c99ecdfb4430fa233e42b3c71ef9c : SUCCESS in 8m 02s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/61f632ae34b1408abc1b0356ed108647 : SUCCESS in 4m 12s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/00fd55a09beb44e59fe142a7af251496 : SUCCESS in 7m 11s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/2abaf2baee374db3bf540cae75874cd2 : SUCCESS in 6m 52s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/2717971f14c0403596c8612787986b5b : SUCCESS in 8m 06s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/c3c3759fbdbf4ecab381d05afaf41498 : SUCCESS in 1h 38m 34s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/4d563fa51c784504b20fafae9fbced45 : SUCCESS in 1h 45m 30s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/7964e5ccca764204a63fd10045a73981 : SUCCESS in 33m 29s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/ca0a84af6a704bb6b3b521c846faab19 : FAILURE in 36m 39s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/660d5ae8196042c880df4a37d3cb11e4 : SUCCESS in 1h 19m 00s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/249b3cd1ce754037a4d841cf1e3b7cf9 : SUCCESS in 30m 03s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/e2701fdb178445cabd34de282a74e2cd : SUCCESS in 1h 39m 11s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/417e533b0a4140a3896a9aff0d28dd29 : SUCCESS in 2h 55m 28s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/73256145227644b5b9a4b37f8a850c99 : SUCCESS in 1h 53m 18s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/0efaaaef29964484830d24670237a3ff : SUCCESS in 1h 52m 00s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/facc15b52bea422a8c19efc1f9d50c78 : SUCCESS in 1h 59m 50s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/3be9046c0854486580e332e38540de68 : SUCCESS in 2h 16m 32s (non-voting)","accounts_in_message":[],"_revision_number":31},{"id":"28c7816ef437ff91c884e093eba6bbe3d96e1145","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2023-02-07 10:47:08.000000000","message":"Patch Set 31:\n\n(1 comment)","accounts_in_message":[],"_revision_number":31},{"id":"a8f3015a45c17b1c55e964b71da8d356cedd467d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-07 13:39:56.000000000","message":"Patch Set 31:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/0ee4cf8d5e2f4703a9872bdbbf0fe694\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/2ea4b097f75f409e84a440d04e942d3f : SUCCESS in 2m 41s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8adbff60595d4d668c292f36b78e7da8 : SUCCESS in 9m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/75090f2902024ddc9ba3da435d53c7ca : SUCCESS in 5m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/89b3fabb93674f278cecafeedf0d9147 : SUCCESS in 5m 03s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6b9b930b9b634c21b7f93061545e8b12 : SUCCESS in 5m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/68900f1a689d4e33afafe5f30554984c : SUCCESS in 17m 30s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fb02c708722b402f901bc3cb3959fabe : SUCCESS in 5m 00s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/82fbd0302b4f42d180102b90bc27ce45 : SUCCESS in 5m 15s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/0dc8c20d926944719b77d5d920040a4d : SUCCESS in 6m 41s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/94aad9bb75694905ba724187d8432978 : SUCCESS in 4m 07s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/91341dc1e47b4c96b97014f1274434fe : SUCCESS in 6m 46s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/19c641a755374205ac0b5f69f88c2c2c : SUCCESS in 7m 39s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/6f8115046bba4b2a9ded8e09d2c8f2f9 : SUCCESS in 8m 51s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/df49c6561aee42f5ae555bf0bf62f8cd : SUCCESS in 1h 28m 43s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/85e7d85314994e759ec5caf78aee38f9 : TIMED_OUT in 2h 33m 58s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/1d94bd6a9f3349578ef330796d6e18ef : SUCCESS in 46m 47s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/4365a4e6baee4b3a971630c3efc39785 : SUCCESS in 53m 19s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/e4ed45cbd492445eb41cafc76e09cffa : SUCCESS in 1h 11m 03s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/be0f2fc0d1a445938e3b738fe4912783 : SUCCESS in 44m 41s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/3c08ed1e98154299b5fe3cf3abe91875 : SUCCESS in 1h 42m 32s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/7e7400b1393944ada7ed206a740e88f3 : SUCCESS in 2h 32m 18s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/580760b500f04b00b42e9069f83c819a : SUCCESS in 1h 46m 01s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/25d21b015fb2493f83c309cee81e3e86 : SUCCESS in 2h 00m 59s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/9a5cf91187b1420d8605178df5cdda8d : SUCCESS in 1h 44m 42s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/39c0755f5f8e46dabc75abeeb8e12f08 : TIMED_OUT in 2h 34m 56s (non-voting)","accounts_in_message":[],"_revision_number":31},{"id":"7b86a51e02c6c547f004807251fc06666a15012a","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2023-02-07 15:22:45.000000000","message":"Patch Set 31:\n\n(1 comment)","accounts_in_message":[],"_revision_number":31},{"id":"9e994fc8c8fad63fe28f7e56d07a023b4677a128","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-07 18:03:16.000000000","message":"Patch Set 31: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/adc912fc53a04c1381f9462c5339ed57\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/6985495e12c049d1bd3792ae27e85e51 : SUCCESS in 2m 02s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/73539c6f0c9b4516911970c0b6c3a62c : SUCCESS in 9m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eafa8fcbd8cb43d181b4c6aafc590329 : SUCCESS in 5m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e741d8cc30634153944f982234d69575 : SUCCESS in 4m 15s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e21c3b6865574eb482d6047b099dd6d1 : SUCCESS in 4m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c839370faf3b4b01975fc19f4ba49fd5 : SUCCESS in 11m 03s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/442150cb1bda4bedb9e53b809168fcf7 : SUCCESS in 4m 46s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/86e18be35332477faba7b84986cf1fd3 : SUCCESS in 5m 16s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/0a0e228f45a043aca2e670fabf7a5c8b : SUCCESS in 8m 32s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/ab59b39f923847be803f420310a1b53e : SUCCESS in 5m 02s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/886eb28c2a5c4fae81300cf73d342f17 : SUCCESS in 7m 40s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/ff62eaf347124a5fa636eff6ce2ed40f : SUCCESS in 8m 12s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/4fa944f32e744be99974e27cd7083772 : SUCCESS in 6m 42s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/a2679f64eeaf46a9b8b2830d35a15abc : SUCCESS in 2h 09m 50s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/f51fb42041c545f2ac3ec70eec51291b : SUCCESS in 1h 37m 58s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/a13c88f4118e4a128d15cd76fd0fb7fb : SUCCESS in 50m 32s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/47abfc80adb140218b6322a8815d5c67 : SUCCESS in 40m 32s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/c1d5e7b218584d7cbedf981e49370241 : SUCCESS in 1h 19m 18s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/ba0f85aafdf4498fb6dad82b6b288279 : SUCCESS in 31m 40s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/8e484a365ec44ab09163d5fba8a5844d : SUCCESS in 2h 33m 14s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/018a607d207a46a8923ebc8ee067beda : SUCCESS in 2h 34m 59s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/c5a407b3bce04eedbf042047dbd14d52 : SUCCESS in 2h 20m 44s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/03e383b9dcc24f2fac189b466ce7cb1e : SUCCESS in 1h 51m 29s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/ef7dc5ca462b498588ea2e52a1779243 : TIMED_OUT in 2h 34m 16s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/9a485bb6812a4b818de3a6e7b70d2b58 : SUCCESS in 2h 33m 23s (non-voting)","accounts_in_message":[],"_revision_number":31},{"id":"bc75519eb5da3d8540546af4593b31b587cc307b","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2023-02-13 10:28:12.000000000","message":"Patch Set 31: Backport-Candidate-2\n\n(3 comments)","accounts_in_message":[],"_revision_number":31},{"id":"85b70907a43d21f4d3bfffe558e0776fd80e3e90","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-03-28 14:53:30.000000000","message":"Patch Set 31: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":31},{"id":"f90c03f4c3d553f06dd9d1999fbff4935b301f20","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-03-31 21:42:47.000000000","message":"Patch Set 31:\n\n(6 comments)","accounts_in_message":[],"_revision_number":31},{"id":"2c04bf78852e895c91570ab1591ea1cf4bfed236","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2023-04-03 09:59:45.000000000","message":"Uploaded patch set 32.","accounts_in_message":[],"_revision_number":32},{"id":"ba804794172fb0e318e7afc19983a68be1dd9e5a","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2023-04-03 10:02:54.000000000","message":"Patch Set 32: Backport-Candidate-2\n\n(8 comments)","accounts_in_message":[],"_revision_number":32},{"id":"569e398a05e2336c00b617911141ef371d2354c0","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-04-03 10:10:25.000000000","message":"Patch Set 32:\n\n(1 comment)","accounts_in_message":[],"_revision_number":32},{"id":"174fd28e8eccf6c5c605cc4777179eef2c4e11d1","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-04-03 10:11:24.000000000","message":"Patch Set 32:\n\n(1 comment)","accounts_in_message":[],"_revision_number":32},{"id":"2dbd75946a0c083cb98b89acc27520e0ed9034d1","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-04-03 10:24:06.000000000","message":"Patch Set 32:\n\n(1 comment)","accounts_in_message":[],"_revision_number":32},{"id":"c33b40086f8629223e84baae107cb9235cc25db9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2023-04-03 12:34:12.000000000","message":"Uploaded patch set 33.","accounts_in_message":[],"_revision_number":33},{"id":"d34ea5e7abfa6251b62450f216046d590641b642","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2023-04-03 12:38:32.000000000","message":"Patch Set 33: Backport-Candidate-2\n\n(2 comments)","accounts_in_message":[],"_revision_number":33},{"id":"172e2fdef1ad534151268ef202ed7fdc8a8069b1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-03 15:34:18.000000000","message":"Patch Set 33: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/1d4459a629514c0e89dfbb0064f52745\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/9e3e639eaa744ef1a79dc19defab2676 : SUCCESS in 2m 01s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/721f7cae4ee84a789b53ebf984dbd029 : SUCCESS in 10m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/53d2b630bd714a7e8fb87dc963d01050 : SUCCESS in 5m 35s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/faacec48a6604450a4865d68b873498b : SUCCESS in 4m 20s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6ede4567586d4fbc96e9ecdd708b9ac8 : SUCCESS in 4m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6cf02a2488c9458abc47176bb13a60fa : SUCCESS in 12m 59s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f949e3efdd944059b27e0e57f9caf4f2 : SUCCESS in 5m 50s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/12cb8706a76a42f2849bd3967632dd7a : SUCCESS in 5m 14s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/890ff0f5523942de8c5f661c1f5754ef : SUCCESS in 8m 44s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/98b0907cc64740c390b07bbd1b36acef : SUCCESS in 4m 46s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/7042b5ded2ca4b1d8056f40ba398af53 : SUCCESS in 7m 28s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/00d62ae387cf47c8b730ff5f095f3dbb : SUCCESS in 6m 56s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/1ba98543786641679ef851649ec43c86 : SUCCESS in 8m 24s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/823506157928452b812a19424e8fd38c : SUCCESS in 1h 51m 14s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/97a45e1317404894b87d56cadb4b6157 : TIMED_OUT in 2h 33m 43s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/c9a891efdba24031b4d31ee09f324057 : SUCCESS in 42m 27s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/63c5296676ae45c59221b9a0cea1eccb : SUCCESS in 48m 11s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/e4fd2bfba6014c4cbbcd2816145aaf59 : SUCCESS in 1h 06m 47s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/c5259d25a0464820b001c89c6e87d9b5 : SUCCESS in 36m 26s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/cf67fa4a36324844adda42fb951923d8 : SUCCESS in 1h 53m 09s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/bdb4637b0a78436ba0cbfa5b50c4d2ed : SUCCESS in 2h 55m 44s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/1c0b9bfabc85488c99b48a0daf1a2d41 : SUCCESS in 2h 07m 07s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/b9580ed7feb242b995f5a8b99b8a3ee1 : TIMED_OUT in 2h 33m 07s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/dd5b99be49a544ff93346a9a033108ff : SUCCESS in 2h 14m 38s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/6eaa6625859b4561871cab1c9b55638d : TIMED_OUT in 2h 33m 40s (non-voting)","accounts_in_message":[],"_revision_number":33},{"id":"44a718747a1031efa4123779aa2bc6d4e088cb50","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2023-04-03 17:45:26.000000000","message":"Patch Set 33:\n\n(1 comment)","accounts_in_message":[],"_revision_number":33},{"id":"cdc7d0b9dd53bbc3162d140c473837cf6d7a45aa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-03 20:23:11.000000000","message":"Patch Set 33: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3c7077ae0cf146b784e4253f9eaf8e5e\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/d1ac748de0a94fad9b1a59fe299b529a : SUCCESS in 2m 03s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/683f98a7285a4bd0ab5ba44186e252b6 : SUCCESS in 9m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cd1390edd33a4cbf92092300d7689816 : SUCCESS in 5m 02s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fa4b0737a121451c901e45534651c5d4 : SUCCESS in 5m 06s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/81176c8b2f4f43cfab049107c16173ea : SUCCESS in 4m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9f2533b78626458bbf86b90975bb3684 : SUCCESS in 10m 27s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9c1876980253420481d485925e3cdf57 : SUCCESS in 5m 23s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/1fb0770a17104a38a748db6ca15bb28a : SUCCESS in 5m 14s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/8da1f2995cc74054ad3dc741265b9d7f : SUCCESS in 8m 41s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/f163e55911ab411692e5e18776091927 : SUCCESS in 4m 18s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/d2523d00c797494290f9e0ee07797a7c : SUCCESS in 7m 34s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/c052fdf3cffa495fadb6598c6e1d353a : SUCCESS in 7m 19s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/1eb4b4d85f1849639fd9af52f5f7684a : SUCCESS in 8m 50s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/696fb740d9b34e3c930b81079add1a43 : SUCCESS in 1h 16m 32s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/d3bd923fca0e467bb9389cb800729e1d : SUCCESS in 2h 00m 21s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/a62d71b976cd4d688f44472275a667fc : SUCCESS in 43m 51s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/4b296d01710c49e4a4e2183f70de508c : SUCCESS in 40m 39s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/7993fb9d1b1c4d89b44870c29a5ae2c0 : SUCCESS in 1h 11m 45s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/5f192249b4bd4233b4f89eaf423a56d2 : SUCCESS in 34m 12s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/b36afb957f9a4d279cf53ba601f91b90 : SUCCESS in 2h 31m 00s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/04343ca04b044697a58fd0f3b7372739 : SUCCESS in 2h 11m 49s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/7a5aa9c4c1b14f0ca4f024f22f4cf6be : SUCCESS in 2h 09m 54s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/e1a3836ebccb49cd843d1855f33531a8 : TIMED_OUT in 2h 33m 59s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/aa1895db2d794ecb842dee69dba96420 : SUCCESS in 1h 56m 11s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/e9d5935468ec4dfcb9e958c94d3da69a : SUCCESS in 2h 26m 16s (non-voting)","accounts_in_message":[],"_revision_number":33},{"id":"eb90acc5cff0501b005323cd18811504199792d0","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-04-03 20:41:01.000000000","message":"Patch Set 33: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":33},{"id":"6b9207fa3ff4dfda3b35534c036a125947aca29f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2023-04-14 11:06:32.000000000","message":"Uploaded patch set 34.\n\nOutdated Votes:\n* Backport-Candidate-2 (copy condition: \"changekind:NO_CHANGE OR changekind:NO_CODE_CHANGE OR changekind:TRIVIAL_REBASE\")\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":34},{"id":"7628d4bede7a1e15baca85000279b04eaeee931f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2023-04-14 11:11:49.000000000","message":"Uploaded patch set 35.","accounts_in_message":[],"_revision_number":35},{"id":"6673ee17a811d3d4a92331f00ca6abe98009b84d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-14 14:04:21.000000000","message":"Patch Set 35: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2cbafb4530aa454fa2715c819da1241a\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/3c615124b51144f4ae69b6b6009fcf54 : SUCCESS in 2m 02s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/05c5e30db94041299911e5772fc78134 : SUCCESS in 9m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d085fd95bb4248c788be7695d38bc0e8 : SUCCESS in 6m 01s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/32256000c4494978a3f9c29154da34fd : SUCCESS in 4m 30s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0b462da41ddd4af08434951e566b44a5 : SUCCESS in 5m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/53b6c43894b24bfea282b434e7d61702 : SUCCESS in 11m 24s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9514ee66778a4ccc87057a80aa5c0956 : SUCCESS in 5m 45s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/5d8dcd96defd49358b7c4f129672eb5c : SUCCESS in 5m 05s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/4701ec9d68e544b99298efcdd12c6f9f : SUCCESS in 8m 34s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/45083ce87f37497fa24f7788785f899d : SUCCESS in 4m 40s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/a54f21d02beb473c8562a3e65384ecab : SUCCESS in 7m 35s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/f492e57d1fe043da9cab081d3e5bc100 : SUCCESS in 7m 07s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/b241148c6a8c4e309102dc83de29119b : SUCCESS in 7m 54s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/cfc041de903c40adbdd390ad7438b506 : SUCCESS in 1h 39m 07s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/16febfbe27684287bc31b7be8deb6fc7 : SUCCESS in 2h 13m 24s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/455f62a2390f41d7b31787aeea0170c4 : SUCCESS in 39m 09s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/50a30c611fd64f5599c6aeb1e400ccde : SUCCESS in 46m 33s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/990ef5fb815b4427bc542f96b432b467 : SUCCESS in 1h 07m 34s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/96de543b10d840ffaad53dd1d65f4174 : SUCCESS in 33m 19s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/2af5981865214251bc73c8777c906f9a : SUCCESS in 2h 18m 44s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/1b64bdc061704127827f048a0897852b : SUCCESS in 2h 49m 32s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/f0539bbf296d43509b831cee96b0b8e4 : SUCCESS in 2h 05m 07s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/034eeadbfc2a42a2b5f8fd2129f8602d : TIMED_OUT in 2h 33m 49s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/a4f12710071145c88c4b92614d718344 : SUCCESS in 2h 21m 28s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/da32dce3b5b34cac8b51cf36009a949a : TIMED_OUT in 2h 33m 55s (non-voting)","accounts_in_message":[],"_revision_number":35},{"id":"2bb3ce65c17bbd5651d66ffba305ecb317f035b8","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2023-04-20 14:56:25.000000000","message":"Patch Set 35: Code-Review+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":35},{"id":"b917f34d168dfcf698903c408cfed6dd81549868","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-04-26 06:47:58.000000000","message":"Patch Set 35: Code-Review+2","accounts_in_message":[],"_revision_number":35},{"id":"854b1cf8b283555845c636d8ef2be36db93f0913","author":{"_account_id":34451,"name":"Fernando Royo","email":"froyo@redhat.com","username":"froyo"},"date":"2023-04-26 14:57:11.000000000","message":"Patch Set 35: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":35},{"id":"e230a8b94feeeca929afd672fe997eca0f2f2c77","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-05-03 16:10:02.000000000","message":"Patch Set 35: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":35},{"id":"663ded920859b49465f8a108c3923360fa744a15","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-03 16:10:41.000000000","message":"Patch Set 35: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":35},{"id":"a6bf5c38a398d01c8e1dd848a034925c10c9fb67","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-03 18:39:16.000000000","message":"Patch Set 35: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/793ff7fb5a434791ad6b22149018ff26\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/b91d4efd4bd84fc4b07e930cb24f72a4 : SUCCESS in 1m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1c1b4d75596445348b8aac64338ab344 : SUCCESS in 5m 39s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0b9b753bf67045b59fb5ca593ffed4b8 : SUCCESS in 5m 01s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/95955da24e7e45fc84b34ad5920a0888 : SUCCESS in 5m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0465d17b2bf54851bad2d9accd179c7a : SUCCESS in 12m 18s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d1b6ed60c5d04f54930886b525f24151 : SUCCESS in 4m 39s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/20789e1c740841c1937f4420a3b3cf61 : SUCCESS in 4m 25s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/a7ec4bb3875c4ab6a0aa98ac2a86a4fd : SUCCESS in 7m 10s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/037a5d9397b24d39bc97f57ecf4eed78 : SUCCESS in 1h 16m 56s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/bc959d177b4c4c0f8734bc06fc2a8d45 : SUCCESS in 2h 28m 29s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/2f9cdee6cd294a33aa534cca0c6b8467 : SUCCESS in 43m 28s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/bf445c3ee0364d1d9f6fbccdaf0179d7 : SUCCESS in 47m 13s","accounts_in_message":[],"_revision_number":35},{"id":"07a59aa962d39005ec2f814526003c1baffac1da","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-03 18:39:18.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":35},{"id":"d3a927f8f4904def8ad1e743e1ddd093e582dbb5","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-03 18:40:41.000000000","message":"Patch Set 35:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/db032e003bd74520a273304b222f86de\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/70cbbf97560942d1b60585085b16d2c5 : SUCCESS in 1m 05s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/53cd519b12fc48508d1caaad4a0ef35a : SUCCESS in 39s","accounts_in_message":[],"_revision_number":35}],"current_revision_number":35,"current_revision":"3e6fd13bfb000785d5eb9efe2139ee739e074460","revisions":{"a9c9c92bc1ae9150132feb1688517e8c6baf5c3d":{"kind":"REWORK","_number":1,"created":"2022-12-01 11:56:28.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/27/866327/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/27/866327/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/27/866327/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/27/866327/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/27/866327/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/27/866327/1"}}},"commit":{"parents":[{"commit":"529f8e2f52010ede293bebdaf7cd20a61758fe58","subject":"Merge \"Fix full graph loadbalancer creation if jobboard is disabled\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/529f8e2f52010ede293bebdaf7cd20a61758fe58"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-12-01 11:56:19.000000000","tz":60},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-12-01 11:56:19.000000000","tz":60},"subject":"WIP remove python-neutronclient","message":"WIP remove python-neutronclient\n\nChange-Id: I686cfdef78de927fa4bc1921c15e8d5853fd2ef9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a9c9c92bc1ae9150132feb1688517e8c6baf5c3d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a9c9c92bc1ae9150132feb1688517e8c6baf5c3d"}]},"branch":"refs/heads/master"},"9dd42843a52e262b0be7a31c72c3f96e0a7f4b69":{"kind":"REWORK","_number":2,"created":"2022-12-01 16:10:21.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/27/866327/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/27/866327/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/27/866327/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/27/866327/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/27/866327/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/27/866327/2"}}},"commit":{"parents":[{"commit":"529f8e2f52010ede293bebdaf7cd20a61758fe58","subject":"Merge \"Fix full graph loadbalancer creation if jobboard is disabled\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/529f8e2f52010ede293bebdaf7cd20a61758fe58"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-12-01 11:56:19.000000000","tz":60},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-12-01 16:10:10.000000000","tz":60},"subject":"WIP remove python-neutronclient","message":"WIP remove python-neutronclient\n\nChange-Id: I686cfdef78de927fa4bc1921c15e8d5853fd2ef9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/9dd42843a52e262b0be7a31c72c3f96e0a7f4b69"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/9dd42843a52e262b0be7a31c72c3f96e0a7f4b69"}]},"branch":"refs/heads/master"},"10c2a5801f57a8f62c65f3f7ffaafb056cf8056d":{"kind":"REWORK","_number":3,"created":"2022-12-05 15:35:07.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/27/866327/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/27/866327/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/27/866327/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/27/866327/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/27/866327/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/27/866327/3"}}},"commit":{"parents":[{"commit":"529f8e2f52010ede293bebdaf7cd20a61758fe58","subject":"Merge \"Fix full graph loadbalancer creation if jobboard is disabled\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/529f8e2f52010ede293bebdaf7cd20a61758fe58"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-12-01 11:56:19.000000000","tz":60},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-12-05 15:34:53.000000000","tz":60},"subject":"WIP remove python-neutronclient","message":"WIP remove python-neutronclient\n\nChange-Id: I686cfdef78de927fa4bc1921c15e8d5853fd2ef9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/10c2a5801f57a8f62c65f3f7ffaafb056cf8056d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/10c2a5801f57a8f62c65f3f7ffaafb056cf8056d"}]},"branch":"refs/heads/master"},"5cb18a12aa60bdd1265fe28df92100e6f11078e2":{"kind":"REWORK","_number":4,"created":"2022-12-07 12:27:00.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/27/866327/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/27/866327/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/27/866327/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/27/866327/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/27/866327/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/27/866327/4"}}},"commit":{"parents":[{"commit":"529f8e2f52010ede293bebdaf7cd20a61758fe58","subject":"Merge \"Fix full graph loadbalancer creation if jobboard is disabled\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/529f8e2f52010ede293bebdaf7cd20a61758fe58"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-12-01 11:56:19.000000000","tz":60},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-12-07 12:26:49.000000000","tz":60},"subject":"WIP remove python-neutronclient","message":"WIP remove python-neutronclient\n\nChange-Id: I686cfdef78de927fa4bc1921c15e8d5853fd2ef9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/5cb18a12aa60bdd1265fe28df92100e6f11078e2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/5cb18a12aa60bdd1265fe28df92100e6f11078e2"}]},"branch":"refs/heads/master"},"010b18222c4585ed18d65dca9875c2e6f5876db0":{"kind":"REWORK","_number":5,"created":"2022-12-12 10:07:46.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/27/866327/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/27/866327/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/27/866327/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/27/866327/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/27/866327/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/27/866327/5"}}},"commit":{"parents":[{"commit":"529f8e2f52010ede293bebdaf7cd20a61758fe58","subject":"Merge \"Fix full graph loadbalancer creation if jobboard is disabled\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/529f8e2f52010ede293bebdaf7cd20a61758fe58"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-12-01 11:56:19.000000000","tz":60},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-12-12 10:07:30.000000000","tz":60},"subject":"WIP remove python-neutronclient","message":"WIP remove python-neutronclient\n\nChange-Id: I686cfdef78de927fa4bc1921c15e8d5853fd2ef9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/010b18222c4585ed18d65dca9875c2e6f5876db0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/010b18222c4585ed18d65dca9875c2e6f5876db0"}]},"branch":"refs/heads/master"},"3fd4861521e34335deb2499e29beb5f5c6ee9e92":{"kind":"REWORK","_number":6,"created":"2022-12-12 18:04:58.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/27/866327/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/27/866327/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/27/866327/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/27/866327/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/27/866327/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/27/866327/6"}}},"commit":{"parents":[{"commit":"529f8e2f52010ede293bebdaf7cd20a61758fe58","subject":"Merge \"Fix full graph loadbalancer creation if jobboard is disabled\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/529f8e2f52010ede293bebdaf7cd20a61758fe58"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-12-01 11:56:19.000000000","tz":60},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-12-12 18:04:47.000000000","tz":60},"subject":"WIP remove python-neutronclient","message":"WIP remove python-neutronclient\n\nChange-Id: I686cfdef78de927fa4bc1921c15e8d5853fd2ef9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/3fd4861521e34335deb2499e29beb5f5c6ee9e92"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/3fd4861521e34335deb2499e29beb5f5c6ee9e92"}]},"branch":"refs/heads/master"},"50d409a6944ca99688ed2f398be9caa9a474fcb7":{"kind":"REWORK","_number":7,"created":"2022-12-15 11:01:29.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/27/866327/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/27/866327/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/27/866327/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/27/866327/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/27/866327/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/27/866327/7"}}},"commit":{"parents":[{"commit":"529f8e2f52010ede293bebdaf7cd20a61758fe58","subject":"Merge \"Fix full graph loadbalancer creation if jobboard is disabled\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/529f8e2f52010ede293bebdaf7cd20a61758fe58"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-12-01 11:56:19.000000000","tz":60},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-12-15 11:01:18.000000000","tz":60},"subject":"WIP remove python-neutronclient","message":"WIP remove python-neutronclient\n\nChange-Id: I686cfdef78de927fa4bc1921c15e8d5853fd2ef9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/50d409a6944ca99688ed2f398be9caa9a474fcb7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/50d409a6944ca99688ed2f398be9caa9a474fcb7"}]},"branch":"refs/heads/master"},"a130c69073af821510046b6d58c78bac47caa1c6":{"kind":"REWORK","_number":8,"created":"2022-12-15 15:41:50.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/27/866327/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/27/866327/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/27/866327/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/27/866327/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/27/866327/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/27/866327/8"}}},"commit":{"parents":[{"commit":"529f8e2f52010ede293bebdaf7cd20a61758fe58","subject":"Merge \"Fix full graph loadbalancer creation if jobboard is disabled\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/529f8e2f52010ede293bebdaf7cd20a61758fe58"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-12-01 11:56:19.000000000","tz":60},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-12-15 15:41:28.000000000","tz":60},"subject":"WIP remove python-neutronclient","message":"WIP remove python-neutronclient\n\nChange-Id: I686cfdef78de927fa4bc1921c15e8d5853fd2ef9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a130c69073af821510046b6d58c78bac47caa1c6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a130c69073af821510046b6d58c78bac47caa1c6"}]},"branch":"refs/heads/master"},"3f0e2b4f7bfbc086173fd35f491239b5cc7e1696":{"kind":"REWORK","_number":9,"created":"2022-12-16 12:32:32.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/27/866327/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/27/866327/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/27/866327/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/27/866327/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/27/866327/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/27/866327/9"}}},"commit":{"parents":[{"commit":"529f8e2f52010ede293bebdaf7cd20a61758fe58","subject":"Merge \"Fix full graph loadbalancer creation if jobboard is disabled\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/529f8e2f52010ede293bebdaf7cd20a61758fe58"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-12-01 11:56:19.000000000","tz":60},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-12-16 12:32:15.000000000","tz":60},"subject":"WIP remove python-neutronclient","message":"WIP remove python-neutronclient\n\nChange-Id: I686cfdef78de927fa4bc1921c15e8d5853fd2ef9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/3f0e2b4f7bfbc086173fd35f491239b5cc7e1696"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/3f0e2b4f7bfbc086173fd35f491239b5cc7e1696"}]},"branch":"refs/heads/master"},"d6cf48a7c7f5f9acb545692eca668923b441516d":{"kind":"REWORK","_number":10,"created":"2022-12-19 15:50:43.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/27/866327/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/27/866327/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/27/866327/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/27/866327/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/27/866327/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/27/866327/10"}}},"commit":{"parents":[{"commit":"529f8e2f52010ede293bebdaf7cd20a61758fe58","subject":"Merge \"Fix full graph loadbalancer creation if jobboard is disabled\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/529f8e2f52010ede293bebdaf7cd20a61758fe58"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-12-01 11:56:19.000000000","tz":60},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-12-19 15:50:35.000000000","tz":60},"subject":"WIP remove python-neutronclient","message":"WIP remove python-neutronclient\n\nChange-Id: I686cfdef78de927fa4bc1921c15e8d5853fd2ef9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d6cf48a7c7f5f9acb545692eca668923b441516d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d6cf48a7c7f5f9acb545692eca668923b441516d"}]},"branch":"refs/heads/master"},"2557b4254eedb07103e17bfd9aa76f80b740b63d":{"kind":"REWORK","_number":11,"created":"2022-12-20 15:59:18.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/27/866327/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/27/866327/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/27/866327/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/27/866327/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/27/866327/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/27/866327/11"}}},"commit":{"parents":[{"commit":"8d27bdb5462474dafc934d164f3a299bfca8dd89","subject":"Merge \"Filter ComputeWaitTimeoutException when jobboard is disabled\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/8d27bdb5462474dafc934d164f3a299bfca8dd89"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-12-01 11:56:19.000000000","tz":60},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-12-20 15:59:09.000000000","tz":60},"subject":"WIP remove python-neutronclient","message":"WIP remove python-neutronclient\n\nChange-Id: I686cfdef78de927fa4bc1921c15e8d5853fd2ef9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/2557b4254eedb07103e17bfd9aa76f80b740b63d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/2557b4254eedb07103e17bfd9aa76f80b740b63d"}]},"branch":"refs/heads/master"},"99f9c052e357f0c0082b8ea23681fa1cafc530e0":{"kind":"TRIVIAL_REBASE","_number":12,"created":"2022-12-20 16:09:25.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/27/866327/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/27/866327/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/27/866327/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/27/866327/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/27/866327/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/27/866327/12"}}},"commit":{"parents":[{"commit":"529f8e2f52010ede293bebdaf7cd20a61758fe58","subject":"Merge \"Fix full graph loadbalancer creation if jobboard is disabled\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/529f8e2f52010ede293bebdaf7cd20a61758fe58"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-12-20 16:08:29.000000000","tz":60},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-12-20 16:08:29.000000000","tz":60},"subject":"WIP remove python-neutronclient","message":"WIP remove python-neutronclient\n\nChange-Id: I686cfdef78de927fa4bc1921c15e8d5853fd2ef9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/99f9c052e357f0c0082b8ea23681fa1cafc530e0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/99f9c052e357f0c0082b8ea23681fa1cafc530e0"}]},"branch":"refs/heads/master"},"b8ba21f50c5b61ea8a6532c75d480e6f5be8fbdf":{"kind":"REWORK","_number":13,"created":"2022-12-21 11:30:58.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/27/866327/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/27/866327/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/27/866327/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/27/866327/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/27/866327/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/27/866327/13"}}},"commit":{"parents":[{"commit":"529f8e2f52010ede293bebdaf7cd20a61758fe58","subject":"Merge \"Fix full graph loadbalancer creation if jobboard is disabled\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/529f8e2f52010ede293bebdaf7cd20a61758fe58"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-12-20 16:08:29.000000000","tz":60},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-12-21 11:30:48.000000000","tz":60},"subject":"WIP remove python-neutronclient","message":"WIP remove python-neutronclient\n\nChange-Id: I686cfdef78de927fa4bc1921c15e8d5853fd2ef9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b8ba21f50c5b61ea8a6532c75d480e6f5be8fbdf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b8ba21f50c5b61ea8a6532c75d480e6f5be8fbdf"}]},"branch":"refs/heads/master"},"bdcc68749d0231e0f7cf62c6a15012e2d64bb454":{"kind":"REWORK","_number":14,"created":"2022-12-21 12:29:48.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/27/866327/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/27/866327/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/27/866327/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/27/866327/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/27/866327/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/27/866327/14"}}},"commit":{"parents":[{"commit":"529f8e2f52010ede293bebdaf7cd20a61758fe58","subject":"Merge \"Fix full graph loadbalancer creation if jobboard is disabled\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/529f8e2f52010ede293bebdaf7cd20a61758fe58"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-12-20 16:08:29.000000000","tz":60},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-12-21 12:29:35.000000000","tz":60},"subject":"WIP remove python-neutronclient","message":"WIP remove python-neutronclient\n\nChange-Id: I686cfdef78de927fa4bc1921c15e8d5853fd2ef9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/bdcc68749d0231e0f7cf62c6a15012e2d64bb454"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/bdcc68749d0231e0f7cf62c6a15012e2d64bb454"}]},"branch":"refs/heads/master"},"89f7a1675cc9e2095996f1ad8a7a11355e8fee3f":{"kind":"TRIVIAL_REBASE","_number":15,"created":"2022-12-29 10:00:56.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/27/866327/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/27/866327/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/27/866327/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/27/866327/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/27/866327/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/27/866327/15"}}},"commit":{"parents":[{"commit":"ac3618b140b1b8ba02f2263abf910c17b3f08aff","subject":"Merge \"Fix typo in docs section title\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/ac3618b140b1b8ba02f2263abf910c17b3f08aff"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-12-20 16:08:29.000000000","tz":60},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-12-29 10:00:41.000000000","tz":60},"subject":"WIP remove python-neutronclient","message":"WIP remove python-neutronclient\n\nChange-Id: I686cfdef78de927fa4bc1921c15e8d5853fd2ef9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/89f7a1675cc9e2095996f1ad8a7a11355e8fee3f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/89f7a1675cc9e2095996f1ad8a7a11355e8fee3f"}]},"branch":"refs/heads/master"},"23e39fffe5b161ff752de28b333d320adf469fed":{"kind":"NO_CODE_CHANGE","_number":16,"created":"2023-01-04 11:47:52.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/27/866327/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/27/866327/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/27/866327/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/27/866327/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/27/866327/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/27/866327/16"}}},"commit":{"parents":[{"commit":"ac3618b140b1b8ba02f2263abf910c17b3f08aff","subject":"Merge \"Fix typo in docs section title\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/ac3618b140b1b8ba02f2263abf910c17b3f08aff"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-12-20 16:08:29.000000000","tz":60},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2023-01-04 11:47:52.000000000","tz":0},"subject":"WIP remove python-neutronclient","message":"WIP remove python-neutronclient\n\n\nStory: 2010509\nTask: 47104\nChange-Id: I686cfdef78de927fa4bc1921c15e8d5853fd2ef9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/23e39fffe5b161ff752de28b333d320adf469fed"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/23e39fffe5b161ff752de28b333d320adf469fed"}]},"branch":"refs/heads/master","description":"Edit commit message"},"c4f2d817640ba8ae048e3395a84fcbd782639a5d":{"kind":"NO_CODE_CHANGE","_number":17,"created":"2023-01-04 11:49:40.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/27/866327/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/27/866327/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/27/866327/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/27/866327/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/27/866327/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/27/866327/17"}}},"commit":{"parents":[{"commit":"ac3618b140b1b8ba02f2263abf910c17b3f08aff","subject":"Merge \"Fix typo in docs section title\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/ac3618b140b1b8ba02f2263abf910c17b3f08aff"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-12-20 16:08:29.000000000","tz":60},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2023-01-04 11:48:57.000000000","tz":60},"subject":"WIP remove python-neutronclient","message":"WIP remove python-neutronclient\n\nStory: 2010509\nTask: 47104\nChange-Id: I686cfdef78de927fa4bc1921c15e8d5853fd2ef9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c4f2d817640ba8ae048e3395a84fcbd782639a5d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c4f2d817640ba8ae048e3395a84fcbd782639a5d"}]},"branch":"refs/heads/master"},"833ec9d23b84e675f2160401f9d8359b8f968e24":{"kind":"REWORK","_number":18,"created":"2023-01-05 11:07:27.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/27/866327/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/27/866327/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/27/866327/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/27/866327/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/27/866327/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/27/866327/18"}}},"commit":{"parents":[{"commit":"ac3618b140b1b8ba02f2263abf910c17b3f08aff","subject":"Merge \"Fix typo in docs section title\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/ac3618b140b1b8ba02f2263abf910c17b3f08aff"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-12-20 16:08:29.000000000","tz":60},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2023-01-05 11:07:10.000000000","tz":60},"subject":"WIP remove python-neutronclient","message":"WIP remove python-neutronclient\n\nStory: 2010509\nTask: 47104\nChange-Id: I686cfdef78de927fa4bc1921c15e8d5853fd2ef9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/833ec9d23b84e675f2160401f9d8359b8f968e24"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/833ec9d23b84e675f2160401f9d8359b8f968e24"}]},"branch":"refs/heads/master"},"b00ca7239d4610a8749e69d9049a58b38e850864":{"kind":"REWORK","_number":19,"created":"2023-01-05 13:30:35.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/27/866327/19","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/27/866327/19","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/27/866327/19 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/27/866327/19 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/27/866327/19 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/27/866327/19"}}},"commit":{"parents":[{"commit":"ac3618b140b1b8ba02f2263abf910c17b3f08aff","subject":"Merge \"Fix typo in docs section title\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/ac3618b140b1b8ba02f2263abf910c17b3f08aff"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-12-20 16:08:29.000000000","tz":60},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-01-05 13:11:52.000000000","tz":-300},"subject":"WIP remove python-neutronclient","message":"WIP remove python-neutronclient\n\nStory: 2010509\nTask: 47104\nChange-Id: I686cfdef78de927fa4bc1921c15e8d5853fd2ef9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b00ca7239d4610a8749e69d9049a58b38e850864"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b00ca7239d4610a8749e69d9049a58b38e850864"}]},"branch":"refs/heads/master"},"8a69ae472d61dabb4f081276fc10533e6ab10cbc":{"kind":"REWORK","_number":20,"created":"2023-01-05 14:25:08.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/27/866327/20","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/27/866327/20","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/27/866327/20 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/27/866327/20 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/27/866327/20 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/27/866327/20"}}},"commit":{"parents":[{"commit":"ac3618b140b1b8ba02f2263abf910c17b3f08aff","subject":"Merge \"Fix typo in docs section title\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/ac3618b140b1b8ba02f2263abf910c17b3f08aff"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-12-20 16:08:29.000000000","tz":60},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-01-05 14:24:57.000000000","tz":60},"subject":"WIP remove python-neutronclient","message":"WIP remove python-neutronclient\n\nStory: 2010509\nTask: 47104\nChange-Id: I686cfdef78de927fa4bc1921c15e8d5853fd2ef9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/8a69ae472d61dabb4f081276fc10533e6ab10cbc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/8a69ae472d61dabb4f081276fc10533e6ab10cbc"}]},"branch":"refs/heads/master"},"aec1e098751494eea0b17ee969a4ae02ae1740aa":{"kind":"REWORK","_number":21,"created":"2023-01-11 01:15:55.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/27/866327/21","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/27/866327/21","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/27/866327/21 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/27/866327/21 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/27/866327/21 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/27/866327/21"}}},"commit":{"parents":[{"commit":"ac3618b140b1b8ba02f2263abf910c17b3f08aff","subject":"Merge \"Fix typo in docs section title\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/ac3618b140b1b8ba02f2263abf910c17b3f08aff"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-12-20 16:08:29.000000000","tz":60},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2023-01-11 01:15:38.000000000","tz":60},"subject":"Remove python-neutronclient","message":"Remove python-neutronclient\n\nThis patch replaces the deprecated library with openstacksdk. It also\ndeprecates the use of the [service_auth] section of the configuration\nfor authentication with Neutron. In a future release these settings\nneed to be part of the [neutron] configuration section.\n\nStory: 2010509\nTask: 47104\nChange-Id: I686cfdef78de927fa4bc1921c15e8d5853fd2ef9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/aec1e098751494eea0b17ee969a4ae02ae1740aa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/aec1e098751494eea0b17ee969a4ae02ae1740aa"}]},"branch":"refs/heads/master"},"c9b5d1b33f82f685e3f8a0dbd243dd7234340a1a":{"kind":"REWORK","_number":22,"created":"2023-01-11 10:16:59.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/27/866327/22","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/27/866327/22","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/27/866327/22 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/27/866327/22 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/27/866327/22 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/27/866327/22"}}},"commit":{"parents":[{"commit":"ac3618b140b1b8ba02f2263abf910c17b3f08aff","subject":"Merge \"Fix typo in docs section title\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/ac3618b140b1b8ba02f2263abf910c17b3f08aff"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-12-20 16:08:29.000000000","tz":60},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2023-01-11 10:16:46.000000000","tz":60},"subject":"Remove python-neutronclient","message":"Remove python-neutronclient\n\nThis patch replaces the deprecated library with openstacksdk. It also\ndeprecates the use of the [service_auth] section of the configuration\nfor authentication with Neutron. In a future release these settings\nneed to be part of the [neutron] configuration section.\n\nStory: 2010509\nTask: 47104\nChange-Id: I686cfdef78de927fa4bc1921c15e8d5853fd2ef9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c9b5d1b33f82f685e3f8a0dbd243dd7234340a1a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c9b5d1b33f82f685e3f8a0dbd243dd7234340a1a"}]},"branch":"refs/heads/master"},"78c2378169742dcc343c0ffc22b72afeaff218ac":{"kind":"REWORK","_number":23,"created":"2023-01-19 15:29:42.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/27/866327/23","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/27/866327/23","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/27/866327/23 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/27/866327/23 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/27/866327/23 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/27/866327/23"}}},"commit":{"parents":[{"commit":"ac3618b140b1b8ba02f2263abf910c17b3f08aff","subject":"Merge \"Fix typo in docs section title\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/ac3618b140b1b8ba02f2263abf910c17b3f08aff"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-12-20 16:08:29.000000000","tz":60},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2023-01-19 15:29:32.000000000","tz":60},"subject":"Remove python-neutronclient","message":"Remove python-neutronclient\n\nThis patch replaces the deprecated library with openstacksdk. It also\ndeprecates the use of the [service_auth] section of the configuration\nfor authentication with Neutron. In a future release these settings\nneed to be part of the [neutron] configuration section.\n\nStory: 2010509\nTask: 47104\nChange-Id: I686cfdef78de927fa4bc1921c15e8d5853fd2ef9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/78c2378169742dcc343c0ffc22b72afeaff218ac"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/78c2378169742dcc343c0ffc22b72afeaff218ac"}]},"branch":"refs/heads/master"},"0aa70b6199c5c343217418441203f56875f12215":{"kind":"REWORK","_number":24,"created":"2023-01-19 16:08:10.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/27/866327/24","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/27/866327/24","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/27/866327/24 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/27/866327/24 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/27/866327/24 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/27/866327/24"}}},"commit":{"parents":[{"commit":"ac3618b140b1b8ba02f2263abf910c17b3f08aff","subject":"Merge \"Fix typo in docs section title\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/ac3618b140b1b8ba02f2263abf910c17b3f08aff"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-12-20 16:08:29.000000000","tz":60},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2023-01-19 16:07:59.000000000","tz":60},"subject":"Remove python-neutronclient","message":"Remove python-neutronclient\n\nThis patch replaces the deprecated library with openstacksdk. It also\ndeprecates the use of the [service_auth] section of the configuration\nfor authentication with Neutron. In a future release these settings\nneed to be part of the [neutron] configuration section.\n\nStory: 2010509\nTask: 47104\nChange-Id: I686cfdef78de927fa4bc1921c15e8d5853fd2ef9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/0aa70b6199c5c343217418441203f56875f12215"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/0aa70b6199c5c343217418441203f56875f12215"}]},"branch":"refs/heads/master"},"a1b161a4f088902d0ef094dae737e6aa13334d3d":{"kind":"REWORK","_number":25,"created":"2023-01-23 08:49:09.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/27/866327/25","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/27/866327/25","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/27/866327/25 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/27/866327/25 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/27/866327/25 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/27/866327/25"}}},"commit":{"parents":[{"commit":"ac3618b140b1b8ba02f2263abf910c17b3f08aff","subject":"Merge \"Fix typo in docs section title\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/ac3618b140b1b8ba02f2263abf910c17b3f08aff"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-12-20 16:08:29.000000000","tz":60},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2023-01-23 08:48:57.000000000","tz":60},"subject":"Remove python-neutronclient","message":"Remove python-neutronclient\n\nThis patch replaces the deprecated library with openstacksdk. It also\ndeprecates the use of the [service_auth] section of the configuration\nfor authentication with Neutron. In a future release these settings\nneed to be part of the [neutron] configuration section.\n\nStory: 2010509\nTask: 47104\nChange-Id: I686cfdef78de927fa4bc1921c15e8d5853fd2ef9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a1b161a4f088902d0ef094dae737e6aa13334d3d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a1b161a4f088902d0ef094dae737e6aa13334d3d"}]},"branch":"refs/heads/master"},"d6f4447a5787fb3ee627ea3d86a91cf54ec2cecc":{"kind":"NO_CODE_CHANGE","_number":26,"created":"2023-01-23 12:27:28.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/27/866327/26","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/27/866327/26","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/27/866327/26 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/27/866327/26 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/27/866327/26 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/27/866327/26"}}},"commit":{"parents":[{"commit":"ac3618b140b1b8ba02f2263abf910c17b3f08aff","subject":"Merge \"Fix typo in docs section title\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/ac3618b140b1b8ba02f2263abf910c17b3f08aff"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-12-20 16:08:29.000000000","tz":60},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2023-01-23 12:26:39.000000000","tz":60},"subject":"Remove python-neutronclient","message":"Remove python-neutronclient\n\nThis patch replaces the deprecated library with openstacksdk. It also\ndeprecates the use of the [service_auth] section of the configuration\nfor authentication with Neutron. In a future release these settings\nneed to be part of the [neutron] configuration section.\n\nUpdate needed on OVN provider side:\nhttps://review.opendev.org/c/openstack/ovn-octavia-provider/+/870514\n\nStory: 2010509\nTask: 47104\nChange-Id: I686cfdef78de927fa4bc1921c15e8d5853fd2ef9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d6f4447a5787fb3ee627ea3d86a91cf54ec2cecc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d6f4447a5787fb3ee627ea3d86a91cf54ec2cecc"}]},"branch":"refs/heads/master"},"a78616aee9fe32bbdbfcc78b5e37cf0f8bcbc3cd":{"kind":"REWORK","_number":27,"created":"2023-01-24 08:17:30.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/27/866327/27","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/27/866327/27","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/27/866327/27 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/27/866327/27 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/27/866327/27 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/27/866327/27"}}},"commit":{"parents":[{"commit":"ac3618b140b1b8ba02f2263abf910c17b3f08aff","subject":"Merge \"Fix typo in docs section title\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/ac3618b140b1b8ba02f2263abf910c17b3f08aff"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-12-20 16:08:29.000000000","tz":60},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2023-01-24 08:17:19.000000000","tz":60},"subject":"Remove python-neutronclient","message":"Remove python-neutronclient\n\nThis patch replaces the deprecated library with openstacksdk. It also\ndeprecates the use of the [service_auth] section of the configuration\nfor authentication with Neutron. In a future release these settings\nneed to be part of the [neutron] configuration section.\n\nUpdate needed on OVN provider side:\nhttps://review.opendev.org/c/openstack/ovn-octavia-provider/+/870514\n\nStory: 2010509\nTask: 47104\nChange-Id: I686cfdef78de927fa4bc1921c15e8d5853fd2ef9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a78616aee9fe32bbdbfcc78b5e37cf0f8bcbc3cd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a78616aee9fe32bbdbfcc78b5e37cf0f8bcbc3cd"}]},"branch":"refs/heads/master"},"3413c84db7f3d567c8c35a77a54a676c3f8a28da":{"kind":"REWORK","_number":28,"created":"2023-01-24 12:29:10.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/27/866327/28","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/27/866327/28","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/27/866327/28 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/27/866327/28 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/27/866327/28 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/27/866327/28"}}},"commit":{"parents":[{"commit":"ac3618b140b1b8ba02f2263abf910c17b3f08aff","subject":"Merge \"Fix typo in docs section title\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/ac3618b140b1b8ba02f2263abf910c17b3f08aff"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-12-20 16:08:29.000000000","tz":60},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2023-01-24 12:28:47.000000000","tz":60},"subject":"Remove python-neutronclient","message":"Remove python-neutronclient\n\nThis patch replaces the deprecated library with openstacksdk. It also\ndeprecates the use of the [service_auth] section of the configuration\nfor authentication with Neutron. In a future release these settings\nneed to be part of the [neutron] configuration section.\n\nUpdate needed on OVN provider side:\nhttps://review.opendev.org/c/openstack/ovn-octavia-provider/+/870514\n\nStory: 2010509\nTask: 47104\nChange-Id: I686cfdef78de927fa4bc1921c15e8d5853fd2ef9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/3413c84db7f3d567c8c35a77a54a676c3f8a28da"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/3413c84db7f3d567c8c35a77a54a676c3f8a28da"}]},"branch":"refs/heads/master"},"259d5c264336abc1670491c183cc6223f3510483":{"kind":"REWORK","_number":29,"created":"2023-01-24 12:37:03.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/27/866327/29","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/27/866327/29","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/27/866327/29 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/27/866327/29 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/27/866327/29 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/27/866327/29"}}},"commit":{"parents":[{"commit":"ac3618b140b1b8ba02f2263abf910c17b3f08aff","subject":"Merge \"Fix typo in docs section title\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/ac3618b140b1b8ba02f2263abf910c17b3f08aff"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-12-20 16:08:29.000000000","tz":60},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2023-01-24 12:36:54.000000000","tz":60},"subject":"Remove python-neutronclient","message":"Remove python-neutronclient\n\nThis patch replaces the deprecated library with openstacksdk. It also\ndeprecates the use of the [service_auth] section of the configuration\nfor authentication with Neutron. In a future release these settings\nneed to be part of the [neutron] configuration section.\n\nUpdate needed on OVN provider side:\nhttps://review.opendev.org/c/openstack/ovn-octavia-provider/+/870514\n\nStory: 2010509\nTask: 47104\nChange-Id: I686cfdef78de927fa4bc1921c15e8d5853fd2ef9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/259d5c264336abc1670491c183cc6223f3510483"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/259d5c264336abc1670491c183cc6223f3510483"}]},"branch":"refs/heads/master"},"87454bc95e403af8c74bb7920e30664f37a007cb":{"kind":"REWORK","_number":30,"created":"2023-01-24 16:43:46.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/27/866327/30","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/27/866327/30","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/27/866327/30 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/27/866327/30 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/27/866327/30 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/27/866327/30"}}},"commit":{"parents":[{"commit":"ac3618b140b1b8ba02f2263abf910c17b3f08aff","subject":"Merge \"Fix typo in docs section title\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/ac3618b140b1b8ba02f2263abf910c17b3f08aff"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-12-20 16:08:29.000000000","tz":60},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2023-01-24 16:43:35.000000000","tz":60},"subject":"Remove python-neutronclient","message":"Remove python-neutronclient\n\nThis patch replaces the deprecated library with openstacksdk. It also\ndeprecates the use of the [service_auth] section of the configuration\nfor authentication with Neutron. In a future release these settings\nneed to be part of the [neutron] configuration section.\n\nUpdate needed on OVN provider side:\nhttps://review.opendev.org/c/openstack/ovn-octavia-provider/+/870514\n\nStory: 2010509\nTask: 47104\nChange-Id: I686cfdef78de927fa4bc1921c15e8d5853fd2ef9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/87454bc95e403af8c74bb7920e30664f37a007cb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/87454bc95e403af8c74bb7920e30664f37a007cb"}]},"branch":"refs/heads/master"},"5654d3d76b7f7852ff6abb90dcbf57957b88fdcc":{"kind":"REWORK","_number":31,"created":"2023-02-06 11:32:04.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/27/866327/31","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/27/866327/31","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/27/866327/31 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/27/866327/31 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/27/866327/31 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/27/866327/31"}}},"commit":{"parents":[{"commit":"c55839caf2a5d17c2da23f6bbdccf08624eac810","subject":"Fix wrong assertion methods","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c55839caf2a5d17c2da23f6bbdccf08624eac810"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-12-20 16:08:29.000000000","tz":60},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2023-02-06 11:31:53.000000000","tz":60},"subject":"Remove python-neutronclient","message":"Remove python-neutronclient\n\nThis patch replaces the deprecated library with openstacksdk. It also\ndeprecates the use of the [service_auth] section of the configuration\nfor authentication with Neutron. In a future release these settings\nneed to be part of the [neutron] configuration section.\n\nUpdate needed on OVN provider side:\nhttps://review.opendev.org/c/openstack/ovn-octavia-provider/+/870514\n\nStory: 2010509\nTask: 47104\nChange-Id: I686cfdef78de927fa4bc1921c15e8d5853fd2ef9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/5654d3d76b7f7852ff6abb90dcbf57957b88fdcc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/5654d3d76b7f7852ff6abb90dcbf57957b88fdcc"}]},"branch":"refs/heads/master"},"872346e33f5f731b8700690d40da31991ced36d1":{"kind":"REWORK","_number":32,"created":"2023-04-03 09:59:45.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/27/866327/32","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/27/866327/32","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/27/866327/32 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/27/866327/32 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/27/866327/32 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/27/866327/32"}}},"commit":{"parents":[{"commit":"c55839caf2a5d17c2da23f6bbdccf08624eac810","subject":"Fix wrong assertion methods","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c55839caf2a5d17c2da23f6bbdccf08624eac810"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-12-20 16:08:29.000000000","tz":60},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2023-04-03 09:59:31.000000000","tz":120},"subject":"Remove python-neutronclient","message":"Remove python-neutronclient\n\nThis patch replaces the deprecated library with openstacksdk. It also\ndeprecates the use of the [service_auth] section of the configuration\nfor authentication with Neutron. In a future release these settings\nneed to be part of the [neutron] configuration section.\n\nUpdate needed on OVN provider side:\nhttps://review.opendev.org/c/openstack/ovn-octavia-provider/+/870514\n\nStory: 2010509\nTask: 47104\nChange-Id: I686cfdef78de927fa4bc1921c15e8d5853fd2ef9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/872346e33f5f731b8700690d40da31991ced36d1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/872346e33f5f731b8700690d40da31991ced36d1"}]},"branch":"refs/heads/master"},"b5451b7a585886c951d1031722c5bbc03015083d":{"kind":"REWORK","_number":33,"created":"2023-04-03 12:34:12.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/27/866327/33","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/27/866327/33","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/27/866327/33 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/27/866327/33 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/27/866327/33 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/27/866327/33"}}},"commit":{"parents":[{"commit":"c55839caf2a5d17c2da23f6bbdccf08624eac810","subject":"Fix wrong assertion methods","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c55839caf2a5d17c2da23f6bbdccf08624eac810"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-12-20 16:08:29.000000000","tz":60},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2023-04-03 12:34:03.000000000","tz":120},"subject":"Remove python-neutronclient","message":"Remove python-neutronclient\n\nThis patch replaces the deprecated library with openstacksdk. It also\ndeprecates the use of the [service_auth] section of the configuration\nfor authentication with Neutron. In a future release these settings\nneed to be part of the [neutron] configuration section.\n\nUpdate needed on OVN provider side:\nhttps://review.opendev.org/c/openstack/ovn-octavia-provider/+/870514\n\nStory: 2010509\nTask: 47104\nChange-Id: I686cfdef78de927fa4bc1921c15e8d5853fd2ef9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b5451b7a585886c951d1031722c5bbc03015083d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b5451b7a585886c951d1031722c5bbc03015083d"}]},"branch":"refs/heads/master"},"5c1eef0c842c785bcfd55c462251140c64cf47a4":{"kind":"REWORK","_number":34,"created":"2023-04-14 11:06:32.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/27/866327/34","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/27/866327/34","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/27/866327/34 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/27/866327/34 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/27/866327/34 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/27/866327/34"}}},"commit":{"parents":[{"commit":"8d70a8050510a0d7fd8ebf78dc79e6986b83189c","subject":"Merge \"allowed_cidr validation for additional_vips\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/8d70a8050510a0d7fd8ebf78dc79e6986b83189c"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-12-20 16:08:29.000000000","tz":60},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2023-04-14 11:05:47.000000000","tz":120},"subject":"Remove python-neutronclient","message":"Remove python-neutronclient\n\nThis patch replaces the deprecated library with openstacksdk. It also\ndeprecates the use of the [service_auth] section of the configuration\nfor authentication with Neutron. In a future release these settings\nneed to be part of the [neutron] configuration section.\n\nUpdate needed on OVN provider side:\nhttps://review.opendev.org/c/openstack/ovn-octavia-provider/+/870514\n\nStory: 2010509\nTask: 47104\nChange-Id: I686cfdef78de927fa4bc1921c15e8d5853fd2ef9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/5c1eef0c842c785bcfd55c462251140c64cf47a4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/5c1eef0c842c785bcfd55c462251140c64cf47a4"}]},"branch":"refs/heads/master"},"3e6fd13bfb000785d5eb9efe2139ee739e074460":{"kind":"REWORK","_number":35,"created":"2023-04-14 11:11:49.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/27/866327/35","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/27/866327/35","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/27/866327/35 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/27/866327/35 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/27/866327/35 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/27/866327/35"}}},"commit":{"parents":[{"commit":"8d70a8050510a0d7fd8ebf78dc79e6986b83189c","subject":"Merge \"allowed_cidr validation for additional_vips\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/8d70a8050510a0d7fd8ebf78dc79e6986b83189c"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-12-20 16:08:29.000000000","tz":60},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2023-04-14 11:11:38.000000000","tz":120},"subject":"Remove python-neutronclient","message":"Remove python-neutronclient\n\nThis patch replaces the deprecated library with openstacksdk. It also\ndeprecates the use of the [service_auth] section of the configuration\nfor authentication with Neutron. In a future release these settings\nneed to be part of the [neutron] configuration section.\n\nUpdate needed on OVN provider side:\nhttps://review.opendev.org/c/openstack/ovn-octavia-provider/+/870514\n\nStory: 2010509\nTask: 47104\nChange-Id: I686cfdef78de927fa4bc1921c15e8d5853fd2ef9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/3e6fd13bfb000785d5eb9efe2139ee739e074460"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/3e6fd13bfb000785d5eb9efe2139ee739e074460"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
