)]}'
{"id":"openstack%2Foctavia~866711","triplet_id":"openstack%2Foctavia~stable%2Fussuri~If7dc46ee18a46e3bd15756839f60ea5b51c9288b","project":"openstack/octavia","branch":"stable/ussuri","attention_set":{},"removed_from_attention_set":{"29244":{"account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"last_update":"2022-12-07 13:58:18.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"If7dc46ee18a46e3bd15756839f60ea5b51c9288b","subject":"Increase TCP buffer maximum and MTU discovery","status":"MERGED","created":"2022-12-06 13:13:26.000000000","updated":"2022-12-07 13:59:38.000000000","submitted":"2022-12-07 13:58:18.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"866711","meta_rev_id":"d46a0b73482df122e0556fcfba9bc8c71b7f5af0","_number":866711,"virtual_id_number":866711,"owner":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-12-07 13:58:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":2,"date":"2022-12-06 15:51:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-12-07 11:03:57.000000000","_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-12-07 11:03:57.000000000","_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-12-06 13:13:26.000000000","updated_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"CC"},{"updated":"2022-12-06 15:27:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-12-06 15:51:30.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2022-12-07 11:03:57.000000000","updated_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"reviewer":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"state":"REVIEWER"}],"messages":[{"id":"2dc60d2641f0af28da8fa10628fbe78ae329b00f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2022-12-06 13:13:26.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0ce8e9b52583b7a2eeb6aa1bd6dc953c470b2792","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-06 15:27:48.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/29b61232dde74ea39fc51422e6fd11f6\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b95d2b2cad2242b5ab296c696a144d1e : SUCCESS in 11m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bcc052c6c598440a897438d01a24131f : SUCCESS in 5m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fedf399733c042649fddc27e47a119d7 : SUCCESS in 5m 21s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/7d977669927b4c6282f000cc9f029e45 : SUCCESS in 4m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aca45e42de364e2894e2a1799063dce2 : SUCCESS in 11m 59s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4969e700f19246a7b427d00c043f7248 : SUCCESS in 4m 43s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/7ba49b0e29164aa0807270aed88e721c : SUCCESS in 4m 57s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/9cf673fee9764c709b1688e05342700c : SUCCESS in 7m 48s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/ce3ede5be45e43958cb8d25979267b4a : SUCCESS in 7m 32s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/e1b19310feca4d6c89323a076bd33cc5 : SUCCESS in 43m 06s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/9f057ba6e2b54fafb788a9efb2a21ec7 : SUCCESS in 1h 43m 20s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/7bdcae9b1ebd43bbbd280877ca8aef91 : SUCCESS in 37m 53s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/ddea72151251472e8584b52a27e2ac10 : SUCCESS in 35m 39s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/ffbc9a89a7c74c80b3fe8ea8f4ebb094 : SUCCESS in 36m 03s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/e79730c7345945729eb895556bb1bb80 : SUCCESS in 2h 02m 34s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/883c74048abf445fbc5e0a9c7b50dd95 : RETRY_LIMIT in 7m 34s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/99f463bb54b7445e83677b2ea20d358e : SUCCESS in 1h 38m 32s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"01ed7efb675c0c37f5ab5d0c640bf77226f2fe71","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2022-12-06 15:51:30.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"5c18f1016f4c3a014b9b617214da025bc923de1f","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-12-07 11:03:57.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"9afa93435916176228ef903c35c369b32a630fcc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-07 11:04:39.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"200d0b90e77fafa95a977e62af32abfa765df371","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-07 13:58:18.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/aca90864d8ed4bfc9c33517c1e1cf0ea\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/22316d9f91e64fed85976a7518b210f2 : SUCCESS in 7m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b6276738a55a4e8f9825bd60a42ec077 : SUCCESS in 4m 31s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/e06e8e0471e64b9faafb887bfec013f3 : SUCCESS in 5m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cb13e85637804da48e81beedf0fbb1f5 : SUCCESS in 12m 36s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/555370196f3841a6a525532c7cd1bfee : SUCCESS in 5m 52s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/a3eb27f210654d668aac3b96601cbf3c : SUCCESS in 6m 47s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/f0e4a08099424de88f7b445919f1e146 : SUCCESS in 58m 22s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/255ad0f6d8ae4d4eb19d8f282de7045c : SUCCESS in 2h 03m 53s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/b2bf46e2ceaa4549ba37d525420d36b1 : SUCCESS in 36m 05s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/d4a8789cb7e64a08a3920ae2659f9bbe : SUCCESS in 33m 14s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/3c2dbac1ad8642668b0bddfb5752e046 : SUCCESS in 34m 16s","accounts_in_message":[],"_revision_number":1},{"id":"8eba4949628ccc48e14a3600ae7a81b93ff31115","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-07 13:58:18.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"d46a0b73482df122e0556fcfba9bc8c71b7f5af0","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-07 13:59:38.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/92d73ff355394a1b92f96274447d9a22\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/838e84c38cb94074b1c068d8b90e20b7 : SUCCESS in 49s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c29ac13ee5734d09b94c927d5501df11 : SUCCESS in 41s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"569cfab950bc3b4e20cc14cc56849420fe1c9151","revisions":{"569cfab950bc3b4e20cc14cc56849420fe1c9151":{"kind":"REWORK","_number":1,"created":"2022-12-06 13:13:26.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/11/866711/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/11/866711/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/11/866711/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/11/866711/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/11/866711/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/11/866711/1"}}},"commit":{"parents":[{"commit":"f1c1869a7adb288ba758b40ac79d6476d1e317e6","subject":"Merge \"Cache subnets validation for batch member update\" into stable/ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/f1c1869a7adb288ba758b40ac79d6476d1e317e6"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2021-09-15 23:48:44.000000000","tz":0},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2022-12-06 13:13:21.000000000","tz":60},"subject":"Increase TCP buffer maximum and MTU discovery","message":"Increase TCP buffer maximum and MTU discovery\n\nThis patch increases the maximum allowed TCP buffer memory to 32MB\nand enabled MTU probe discovery when an ICMP black hole is detected.\n\nThis should improve reliability in some situations when the amphora are\nprovisioned on hosts with 10gbps or faster network adapters.\n\nChange-Id: If7dc46ee18a46e3bd15756839f60ea5b51c9288b\n(cherry picked from commit 421eecd1e5e6d86327278dd8620069297e2d9843)\n(cherry picked from commit fc55d6cfc86df260a97eff77ed39ff123ad060a1)\n(cherry picked from commit 31291c488f0392e11d82f8889bf66e8c07972727)\n(cherry picked from commit 0535933986fac1dcd9865477c21e90acf6885333)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/569cfab950bc3b4e20cc14cc56849420fe1c9151"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/569cfab950bc3b4e20cc14cc56849420fe1c9151"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[]}
