)]}'
{"id":"openstack%2Foctavia~866712","triplet_id":"openstack%2Foctavia~stable%2Ftrain~If7dc46ee18a46e3bd15756839f60ea5b51c9288b","project":"openstack/octavia","branch":"stable/train","attention_set":{},"removed_from_attention_set":{"29244":{"account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"last_update":"2022-12-07 13:58:20.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"If7dc46ee18a46e3bd15756839f60ea5b51c9288b","subject":"Increase TCP buffer maximum and MTU discovery","status":"MERGED","created":"2022-12-06 13:13:36.000000000","updated":"2022-12-07 13:59:39.000000000","submitted":"2022-12-07 13:58:20.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"866712","meta_rev_id":"6674118dc735aae067d24c86fa23d2cd54f94adc","_number":866712,"virtual_id_number":866712,"owner":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-12-07 13:58:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":2,"date":"2022-12-06 15:51:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-12-07 11:01:27.000000000","_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-12-07 11:04:18.000000000","_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-12-06 13:13:36.000000000","updated_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"CC"},{"updated":"2022-12-06 15:40:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-12-06 15:51:39.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2022-12-07 11:01:27.000000000","updated_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"reviewer":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"state":"REVIEWER"}],"messages":[{"id":"f0364ab31d0e81ed0810052ae850dea10e684e53","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2022-12-06 13:13:36.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a5d051fd5aac9534c9c2d12672547f94f8f91d37","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-06 15:40:06.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5d50298449914d7589d22ff9ae947773\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/530112909760468e8d137d1f14071908 : SUCCESS in 7m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/16d78aec82b4473c95f8a1073dc87a81 : SUCCESS in 6m 23s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/66015afae5a74c6f8e7d5e11c06b47bb : SUCCESS in 4m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cbddcb2e10654b37922904aadeb0a192 : SUCCESS in 4m 44s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/569b955d3a0941a1a0d099b43a02609c : SUCCESS in 4m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0e210ca9f8ce46c99f2e4721e2b8aa53 : SUCCESS in 10m 29s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0b3ffdafaec34ae5a25848ec783bdf9e : SUCCESS in 3m 56s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/747e192ea950440caac20cd7c0d8de95 : SUCCESS in 8m 21s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/a360074e49514430abb3d190cae625c9 : SUCCESS in 6m 42s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/9930c05d50484374a19341236a7f1856 : SUCCESS in 6m 54s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/7308903607f94518839954213df91243 : SUCCESS in 5m 52s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/e5fa513828a24287ae035706dbcbbde4 : SUCCESS in 46m 18s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/f7b6663930ad43e79ebc8190ff8172dc : SUCCESS in 51m 32s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/7ee8d59d932b4a37a41b6e2bce66b4d6 : SUCCESS in 1h 27m 06s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/40d4da6cec6a4e7a9cc95d4cf68adecf : SUCCESS in 2h 01m 43s (non-voting)\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/5689a51e206f4694beb21afe8517af99 : FAILURE in 15m 36s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/94ec191c55a54d4d95dd04a29de6b00c : RETRY_LIMIT in 5m 20s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/41392d4978214fa7bf0a4c85dfdff418 : RETRY_LIMIT in 5m 12s (non-voting)\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/f6d22a2bc0b847db8d19790dbbc8391a : SUCCESS in 35m 00s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/35349706c5774854b7fa3d2dab54b083 : SUCCESS in 1h 30m 45s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/151f7173659948c6afabe2f3ae509382 : FAILURE in 32m 12s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/047e5c90d18a4933a6f931522f2bf4a3 : FAILURE in 30m 08s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/8169dc935971466c81817242c935ff3e : FAILURE in 14m 07s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f0f23a8107ea4ead8ed30f4b18fdd1f8dbfc316f","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2022-12-06 15:51:39.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"3f2c4d08c7327e8f2e09f944bd07e8209059ed0d","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-12-07 11:01:27.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"d26c12845de8fc9ec292fef118aa3c6dc395d5c8","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-12-07 11:04:18.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"0bf5ec0d3eee7c40c42b9590c4b01c7f8d5a0c65","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-07 11:04:58.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b76ceea46e71a1c80218320d507b08e434fcf272","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-07 13:58:20.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/de5859465d3d4a0ba5eced449f5dbe10\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3c309f771e304b5f8e13b07f656b2f42 : SUCCESS in 4m 27s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/9427aa8e48c644928049f165c566d8fc : SUCCESS in 4m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4238a1833bb848379a44db8e870b98c9 : SUCCESS in 5m 59s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/8fa217f5b3e44eecb0b6b513b06e71a5 : SUCCESS in 6m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/69e9398da8d440a4bfd9c94564a151ac : SUCCESS in 9m 44s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/033e21baaeb146dfa255beb5b34eee33 : SUCCESS in 4m 48s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/7570291b1e764e7881cfa048dc1a694a : SUCCESS in 5m 55s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/f6aa77ddc283482581a2bb2b905b8017 : SUCCESS in 7m 37s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/8091548139e24e5c939746333f5957d6 : SUCCESS in 46m 28s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/296d3d8e56f24397a4ca82db11becfbd : SUCCESS in 1h 00m 20s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/e12ed147ef04497d9de356c22af7cd68 : SUCCESS in 2h 03m 56s","accounts_in_message":[],"_revision_number":1},{"id":"003f098e4e04f36baf06231f3e71ab23323ed706","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-07 13:58:20.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"6674118dc735aae067d24c86fa23d2cd54f94adc","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-07 13:59:39.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/345ea5f253554f8e8f4d794df7d5d1e1\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c176cc5ccc1c4f9291ae5cf1d64c33a9 : SUCCESS in 53s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/37f99b8cce394d62852a5ec930f6bbb0 : SUCCESS in 46s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"cfc116409286d6d22a38a05468ea80a8ef81308d","revisions":{"cfc116409286d6d22a38a05468ea80a8ef81308d":{"kind":"REWORK","_number":1,"created":"2022-12-06 13:13:36.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/12/866712/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/12/866712/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/866712/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/866712/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/866712/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/12/866712/1"}}},"commit":{"parents":[{"commit":"e52a2ebabd921e0b0ea5b3fef03a8e92cab854a3","subject":"Merge \"Handle feature compatibility of HAProxy server-state-file option\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e52a2ebabd921e0b0ea5b3fef03a8e92cab854a3"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2021-09-15 23:48:44.000000000","tz":0},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2022-12-06 13:13:32.000000000","tz":60},"subject":"Increase TCP buffer maximum and MTU discovery","message":"Increase TCP buffer maximum and MTU discovery\n\nThis patch increases the maximum allowed TCP buffer memory to 32MB\nand enabled MTU probe discovery when an ICMP black hole is detected.\n\nThis should improve reliability in some situations when the amphora are\nprovisioned on hosts with 10gbps or faster network adapters.\n\nChange-Id: If7dc46ee18a46e3bd15756839f60ea5b51c9288b\n(cherry picked from commit 421eecd1e5e6d86327278dd8620069297e2d9843)\n(cherry picked from commit fc55d6cfc86df260a97eff77ed39ff123ad060a1)\n(cherry picked from commit 31291c488f0392e11d82f8889bf66e8c07972727)\n(cherry picked from commit 0535933986fac1dcd9865477c21e90acf6885333)\n(cherry picked from commit 569cfab950bc3b4e20cc14cc56849420fe1c9151)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/cfc116409286d6d22a38a05468ea80a8ef81308d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/cfc116409286d6d22a38a05468ea80a8ef81308d"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[]}
