)]}'
{"id":"openstack%2Foctavia~868019","triplet_id":"openstack%2Foctavia~stable%2Fyoga~Id328b27db1ec9c8e17bf18120259e41e75dab3b9","project":"openstack/octavia","branch":"stable/yoga","topic":"story/2010335-stable/yoga","attention_set":{},"removed_from_attention_set":{"31664":{"account":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"last_update":"2022-12-20 09:58:37.000000000","reason":"\u003cGERRIT_ACCOUNT_31664\u003e replied on the change","reason_account":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"}},"6577":{"account":{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},"last_update":"2023-01-16 11:23:04.000000000","reason":"Change was submitted"},"29244":{"account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"last_update":"2023-01-16 09:24:42.000000000","reason":"\u003cGERRIT_ACCOUNT_29244\u003e replied on the change","reason_account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}},"34429":{"account":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"last_update":"2022-12-27 10:35:24.000000000","reason":"\u003cGERRIT_ACCOUNT_34429\u003e replied on the change","reason_account":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}}},"hashtags":[],"change_id":"Id328b27db1ec9c8e17bf18120259e41e75dab3b9","subject":"Fix full graph loadbalancer creation if jobboard is disabled","status":"MERGED","created":"2022-12-19 12:12:29.000000000","updated":"2023-01-16 11:24:20.000000000","submitted":"2023-01-16 11:23:04.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"868019-story/2010335-stable/yoga","cherry_pick_of_change":859710,"cherry_pick_of_patch_set":2,"meta_rev_id":"cd6c59f73bbb9d21d4b034b8851a9eff9d88f637","_number":868019,"virtual_id_number":868019,"owner":{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-01-16 11:23:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"recommended":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"all":[{"value":1,"date":"2022-12-20 09:58:37.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":2,"date":"2023-01-16 09:24:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-12-27 10:35:24.000000000","_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"all":[{"value":0,"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":1,"date":"2023-01-16 09:24:42.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-12-19 15:02:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-12-20 07:06:28.000000000","updated_by":{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},"reviewer":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"state":"REVIEWER"},{"updated":"2022-12-20 07:06:28.000000000","updated_by":{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},"reviewer":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"state":"REVIEWER"},{"updated":"2022-12-20 07:06:28.000000000","updated_by":{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"REVIEWER"}],"messages":[{"id":"52e57a5bf2ba552438aebad8277f013f2a1fd5bd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},"date":"2022-12-19 12:12:29.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"0e7a64c0537e2327320a03137811e5808df4ddfa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-19 15:02:48.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/09bd2ae871684176b526aaaf42612852\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8475542e17ea480391709245ecdab933 : SUCCESS in 9m 03s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/964758bcbfc1429fbbec6dd78d131aa3 : SUCCESS in 8m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c7128d74631a41ce9d994b476b9cc01a : SUCCESS in 6m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/63087ccf66194cb1872a97784612e884 : SUCCESS in 5m 02s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e43f972e71f2449e93e7fe4a95944acf : SUCCESS in 4m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/06b7c3ad381140e9b7910ea812565719 : SUCCESS in 12m 17s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7416a165a76a4d11ac68e5574b4d6137 : SUCCESS in 4m 27s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/067a832b88014e43830cf9d8ccf4bd87 : SUCCESS in 5m 30s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/338946c893f549ca8c8faeef0d0b4d50 : SUCCESS in 4m 38s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/8379a631c511444dbe5852b4999fc14d : SUCCESS in 5m 15s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/0705a76c5d3848dc9e31fa29696433c7 : SUCCESS in 6m 24s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/fece2209908d49259b86ddfc102e9ec7 : SUCCESS in 1h 34m 43s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/fe6dd534c99e419f97302223dffaac8b : SUCCESS in 1h 48m 50s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/b63dd41751de4ffd8b6900ec05743b50 : SUCCESS in 41m 01s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/f437f2a035cb428e9a2eda452e3a9b93 : SUCCESS in 59m 37s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/8920875b8e484425bd833619566d44ae : FAILURE in 1h 19m 11s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/0a0bb30d628844a7aa5c3b527ec8805b : SUCCESS in 34m 06s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/adc7fe25af484f31a88afbb67b28ef54 : SUCCESS in 1h 51m 10s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/37833441125b4059ac09ba98a571df18 : SUCCESS in 2h 42m 28s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/68b43a56eac14acba44027a7c94069ed : SUCCESS in 2h 13m 33s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/c588229f651d450e9fbcfb22befae8d4 : SUCCESS in 2h 02m 26s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/4572c828133e456e911965bb7d7170bf : SUCCESS in 2h 28m 47s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8-stream https://zuul.opendev.org/t/openstack/build/dde8ea4e874d46fca48802a137e36eb1 : FAILURE in 35m 43s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d3305cf99b9132fde5aa9c5f579d717db9f8db0e","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2022-12-20 09:58:37.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"e21c4da3e153e374ceba180db400bd735461817d","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-12-27 10:35:24.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"8a671db0814dba08d88bf17e630a17f93d7906fb","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-01-16 09:24:42.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"b8ff19d723b57c01357bfa50083dba90b90ead5f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-16 09:25:27.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"8fba5664d915cb881f2d5be28e970d2bee5df450","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-16 11:23:04.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/920e72bf84954d71a3947d9be3e7d2d9\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1bfc0dd6b6094baa877107c33efaa3f5 : SUCCESS in 7m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c42cedf13c1844f5bc4eab9cbe1d3492 : SUCCESS in 7m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c1fd72bf1f2f4bf2be52196c69e0447e : SUCCESS in 4m 55s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b94966a59e864f729fad547d7f9a77de : SUCCESS in 5m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/45d84baf1d964d1bbaaa2b18824756f6 : SUCCESS in 11m 38s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d5180275b1964e1fb9299f56dcf63e21 : SUCCESS in 5m 18s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/5b583a736968496892e5a0579ab6fa4e : SUCCESS in 5m 46s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/0306db5300b849c994c956249aff6259 : SUCCESS in 9m 19s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/39773c8c22a943449c1b100f1f00b69e : SUCCESS in 1h 35m 21s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/c4153da1ce284944a254dafc8b2e9e62 : SUCCESS in 1h 56m 38s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/d12757ae9d1b4892882a20817483e178 : SUCCESS in 36m 46s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/99e59fd15fc64cdd8eaf7bfcbecc4c20 : SUCCESS in 43m 04s","accounts_in_message":[],"_revision_number":1},{"id":"7ea3295fef15efa268578d04a0ea540eb685dff5","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-16 11:23:04.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"cd6c59f73bbb9d21d4b034b8851a9eff9d88f637","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-16 11:24:20.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d0cb34984a3446938a68207cfe749f51\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dfdb4c33d6f445a59e5761b26b218afb : SUCCESS in 44s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/eb927bf5d6fd441c8539f46b3cde1741 : SUCCESS in 39s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"c662e42b849aa962a814fe756e109950e794277c","revisions":{"c662e42b849aa962a814fe756e109950e794277c":{"kind":"REWORK","_number":1,"created":"2022-12-19 12:12:29.000000000","uploader":{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},"ref":"refs/changes/19/868019/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/19/868019/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/19/868019/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/19/868019/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/19/868019/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/19/868019/1"}}},"commit":{"parents":[{"commit":"8fa8d8f071785d82fefe3ff88eea2935d4144e3f","subject":"Merge \"Fix plugging member subnets on existing networks\" into stable/yoga","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/8fa8d8f071785d82fefe3ff88eea2935d4144e3f"}]}],"author":{"name":"Anton Kurbatov","email":"Anton.Kurbatov@acronis.com","date":"2022-09-28 18:48:45.000000000","tz":0},"committer":{"name":"Sergey Kraynev","email":"sergejyit@gmail.com","date":"2022-12-19 12:12:29.000000000","tz":0},"subject":"Fix full graph loadbalancer creation if jobboard is disabled","message":"Fix full graph loadbalancer creation if jobboard is disabled\n\nThis patch fixes the creation of a full loadbalancer graph in case the\njobboard_enabled config value is False, when no \"listeners\" parameter was\npassed to the get_create_load_balancer_flow function call: use kwargs as the\nflow creation function parameters and handle \"store\" separately.\n\nStory 2010335\nTask 46462\n\nSigned-off-by: Anton Kurbatov \u003cAnton.Kurbatov@acronis.com\u003e\nChange-Id: Id328b27db1ec9c8e17bf18120259e41e75dab3b9\n(cherry picked from commit 216cce39c9c22831d725703f41c54b9305d65dc8)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c662e42b849aa962a814fe756e109950e794277c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c662e42b849aa962a814fe756e109950e794277c"}]},"branch":"refs/heads/stable/yoga"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[]}
