)]}'
{"id":"openstack%2Foctavia~868020","triplet_id":"openstack%2Foctavia~stable%2Fzed~Id328b27db1ec9c8e17bf18120259e41e75dab3b9","project":"openstack/octavia","branch":"stable/zed","topic":"story/2010335-stable/zed","attention_set":{},"removed_from_attention_set":{"31664":{"account":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"last_update":"2022-12-20 09:57:17.000000000","reason":"\u003cGERRIT_ACCOUNT_31664\u003e replied on the change","reason_account":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"}},"6577":{"account":{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},"last_update":"2023-01-16 10:55:50.000000000","reason":"Change was submitted"},"29244":{"account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"last_update":"2023-01-16 09:24:30.000000000","reason":"\u003cGERRIT_ACCOUNT_29244\u003e replied on the change","reason_account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}},"34429":{"account":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"last_update":"2022-12-27 10:34:36.000000000","reason":"\u003cGERRIT_ACCOUNT_34429\u003e replied on the change","reason_account":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}}},"hashtags":[],"change_id":"Id328b27db1ec9c8e17bf18120259e41e75dab3b9","subject":"Fix full graph loadbalancer creation if jobboard is disabled","status":"MERGED","created":"2022-12-19 12:13:13.000000000","updated":"2023-01-16 10:56:57.000000000","submitted":"2023-01-16 10:55:50.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"868020-story/2010335-stable/zed","cherry_pick_of_change":859710,"cherry_pick_of_patch_set":2,"meta_rev_id":"8b086dbf8d0c4db214e673fe4b1b385cef9d45bf","_number":868020,"virtual_id_number":868020,"owner":{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-01-16 10:55:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"recommended":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"all":[{"value":1,"date":"2022-12-20 09:57:17.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":2,"date":"2023-01-16 09:24:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-12-27 10:34:36.000000000","_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"all":[{"value":0,"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":1,"date":"2023-01-16 09:24:30.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-12-19 14:54:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-12-20 07:06:54.000000000","updated_by":{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},"reviewer":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"state":"REVIEWER"},{"updated":"2022-12-20 07:06:54.000000000","updated_by":{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},"reviewer":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"state":"REVIEWER"},{"updated":"2022-12-20 07:06:54.000000000","updated_by":{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"REVIEWER"}],"messages":[{"id":"369c09fac2d6b3adf191a4fdd6d180f5aead7d1b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},"date":"2022-12-19 12:13:13.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"76830106815e6c79b049232efe87e2cb25695a2f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-19 14:54:12.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8fbb88be5bf34254ae87c38b75f5bd12\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c3a54fd34af846a89b2d5b1fc4ad72e0 : SUCCESS in 11m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0d47c9ab2d4d47b19599155e9bf0e746 : SUCCESS in 7m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bd4cc2ec7b644c5c977c694ba5c286ca : SUCCESS in 4m 55s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4e0d0d7520c24e8c867f07f2d88c1380 : SUCCESS in 5m 51s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/88efeb1de53d423b8f8341dd3fe510eb : SUCCESS in 5m 13s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6085dda12e0e4ccc82c8176398ce6fb1 : SUCCESS in 12m 50s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/edb0c595621e48d9939911eecaf3f5b8 : SUCCESS in 5m 18s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/5f7cd673327c47278688a5bf9c2da467 : SUCCESS in 6m 40s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/9f7d8d47c0b142be89268387d0a7e4a3 : SUCCESS in 8m 47s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/11641910bfaf4ea2890b28b10438b006 : SUCCESS in 5m 06s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/8cd750da57744f9c809af4c5ae2ee612 : SUCCESS in 8m 09s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/e306a6b2b77a4a0f9b2e94e447bacc6d : SUCCESS in 7m 06s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/9aa58e2dcdfd45ef8728e3941f2a5cf8 : SUCCESS in 7m 35s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/7b1a6acc295c4b1f909632dfab0592eb : SUCCESS in 1h 32m 20s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/d9b3a6a9aa5f4d999a68bb2ea6f95727 : SUCCESS in 1h 57m 05s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/410b4bef308044f384bbc99b9e43ef1a : SUCCESS in 44m 03s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/03f26510fdc84f97bae4136d9df7627c : SUCCESS in 1h 11m 18s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/2016713dea67413c80fe5c15a25303ff : FAILURE in 1h 25m 19s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/6d27123ab7df432d9e2d0319f321dcd8 : SUCCESS in 31m 27s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/b85eb68485fc4939ba487986beedaed2 : SUCCESS in 1h 47m 59s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/f2bad4e119a04902b0037f7f73698dc1 : SUCCESS in 2h 15m 37s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/58641a348d094d33b8b22bf42f6908b3 : SUCCESS in 2h 19m 07s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/bc0e03589f3c459c9e7fd0449b12450d : SUCCESS in 1h 58m 30s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/cf224e4620614260b57c2b63a14be632 : SUCCESS in 1h 54m 25s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/f71d8c01b31544748c12515b470885eb : TIMED_OUT in 2h 34m 34s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"0944eddca3ba06874cc0365af6804644ecfab622","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2022-12-20 09:57:17.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"b1bb00646be4e0c5048a5103f932ebd0b581c38e","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-12-27 10:34:36.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"f40dae1e49307af73fd6a36b4caf6873466fcf59","author":{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},"date":"2022-12-28 13:37:19.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"debcbc9e1da790e99ad2832e6e84d4a6842d323e","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-12-28 14:55:18.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"6c12cdddf2677ab210694c86e8fbf839105791fd","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-01-16 09:24:30.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"c60a7f84263dcf0ecacbb8fe5c84419c10bef00f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-16 09:25:15.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"c8e7ab6f705c1aaa630e52f7bcb4be6bb291112b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-16 10:55:49.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/531b0473dfcd4e3094e10ea32e11e8ad\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0e234bbd9e2a4893b791c0a31d25cba0 : SUCCESS in 6m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4118d375acbd454a9a58255143294e60 : SUCCESS in 4m 47s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d045e903aa5d4925a5f9ac2785259b19 : SUCCESS in 4m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d1c2c6894e984b18a6e00df4f2a0ec2e : SUCCESS in 11m 50s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f34ba69dbe9643fcb0f5c4b30d5f45c2 : SUCCESS in 5m 28s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/2ad40d9261bd46089aea2b58039c198f : SUCCESS in 5m 27s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/1da1372431c044a68a3a44cec923d579 : SUCCESS in 7m 32s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/1f93e8010f244110bb21f05b4b4ee88e : SUCCESS in 1h 28m 14s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/ca85f6baaf394deebe0ecedfe9276221 : SUCCESS in 1h 29m 56s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/6c33a9494b85403d9c4a8bb2efb86ed9 : SUCCESS in 35m 36s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/e360f289d07040ebb90ad286a5ac0859 : SUCCESS in 50m 49s","accounts_in_message":[],"_revision_number":1},{"id":"1dc02a86a99c08fd806ff8b5573969daced51ab4","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-16 10:55:50.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"8b086dbf8d0c4db214e673fe4b1b385cef9d45bf","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-16 10:56:57.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c94233abbcd94f9ba99c8ea3f1751a9c\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/152f321171544dafb1cea20c0462da0d : SUCCESS in 47s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5681a4e0c34c4a059c25f75b6b8da7ad : SUCCESS in 44s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"cfd87ccba1a5a667ce6a917c6bb741a53c8a8ff6","revisions":{"cfd87ccba1a5a667ce6a917c6bb741a53c8a8ff6":{"kind":"REWORK","_number":1,"created":"2022-12-19 12:13:13.000000000","uploader":{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},"ref":"refs/changes/20/868020/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/20/868020/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/20/868020/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/20/868020/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/20/868020/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/20/868020/1"}}},"commit":{"parents":[{"commit":"572c0dace98320a2ff925bf91e5ff07a4cfd9f4f","subject":"Add a newline when writing the server state file","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/572c0dace98320a2ff925bf91e5ff07a4cfd9f4f"}]}],"author":{"name":"Anton Kurbatov","email":"Anton.Kurbatov@acronis.com","date":"2022-09-28 18:48:45.000000000","tz":0},"committer":{"name":"Sergey Kraynev","email":"sergejyit@gmail.com","date":"2022-12-19 12:13:13.000000000","tz":0},"subject":"Fix full graph loadbalancer creation if jobboard is disabled","message":"Fix full graph loadbalancer creation if jobboard is disabled\n\nThis patch fixes the creation of a full loadbalancer graph in case the\njobboard_enabled config value is False, when no \"listeners\" parameter was\npassed to the get_create_load_balancer_flow function call: use kwargs as the\nflow creation function parameters and handle \"store\" separately.\n\nStory 2010335\nTask 46462\n\nSigned-off-by: Anton Kurbatov \u003cAnton.Kurbatov@acronis.com\u003e\nChange-Id: Id328b27db1ec9c8e17bf18120259e41e75dab3b9\n(cherry picked from commit 216cce39c9c22831d725703f41c54b9305d65dc8)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/cfd87ccba1a5a667ce6a917c6bb741a53c8a8ff6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/cfd87ccba1a5a667ce6a917c6bb741a53c8a8ff6"}]},"branch":"refs/heads/stable/zed"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[]}
