)]}'
{"id":"openstack%2Foctavia~868423","triplet_id":"openstack%2Foctavia~master~I98cce51a2536bcf204e5753ce4984af57e7e69e3","project":"openstack/octavia","branch":"master","attention_set":{},"removed_from_attention_set":{"29244":{"account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"last_update":"2023-01-11 19:25:30.000000000","reason":"Change was submitted"},"34429":{"account":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"last_update":"2023-01-02 11:27:22.000000000","reason":"\u003cGERRIT_ACCOUNT_34429\u003e replied on the change","reason_account":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}}},"hashtags":[],"change_id":"I98cce51a2536bcf204e5753ce4984af57e7e69e3","subject":"Fix listener creation with fully-populated API","status":"MERGED","created":"2022-12-22 15:53:40.000000000","updated":"2023-01-11 19:26:47.000000000","submitted":"2023-01-11 19:25:30.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":7,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"868423","meta_rev_id":"9767754fcb22f42230ebe955f2456f667cb350c3","_number":868423,"virtual_id_number":868423,"owner":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-01-11 19:25:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"recommended":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"all":[{"value":1,"date":"2023-01-04 16:22:38.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":2,"date":"2023-01-11 16:44:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":2,"date":"2023-01-02 11:27:22.000000000","_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":1,"date":"2023-01-11 16:44:55.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"approved":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"all":[{"value":0,"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":2,"date":"2023-01-02 12:12:26.000000000","_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-12-22 18:30:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-12-23 11:03:25.000000000","updated_by":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"reviewer":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"state":"REVIEWER"},{"updated":"2022-12-28 12:03:30.000000000","updated_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"reviewer":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"state":"REVIEWER"},{"updated":"2023-01-11 16:44:55.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"}],"messages":[{"id":"3200135d9299a31b9de54241b1c6bd5c5d24e6dd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2022-12-22 15:53:40.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3d1aa52d85758a5ac703a25bc02ce82208d0c9d8","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2022-12-22 15:59:21.000000000","message":"Patch Set 1: Backport-Candidate+2","accounts_in_message":[],"_revision_number":1},{"id":"7e9e719a6e9b9811ded911191a5f628d0ff09e46","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-22 18:30:28.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/934733823c134e5ea7987c9ceff3e2db\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/73b7e53b7af84a3ba39e79125a5bc8be : SUCCESS in 11m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f5a9b576596242099b56710b6cf7d651 : SUCCESS in 5m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/64b902520d504e3e9807737f5c0d171d : SUCCESS in 5m 10s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3f664d1857f04c09be8c58249a26c409 : SUCCESS in 5m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/afd0ffa3a6af45c8b455353f79be408d : SUCCESS in 11m 43s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/140f2f61acfa4e69ad4bc12d51bfc36c : SUCCESS in 4m 42s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/098538b0b0bb47a684b7a8e16d5c5cd0 : SUCCESS in 5m 30s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/2df3640c943c4b46a5797f0578a1105a : SUCCESS in 8m 05s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/cb6b8ff9d21b4e9c9dc0b6742129e79c : SUCCESS in 5m 14s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/2a204ba25e5a4d7e9ab48ed5898bc809 : SUCCESS in 6m 50s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/9261b66b31e74fdd8c91bd824a1bf1ef : SUCCESS in 7m 18s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/63bca04a6117468fbde64e5f3a71a280 : SUCCESS in 10m 48s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/8416b855982d4908acd42226e579746a : SUCCESS in 1h 28m 16s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/0683d6c45ef44a2d8467f88de51fbf58 : SUCCESS in 1h 37m 13s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/784317e68c284f94b5a7f9088935e365 : SUCCESS in 33m 28s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/7c413270d020455c9905c6cc22dccc2a : SUCCESS in 48m 07s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/5b18cafc294c49abb2d078ff8e252372 : FAILURE in 1h 31m 35s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/f2e0ce37665e495faba6fd4485a368be : SUCCESS in 36m 18s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/da923971a850411f86189daadfe9907a : SUCCESS in 1h 28m 47s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/0106f8a91cd1452eaa2472bba9487a13 : SUCCESS in 2h 06m 01s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/bebcf0322d7242a78fbc288edb126290 : SUCCESS in 1h 50m 59s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/6bcb8e1cfb194101b4a3e24671054728 : SUCCESS in 1h 56m 56s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/6a629f329e4d4fdf94dcb04f32c6a126 : SUCCESS in 1h 29m 52s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/c1133ce56f3a4dce9cf30074ad339ad8 : SUCCESS in 2h 29m 49s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"e5087312ebda09febf6990ab5984acd349ef06a0","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2022-12-23 11:03:25.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"4a9740c467ad97c5e5f3053262a10a5c55cf3554","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-12-28 12:03:30.000000000","message":"Patch Set 1: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"5609fbfa16fdca42e7dbca1e0919a621bc267b1c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-01-02 10:38:56.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"a8b45738129ae5c3c68237449e86e70a30d952fa","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-01-02 10:39:00.000000000","message":"Patch Set 1: Backport-Candidate+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"45a8c4d688abe4267e0b67c5c122757f1612205e","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2023-01-02 11:27:22.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"07823225c7d79e2f4970e44c4f8e9fe37909fa37","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2023-01-02 12:12:26.000000000","message":"Patch Set 2: Backport-Candidate+2","accounts_in_message":[],"_revision_number":2},{"id":"7f21f841448aca99f258ecf9aa6b4dd982bf5407","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-02 13:15:41.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/789f9acb79e842b5948e3fe4fec808f7\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/34dd8f84b4e24324a30420399d05fd99 : FAILURE in 4m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/faaf6b46b9b7440cac45f890dddc58cb : FAILURE in 4m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9c87d646d7594a068cc71f86558747b2 : FAILURE in 4m 28s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/13741702ce374404823aff55d7bc1bfd : FAILURE in 4m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ae8cb2bc034042b2b6c489e248eea33c : FAILURE in 4m 52s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/978f99ec7093449591dd3af08fe262ac : SUCCESS in 4m 45s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/53383af88e414207af5f46776ef9d7f8 : FAILURE in 4m 23s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/0f25882275ac4ebfb54c5b95f57a1ac2 : FAILURE in 7m 00s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/105496ae689e44048061afa8b89d88da : SUCCESS in 3m 41s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/5ff5623a699743d0af55c010181686b2 : FAILURE in 4m 12s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/38885f1ccd064518ad400b8be0e882c0 : FAILURE in 6m 36s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/6a1e86fce3fe473c84f8b1e311e9f274 : FAILURE in 4m 38s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/16746e7bbe704a15a9b18a243994dad2 : SUCCESS in 1h 19m 07s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/9787ad227dc74f0ba3e825367a61e947 : SUCCESS in 1h 41m 36s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/31b5343ef1f64fe79953cbb2200d7d0a : SUCCESS in 32m 43s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/44beb4ee9a05437db362e01fb06325c0 : SUCCESS in 49m 52s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/a8569f6c5bd8474b82b59e9dd7580021 : FAILURE in 1h 24m 27s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/be6e785d7aa44681803f4e436c4e2935 : SUCCESS in 30m 00s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/6f1360f720a1451b8c6403ec27871080 : SUCCESS in 1h 37m 23s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/7a3b95f28ca140eca4c6601c6976d70f : SUCCESS in 2h 04m 12s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/f8fcbd6c1b3d4ec88bbd35bf29e3d45a : SUCCESS in 2h 15m 02s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/6158e167ec794f02a4426f633a3eb623 : TIMED_OUT in 2h 33m 45s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/9cf54b4742594946bf32627697d11333 : SUCCESS in 1h 42m 47s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/235b59285c0a42c0894651716b85969b : SUCCESS in 2h 14m 08s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"d152c3bf0cda932593e4a967b503cce23a54d946","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-01-04 16:17:12.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"9e75403df2b2b87175aae761b84853033afa38eb","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2023-01-04 16:22:38.000000000","message":"Patch Set 2: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"afe86cfe23e54722e25a7923df7df9d566903464","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-04 18:31:43.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dd2504aa1f4542c6870cb86848cedfad\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6c3c7be526af4a4dbc26ffc65f95d199 : SUCCESS in 12m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/562f7790f5e84eb0a032f968aba073d4 : SUCCESS in 5m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c681ba6e5e394d42af78237725791a20 : SUCCESS in 5m 18s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2a4bb2bd3a894c17b66b13ff4304e6c6 : SUCCESS in 5m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/518c3c7893c642ebb3d419a39cd45765 : SUCCESS in 13m 35s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0ad78bcefa5a49c4b5443e136aa6fcf6 : SUCCESS in 5m 33s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/c9e8861dfc1c4ffc9c8cc72bbc99dfc1 : SUCCESS in 5m 36s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/1d4ee92f9a1d42f39df59512777ddc4b : SUCCESS in 8m 56s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/8e8a5070521c4ba49426d53b5965ef00 : SUCCESS in 5m 15s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/0572c8fbd92f4de3b39d39d0aed372fd : SUCCESS in 8m 33s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/b1db397566fc4be6ab968b437e302bc6 : SUCCESS in 8m 43s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/5922053dfe2c4509ae5d6fedf4d40aa0 : SUCCESS in 9m 18s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/d9f1bb9464ab4936bbc4722798523c4c : SUCCESS in 1h 39m 32s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/8c612f4d23054a92b660d3f7fc3c8714 : SUCCESS in 1h 41m 33s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/d3ec882cc11b4ebbbf2fa8f918ffa101 : SUCCESS in 33m 48s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/68c55962effd49e182b7a1bfd0a7daff : SUCCESS in 49m 18s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/a8e20f4d050743629e76f294b501bf51 : SUCCESS in 1h 12m 21s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/f0d7b8eb4e28418ba24aa1ab4ae1e506 : SUCCESS in 30m 12s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/93575e2cc9fc440eb6d9764dff7b7fc6 : SUCCESS in 1h 45m 14s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/49b03bdd2def45d382b5ea711fdf030f : SUCCESS in 2h 12m 44s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/29aa82a33f274c6d99752c35ca520428 : SUCCESS in 1h 37m 36s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/2201ef7c01454d8d8c077ed5dc62d65d : SUCCESS in 1h 45m 01s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/bf7f8772b9b6432c8604f18988737819 : SUCCESS in 1h 40m 38s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/830cc22761744cbd804801c9d0dcca05 : SUCCESS in 2h 05m 24s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"6f3068a62123f0fdbe761eb2f5216e2b8a7a0c5a","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-01-11 16:44:55.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"f94efdf09e468121f03f61551e07a85bf1cfc212","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-11 16:45:35.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"d7cbe96023ac0c61d5c5f5db5650684e192f1d82","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-11 19:25:30.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7def95f995474acda2b4939e43eb2491\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e7e8d176a391434f9366d35353deb84b : SUCCESS in 5m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f9250cd58bd6427f951a592137f047a8 : SUCCESS in 7m 10s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6f8313c88a2e4c06adcdbde620c47439 : SUCCESS in 4m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/263594a224044793a0a97eb9e85ff446 : SUCCESS in 13m 46s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/13ae6a526a5048e79c5e4bd7924187b3 : SUCCESS in 4m 43s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/8037ecebafd44452b9d9e1ba2cfb885e : SUCCESS in 6m 09s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/3c4db6ffd359493191dc23c2a738bc00 : SUCCESS in 7m 52s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/9ba32077a51c4349afecf92608f8d386 : SUCCESS in 1h 29m 25s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/51618532303446eba63191488d786e43 : SUCCESS in 2h 28m 51s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/958c7487872d477e84e8a1e9aee7b1a4 : SUCCESS in 43m 43s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/68a81ad6e52e4b32b1675416f36da23a : SUCCESS in 58m 26s","accounts_in_message":[],"_revision_number":2},{"id":"506ccfb22bc0438fddabab6de6af7ba9d541ff51","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-11 19:25:30.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"9767754fcb22f42230ebe955f2456f667cb350c3","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-11 19:26:47.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/37997874c9d142f9ba1863cb9e9709ff\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/85862abb95a14566818eed94bdfec648 : SUCCESS in 58s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/58f203a0f7cf4b6f9b99978ecde5d647 : SUCCESS in 39s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"d0a0f53dc3522cdc334afebbc42e70ac3f198530","revisions":{"73592ac59096ba782f10cec5a92264100011c27f":{"kind":"REWORK","_number":1,"created":"2022-12-22 15:53:40.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/23/868423/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/23/868423/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/23/868423/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/23/868423/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/23/868423/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/23/868423/1"}}},"commit":{"parents":[{"commit":"ac3618b140b1b8ba02f2263abf910c17b3f08aff","subject":"Merge \"Fix typo in docs section title\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/ac3618b140b1b8ba02f2263abf910c17b3f08aff"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2022-12-22 15:30:57.000000000","tz":-300},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2022-12-22 15:30:59.000000000","tz":-300},"subject":"Fix listener creation with fully-populated API","message":"Fix listener creation with fully-populated API\n\nCreating a listener with the fully-populated API was broken in SINGLE\ntopology, one required parameter was missing from the flow.\n\nStory 2010488\nTask 47060\n\nChange-Id: I98cce51a2536bcf204e5753ce4984af57e7e69e3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/73592ac59096ba782f10cec5a92264100011c27f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/73592ac59096ba782f10cec5a92264100011c27f"}]},"branch":"refs/heads/master"},"d0a0f53dc3522cdc334afebbc42e70ac3f198530":{"kind":"REWORK","_number":2,"created":"2023-01-02 10:38:56.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/23/868423/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/23/868423/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/23/868423/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/23/868423/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/23/868423/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/23/868423/2"}}},"commit":{"parents":[{"commit":"ac3618b140b1b8ba02f2263abf910c17b3f08aff","subject":"Merge \"Fix typo in docs section title\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/ac3618b140b1b8ba02f2263abf910c17b3f08aff"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2022-12-22 15:30:57.000000000","tz":-300},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-01-02 10:38:35.000000000","tz":60},"subject":"Fix listener creation with fully-populated API","message":"Fix listener creation with fully-populated API\n\nCreating a listener with the fully-populated API was broken in SINGLE\ntopology, one required parameter was missing from the flow.\n\nStory 2010488\nTask 47060\n\nChange-Id: I98cce51a2536bcf204e5753ce4984af57e7e69e3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d0a0f53dc3522cdc334afebbc42e70ac3f198530"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d0a0f53dc3522cdc334afebbc42e70ac3f198530"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}}]}],"submit_requirements":[]}
