)]}'
{"id":"openstack%2Foctavia~868765","triplet_id":"openstack%2Foctavia~master~I5ac930e0d3928bb8acba4212dfff25f788acd442","project":"openstack/octavia","branch":"master","topic":"tox4","attention_set":{},"removed_from_attention_set":{"29244":{"account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"last_update":"2023-01-04 00:19:34.000000000","reason":"Change was submitted"},"34429":{"account":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"last_update":"2023-01-03 17:11:18.000000000","reason":"\u003cGERRIT_ACCOUNT_34429\u003e replied on the change","reason_account":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}}},"hashtags":[],"change_id":"I5ac930e0d3928bb8acba4212dfff25f788acd442","subject":"Fix broken tox.ini with tox4","status":"MERGED","created":"2022-12-28 15:09:32.000000000","updated":"2023-01-05 08:45:35.000000000","submitted":"2023-01-04 00:19:34.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":7,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"868765-tox4","meta_rev_id":"3be71e29e38dde2858ad0f349feb0650b5424408","_number":868765,"virtual_id_number":868765,"owner":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-01-04 00:19:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":2,"date":"2023-01-03 17:11:18.000000000","_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":1,"date":"2023-01-03 22:21:28.000000000","_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"CC":[{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-12-28 19:19:02.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-01-02 13:13:30.000000000","updated_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"REVIEWER"},{"updated":"2023-01-03 10:04:40.000000000","updated_by":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"reviewer":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"state":"CC"},{"updated":"2023-01-05 08:45:35.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"CC"}],"messages":[{"id":"2cde4611444ae95b02810c9a41b087eae5064d6a","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-12-28 15:09:32.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2305ec39661023eab2132107820f853b90307bff","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-12-28 15:15:14.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"fcefb66c8498a6b3ef980c63cd660ad1b2ab2dd9","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-12-28 15:29:46.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"41998cd76ecfa90c39acbad85a7359c9e90dd8a5","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-12-28 15:45:48.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"9de0c750462b8ae54635d80608ac2fffebfbbce6","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-12-28 16:58:36.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"d47acd71b515a9df455bf8353f7f0238a07482cf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-28 19:19:02.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bf1c9fbd25544520942ebbe00e2bf021\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/d4416aae3d364c05a4c8677fb27e37fe : SUCCESS in 4m 39s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/04fa72ca064342e5b69015ef02fd122b : SUCCESS in 10m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3a98a4fac31d4183844dcd906225bfd7 : SUCCESS in 5m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/67607af4ce8e417a9dee1371ceaf0fce : SUCCESS in 4m 45s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0564b8eacde1493896edc0314dd7247e : SUCCESS in 4m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ab6abb1164754ec9901e41ef24db7816 : SUCCESS in 12m 08s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8f656917d5ef414fb3641ef85e1727b2 : SUCCESS in 4m 46s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/c40990ac347c482886b02ae9f1e5b7c1 : SUCCESS in 5m 25s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/0e2c7f7856a04fbab25a819e797a857d : SUCCESS in 7m 39s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/3293d9c66f4e4c1497f272348aaa247c : SUCCESS in 4m 32s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/e84bf5a723724144ab7d41d551c44cdb : SUCCESS in 7m 02s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/5095489d8be64fb48cbc0db56a23107a : SUCCESS in 6m 48s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/bdf39caa348a4fc4bee03d0f6fc5e89e : SUCCESS in 7m 41s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/19e75b828bfc48c985a46c326c5388f2 : SUCCESS in 1h 34m 44s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/5ebf6a48ce664efa9c6ef6fba74c2add : SUCCESS in 2h 14m 02s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/a317e870d73243ed84f3fe6f8629687e : SUCCESS in 35m 33s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/ec6377bc8562461895147b7d66db6631 : FAILURE in 1h 17m 17s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/5f69b74e9c834025b0aaeee71708f2d8 : SUCCESS in 33m 49s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/12fae6fadde94866a4051bfb8ea3447a : SUCCESS in 2h 04m 04s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/04cd2dd53e3b457c8e079ff1c40c4465 : SUCCESS in 2h 17m 10s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/2a23bf2639334c77a20256ddca36961e : SUCCESS in 2h 13m 37s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/834032bb177848afba7a50e2ff9da226 : SUCCESS in 2h 12m 21s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/4a26c56912e54f15ab86e27379e0eb2c : SUCCESS in 1h 58m 08s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/c701ba2618cb41c5a6b242713b197f99 : SUCCESS in 2h 05m 47s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"1679193931987b1d5595dfcc16749466bba8c031","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-12-28 21:41:38.000000000","message":"Patch Set 4:\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":4},{"id":"bded88db1919b7676ce89c70d0db07d5d354c341","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-12-29 08:22:54.000000000","message":"Patch Set 5: Commit message was updated.","accounts_in_message":[],"_revision_number":5},{"id":"702555bdd38494120dc9db581598616b2866d3ce","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-29 11:00:32.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6d099dbf13da41708cdf166f7b8a93ce\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/ab16d0c869d44085a4de6d19cc55ddea : SUCCESS in 4m 44s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f442999bec314ef8a728cafde5d3821b : SUCCESS in 11m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fb463b75667b4cb0be6d56c82f3a8d15 : SUCCESS in 5m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e270fc67109a40ae940e9883a97eb831 : SUCCESS in 5m 01s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7f3534be6755483790c4a1b5df58edbc : SUCCESS in 4m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c620b807fc2e4744a6191c08610e2f41 : SUCCESS in 11m 46s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/28e5bab5439e436fa496d1315cd0e1ec : SUCCESS in 4m 35s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/0c1d553d990241178ee6c812849b2653 : SUCCESS in 4m 57s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/5a1b8c10551649d5a94a96fc1e5bdbb1 : SUCCESS in 9m 07s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/e29d817a0bb8425b9a6351d46416d371 : SUCCESS in 4m 01s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/49f643412e9a499e9a353656cb61f28b : SUCCESS in 7m 55s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/389bab53a9ef428b9c63c3543f0ad335 : SUCCESS in 6m 57s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/c99f2d8281ac43adaa0577dbc1090fa2 : SUCCESS in 8m 51s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/56b04170c09f42198855e9d96b13d7b4 : SUCCESS in 1h 25m 15s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/cd4c926d6f5742989891a90123b4ba34 : SUCCESS in 1h 43m 27s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/7c85fef3ff274b16838d716520f828fe : SUCCESS in 32m 51s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/9733a28a10fd45c39a1c49236e732569 : FAILURE in 1h 17m 28s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/9ba6979e871746ffa49604b5133c7ef8 : SUCCESS in 30m 16s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/f3a2dd403820411eb8679a41d4985753 : SUCCESS in 1h 41m 29s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/ec26c4e0c6af4c58ad50264e1da3f70f : SUCCESS in 2h 07m 01s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/6fcf4fe78f2348aa881ca80e7c236737 : SUCCESS in 1h 35m 23s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/c2632e296cf24ea7940da3b2c5ecc081 : SUCCESS in 1h 42m 42s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/ad6c0718a5b841039f6248adf73c3e36 : SUCCESS in 1h 28m 48s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/a654ed38f7bb425c8c65d5c306679c12 : SUCCESS in 2h 20m 32s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"4b591307411e4d1addce2b4c0a74bf610a17b5dd","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-01-02 13:13:30.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"5eadf2d91507ac864c6043f3dfb9a75ee412590d","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-01-02 13:44:00.000000000","message":"Patch Set 5: Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"e9280df8d1c119d7bc0a19d1f71f7f26c413b50e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-02 13:44:39.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"603f1e7d5cfbd9d2507634cec94bb614174d7b13","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-02 13:51:12.000000000","message":"Patch Set 5: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/4a507d674645461aa308df0b422b49eb\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/987ecb4738e346d281f553a7ee2ad671 : CANCELED\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5ceee7762c3a4decb9da380677507a80 : CANCELED\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5b985751cd3c4a7d878c1220f5fc1bc4 : CANCELED\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/58949d876aa0448aba85d2750ec5f992 : CANCELED\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0e29fd7dd9594dd19727b063bd50eeb8 : CANCELED\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e5fc31dadd634c5a832fd0ddc93ee7d8 : CANCELED\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/1f62372e5f82457ca96a9989b7fe7827 : CANCELED\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/49a27ab9819e4d44ba456ac840103f54 : FAILURE in 2m 19s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/2175e13bbe4349f6833af6ea37f55799 : CANCELED\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/c06fc5fb20a042b38772ad9aae1306d0 : CANCELED\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/2738bc08cad04736809610a19f55f765 : CANCELED\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/7e9ff156bc714b6aacdfed114e6cbe10 : CANCELED","accounts_in_message":[],"_revision_number":5},{"id":"49d4ed77bb15448ee27380f2f491ae1d5324d545","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2023-01-02 13:59:47.000000000","message":"Topic fix-broken-pep8 removed","accounts_in_message":[],"_revision_number":5},{"id":"32a6a68ddfb0f1e640688444bf67baa55516b11a","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2023-01-02 13:59:53.000000000","message":"Topic set to tox4","accounts_in_message":[],"_revision_number":5},{"id":"6017ebd65496c3e050fd53a135f9857cbfef4b43","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2023-01-02 14:08:35.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"ffbff0ba62f8482f68a4f122b1cf2b8a6746a773","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2023-01-02 14:38:40.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"85035d646a432c05f640fbc664e94b5055b1b163","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2023-01-02 14:53:01.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"2f25046d1f81eb37a67523356edbc961deb7a2bf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2023-01-02 14:53:39.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"71e302658a2575ff5dfbaa3c0c957df7e3a6e735","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-02 17:33:33.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/d20360b67ed341a5ace1fd20edc53ab2\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/48a1cc8a7fa34e8ea7f0b02a9cd54720 : FAILURE in 3m 01s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4d60775ef87c4308935b1367b57de7f5 : FAILURE in 2m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d9db3728a0214bdaa324c2a6c4325456 : FAILURE in 2m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7d62256445bf4c67a052a08dadd4336d : FAILURE in 2m 17s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/86df0881740f46908442134c4c5b2478 : FAILURE in 2m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4e9fd15fc83543dd96eda17388cbcf19 : FAILURE in 3m 47s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7a46f277db454715b85820b7e2630824 : SUCCESS in 4m 24s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/1aa75004150744c2ad972103125cefd1 : FAILURE in 3m 37s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/ec89a7c5757347a7be25c5d925d2bb72 : FAILURE in 2m 20s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/7d5672a917634462a14706ace3c8cf21 : FAILURE in 2m 26s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/9564a743efc04263b9b7910bff0e67b9 : FAILURE in 2m 18s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/e1199623a62745439be6b9a4bf8ea49d : FAILURE in 2m 26s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/690a950104384dca81e986a6b7ce2364 : FAILURE in 2m 29s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/bc1c4172a3174103b4e3d2ada5945a9a : SUCCESS in 1h 29m 11s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/35950b8e11914ba7a077dcc565381fb3 : SUCCESS in 1h 48m 21s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/62b0cbc554414f5c9536e1b09aa1053d : SUCCESS in 33m 48s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/aff0dbacf1044cd4bd0892f3586f8f1a : FAILURE in 1h 08m 09s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/423b3fffd88a4f478c4171a2293b0d19 : SUCCESS in 37m 06s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/6b0a1a13c82f475d9e8940a17b511d74 : SUCCESS in 2h 15m 14s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/a4cadb7cdb4f423e86ee06846e743d49 : SUCCESS in 2h 37m 10s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/9a675fb72dc645a493d021f0a788a5ce : SUCCESS in 2h 31m 27s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/9e49b94ece05450e8eb2625b2737c010 : SUCCESS in 1h 56m 41s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/ddf1d0b5dd9d47749ce2b41b0a8da890 : SUCCESS in 2h 05m 11s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/af038e63ab204707aa16f7547539a8a0 : SUCCESS in 2h 20m 27s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"4e7bd9d6bc62e4c42fb8be4e1816ba2690efbf1c","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2023-01-03 08:14:44.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"4f1e415eeb727a7c6e0cfde639e478875ed15561","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2023-01-03 08:50:40.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"5ff923d7c19f3d90f7ac9b9a115869fe6628cc35","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2023-01-03 10:04:40.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"19cdd43b58a1cecd94782de0e250c48298b06f79","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2023-01-03 10:05:14.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"fe0f4fe14b84dbe6d40225b76ed134ec137fff1d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-03 11:32:21.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/f6514ab6d6c74100b262b910974cb42d\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/7afdd46811174d84a71c540b9caab1c0 : FAILURE in 3m 01s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/86fb2a6b16014bea967094d7bf881ae6 : FAILURE in 1m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/63aed70520d4454b894e2f4e58cf6356 : FAILURE in 2m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3546c771e9b345e6bacfce0fa9aa6563 : FAILURE in 2m 22s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ba168d48367c4341846d09279f7e9534 : FAILURE in 2m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/54dfb32e5fe340aa8b3496bba9055560 : FAILURE in 3m 18s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/23999f1a1e234791883c7f3aec043e9a : SUCCESS in 5m 41s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/7bc174d099734102a77b47a16fa773ac : FAILURE in 2m 04s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/ad49ea94dbfb4f19a4f5e91a473b446a : FAILURE in 2m 34s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/58df26e636d84eaaa59c1441db747be0 : FAILURE in 3m 42s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/469344dc79e141c2b47994d39caabebe : FAILURE in 2m 45s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/cc0ce1ebf60b4ed98d9eb0731e3c4b3a : FAILURE in 3m 05s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/2386bb49bf3f4b7db25fc09246eca229 : FAILURE in 2m 14s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/4e44ce0a23fe47dda488e989edd072f5 : SUCCESS in 1h 22m 43s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/becb1897830f4aeb9d20807f5b9305bb : SUCCESS in 1h 55m 35s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/ffc992a372254b21bd6ec7dd1c89a3ba : SUCCESS in 37m 18s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/4124c1f497c14c99a421b3968187fac9 : FAILURE in 1h 24m 15s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/6ee9cd4d92064adf822f75c026c27438 : SUCCESS in 39m 45s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/664b754fe37647668d6e16960ea501d2 : SUCCESS in 2h 18m 21s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/da26bb26faf947a2988bb75b1af3ccc2 : SUCCESS in 2h 39m 17s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/8cf2917da5884d9783ccf7690fc878ec : SUCCESS in 2h 00m 52s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/6f2e85aba9d8433bb36e60a5608e7906 : TIMED_OUT in 2h 33m 41s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/b8ddfbc156814392928fca7a4af08a04 : SUCCESS in 2h 27m 49s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/a4755db4a5a345e980ca630a00425b1f : TIMED_OUT in 2h 34m 45s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"11cf90a50e7334330b9f3311b3e82933a47b5838","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-01-03 13:49:16.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"702d4c1068867afd50785630b2d58fc3afe1dadc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-03 13:54:28.000000000","message":"Patch Set 11: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/5057fde0f3944df295c5a7467e99bc41\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/a2a404fbe47947778f74e01e476a984a : FAILURE in 3m 34s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/1df2316be0ac4efb8bd0d09df08f9fec : FAILURE in 2m 37s","accounts_in_message":[],"_revision_number":11},{"id":"329078794dab787312f0849154721b9fbe2c8a45","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-01-03 14:10:37.000000000","message":"Patch Set 11:\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"e5286be8c42b22e169449d143a2f02c919b9f2a2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-03 14:15:17.000000000","message":"Patch Set 11:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/27eadbaa444a4c078d8a653567bd1ad4\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/f1ad07aafe0441d496e03659f0fd69e9 : FAILURE in 3m 21s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/f2578c3d0c3147cbb8176358cff6e291 : FAILURE in 2m 55s","accounts_in_message":[],"_revision_number":11},{"id":"b8dda6e1b07fb0da2b41b11d79d87e2c5c2d9ce4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-01-03 15:56:07.000000000","message":"Uploaded patch set 12.","accounts_in_message":[],"_revision_number":12},{"id":"6e857ef336a3b5e6a9781db327a03355028647d8","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2023-01-03 17:11:18.000000000","message":"Patch Set 12: Code-Review+2","accounts_in_message":[],"_revision_number":12},{"id":"1d112803a5b309a7f8d992595540bf9c7fb9ef6d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-03 18:20:22.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/df6afebace694c3b8357de9a508b8400\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/e03876fe44f1444fadec47b8417470c3 : SUCCESS in 5m 14s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1c7801684b884def8d52e6a155a1a6d6 : SUCCESS in 13m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8df639635985438eb6fc8b4135c15d15 : SUCCESS in 5m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c8c20d4e293b4ab386e89efbc05be5ef : SUCCESS in 4m 33s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/bf0b8fd52db142f1969d220a7c00780a : SUCCESS in 4m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e4fcc40c4d0d483a843d84c9b377923d : SUCCESS in 13m 29s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f11339b53aa94031ac8cf950a816c898 : SUCCESS in 5m 11s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/e3191afad4174fbab54ac0f52c3d798c : SUCCESS in 4m 31s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/dea99b9e959948aa817caad5a3f0b90c : SUCCESS in 7m 03s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/4e1b0540f8b14dabb88716f6ec2c6585 : SUCCESS in 4m 14s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/950e643c63a3459fb45ea23b594ec3aa : SUCCESS in 7m 52s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/16694216603741e6837dd82d7a77ebe5 : SUCCESS in 7m 50s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/1eddc0ce6dbf4f2ca5e0ec4b6c5b825b : SUCCESS in 8m 44s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/4c45872499df4480864148b1e0df4b4a : SUCCESS in 1h 21m 54s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/629740171bf74c5bb2ec3f3a4398dfe6 : SUCCESS in 1h 28m 01s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/ea4292d682044ce5975fba8116c67258 : SUCCESS in 34m 17s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/876e5aa7dd344908ba2fb0a5547ac556 : SUCCESS in 1h 15m 55s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/d0978fdb31fe4b81abc97880406379ef : SUCCESS in 32m 21s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/63e9093dd1a14fa8a2888d629ffccaa1 : SUCCESS in 1h 32m 08s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/6579b11c590a4e8e96a087ffbbca9c6a : SUCCESS in 1h 59m 53s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/ee5765cd480a489d94331be6d74bcec0 : SUCCESS in 1h 50m 00s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/b4ca9d4e0f5d4031a5e17da8d6322e40 : SUCCESS in 1h 50m 39s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/94aed487c975490495174f9b0fde6bcd : SUCCESS in 2h 01m 38s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/d25dea02bae3423598587795a8e2695c : SUCCESS in 2h 22m 29s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"b6dd861f5826a9e52a97b0db08bc62ff3e84a56c","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2023-01-03 22:21:28.000000000","message":"Patch Set 12: Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"0b640b5d7d12a1cf8610b47ef51d3d824d4a172d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-03 22:22:04.000000000","message":"Patch Set 12: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":12},{"id":"b1325044ccfddd0f9c311f988d8aecc8755b7485","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-04 00:19:34.000000000","message":"Patch Set 12: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c66997e44ad04189b9d46b2bc5462bdf\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/4ad604ed5d7a401b97d3c2d449982477 : SUCCESS in 5m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2d14d2ff536644c5801e404c5e9515ef : SUCCESS in 5m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c304e7714564485ca99248634056f1d5 : SUCCESS in 5m 06s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7956653ec9684e84b30f6984673eb9a4 : SUCCESS in 4m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3f9a97e2656546b99f31ddd8f5c6ed2c : SUCCESS in 12m 18s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/402f6dad559e4892bbb4f1a2334af578 : SUCCESS in 4m 55s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/7070f59d86c541a9b31288049d0754c8 : SUCCESS in 4m 17s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/97e276e67f8741348114d746417f3da7 : SUCCESS in 6m 52s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/f395afd65c0c46cdaa822fea2580f09c : SUCCESS in 1h 22m 27s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/46fc158e94be42c487918a1a561dcaba : SUCCESS in 1h 56m 44s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/fca829e229674481919ae37ea87c4296 : SUCCESS in 39m 48s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/72fa6ad430824984b67fb277f0e73bb6 : SUCCESS in 1h 07m 13s","accounts_in_message":[],"_revision_number":12},{"id":"5faf75a1136c15981b5555a7c1ed8f6600e92ced","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-04 00:19:34.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":12},{"id":"95b0e0c852e48e58e055ead8b3ae18a728953f90","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-04 00:20:57.000000000","message":"Patch Set 12:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/eae8753acaf0405183f074701fa8df40\n\n- promote-api-ref https://zuul.opendev.org/t/openstack/build/c4cb51c1c5324340ba123c4ff0f3dd82 : SUCCESS in 47s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/387576478b3949b3b1ba21a535e660e0 : SUCCESS in 1m 01s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/006a19ece30542a2bcf36b1f0bdce2f3 : SUCCESS in 51s","accounts_in_message":[],"_revision_number":12}],"current_revision_number":12,"current_revision":"41ee7e7aa1117ae42ea0fd1ea7b3ea8ebba27476","revisions":{"24881f4545c362927e302869330d8296374eaab5":{"kind":"REWORK","_number":1,"created":"2022-12-28 15:09:32.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/65/868765/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/65/868765/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/65/868765/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/65/868765/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/65/868765/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/65/868765/1"}}},"commit":{"parents":[{"commit":"ac3618b140b1b8ba02f2263abf910c17b3f08aff","subject":"Merge \"Fix typo in docs section title\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/ac3618b140b1b8ba02f2263abf910c17b3f08aff"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-12-28 15:09:12.000000000","tz":60},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-12-28 15:09:12.000000000","tz":60},"subject":"Fix broken pep8 CI job by allowing the coding-checks.sh in tox.ini","message":"Fix broken pep8 CI job by allowing the coding-checks.sh in tox.ini\n\nChange-Id: I5ac930e0d3928bb8acba4212dfff25f788acd442\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/24881f4545c362927e302869330d8296374eaab5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/24881f4545c362927e302869330d8296374eaab5"}]},"branch":"refs/heads/master"},"f240d0bdb8d97757c1cfc22f30b75a73545526c9":{"kind":"REWORK","_number":2,"created":"2022-12-28 15:15:14.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/65/868765/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/65/868765/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/65/868765/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/65/868765/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/65/868765/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/65/868765/2"}}},"commit":{"parents":[{"commit":"ac3618b140b1b8ba02f2263abf910c17b3f08aff","subject":"Merge \"Fix typo in docs section title\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/ac3618b140b1b8ba02f2263abf910c17b3f08aff"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-12-28 15:09:12.000000000","tz":60},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-12-28 15:15:03.000000000","tz":60},"subject":"Fix broken pep8 CI job by allowing the coding-checks.sh in tox.ini","message":"Fix broken pep8 CI job by allowing the coding-checks.sh in tox.ini\n\nChange-Id: I5ac930e0d3928bb8acba4212dfff25f788acd442\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/f240d0bdb8d97757c1cfc22f30b75a73545526c9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/f240d0bdb8d97757c1cfc22f30b75a73545526c9"}]},"branch":"refs/heads/master"},"853d5b50d6271bfd266579e98f83b1f4a1eef2dc":{"kind":"REWORK","_number":3,"created":"2022-12-28 15:45:48.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/65/868765/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/65/868765/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/65/868765/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/65/868765/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/65/868765/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/65/868765/3"}}},"commit":{"parents":[{"commit":"ac3618b140b1b8ba02f2263abf910c17b3f08aff","subject":"Merge \"Fix typo in docs section title\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/ac3618b140b1b8ba02f2263abf910c17b3f08aff"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-12-28 15:09:12.000000000","tz":60},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-12-28 15:45:37.000000000","tz":60},"subject":"Fix broken pep8 CI job by allowing the coding-checks.sh in tox.ini","message":"Fix broken pep8 CI job by allowing the coding-checks.sh in tox.ini\n\nChange-Id: I5ac930e0d3928bb8acba4212dfff25f788acd442\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/853d5b50d6271bfd266579e98f83b1f4a1eef2dc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/853d5b50d6271bfd266579e98f83b1f4a1eef2dc"}]},"branch":"refs/heads/master"},"e970f5b614658436c1a5fadc0a06ca2fb93c0dc8":{"kind":"REWORK","_number":4,"created":"2022-12-28 16:58:36.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/65/868765/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/65/868765/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/65/868765/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/65/868765/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/65/868765/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/65/868765/4"}}},"commit":{"parents":[{"commit":"ac3618b140b1b8ba02f2263abf910c17b3f08aff","subject":"Merge \"Fix typo in docs section title\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/ac3618b140b1b8ba02f2263abf910c17b3f08aff"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-12-28 15:09:12.000000000","tz":60},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-12-28 16:56:12.000000000","tz":60},"subject":"Fix broken tox.ini with tox4","message":"Fix broken tox.ini with tox4\n\n- scripts were not in allowlist for pep8 environment\n- sdist setting breaks the docs environment\n\nChange-Id: I5ac930e0d3928bb8acba4212dfff25f788acd442\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e970f5b614658436c1a5fadc0a06ca2fb93c0dc8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e970f5b614658436c1a5fadc0a06ca2fb93c0dc8"}]},"branch":"refs/heads/master"},"5d04ee27ea61985ebe91ca93a3d38228cb852690":{"kind":"NO_CODE_CHANGE","_number":5,"created":"2022-12-29 08:22:54.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/65/868765/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/65/868765/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/65/868765/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/65/868765/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/65/868765/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/65/868765/5"}}},"commit":{"parents":[{"commit":"ac3618b140b1b8ba02f2263abf910c17b3f08aff","subject":"Merge \"Fix typo in docs section title\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/ac3618b140b1b8ba02f2263abf910c17b3f08aff"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-12-28 15:09:12.000000000","tz":60},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-12-29 08:22:54.000000000","tz":0},"subject":"Fix broken tox.ini with tox4","message":"Fix broken tox.ini with tox4\n\n- scripts were not in allowlist for pep8 environment\n- skipsdist setting breaks the docs environment\n\nChange-Id: I5ac930e0d3928bb8acba4212dfff25f788acd442\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/5d04ee27ea61985ebe91ca93a3d38228cb852690"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/5d04ee27ea61985ebe91ca93a3d38228cb852690"}]},"branch":"refs/heads/master","description":"Edit commit message"},"32ad26ad041b7017990bdc9a236f8681b7e2fe47":{"kind":"REWORK","_number":6,"created":"2023-01-02 14:08:35.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/65/868765/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/65/868765/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/65/868765/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/65/868765/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/65/868765/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/65/868765/6"}}},"commit":{"parents":[{"commit":"ac3618b140b1b8ba02f2263abf910c17b3f08aff","subject":"Merge \"Fix typo in docs section title\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/ac3618b140b1b8ba02f2263abf910c17b3f08aff"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-12-28 15:09:12.000000000","tz":60},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2023-01-02 14:04:04.000000000","tz":60},"subject":"Fix broken tox.ini with tox4","message":"Fix broken tox.ini with tox4\n\n- scripts were not in allowlist for pep8 environment\n- skipsdist setting breaks the docs environment\n- fix that tox couldn\u0027t find Python interpreter\n\nChange-Id: I5ac930e0d3928bb8acba4212dfff25f788acd442\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/32ad26ad041b7017990bdc9a236f8681b7e2fe47"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/32ad26ad041b7017990bdc9a236f8681b7e2fe47"}]},"branch":"refs/heads/master"},"3dac2b925d51677a3bc911e699721f653755f6db":{"kind":"REWORK","_number":7,"created":"2023-01-02 14:38:40.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/65/868765/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/65/868765/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/65/868765/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/65/868765/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/65/868765/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/65/868765/7"}}},"commit":{"parents":[{"commit":"ac3618b140b1b8ba02f2263abf910c17b3f08aff","subject":"Merge \"Fix typo in docs section title\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/ac3618b140b1b8ba02f2263abf910c17b3f08aff"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-12-28 15:09:12.000000000","tz":60},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2023-01-02 14:38:25.000000000","tz":60},"subject":"Fix broken tox.ini with tox4","message":"Fix broken tox.ini with tox4\n\n- scripts were not in allowlist for pep8 environment\n- skipsdist setting breaks the docs environment\n\nChange-Id: I5ac930e0d3928bb8acba4212dfff25f788acd442\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/3dac2b925d51677a3bc911e699721f653755f6db"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/3dac2b925d51677a3bc911e699721f653755f6db"}]},"branch":"refs/heads/master"},"e3d241e84b64c6b9fa0b1adef0138a89f64a2245":{"kind":"REWORK","_number":8,"created":"2023-01-02 14:53:01.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/65/868765/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/65/868765/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/65/868765/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/65/868765/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/65/868765/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/65/868765/8"}}},"commit":{"parents":[{"commit":"ac3618b140b1b8ba02f2263abf910c17b3f08aff","subject":"Merge \"Fix typo in docs section title\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/ac3618b140b1b8ba02f2263abf910c17b3f08aff"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-12-28 15:09:12.000000000","tz":60},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2023-01-02 14:52:45.000000000","tz":60},"subject":"Fix broken tox.ini with tox4","message":"Fix broken tox.ini with tox4\n\n- scripts were not in allowlist for pep8 environment\n- skipsdist setting breaks the docs environment\n\nChange-Id: I5ac930e0d3928bb8acba4212dfff25f788acd442\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e3d241e84b64c6b9fa0b1adef0138a89f64a2245"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e3d241e84b64c6b9fa0b1adef0138a89f64a2245"}]},"branch":"refs/heads/master"},"a2d8fe0f4d1ed3fc9ce882d4495260c05709b67a":{"kind":"REWORK","_number":9,"created":"2023-01-02 14:53:39.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/65/868765/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/65/868765/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/65/868765/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/65/868765/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/65/868765/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/65/868765/9"}}},"commit":{"parents":[{"commit":"ac3618b140b1b8ba02f2263abf910c17b3f08aff","subject":"Merge \"Fix typo in docs section title\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/ac3618b140b1b8ba02f2263abf910c17b3f08aff"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-12-28 15:09:12.000000000","tz":60},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2023-01-02 14:53:27.000000000","tz":60},"subject":"Fix broken tox.ini with tox4","message":"Fix broken tox.ini with tox4\n\n- scripts were not in allowlist for pep8 environment\n- skipsdist setting breaks the docs environment\n\nChange-Id: I5ac930e0d3928bb8acba4212dfff25f788acd442\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a2d8fe0f4d1ed3fc9ce882d4495260c05709b67a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a2d8fe0f4d1ed3fc9ce882d4495260c05709b67a"}]},"branch":"refs/heads/master"},"9352563db15b7879e8988de685a061d32ecf1a69":{"kind":"REWORK","_number":10,"created":"2023-01-03 08:50:40.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/65/868765/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/65/868765/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/65/868765/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/65/868765/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/65/868765/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/65/868765/10"}}},"commit":{"parents":[{"commit":"ac3618b140b1b8ba02f2263abf910c17b3f08aff","subject":"Merge \"Fix typo in docs section title\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/ac3618b140b1b8ba02f2263abf910c17b3f08aff"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-12-28 15:09:12.000000000","tz":60},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2023-01-03 08:50:32.000000000","tz":60},"subject":"Fix broken tox.ini with tox4","message":"Fix broken tox.ini with tox4\n\n- scripts were not in allowlist for pep8 environment\n- skipsdist setting breaks the docs environment\n\nChange-Id: I5ac930e0d3928bb8acba4212dfff25f788acd442\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/9352563db15b7879e8988de685a061d32ecf1a69"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/9352563db15b7879e8988de685a061d32ecf1a69"}]},"branch":"refs/heads/master"},"02d40b70851dd72d288cc502d06c28657fc10a3f":{"kind":"REWORK","_number":11,"created":"2023-01-03 13:49:16.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/65/868765/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/65/868765/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/65/868765/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/65/868765/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/65/868765/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/65/868765/11"}}},"commit":{"parents":[{"commit":"ac3618b140b1b8ba02f2263abf910c17b3f08aff","subject":"Merge \"Fix typo in docs section title\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/ac3618b140b1b8ba02f2263abf910c17b3f08aff"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-12-28 15:09:12.000000000","tz":60},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-01-03 13:48:49.000000000","tz":60},"subject":"Fix broken tox.ini with tox4","message":"Fix broken tox.ini with tox4\n\n- scripts were not in allowlist for pep8 environment\n- skipsdist setting breaks the docs environment\n\nChange-Id: I5ac930e0d3928bb8acba4212dfff25f788acd442\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/02d40b70851dd72d288cc502d06c28657fc10a3f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/02d40b70851dd72d288cc502d06c28657fc10a3f"}]},"branch":"refs/heads/master"},"41ee7e7aa1117ae42ea0fd1ea7b3ea8ebba27476":{"kind":"REWORK","_number":12,"created":"2023-01-03 15:56:07.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/65/868765/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/65/868765/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/65/868765/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/65/868765/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/65/868765/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/65/868765/12"}}},"commit":{"parents":[{"commit":"ac3618b140b1b8ba02f2263abf910c17b3f08aff","subject":"Merge \"Fix typo in docs section title\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/ac3618b140b1b8ba02f2263abf910c17b3f08aff"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-12-28 15:09:12.000000000","tz":60},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-01-03 15:14:38.000000000","tz":60},"subject":"Fix broken tox.ini with tox4","message":"Fix broken tox.ini with tox4\n\n- scripts were not in allowlist for pep8 environment\n- skipsdist setting breaks the docs environment\n\nChange-Id: I5ac930e0d3928bb8acba4212dfff25f788acd442\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/41ee7e7aa1117ae42ea0fd1ea7b3ea8ebba27476"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/41ee7e7aa1117ae42ea0fd1ea7b3ea8ebba27476"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[]}
