)]}'
{"id":"openstack%2Foctavia~868994","triplet_id":"openstack%2Foctavia~master~Ibe135ccdb47fdde5be92cbb1488c942cb142c40a","project":"openstack/octavia","branch":"master","attention_set":{"29244":{"account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"last_update":"2023-01-10 13:35:14.000000000","reason":"\u003cGERRIT_ACCOUNT_31664\u003e replied on the change","reason_account":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"}}},"removed_from_attention_set":{"31664":{"account":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"last_update":"2024-06-07 11:39:15.000000000","reason":"\u003cGERRIT_ACCOUNT_31664\u003e replied on the change","reason_account":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"}},"34429":{"account":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"last_update":"2024-06-11 06:25:19.000000000","reason":"\u003cGERRIT_ACCOUNT_34429\u003e replied on the change","reason_account":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}}},"hashtags":[],"change_id":"Ibe135ccdb47fdde5be92cbb1488c942cb142c40a","subject":"Fix qos_policy_id on VRRP ports after failover","status":"NEW","created":"2023-01-02 10:36:12.000000000","updated":"2024-06-11 06:25:19.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":6,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"927467ab64eea26883eb6160cf52fe47aa4dba79","_number":868994,"virtual_id_number":868994,"owner":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"tag":"autogenerated:zuul:check","value":1,"date":"2024-06-07 14:44:49.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"recommended":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"all":[{"value":1,"date":"2024-06-07 11:39:15.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-06-11 06:25:19.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"approved":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-06-11 06:25:19.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-01-02 12:05:38.000000000","updated_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"reviewer":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"state":"REVIEWER"},{"updated":"2023-01-02 12:48:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-01-09 16:02:15.000000000","updated_by":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"reviewer":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"state":"REVIEWER"}],"messages":[{"id":"b8c6330ce2f14789e9cd202a66ec7d76b0c57f37","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-01-02 10:36:12.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f254380440e6ec58321faa7e6d8a4dc7b635c9a0","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2023-01-02 12:05:38.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"7a58ab1c8ad7ce5f94c188d5eba236370a3e9756","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-02 12:48:33.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/3a909eac3481427e8ef5d66237e5d759\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a5166e45089e414aa2d16252d7562346 : FAILURE in 4m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/497a5960044d45febd2026f47b474d44 : FAILURE in 3m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9f963154e736454482c0a50245cfacb1 : FAILURE in 4m 03s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/750fd976f4af489bbc4c89357e0697ab : FAILURE in 3m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/89b043487bdd403f8725a9a54a93c242 : FAILURE in 4m 09s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/103d96e77119492da5fe95be65f1e132 : SUCCESS in 4m 10s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/879fd1da971c4adda2e834fbaca84f2a : FAILURE in 4m 16s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/fa55234bfd7c44e59e580bf08cd16d14 : FAILURE in 4m 54s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/3e54811775b8461bbb6dbc9f6fdcce58 : SUCCESS in 3m 15s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/c5ff0e36131642d9909f592839091b14 : FAILURE in 5m 29s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/fbb2693fb0bf480593a04bb5152a9eb1 : FAILURE in 4m 19s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/4418469ec6c4466ca36cae46309baa45 : FAILURE in 5m 37s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/5148108ca2ab425bad7dd7340a2a99de : SUCCESS in 1h 07m 35s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/d879fb41450e4de28984cd140f7eb077 : SUCCESS in 2h 04m 28s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/f1810ee7c42546f0884b1b272ef511bf : SUCCESS in 33m 25s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/86e7cf57b9de44688434b27a902cb64b : SUCCESS in 1h 07m 33s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/c2814a9f3c0740bbbc3dfe88e1e6eefc : FAILURE in 1h 19m 05s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/91c3856ffd4b48e1bddc5842f1f3b7c0 : SUCCESS in 35m 09s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/e1691927cc61498c86f6f6627ded7599 : SUCCESS in 1h 49m 01s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/95a686a5056a45c584b4b6c5430afa3f : SUCCESS in 2h 00m 13s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/4be228f413bb40d8b4cefce348bd4d58 : SUCCESS in 1h 51m 42s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/df8729f4571d4fd3ac1ef8e3bf2e46b2 : SUCCESS in 2h 10m 59s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/1b3ac0374c124afd9b1f4999b2c3c0a7 : SUCCESS in 1h 35m 12s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/90c842c3a7544cfbb22d90a2063268f6 : SUCCESS in 1h 59m 59s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"1eec7f225ce339bb1f18499ce8d9361c3089ecac","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-01-04 16:17:46.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"bb493acc48e80806168defbca8c1e492cce56ffb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-04 18:27:00.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/703c4ed57e8d493fab1e8d85e52ccb11\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4157f9e95ac648bf8f07df1be95fe7e3 : SUCCESS in 11m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b72ee848fdf74a6a9cf52f643eb390e6 : SUCCESS in 6m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a3fc538cf8cb4128b6f5dcd79011af60 : SUCCESS in 4m 34s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2feb90f0facb45e9896ad04953f5c190 : SUCCESS in 5m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f421dc1aef2549f9bab0d54991065d5c : SUCCESS in 10m 31s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c873b7d40f9248e591ea9e995f415607 : SUCCESS in 4m 55s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/f48771777c0b4b15becc58ec4fbc4f1f : SUCCESS in 5m 38s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/e95c4919d2cc48e1b61a1fb320b335a2 : SUCCESS in 8m 14s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/2b0412e97f1f4545bd47c23266adb863 : SUCCESS in 4m 46s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/047b94de3d024b41b8f0b3b43313e331 : SUCCESS in 6m 24s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/ae8641a7fcc94403858376b973d97d4b : SUCCESS in 5m 38s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/1ee188d4037c43caa09e9861f15fbc46 : SUCCESS in 8m 27s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/fc3e59dc7df24384891da415eecaba91 : SUCCESS in 1h 32m 03s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/c8e6bcb34ba7498aa9879a1c2da386a6 : SUCCESS in 1h 51m 01s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/75378c7856f943b18d8aec0dc08955c4 : SUCCESS in 36m 41s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/acd51bff667644ecbfecff3904f181aa : SUCCESS in 1h 09m 49s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/eb9d65b8b1ab4945a7272f337a358843 : SUCCESS in 1h 01m 01s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/a4761c89709a49cf8ccce0ad021a0207 : SUCCESS in 34m 49s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/baf5a63d1df54605a7130939e5df96b6 : SUCCESS in 1h 40m 45s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/e9747757c43c4622baec4e6bb2abdac4 : SUCCESS in 2h 07m 18s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/ada3dcfc2b3b4f5998a8096f6e9ab709 : SUCCESS in 1h 46m 18s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/f3236966d87a4e129761f5973ccadd4a : SUCCESS in 1h 49m 05s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/ea22491d12f64a2aa5471ab9cc8a67a5 : SUCCESS in 1h 32m 04s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/111a61c560104b1a8c8bf8fa1f90b0b7 : SUCCESS in 2h 06m 17s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"e648dffff3ab3461717c67f0cb319c310bf82cd6","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2023-01-09 16:02:15.000000000","message":"Patch Set 1: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"3f213ce12af448fcb75de40705dec1f6ce83b86f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-01-10 13:33:29.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"7961588720019275bae5ee8eca6ce362619d68a4","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-01-10 13:33:52.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"e8a90ba07b7cd707d15d102c74fac17855006eaa","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2023-01-10 13:35:14.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"071b93c3ee09bdf6c4c383d59aed3fd04c6743ab","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-10 15:57:30.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/5c995dce0e2e410a9b7afad0a0bcb5fb\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/051cee131e2149618210b1e8609b0d28 : SUCCESS in 10m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bfdd77c4eeb643e7969a99cdba78eaff : SUCCESS in 5m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6c85971084d14f92b44c919bb37ed9e8 : SUCCESS in 5m 23s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a775822a70f044b5ad95af9b2416a454 : SUCCESS in 5m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ab12c20c752d46548c494b9c1cfd1ad6 : SUCCESS in 12m 08s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8da93195b180424a84dc986a46b52029 : SUCCESS in 4m 47s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/9a346e071376456aa9a0025561706633 : SUCCESS in 4m 33s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/97b539cb276a40309f33eb4068bab081 : FAILURE in 2m 09s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/76f06afdc59b4050ae8a84ce0f239de4 : SUCCESS in 4m 29s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/5c02faba13204fd7b2f759a934eedbb2 : FAILURE in 2m 10s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/1648fca3e3aa417a940b38731832705f : FAILURE in 2m 10s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/06c2a0ed8c7c409ea685f6d48e7db6d6 : FAILURE in 2m 18s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/3455b8c2b7a4463bb28bd9d3e2ea0b28 : SUCCESS in 1h 13m 41s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/76d3026925c2480f822caea58e761098 : SUCCESS in 1h 45m 03s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/f0595f910432409298047b3a73bc9266 : SUCCESS in 37m 11s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/740d1723421f4483b94bfb07f637226d : SUCCESS in 54m 12s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/9499680d734e4996a3d565c55ec2743d : FAILURE in 41m 38s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/e89fad31d2f848dc819ec7935ac18c71 : SUCCESS in 34m 36s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/2731821ed87b4fc48b3636dbafab2506 : SUCCESS in 1h 45m 16s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/49734c4641d64e1f8b91e5064c9d8fc8 : SUCCESS in 1h 55m 21s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/eecab4d4876f49cd9fcbfe5b5e0472af : SUCCESS in 1h 54m 15s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/050e01cf193047d491f1b6da1ea7b653 : SUCCESS in 2h 00m 09s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/2615f5a9ce834bfa9f4d1963033e0b88 : SUCCESS in 1h 54m 32s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/8428f993f09644b7a98a81c622389bc8 : SUCCESS in 2h 22m 04s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"ff99bb7ad9f46d85c271f5f2c465f2fd994fc417","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2023-01-13 12:05:30.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"5ebd7a95d8d21c279806080a0b571dc02bdc948b","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2023-01-13 12:11:20.000000000","message":"Patch Set 2: Backport-Candidate+2 Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"d893bc56edf65fd086bc90c94c1633a155f7c3a3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-13 14:37:56.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e88d244c0dfb4b39b69ccc77cd17a13c\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9efe3ccb6b0c4666a588281aba57ac8a : SUCCESS in 13m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7fbd4cf544f546d68b4e1917093323b4 : SUCCESS in 4m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7722d7f0503b4d1e9e3494565a18d11d : SUCCESS in 4m 20s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a2d787d423fc42c1a2b5f07051ad070a : SUCCESS in 4m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8599255ec5a44843911161098d8b1c06 : SUCCESS in 11m 06s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f831340bde1d4587b245b97b7dff2a09 : SUCCESS in 4m 32s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/f18976a47eaa45fd812844e4ad47eb6b : SUCCESS in 5m 09s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/e0b0af6d64f84d469211d348ca59b8e8 : SUCCESS in 7m 46s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/9d2b61eeb49e489c999fa70706f3b05d : SUCCESS in 4m 11s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/d43d926d60f84daebaef9b05ddbac8a9 : SUCCESS in 8m 22s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/44ed911634f1489fb78bfcdc1ed98921 : SUCCESS in 8m 48s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/94438b12f9354e3e81e69b88ddfc4090 : SUCCESS in 7m 03s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/7c6de1e5f5ce42e6ad049a6a650f575d : SUCCESS in 1h 52m 35s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/b2886779f77744df89d4f9d4e5735433 : SUCCESS in 1h 46m 48s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/808b68fca02b491fa75bd6a15eb94cb7 : SUCCESS in 32m 26s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/acbebe99c71b414b8ef89bff2fd1474d : SUCCESS in 1h 10m 13s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/dd7a154fe1a84903997b28569cb4d5b2 : SUCCESS in 1h 13m 46s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/0aade1db5ff64f70822b9ed481a3f053 : SUCCESS in 31m 52s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/648c5c59bcd546eca7a0e0a04cf6b4eb : SUCCESS in 1h 42m 09s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/ff219c755734494f80e54da5728cc1d9 : FAILURE in 2h 05m 13s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/2f84856529044123bb2d82cec16b2c2a : SUCCESS in 1h 47m 19s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/06624bd4006f40da8b34d212fe57cfc5 : SUCCESS in 1h 51m 54s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/961104c7284d45409b1664222d98927e : SUCCESS in 1h 55m 35s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/5389dfb978604a4e8e882df7407cea92 : SUCCESS in 2h 27m 22s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"1b77e26a82c73987d2754784ea67dc2ed905e2ce","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2024-06-06 10:17:54.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Backport-Candidate+2 (copy condition: \"changekind:NO_CODE_CHANGE OR changekind:TRIVIAL_REBASE\")\n* Code-Review+1, Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"b5afad9a23c3c4d2616cfc1bff47cb9d92a481d7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-06 13:24:49.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/a9b996a27c0047a781a3347eb6e1888f\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9a8018dfb19d424fa03e420c3fe75f50 : SUCCESS in 11m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7820f339d9ca4d80b1c1bca086f8fe3b : SUCCESS in 4m 40s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4ba81f12f63d4b5b8a0bc25334d18f5a : SUCCESS in 4m 51s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/320c41a920da4d86880dc75d8a52c8cb : SUCCESS in 4m 40s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/7457befb3e1a4f69871c06e85c9a0996 : SUCCESS in 9m 35s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fc1fb3b43c18409595cf39dd10c6865d : SUCCESS in 11m 55s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ac46871fb7374b92aad1520a92b7991d : SUCCESS in 5m 44s\n- octavia-tox-py311-tips https://zuul.opendev.org/t/openstack/build/779dfb34fbb34c7ba36db43772cf723c : SUCCESS in 5m 41s\n- octavia-tox-functional-py311-tips https://zuul.opendev.org/t/openstack/build/6ca52a79f7fb443cbb51860c3de08fec : SUCCESS in 6m 21s\n- octavia-tox-functional-py311-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/41af875ead894e5e864bf4707c882a41 : SUCCESS in 6m 46s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/4e5844b5fe3e409bb2a1b5da267a9ff2 : SUCCESS in 4m 41s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/2274420261294662afcdff5a272531ed : SUCCESS in 7m 23s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/ce2fc651c5d94719ad0f944a0f614d40 : SUCCESS in 6m 59s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/4d1f3fe491ce4909894628de78977dcf : SUCCESS in 6m 31s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/557b2060d9e04335be05c702492938c2 : SUCCESS in 1h 18m 30s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/2fb6ee39f9f84c5dacebd420496e7dcd : TIMED_OUT in 2h 34m 06s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/74c5e8a181fc435489a2a640695bfeea : SUCCESS in 54m 36s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/b4c004f3045a464ba9b8971cc113ac18 : SUCCESS in 1h 53m 34s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/333f2ffbc45b4791828fa7961f763afc : SUCCESS in 1h 08m 44s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/e0ea2ee0d6ba4049a9d06e4311500606 : SUCCESS in 37m 56s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/fda57fce6c41496c8b50da8de99af71a : SUCCESS in 55m 28s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/3ecacca669c34a8cacf9895fcc434715 : SUCCESS in 1h 14m 32s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/afa628ea05b04830b0f16640e5d9d9bf : SUCCESS in 35m 32s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/dc0ad1e0a6ae47099c36498dbad58e50 : SUCCESS in 2h 32m 51s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/5b261e4a45734b64a12eea550ec90823 : TIMED_OUT in 3h 03m 38s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/730c63b6d72f4952b53b299970c77cff : FAILURE in 21m 19s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/83404b0fb2b14fc29ad3c728f892aa4f : TIMED_OUT in 2h 34m 56s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"f7310b863c2c3f07f1048d830fca324dd31fd366","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2024-06-07 11:39:15.000000000","message":"Patch Set 3: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"60dc647576d42fb9b5b665eddd0fae4b9d606ddb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-07 14:44:49.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dd78258da5154959bb59257a671f6859\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bcec1633d760477aadb5b08f6eb6be5d : SUCCESS in 13m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/122aa7e8b4eb4a758355e7de5b006a8c : SUCCESS in 4m 31s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/df9f24c4fd14480387b5c66cbd22d6de : SUCCESS in 5m 13s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/eda0cdfa3c00456d9e8a3637c9d6844b : SUCCESS in 4m 34s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/3c814ec9452440f2a8412194c076d421 : SUCCESS in 8m 19s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/88cc8803fe8346deb794a1c10a948379 : SUCCESS in 10m 11s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ada1cb4758a94204a761e5103aaa72d3 : SUCCESS in 5m 17s\n- octavia-tox-py311-tips https://zuul.opendev.org/t/openstack/build/d20e525389914c4cb689b0b6039f8939 : SUCCESS in 5m 02s\n- octavia-tox-functional-py311-tips https://zuul.opendev.org/t/openstack/build/1cb0487f4454472092f41ce0c2232205 : SUCCESS in 5m 03s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/9a7675a8725e43df87dc6938f619f7ea : SUCCESS in 4m 27s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/69209db508984b8c8edfc832358a3f70 : SUCCESS in 6m 57s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/3dc16e37971e4bd48252aa81862eec8f : SUCCESS in 6m 33s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/b6d860c1d5744f79bbdda870e9930bb1 : SUCCESS in 4m 53s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/9f06ebf431fa4eeb933ae861bff0b126 : SUCCESS in 1h 29m 53s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/bce911a2d43b4c7ea894d768338723ee : SUCCESS in 1h 49m 36s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/e4228577b7c64742a9db41d2a446d467 : SUCCESS in 1h 10m 52s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/865facfae002411c905ad2856dec4472 : SUCCESS in 1h 58m 02s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/1558ed286bac45e8ba2dcab11b9839fb : SUCCESS in 1h 04m 57s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/97eedd238e9d48c1b6f97485e125bc4e : SUCCESS in 43m 40s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/f536edf1736f457ebc30d12819b13df8 : SUCCESS in 46m 56s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/4a62dc94523f45948e8431d803ade111 : SUCCESS in 1h 02m 21s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/0a06c44018be479496051fee271b0432 : SUCCESS in 35m 04s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/668c9d7ecdd945d28bedd66d09161052 : TIMED_OUT in 2h 32m 54s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/d5c59bf043124bb9923ef7831e1130c1 : TIMED_OUT in 3h 02m 57s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/4e254cb2046f45688a4ebfa749b67aef : FAILURE in 30m 02s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/0fba924eb99a403183095d4744cdb585 : TIMED_OUT in 2h 33m 45s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"927467ab64eea26883eb6160cf52fe47aa4dba79","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2024-06-11 06:25:19.000000000","message":"Patch Set 3: Backport-Candidate+2 Code-Review+2","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"132fc6f94f53e4c881d77f080d4f232314f05eb7","revisions":{"02e1b220789e3a7e0153f6982d96ee0cc61dadb3":{"kind":"REWORK","_number":1,"created":"2023-01-02 10:36:12.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/94/868994/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/94/868994/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/94/868994/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/94/868994/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/94/868994/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/94/868994/1"}}},"commit":{"parents":[{"commit":"ac3618b140b1b8ba02f2263abf910c17b3f08aff","subject":"Merge \"Fix typo in docs section title\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/ac3618b140b1b8ba02f2263abf910c17b3f08aff"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-01-02 10:16:01.000000000","tz":-300},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-01-02 10:17:51.000000000","tz":-300},"subject":"Fix qos_policy_id on VRRP ports after failover","message":"Fix qos_policy_id on VRRP ports after failover\n\nThe qos_policy_id attribute was not applied to the new VRRP ports\ncreated after the failover of a load balancer.\n\nThis commit adds 2 things:\n- The qos_policy_id is applied to the VIP port: this adds consistency\n  with the VRRP ports, it doesn\u0027t do anything on this port because it is\n  not used but the attributes of this port are used to create the VRRP\n  port in the failover flow.\n- The allocate_vip function of the allowed_address_pairs driver sets the\n  qos_policy_id when creating a port if it is set for the VIP object.\n\nStory 2010498\nTask 47078\n\nChange-Id: Ibe135ccdb47fdde5be92cbb1488c942cb142c40a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/02e1b220789e3a7e0153f6982d96ee0cc61dadb3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/02e1b220789e3a7e0153f6982d96ee0cc61dadb3"}]},"branch":"refs/heads/master"},"e9d3f9cd7f2e1564c6a28f3ade15c4f794e179ae":{"kind":"REWORK","_number":2,"created":"2023-01-10 13:33:29.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/94/868994/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/94/868994/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/94/868994/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/94/868994/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/94/868994/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/94/868994/2"}}},"commit":{"parents":[{"commit":"ac3618b140b1b8ba02f2263abf910c17b3f08aff","subject":"Merge \"Fix typo in docs section title\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/ac3618b140b1b8ba02f2263abf910c17b3f08aff"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-01-02 10:16:01.000000000","tz":-300},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-01-10 13:32:21.000000000","tz":60},"subject":"Fix qos_policy_id on VRRP ports after failover","message":"Fix qos_policy_id on VRRP ports after failover\n\nThe qos_policy_id attribute was not applied to the new VRRP ports\ncreated after the failover of a load balancer.\n\nThis commit adds 2 things:\n- The qos_policy_id is applied to the VIP port: this adds consistency\n  with the VRRP ports, it doesn\u0027t do anything on this port because it is\n  not used but the attributes of this port are used to create the VRRP\n  port in the failover flow.\n- The allocate_vip function of the allowed_address_pairs driver sets the\n  qos_policy_id when creating a port if it is set for the VIP object.\n\nStory 2010498\nTask 47078\n\nChange-Id: Ibe135ccdb47fdde5be92cbb1488c942cb142c40a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e9d3f9cd7f2e1564c6a28f3ade15c4f794e179ae"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e9d3f9cd7f2e1564c6a28f3ade15c4f794e179ae"}]},"branch":"refs/heads/master"},"132fc6f94f53e4c881d77f080d4f232314f05eb7":{"kind":"REWORK","_number":3,"created":"2024-06-06 10:17:54.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/94/868994/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/94/868994/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/94/868994/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/94/868994/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/94/868994/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/94/868994/3"}}},"commit":{"parents":[{"commit":"541682314693ab543b148a9ed7703192f1673e00","subject":"Merge \"String conversion to f-strings\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/541682314693ab543b148a9ed7703192f1673e00"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-01-02 10:16:01.000000000","tz":-300},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2024-06-06 10:15:11.000000000","tz":120},"subject":"Fix qos_policy_id on VRRP ports after failover","message":"Fix qos_policy_id on VRRP ports after failover\n\nThe qos_policy_id attribute was not applied to the new VRRP ports\ncreated after the failover of a load balancer.\n\nThis commit adds 2 things:\n- The qos_policy_id is applied to the VIP port: this adds consistency\n  with the VRRP ports, it doesn\u0027t do anything on this port because it is\n  not used but the attributes of this port are used to create the VRRP\n  port in the failover flow.\n- The allocate_vip function of the allowed_address_pairs driver sets the\n  qos_policy_id when creating a port if it is set for the VIP object.\n\nStory 2010498\nTask 47078\n\nChange-Id: Ibe135ccdb47fdde5be92cbb1488c942cb142c40a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/132fc6f94f53e4c881d77f080d4f232314f05eb7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/132fc6f94f53e4c881d77f080d4f232314f05eb7"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}},{"label":"Workflow","status":"MAY"},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
