)]}'
{"id":"openstack%2Foctavia~869153","triplet_id":"openstack%2Foctavia~stable%2Fzed~I5ac930e0d3928bb8acba4212dfff25f788acd442","project":"openstack/octavia","branch":"stable/zed","attention_set":{},"removed_from_attention_set":{"34429":{"account":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"last_update":"2023-01-23 13:24:17.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I5ac930e0d3928bb8acba4212dfff25f788acd442","subject":"Fix broken tox.ini with tox4","status":"ABANDONED","created":"2023-01-04 09:18:33.000000000","updated":"2023-01-23 13:24:17.000000000","total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"eec1a19b2d6daf14e0412d4fe60e524ec51a4f9e","_number":869153,"virtual_id_number":869153,"owner":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"tag":"autogenerated:zuul:check","value":1,"date":"2023-01-04 11:58:52.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"rejected":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"all":[{"value":-1,"date":"2023-01-06 14:16:41.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-01-04 11:58:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-01-06 14:16:41.000000000","updated_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"REVIEWER"}],"messages":[{"id":"db87829e0dcf317ff6755ab4781b5227ff49eb80","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2023-01-04 09:18:33.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7d8e57215975bae219786164339accd923890c38","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-04 11:58:52.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e19c1d26a85649038ca9c741209f78d6\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/edd77a4ff2d04f94b55e68963b514bf0 : SUCCESS in 11m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e7edd58fd054440795394c42c40768aa : SUCCESS in 6m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4ca75ca7873d4affaa8ab9f4c8997077 : SUCCESS in 4m 59s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/dc98489014a24e9688054c1729915670 : SUCCESS in 5m 06s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/821085765c174a71aa3bc248b8f223b8 : SUCCESS in 4m 50s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0f05ad112a734088bb0e0993df70087e : SUCCESS in 10m 48s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/62ee284be19346caa157aa90f0ea825a : SUCCESS in 4m 59s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/66ebf07db3974cd5ad645ef767833c16 : SUCCESS in 4m 50s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/4cde0e8305f54d9bab0479193d1fda8e : SUCCESS in 7m 12s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/8d34ccc8353b44f0bae930d08978385d : SUCCESS in 4m 19s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/5ff37984520c469bbb216b7e5b70ff33 : SUCCESS in 8m 15s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/3c58ac6bec584ba1958abab24c18b8c0 : SUCCESS in 6m 34s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/fd53fdc2a2374690b66d3881a6f749f5 : SUCCESS in 8m 34s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/08ece4ae803244ad91c5240d3e819f6d : SUCCESS in 1h 32m 01s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/ae39222279e4404aae4df6c24936b40f : SUCCESS in 2h 19m 29s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/74226e8d315542d49a5f1f145b0586c0 : SUCCESS in 38m 15s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/aab16d330f56460d92b6c3851884ee60 : SUCCESS in 1h 24m 19s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/4ffab92ba52a4e36a17943e22d627cb8 : SUCCESS in 33m 56s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/f404900f10c549d6abf4940e60bdff85 : SUCCESS in 2h 00m 20s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/131fc3db446d4fd7bb35bf48ca0cac3e : SUCCESS in 2h 39m 28s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/26c666c7749e4a4590eb4983879f049a : TIMED_OUT in 2h 33m 52s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/10fbfef553144bd3a8ef860b0fb8d8ac : SUCCESS in 2h 30m 05s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/09b7200c3543435f9446bf0054587e9b : SUCCESS in 2h 09m 14s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/d11c2b411d0544e5afb2ead5faa00181 : SUCCESS in 2h 25m 14s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d6668009012dfe94330ee0431fc5dc7c2adfcef3","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-01-06 14:16:41.000000000","message":"Patch Set 1: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"eec1a19b2d6daf14e0412d4fe60e524ec51a4f9e","tag":"autogenerated:gerrit:abandon","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2023-01-23 13:24:17.000000000","message":"Abandoned\n\nFix not necessary because tox version gets pinned to \u003c4.0 on stable branches.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"946b8b7cdb79d96266435c2f32a102c08474e5bc","revisions":{"946b8b7cdb79d96266435c2f32a102c08474e5bc":{"kind":"REWORK","_number":1,"created":"2023-01-04 09:18:33.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/53/869153/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/53/869153/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/53/869153/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/53/869153/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/53/869153/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/53/869153/1"}}},"commit":{"parents":[{"commit":"572c0dace98320a2ff925bf91e5ff07a4cfd9f4f","subject":"Add a newline when writing the server state file","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/572c0dace98320a2ff925bf91e5ff07a4cfd9f4f"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-12-28 15:09:12.000000000","tz":60},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2023-01-04 09:18:15.000000000","tz":60},"subject":"Fix broken tox.ini with tox4","message":"Fix broken tox.ini with tox4\n\n- scripts were not in allowlist for pep8 environment\n- skipsdist setting breaks the docs environment\n\nChange-Id: I5ac930e0d3928bb8acba4212dfff25f788acd442\n(cherry picked from commit 41ee7e7aa1117ae42ea0fd1ea7b3ea8ebba27476)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/946b8b7cdb79d96266435c2f32a102c08474e5bc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/946b8b7cdb79d96266435c2f32a102c08474e5bc"}]},"branch":"refs/heads/stable/zed"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
