)]}'
{"id":"openstack%2Foctavia~869155","triplet_id":"openstack%2Foctavia~stable%2Fyoga~I5ac930e0d3928bb8acba4212dfff25f788acd442","project":"openstack/octavia","branch":"stable/yoga","attention_set":{},"removed_from_attention_set":{"34429":{"account":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"last_update":"2023-01-23 13:24:51.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I5ac930e0d3928bb8acba4212dfff25f788acd442","subject":"Fix broken tox.ini with tox4","status":"ABANDONED","created":"2023-01-04 09:20:29.000000000","updated":"2023-01-23 13:24:51.000000000","total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"4666a75fd7037307fb415fd5a5f5ad016b677fca","_number":869155,"virtual_id_number":869155,"owner":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"tag":"autogenerated:zuul:check","value":1,"date":"2023-01-04 12:00:59.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"rejected":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"all":[{"value":-1,"date":"2023-01-06 14:16:48.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-01-04 12:00:59.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-01-06 14:16:48.000000000","updated_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"REVIEWER"}],"messages":[{"id":"ab09c82dd58828589851ee98f8dc10a9761c29a9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2023-01-04 09:20:29.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"69f7b5bb6ac7aa90f3f92cfb1528407d0e283871","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2023-01-04 09:25:36.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"0168c5c8168b5e1af30669050af38e0d40f8204f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2023-01-04 09:27:03.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"177f5c68977a0127f0bef1b99ca80f29cc4993e5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-04 12:00:59.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/47dcd8ad0d434799985f3ffaf749ad2d\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ff5c50e83296417b80c25a0c54cdf613 : SUCCESS in 14m 45s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/051140050ab242629b4d5809ee20f808 : SUCCESS in 8m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5ed79cbd92234e408e1c5d9a1cd27832 : SUCCESS in 7m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e3fcd8d568d54c3580f464e40228c0d6 : SUCCESS in 5m 15s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f9e67062af00428487a341fab3d5027d : SUCCESS in 4m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/43952c63282c4174af103e0e8ad831ac : SUCCESS in 13m 03s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ee42ecc7f8814205b70a517c58c57225 : SUCCESS in 5m 32s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/dc443badf0874a25bc4368cabe540bfa : SUCCESS in 6m 01s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/c9a69cfda6bc40cb8f05a5ac837d8015 : SUCCESS in 4m 59s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/72987e207e8149fdbcaefaaf904882dc : SUCCESS in 5m 13s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/00836629bef540ac97a5b6f457e5def8 : SUCCESS in 7m 19s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/dc0b62b977dd4b4bba212bb28852eb9e : SUCCESS in 1h 25m 00s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/005d63e87b154e7fae85d52c0cda1d46 : SUCCESS in 1h 41m 36s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/cc5fdf9da13a43d08b05d0fae5182010 : SUCCESS in 32m 19s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/02400d295690434b8d7b9b9c8e89eaf7 : SUCCESS in 1h 18m 12s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/99497c946f4f45d58c4ce21349e4264c : SUCCESS in 29m 06s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/22a241bac0ac45479c4a35bec8644c69 : SUCCESS in 2h 31m 49s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/263c93cfb2d44d439cedd6a520e866ab : SUCCESS in 2h 25m 35s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/5d5a68c5dc374d62b983341628aac775 : SUCCESS in 2h 32m 58s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/1b3d6e81121b494d94684bdf68b62c96 : SUCCESS in 2h 29m 21s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/22d9bdb16b9c4192ae45f39dcd38d1d7 : SUCCESS in 2h 02m 26s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8-stream https://zuul.opendev.org/t/openstack/build/4789a72fcd0b40c88a7d9e3483498458 : FAILURE in 38m 40s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"4bef7f6ccc51b9a1e8daca94dd352e053cacdfe2","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-01-06 14:16:48.000000000","message":"Patch Set 3: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"4666a75fd7037307fb415fd5a5f5ad016b677fca","tag":"autogenerated:gerrit:abandon","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2023-01-23 13:24:51.000000000","message":"Abandoned\n\nFix not necessary because tox version gets pinned to \u003c4.0 on stable branches.","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"a57a6a1232dc432880fc1a72895ef2e4aba8344d","revisions":{"1a16cab6c014772ad56d5b19f1031a2fec601715":{"kind":"REWORK","_number":1,"created":"2023-01-04 09:20:29.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/55/869155/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/55/869155/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/55/869155/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/55/869155/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/55/869155/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/55/869155/1"}}},"commit":{"parents":[{"commit":"8fa8d8f071785d82fefe3ff88eea2935d4144e3f","subject":"Merge \"Fix plugging member subnets on existing networks\" into stable/yoga","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/8fa8d8f071785d82fefe3ff88eea2935d4144e3f"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-12-28 15:09:12.000000000","tz":60},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2023-01-04 09:20:03.000000000","tz":60},"subject":"Fix broken tox.ini with tox4","message":"Fix broken tox.ini with tox4\n\n- scripts were not in allowlist for pep8 environment\n- skipsdist setting breaks the docs environment\n\nConflicts:\n\ttox.ini\n\nChange-Id: I5ac930e0d3928bb8acba4212dfff25f788acd442\n(cherry picked from commit 41ee7e7aa1117ae42ea0fd1ea7b3ea8ebba27476)\n(cherry picked from commit 946b8b7cdb79d96266435c2f32a102c08474e5bc)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/1a16cab6c014772ad56d5b19f1031a2fec601715"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/1a16cab6c014772ad56d5b19f1031a2fec601715"}]},"branch":"refs/heads/stable/yoga"},"c7706385d6912c2da66f6ef310ce9800b52a6a3e":{"kind":"REWORK","_number":2,"created":"2023-01-04 09:25:36.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/55/869155/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/55/869155/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/55/869155/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/55/869155/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/55/869155/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/55/869155/2"}}},"commit":{"parents":[{"commit":"8fa8d8f071785d82fefe3ff88eea2935d4144e3f","subject":"Merge \"Fix plugging member subnets on existing networks\" into stable/yoga","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/8fa8d8f071785d82fefe3ff88eea2935d4144e3f"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-12-28 15:09:12.000000000","tz":60},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2023-01-04 09:25:25.000000000","tz":60},"subject":"Fix broken tox.ini with tox4","message":"Fix broken tox.ini with tox4\n\n- scripts were not in allowlist for pep8 environment\n- skipsdist setting breaks the docs environment\n\nConflicts:\n\ttox.ini\n\nChange-Id: I5ac930e0d3928bb8acba4212dfff25f788acd442\n(cherry picked from commit 41ee7e7aa1117ae42ea0fd1ea7b3ea8ebba27476)\n(cherry picked from commit 946b8b7cdb79d96266435c2f32a102c08474e5bc)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c7706385d6912c2da66f6ef310ce9800b52a6a3e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c7706385d6912c2da66f6ef310ce9800b52a6a3e"}]},"branch":"refs/heads/stable/yoga"},"a57a6a1232dc432880fc1a72895ef2e4aba8344d":{"kind":"REWORK","_number":3,"created":"2023-01-04 09:27:03.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/55/869155/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/55/869155/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/55/869155/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/55/869155/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/55/869155/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/55/869155/3"}}},"commit":{"parents":[{"commit":"8fa8d8f071785d82fefe3ff88eea2935d4144e3f","subject":"Merge \"Fix plugging member subnets on existing networks\" into stable/yoga","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/8fa8d8f071785d82fefe3ff88eea2935d4144e3f"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-12-28 15:09:12.000000000","tz":60},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2023-01-04 09:26:47.000000000","tz":60},"subject":"Fix broken tox.ini with tox4","message":"Fix broken tox.ini with tox4\n\n- scripts were not in allowlist for pep8 environment\n- skipsdist setting breaks the docs environment\n\nConflicts:\n\ttox.ini\n\nChange-Id: I5ac930e0d3928bb8acba4212dfff25f788acd442\n(cherry picked from commit 41ee7e7aa1117ae42ea0fd1ea7b3ea8ebba27476)\n(cherry picked from commit 946b8b7cdb79d96266435c2f32a102c08474e5bc)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a57a6a1232dc432880fc1a72895ef2e4aba8344d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a57a6a1232dc432880fc1a72895ef2e4aba8344d"}]},"branch":"refs/heads/stable/yoga"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
