)]}'
{"id":"openstack%2Foctavia~869158","triplet_id":"openstack%2Foctavia~stable%2Fwallaby~I5ac930e0d3928bb8acba4212dfff25f788acd442","project":"openstack/octavia","branch":"stable/wallaby","attention_set":{},"removed_from_attention_set":{"34429":{"account":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"last_update":"2023-01-23 13:24:59.000000000","reason":"\u003cGERRIT_ACCOUNT_34429\u003e replied on the change","reason_account":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}}},"hashtags":[],"change_id":"I5ac930e0d3928bb8acba4212dfff25f788acd442","subject":"Fix broken tox.ini with tox4","status":"ABANDONED","created":"2023-01-04 09:28:30.000000000","updated":"2023-04-13 15:48:52.000000000","total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"048b745af62c4c587812763af0d47d15fb3f28d7","_number":869158,"virtual_id_number":869158,"owner":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"tag":"autogenerated:zuul:check","value":1,"date":"2023-01-04 13:19:37.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"rejected":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"all":[{"value":-1,"date":"2023-01-06 14:16:30.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-01-04 13:19:37.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-01-06 14:16:30.000000000","updated_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"REVIEWER"}],"messages":[{"id":"4b55545252f67edc7e0e21260a70ce4d32a80aa1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2023-01-04 09:28:30.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"bc6000c9281b169c8f319a282e0b98c3e7e98df4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2023-01-04 10:48:04.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"640aa44ed8daddd90e35b0f3f83d30ad6ca57ba9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-04 13:19:37.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bc8fddb76fc44d0ea1de3ce8b3a471e2\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/abde65ada36d4c0187726549d143467d : SUCCESS in 8m 42s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/90cf34a68082401eb6c3d5a409461e85 : SUCCESS in 6m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/08af563a4edf45378e4023a40201ce76 : SUCCESS in 4m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9b181a365fdd4ef49824bfd7f5ef17b4 : SUCCESS in 4m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/39f0d800cff34bafb03be29ca90aa531 : SUCCESS in 4m 59s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9795dd6ef7d643aaac997b3e95949fe0 : SUCCESS in 4m 13s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/26623f75ae7c47be9fb31af784a5e306 : SUCCESS in 11m 56s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/120fdff05e2b4554802c5c248606fc83 : SUCCESS in 4m 36s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/b96f5c88059d4c4b8a64dd40cc760ef0 : SUCCESS in 4m 57s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/a0644c6e00a64caba48b3e4d1f5e3d27 : SUCCESS in 7m 36s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/7798914a8a60438f9810cf45a5d7c2fe : SUCCESS in 4m 15s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/2e9faa0a966649bebb30640d92659a29 : SUCCESS in 7m 21s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/4f2949d228ae41ba8190c728eb4d0fbf : SUCCESS in 1h 03m 07s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/39ff8fa843614b228d515c1081651cd5 : SUCCESS in 1h 56m 34s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/35b9fe869dd94a3ebc4ba91a6ecae77b : SUCCESS in 45m 54s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/3ea1f10556d243fbb31ad56d752ecd7d : SUCCESS in 35m 06s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/48155428bb7f4c2d9e2421c212498455 : SUCCESS in 42m 39s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/ff00c88cdd13451083716452acee437b : FAILURE in 11m 38s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/2aaf7bcfd8af40d3ad3fd8f2c78a2d06 : SUCCESS in 1h 19m 08s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/62c0c24b1ca74a3d8754e57b2e5f108e : FAILURE in 34m 47s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/2531be13d9dd44f1997bfbc7d9aefca5 : SUCCESS in 1h 52m 25s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/24a285d021e84460a05a489266754055 : SUCCESS in 1h 48m 56s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/60a629766c6f4e3b971a8775cd0dea87 : SUCCESS in 2h 29m 16s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/9e90afde09ad486e9bad085886effc47 : NODE_FAILURE Node request 300-0020091093 failed in 0s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"2a1fbd33619c131876295340c1756354fa425204","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-01-06 14:16:30.000000000","message":"Patch Set 2: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"8a7857448c81814ee779ce7cabc76d9d66d8af85","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2023-01-23 13:24:59.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"048b745af62c4c587812763af0d47d15fb3f28d7","tag":"autogenerated:gerrit:abandon","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2023-04-13 15:48:52.000000000","message":"Abandoned\n\nnot needed anymore","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"72400f9c68683d8e66efce18cb2a9780c03c9c58","revisions":{"ef27989cd6117dfa22b0a33b91b9ff5b361d42a3":{"kind":"REWORK","_number":1,"created":"2023-01-04 09:28:30.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/58/869158/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/58/869158/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/58/869158/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/58/869158/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/58/869158/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/58/869158/1"}}},"commit":{"parents":[{"commit":"b0379d6a8411ae65413579bdd7927bfa07e51b5b","subject":"Merge \"Fix plugging member subnets on existing networks\" into stable/wallaby","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b0379d6a8411ae65413579bdd7927bfa07e51b5b"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-12-28 15:09:12.000000000","tz":60},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2023-01-04 09:28:17.000000000","tz":60},"subject":"Fix broken tox.ini with tox4","message":"Fix broken tox.ini with tox4\n\n- scripts were not in allowlist for pep8 environment\n- skipsdist setting breaks the docs environment\n\nConflicts:\n\ttox.ini\n\nChange-Id: I5ac930e0d3928bb8acba4212dfff25f788acd442\n(cherry picked from commit 41ee7e7aa1117ae42ea0fd1ea7b3ea8ebba27476)\n(cherry picked from commit 946b8b7cdb79d96266435c2f32a102c08474e5bc)\n(cherry picked from commit a57a6a1232dc432880fc1a72895ef2e4aba8344d)\n(cherry picked from commit 3833da50a715eba2b90f280325e868fb47e2b467)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/ef27989cd6117dfa22b0a33b91b9ff5b361d42a3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/ef27989cd6117dfa22b0a33b91b9ff5b361d42a3"}]},"branch":"refs/heads/stable/wallaby"},"72400f9c68683d8e66efce18cb2a9780c03c9c58":{"kind":"REWORK","_number":2,"created":"2023-01-04 10:48:04.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/58/869158/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/58/869158/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/58/869158/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/58/869158/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/58/869158/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/58/869158/2"}}},"commit":{"parents":[{"commit":"b0379d6a8411ae65413579bdd7927bfa07e51b5b","subject":"Merge \"Fix plugging member subnets on existing networks\" into stable/wallaby","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b0379d6a8411ae65413579bdd7927bfa07e51b5b"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-12-28 15:09:12.000000000","tz":60},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2023-01-04 10:47:55.000000000","tz":60},"subject":"Fix broken tox.ini with tox4","message":"Fix broken tox.ini with tox4\n\n- scripts were not in allowlist for pep8 environment\n- skipsdist setting breaks the docs environment\n\nConflicts:\n\ttox.ini\n\nChange-Id: I5ac930e0d3928bb8acba4212dfff25f788acd442\n(cherry picked from commit 41ee7e7aa1117ae42ea0fd1ea7b3ea8ebba27476)\n(cherry picked from commit 946b8b7cdb79d96266435c2f32a102c08474e5bc)\n(cherry picked from commit a57a6a1232dc432880fc1a72895ef2e4aba8344d)\n(cherry picked from commit 3833da50a715eba2b90f280325e868fb47e2b467)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/72400f9c68683d8e66efce18cb2a9780c03c9c58"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/72400f9c68683d8e66efce18cb2a9780c03c9c58"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":["label:Workflow\u003dMIN"],"failing_atoms":["label:Workflow\u003dMAX"],"atom_explanations":{}}}]}
