)]}'
{"id":"openstack%2Foctavia~869159","triplet_id":"openstack%2Foctavia~stable%2Fvictoria~I5ac930e0d3928bb8acba4212dfff25f788acd442","project":"openstack/octavia","branch":"stable/victoria","attention_set":{},"removed_from_attention_set":{"34429":{"account":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"last_update":"2023-01-23 13:24:00.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I5ac930e0d3928bb8acba4212dfff25f788acd442","subject":"Fix broken tox.ini with tox4","status":"ABANDONED","created":"2023-01-04 09:28:44.000000000","updated":"2023-01-23 13:24:00.000000000","total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"b28e0f941cac17416dd0b0be199518dfcb04d8bf","_number":869159,"virtual_id_number":869159,"owner":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"tag":"autogenerated:zuul:check","value":1,"date":"2023-01-04 12:20:04.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"rejected":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"all":[{"value":-1,"date":"2023-01-06 14:17:05.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-01-04 12:20:04.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-01-06 14:17:05.000000000","updated_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"REVIEWER"}],"messages":[{"id":"eeb2424c5b7ee37ff213806aa007233094c75c1a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2023-01-04 09:28:44.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"82f1b82efa6f9ca308391bbe0abeffe1c3800719","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2023-01-04 10:40:28.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"d4caad231cbaf02b9287839a463dd89f6ca643bb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-04 12:20:04.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e49f3708ca5c407b895edee1600ec7a0\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2473373f8c6f4a2bbf0b9f304078eba8 : SUCCESS in 9m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/862bb1501fb4465faa1878716d9dab7c : SUCCESS in 5m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d56d0cfeacc54ca788c891d62f9eba80 : SUCCESS in 4m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a95d78a8d2ef4334935c697108eedd2b : SUCCESS in 5m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4fb7cb08abfc4e5582b3a5e83aa64264 : SUCCESS in 11m 37s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/63a468ba82684b71a95576506a04e7b8 : SUCCESS in 4m 22s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/bbe41046fe75478ba78d1af40d3c2f57 : SUCCESS in 6m 01s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/ff8f8229ce594c4fa379bd329ad4f414 : SUCCESS in 7m 47s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/0960c6b39f81444eb57053f9e9c58a63 : SUCCESS in 4m 21s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/75f6d75b0c784e4d9aa82fd974188bb1 : SUCCESS in 9m 11s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/14fa18e2a2cf48278b2062452fb353cd : SUCCESS in 1h 12m 22s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/f7c8c7b9e3614c96871ba1985f61696f : SUCCESS in 1h 24m 34s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/5b4d81cac1a04b27b220d25a7fb7c780 : SUCCESS in 32m 12s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/7854de9dbbe84180bd3fe5a8020dc4d9 : SUCCESS in 28m 49s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/3779899266274a969cda85f6e8036e20 : SUCCESS in 29m 43s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/0b1ea1f94b504809bea8efa65e6038bb : SUCCESS in 1h 32m 25s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/f46f5272ead9472d8e15881f240a95fd : FAILURE in 28m 16s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/52b873bc8d6341f485456f9dc706d93a : SUCCESS in 1h 38m 04s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/8c7fc64c84eb45b98fb2eee280c3a60f : FAILURE in 1h 30m 05s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/9d9c295718d141518714804ac45d7f3a : FAILURE in 1h 18m 46s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/246547b58ce64f61884b3e7ef8597822 : NODE_FAILURE Node request 300-0020091023 failed in 0s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"2c8bd1e6c3010fb0609947a23345dc2d583c787c","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-01-06 14:17:05.000000000","message":"Patch Set 2: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"b28e0f941cac17416dd0b0be199518dfcb04d8bf","tag":"autogenerated:gerrit:abandon","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2023-01-23 13:24:00.000000000","message":"Abandoned\n\nFix not necessary because tox version gets pinned to \u003c4.0 on stable branches.","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"edac51f2097a11710ed94f902e7d9179e1348e99","revisions":{"eff9a18e52036c8d995a22be5cf25edbb667b8fd":{"kind":"REWORK","_number":1,"created":"2023-01-04 09:28:44.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/59/869159/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/59/869159/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/59/869159/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/59/869159/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/59/869159/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/59/869159/1"}}},"commit":{"parents":[{"commit":"1f617ff40aaf20b3247be6f8f90ca9ef684766b9","subject":"Merge \"Fix bug when rolling back prov and op status for some API calls\" into stable/victoria","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/1f617ff40aaf20b3247be6f8f90ca9ef684766b9"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-12-28 15:09:12.000000000","tz":60},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2023-01-04 09:28:31.000000000","tz":60},"subject":"Fix broken tox.ini with tox4","message":"Fix broken tox.ini with tox4\n\n- scripts were not in allowlist for pep8 environment\n- skipsdist setting breaks the docs environment\n\nConflicts:\n\ttox.ini\n\nChange-Id: I5ac930e0d3928bb8acba4212dfff25f788acd442\n(cherry picked from commit 41ee7e7aa1117ae42ea0fd1ea7b3ea8ebba27476)\n(cherry picked from commit 946b8b7cdb79d96266435c2f32a102c08474e5bc)\n(cherry picked from commit a57a6a1232dc432880fc1a72895ef2e4aba8344d)\n(cherry picked from commit 3833da50a715eba2b90f280325e868fb47e2b467)\n(cherry picked from commit ef27989cd6117dfa22b0a33b91b9ff5b361d42a3)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/eff9a18e52036c8d995a22be5cf25edbb667b8fd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/eff9a18e52036c8d995a22be5cf25edbb667b8fd"}]},"branch":"refs/heads/stable/victoria"},"edac51f2097a11710ed94f902e7d9179e1348e99":{"kind":"REWORK","_number":2,"created":"2023-01-04 10:40:28.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/59/869159/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/59/869159/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/59/869159/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/59/869159/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/59/869159/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/59/869159/2"}}},"commit":{"parents":[{"commit":"1f617ff40aaf20b3247be6f8f90ca9ef684766b9","subject":"Merge \"Fix bug when rolling back prov and op status for some API calls\" into stable/victoria","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/1f617ff40aaf20b3247be6f8f90ca9ef684766b9"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-12-28 15:09:12.000000000","tz":60},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2023-01-04 10:40:20.000000000","tz":60},"subject":"Fix broken tox.ini with tox4","message":"Fix broken tox.ini with tox4\n\n- scripts were not in allowlist for pep8 environment\n- skipsdist setting breaks the docs environment\n\nConflicts:\n\ttox.ini\n\nChange-Id: I5ac930e0d3928bb8acba4212dfff25f788acd442\n(cherry picked from commit 41ee7e7aa1117ae42ea0fd1ea7b3ea8ebba27476)\n(cherry picked from commit 946b8b7cdb79d96266435c2f32a102c08474e5bc)\n(cherry picked from commit a57a6a1232dc432880fc1a72895ef2e4aba8344d)\n(cherry picked from commit 3833da50a715eba2b90f280325e868fb47e2b467)\n(cherry picked from commit ef27989cd6117dfa22b0a33b91b9ff5b361d42a3)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/edac51f2097a11710ed94f902e7d9179e1348e99"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/edac51f2097a11710ed94f902e7d9179e1348e99"}]},"branch":"refs/heads/stable/victoria"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
