)]}'
{"id":"openstack%2Foctavia~869161","triplet_id":"openstack%2Foctavia~stable%2Ftrain~I5ac930e0d3928bb8acba4212dfff25f788acd442","project":"openstack/octavia","branch":"stable/train","attention_set":{},"removed_from_attention_set":{"34429":{"account":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"last_update":"2023-01-23 13:23:45.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I5ac930e0d3928bb8acba4212dfff25f788acd442","subject":"Fix broken tox.ini with tox4","status":"ABANDONED","created":"2023-01-04 09:37:21.000000000","updated":"2023-01-23 13:23:45.000000000","total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"2595e6ca8820cb313d0bb63b9cb401fa07fb1a6c","_number":869161,"virtual_id_number":869161,"owner":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"tag":"autogenerated:zuul:check","value":1,"date":"2023-01-04 11:51:27.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"rejected":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"all":[{"value":-1,"date":"2023-01-06 14:17:15.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-01-04 11:51:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-01-06 14:17:15.000000000","updated_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"REVIEWER"}],"messages":[{"id":"57deb23b179043ada82f77b8b2d72983ec117f71","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2023-01-04 09:37:21.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1e159cabe43c7bb927acec9b88f2f33fd2c12079","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-04 11:51:27.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9fc0a2e04fdf480a9b366b69faa961a8\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/66c06c4d318447109c3d333e690fd425 : SUCCESS in 9m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/242b7196b8ea4fcc955079773942f7b6 : SUCCESS in 5m 53s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/e65c3b66c60341078575ff7c8622e881 : SUCCESS in 4m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/48ba1d3effae4f77bf3a1e211a47baa2 : SUCCESS in 4m 11s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/7c8f693b8a154aa394263280afa41817 : SUCCESS in 4m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1c22bbf0bf7945e29e9f65e37620e7b4 : SUCCESS in 10m 28s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9f6297c4a12b4333829e0eaaa9ac3642 : SUCCESS in 4m 27s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/52159c2e1b064b368e8f62797f0d6b1c : SUCCESS in 4m 34s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/2a6acdad34164bcfadf2029061b1d237 : SUCCESS in 6m 38s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/64c7b29792fa4487a489af61a8338bcd : SUCCESS in 6m 25s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/7d3a454f236b4479bf4f40bdf461ee8b : SUCCESS in 5m 51s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/2b91358079d64ad5a1480a66d46ea298 : SUCCESS in 44m 38s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/99c3ab94af794068b1d5c311afc25df7 : SUCCESS in 41m 56s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/fa11ebf0235544628bf0320e414524ca : SUCCESS in 1h 31m 01s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/46380d8f871d4c8db3159920526f38ca : SUCCESS in 2h 01m 22s (non-voting)\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/eb183773473a487381f9539911519040 : FAILURE in 23m 52s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/d6696bfba64f48caa34f99146a685e24 : RETRY_LIMIT in 5m 07s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/b30d82852ee14374a04d828c3c8680ad : RETRY_LIMIT in 5m 20s (non-voting)\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/0d5eecfafe564ad9a155fec0a313bb06 : SUCCESS in 57m 08s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/9dbbd5a11c8148a399d5474452b6faea : SUCCESS in 2h 02m 05s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/670dbdc4f3a64c6e81fabf98e9e3272e : FAILURE in 47m 08s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/53393ee9b9ac48219a36305343710105 : FAILURE in 49m 12s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/b1608b58f2b74f25b682554dbb67fa98 : FAILURE in 21m 56s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"5082cc926b9d26a5a8765a965bf38a829ac1e212","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-01-06 14:17:15.000000000","message":"Patch Set 1: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"2595e6ca8820cb313d0bb63b9cb401fa07fb1a6c","tag":"autogenerated:gerrit:abandon","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2023-01-23 13:23:45.000000000","message":"Abandoned\n\nFix not necessary because tox version gets pinned to \u003c4.0 on stable branches.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"b7df3049e2f0462a9f70b08bcc76d6e757e284eb","revisions":{"b7df3049e2f0462a9f70b08bcc76d6e757e284eb":{"kind":"REWORK","_number":1,"created":"2023-01-04 09:37:21.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/61/869161/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/61/869161/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/61/869161/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/61/869161/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/61/869161/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/61/869161/1"}}},"commit":{"parents":[{"commit":"1b737b870c2df11423bdf0858041df8483a4919a","subject":"Merge \"Ignore status update on deleted objects in driver-agent\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/1b737b870c2df11423bdf0858041df8483a4919a"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-12-28 15:09:12.000000000","tz":60},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2023-01-04 09:36:47.000000000","tz":60},"subject":"Fix broken tox.ini with tox4","message":"Fix broken tox.ini with tox4\n\n- scripts were not in allowlist for pep8 environment\n- skipsdist setting breaks the docs environment\n\nConflicts:\n\ttox.ini\n\nChange-Id: I5ac930e0d3928bb8acba4212dfff25f788acd442\n(cherry picked from commit 41ee7e7aa1117ae42ea0fd1ea7b3ea8ebba27476)\n(cherry picked from commit 946b8b7cdb79d96266435c2f32a102c08474e5bc)\n(cherry picked from commit a57a6a1232dc432880fc1a72895ef2e4aba8344d)\n(cherry picked from commit 3833da50a715eba2b90f280325e868fb47e2b467)\n(cherry picked from commit ef27989cd6117dfa22b0a33b91b9ff5b361d42a3)\n(cherry picked from commit eff9a18e52036c8d995a22be5cf25edbb667b8fd)\n(cherry picked from commit 3009605831bfbb0ec313660bfca6bfc198ffdc69)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b7df3049e2f0462a9f70b08bcc76d6e757e284eb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b7df3049e2f0462a9f70b08bcc76d6e757e284eb"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
