)]}'
{"id":"openstack%2Foctavia~869671","triplet_id":"openstack%2Foctavia~master~I5fdb111cae9a801bb760eeb058ef8e9ff952318d","project":"openstack/octavia","branch":"master","attention_set":{},"removed_from_attention_set":{"29244":{"account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"last_update":"2023-01-11 05:04:02.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I5fdb111cae9a801bb760eeb058ef8e9ff952318d","subject":"Fixing tests with tox 4.2.6","status":"MERGED","created":"2023-01-10 10:33:25.000000000","updated":"2023-01-11 05:05:41.000000000","submitted":"2023-01-11 05:04:02.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"869671","meta_rev_id":"0d88ace61596e4c294e6704355d5e47de1cad86c","_number":869671,"virtual_id_number":869671,"owner":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-01-11 05:04:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-01-10 13:55:06.000000000","_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-01-10 13:55:06.000000000","_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"CC":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-01-10 10:39:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-01-10 13:55:06.000000000","updated_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"reviewer":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"state":"REVIEWER"},{"updated":"2023-01-10 18:56:34.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"CC"}],"messages":[{"id":"12279e6ce7e3a61f65bbb25f7fc6e4347510f30c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-01-10 10:33:25.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d52c392609e234dde3765fc86904ff51e7011fa8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-10 10:39:14.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/544dee0821da472589f7cca6717fd421\n\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/3c387aad70b047c8a83e1d3ee955fdab : FAILURE in 2m 55s","accounts_in_message":[],"_revision_number":1},{"id":"a463f4942f54c679b152ce36258f3eda5f7f25d2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-01-10 10:55:03.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"6931cf610eef565728782df098b0dda542e76dca","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-10 11:04:07.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/936952ec9dc542798502125485df0532\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/e028b508566947329327094b0be6620f : SUCCESS in 5m 08s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/e8dba529a6f8404f89ac09b2225bcf32 : SUCCESS in 6m 55s","accounts_in_message":[],"_revision_number":2},{"id":"b58a580edcd5bb32d314b450f618104320c4d56e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-01-10 13:25:17.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"f7551e295144de9b78213ca87bb0016f3796687d","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2023-01-10 13:55:06.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"8258cb7459d2cc556cc426d597351d20bb02c73d","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2023-01-10 14:09:40.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"c58e97190b97a96c56b7d152696d3819e238e05c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-10 16:01:12.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/f1e590c6d33e42ed982a7fbc0b1438a3\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/1dacb219db2e4e868450ea9246290643 : SUCCESS in 5m 01s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/dd580814e0054a2787fd8dc227abeea3 : SUCCESS in 10m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0addfd6e046645bc826e1bd833229009 : SUCCESS in 6m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/49ceb98e434041d4a215fac7bbd3cca4 : SUCCESS in 4m 22s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/61fe4f0831f9432081645c7c5bd57cdb : SUCCESS in 4m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c7012a7fded946a49b876bd6edf27bbf : SUCCESS in 12m 08s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f6023767f30a4f89b978ef489ef1756d : SUCCESS in 6m 29s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/7acaa983cba047119b7c76d7c594016d : SUCCESS in 6m 11s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/fa77329b9d214adb8e0b7276b93852b6 : SUCCESS in 7m 58s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/c3f1bb3803eb44bf8be99f35b60b54aa : SUCCESS in 5m 23s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/9e3b4ffb3084435995abc5ba94ca3074 : SUCCESS in 6m 44s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/008ee46cd7794c1ea586ec1b91c3b1dd : SUCCESS in 8m 54s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/0ef987e5cb234ae397179455877022c6 : SUCCESS in 8m 10s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/d43b190998da4105a83f87f1f8991304 : SUCCESS in 1h 19m 01s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/8095a12a5fb54320ab584dd53cd757f7 : SUCCESS in 2h 15m 47s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/7203a637329749de8793cad45f74b878 : FAILURE in 24m 35s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/e24f671461ce49bd86c96259a6c3b932 : SUCCESS in 1h 19m 39s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/f90bd251aa99458bb1cffe0b7ed08b32 : SUCCESS in 39m 46s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/b41dabd8af364425836f7858e606ba52 : FAILURE in 22m 26s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/b52b09805b794001bdd0f837f37434fa : SUCCESS in 1h 59m 58s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/3c82981497d043f295b92ff21b95b2e1 : TIMED_OUT in 2h 33m 43s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/fae6624737ff476f95b1f5a7009de49f : SUCCESS in 1h 47m 59s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/2e5e0858e0054be0855a98ce1e18b1ec : SUCCESS in 1h 45m 54s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/53c635eb2e3a42c7978a181d78a0ac1b : SUCCESS in 2h 10m 56s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"dfd764a64a26b966b95dbf9ff5d74150bb38a8d5","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2023-01-10 16:19:21.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"d0e9d2d46593d065719626e1bd510fbd94189353","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-10 18:44:27.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/8c22cf1dec794bea96778d6b4f84da67\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/759cd1745c8c433488974a977598dc54 : SUCCESS in 5m 05s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/96d1591b26ed4c13940ab43deb948d7e : SUCCESS in 12m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7353c22ca1e84720b1d251f21f5e5fdc : SUCCESS in 5m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d9834b2142804e529ab24e0cecc5319b : SUCCESS in 4m 06s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/467e2b7a7a664fcab69c605ec96b6181 : POST_FAILURE in 34m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e281367cf48f47aba8b22f44b653a347 : SUCCESS in 10m 34s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b0340952c266464293c8432b02259d24 : SUCCESS in 4m 23s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/5ba991879e19409088f8941dbe438e2b : SUCCESS in 5m 17s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/5d92339b88604c3c95fb1754e5c2cd24 : SUCCESS in 7m 40s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/e562307073b44fc5a0668d892ef97c9a : SUCCESS in 4m 33s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/8d392f72793c4d899f46e49cda4ea48c : SUCCESS in 7m 52s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/dae6ec23e7754e9ebd4dcbdcf9c3b68d : SUCCESS in 6m 39s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/7cf607be53774e1485e72e12ae5583cb : SUCCESS in 7m 26s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/e0ddc047c8f2484bb70fcded7ef8125e : SUCCESS in 2h 03m 25s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/b874f8bf0ac14155bde14719ae0b5ed0 : SUCCESS in 1h 55m 24s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/85360d40b7b946798992b518126598d3 : SUCCESS in 33m 58s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/bb02b0d2ccc144b0825f2ebdc4e376c6 : SUCCESS in 1h 12m 40s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/9c665e11f3944203b8783f4a380c79bb : SUCCESS in 30m 59s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/752c9ad33ad1423d9dbd4af44d920469 : SUCCESS in 1h 40m 24s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/5d735a1ba3b046b8b876a929cd240191 : SUCCESS in 1h 51m 17s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/2637bd64481d4e23b7aafe9491b84106 : SUCCESS in 2h 03m 09s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/9dec5ba9ed994e46add54bcad55ef21e : SUCCESS in 1h 32m 51s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/0234526e8c654ccd85dc30de8a62d878 : SUCCESS in 1h 42m 24s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/3f5a62dd683045d7aa5458633c3027fb : SUCCESS in 2h 21m 49s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"3dc4572ffb6896e0df09f77380004bb25c001b96","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-01-10 18:56:34.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"9656b0b65e1f42f7d8c237b4faecb84e048c994b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-10 21:26:12.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/37ed9702693f4f47910a31c8a14963b7\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/360d495323ca4479bd577d31d42c4918 : SUCCESS in 5m 20s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4fc4630603d043748f58a0508e6ab393 : SUCCESS in 10m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a59f1fdd51c847d79438c674131dc436 : SUCCESS in 4m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/522529e028a441f7bcf958ed03c8e647 : SUCCESS in 4m 36s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8494d4706a1c413c9ecbde5bccdc8b7f : SUCCESS in 4m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a585e7416e42412ea1c720022957c0e3 : SUCCESS in 11m 15s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/310e73c0dc604ff296aec6847d03f17f : SUCCESS in 4m 48s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/b29227546e1642619a7535f3cf947840 : SUCCESS in 4m 56s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/aa5df0dfbaa94ae1a40efad15f1ce887 : SUCCESS in 6m 37s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/493f0a1a64424f7e8782fa6a80e1db58 : SUCCESS in 4m 40s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/95ccf910e10c4d94a787cab2cd82009a : SUCCESS in 8m 41s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/17fec7e4c0434bbdb3c2c88ca485a034 : SUCCESS in 7m 26s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/bfc95eeb151d46acae4f1b4bd1646069 : SUCCESS in 8m 51s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/8b429329e02f47cd95749803a51f9cca : SUCCESS in 1h 40m 24s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/e82597dd906940e0b47e2b613f971241 : SUCCESS in 2h 07m 27s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/3a820a7fe4ec41d4b834600447fe22c3 : SUCCESS in 37m 51s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/5154a84340cf48bfa381d4b419c2374f : SUCCESS in 1h 10m 10s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/d852d93b00134d8b8a4c539eb2ac114e : SUCCESS in 33m 04s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/746240bae4f041b3a634053429ff5865 : SUCCESS in 2h 01m 35s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/d7d5e238215a469cb1acdbc3bcc3a79c : SUCCESS in 2h 26m 22s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/c2ac600660e248fd9ef5891f9e6b8921 : SUCCESS in 1h 56m 53s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/d1acf9e43b8f4132ac8dad096f0ae087 : FAILURE in 18m 04s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/9f2ed865f00a49f8b5cde1ca2b38e1ce : SUCCESS in 1h 58m 25s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/b8e695b8594d47a7908ca55ba873c82d : SUCCESS in 2h 09m 14s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"1ce984b412fb4b922117ae403c6e46fe5a652d66","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-10 21:26:49.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"b8ca496956b1761a575b983815bfc01959fd279f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-10 21:57:22.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/e519b617260d41c786ae476be9583c3a\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/dae504cad33042129a0333b555631ee0 : SUCCESS in 5m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/568bc03d3cef4193810d1ce5b7d9c5a3 : SUCCESS in 5m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/af762e1b33544e3383e055ebf0f7b201 : SUCCESS in 4m 52s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e44baef4a18f476d8cbf28b635b8059b : SUCCESS in 4m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/23f07750eaa449e993d64d6f9362ae7a : SUCCESS in 11m 57s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7d6f2acaf4414a55bb0ad584a7782a99 : SUCCESS in 4m 53s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/478855a94d784176bbff0cfd1d98272e : SUCCESS in 4m 54s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/3d64a64c405a496b8e53655dc7ae6422 : SUCCESS in 7m 15s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/72929b6ef92541e8be25b869a36af5e2 : CANCELED\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/8723e03bb08a41338b15305d8b72ca88 : CANCELED\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/f100408a2b5e428b949d3405950116f2 : POST_FAILURE in 29m 16s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/9e81888b00a64aa59c1ddd0c7681edf6 : CANCELED","accounts_in_message":[],"_revision_number":3},{"id":"3be800a7d1dda760323bbd42813eb1036487c17f","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2023-01-11 00:04:04.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"6f18ecbda7cc559cd7108eee00ee3926ff3659c1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-11 02:51:29.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/13bc3743920a44fba19043bf99f0da5e\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/fd069bfe871f4a719e5ddeb763eebac9 : SUCCESS in 5m 20s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0698edf3e73d43b0a7567a2b48a1f7d5 : SUCCESS in 10m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a6d2206a06f649738b9c7ac5032d5bba : SUCCESS in 5m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9bfcea67ace44810912cbd86a35959af : SUCCESS in 3m 57s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c3307e1f35954894a47abe73376f6b21 : SUCCESS in 4m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/48a88b6f21a34e8f82937a5c509467d5 : SUCCESS in 9m 27s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/19225e4be31549cd8e19743d87b98ec0 : SUCCESS in 3m 46s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/0446d3390e344b9ab1be03b3fc538847 : SUCCESS in 4m 52s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/102e96ab76ce49708939bc8c4acad823 : SUCCESS in 8m 10s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/cbe6d8207c79492c8d07e31b0300a499 : SUCCESS in 3m 40s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/b218f1c69bf94c58b25690309f324775 : SUCCESS in 6m 41s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/17912f3cd9c7429cb93414d007fa5990 : SUCCESS in 5m 28s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/99c32d44dd35405c8462ee4a5ed4f0e9 : SUCCESS in 7m 49s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/241d158ef6a34822843b25d73e269b5c : SUCCESS in 1h 22m 42s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/c709fbbfcf574b6c9c672dffd3ca1612 : SUCCESS in 1h 40m 22s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/c39bc38cb9d04e37a18f4fab723bfd01 : SUCCESS in 34m 23s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/9c5866d597814c58b561f25b826b0a02 : SUCCESS in 1h 12m 47s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/987c8a4109c94df99a146319c3cfb425 : SUCCESS in 31m 07s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/3cee4e60fbaf4c34b08b4cebf55f2fd6 : SUCCESS in 1h 30m 00s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/6ad172d4579f4faea838ba7db6ad6e47 : SUCCESS in 2h 17m 35s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/418a4d9dba3648c1a39b28b953fba2d3 : SUCCESS in 1h 37m 17s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/f74c8684edba4191b9590b93e76abf3e : SUCCESS in 1h 35m 15s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/b1898f7fc9884e8389092433287f91a0 : SUCCESS in 1h 41m 15s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/2376cf097b95427d838826cf4b1c09cb : SUCCESS in 2h 32m 44s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"25138a6613064c11f77bd1747cd040db9720cb7c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-11 02:52:12.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"e37ed4b4912bf571b5255bf89cb135cb83a8f494","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-11 05:04:02.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a2a82ab78fd74bb8a950a26ec1fb0272\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/90705e897f174980ba9324e3b14d069f : SUCCESS in 6m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/be0bcf224cdb4e538cdd0e1e700bd880 : SUCCESS in 6m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b4d965c0c2d64d0f993e7427b0011056 : SUCCESS in 5m 32s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2c2fb244075e4c2ea8d54c05e43e607b : SUCCESS in 4m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f18a270f940b43e0a1714478c75b7a4a : SUCCESS in 12m 12s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/92155816085f405090db458312e6b946 : SUCCESS in 5m 08s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/90e0782634154749b3eba3e87030aecd : SUCCESS in 5m 19s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/4cfa3571ced741c4a3cbbb149d88ce6e : SUCCESS in 7m 28s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/49d9b0e272754ed892ade1d2f5d42588 : SUCCESS in 1h 26m 06s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/addef71031ae4c1f8e562e4efa7fe272 : SUCCESS in 2h 10m 04s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/4f54860f9458480295d1c02e36e90cad : SUCCESS in 48m 26s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/49583fb078a44fd0b544c4f8ca071e92 : SUCCESS in 1h 04m 33s","accounts_in_message":[],"_revision_number":3},{"id":"04dd8d2f10eafaa77dd5945ae36228ccef67212f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-11 05:04:02.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"0d88ace61596e4c294e6704355d5e47de1cad86c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-11 05:05:41.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/645692cd354d408fa6eede627006677d\n\n- promote-api-ref https://zuul.opendev.org/t/openstack/build/1a8a6b684c5e416298b80102c699583d : SUCCESS in 47s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bf5407e55cd0487a8a27c078677138ca : SUCCESS in 1m 19s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8d651e4e83ea4338b48e71dca74dbae7 : SUCCESS in 48s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"b28be698f87b7b75595b44d0254e931bbd7cd745","revisions":{"6cf0f52b0e0503894b927d94fa7515aae6ab7b3f":{"kind":"REWORK","_number":1,"created":"2023-01-10 10:33:25.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/71/869671/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/71/869671/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/71/869671/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/71/869671/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/71/869671/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/71/869671/1"}}},"commit":{"parents":[{"commit":"c47bc0d90cd9a4be60c87103fe4cfaf7b6dca076","subject":"Merge \"Remove i386 amphora build support\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c47bc0d90cd9a4be60c87103fe4cfaf7b6dca076"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-01-10 10:28:55.000000000","tz":60},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-01-10 10:32:45.000000000","tz":60},"subject":"DNM/WIP Testing functional-py38","message":"DNM/WIP Testing functional-py38\n\nChange-Id: I5fdb111cae9a801bb760eeb058ef8e9ff952318d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/6cf0f52b0e0503894b927d94fa7515aae6ab7b3f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/6cf0f52b0e0503894b927d94fa7515aae6ab7b3f"}]},"branch":"refs/heads/master"},"68609f59610d7d597701d2f40eeaa6af992a147b":{"kind":"REWORK","_number":2,"created":"2023-01-10 10:55:03.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/71/869671/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/71/869671/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/71/869671/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/71/869671/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/71/869671/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/71/869671/2"}}},"commit":{"parents":[{"commit":"c47bc0d90cd9a4be60c87103fe4cfaf7b6dca076","subject":"Merge \"Remove i386 amphora build support\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c47bc0d90cd9a4be60c87103fe4cfaf7b6dca076"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-01-10 10:28:55.000000000","tz":60},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-01-10 10:54:56.000000000","tz":60},"subject":"DNM/WIP Testing functional-py38","message":"DNM/WIP Testing functional-py38\n\nChange-Id: I5fdb111cae9a801bb760eeb058ef8e9ff952318d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/68609f59610d7d597701d2f40eeaa6af992a147b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/68609f59610d7d597701d2f40eeaa6af992a147b"}]},"branch":"refs/heads/master"},"b28be698f87b7b75595b44d0254e931bbd7cd745":{"kind":"REWORK","_number":3,"created":"2023-01-10 13:25:17.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/71/869671/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/71/869671/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/71/869671/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/71/869671/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/71/869671/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/71/869671/3"}}},"commit":{"parents":[{"commit":"c47bc0d90cd9a4be60c87103fe4cfaf7b6dca076","subject":"Merge \"Remove i386 amphora build support\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c47bc0d90cd9a4be60c87103fe4cfaf7b6dca076"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-01-10 10:28:55.000000000","tz":60},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-01-10 13:24:16.000000000","tz":60},"subject":"Fixing tests with tox 4.2.6","message":"Fixing tests with tox 4.2.6\n\nChange-Id: I5fdb111cae9a801bb760eeb058ef8e9ff952318d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b28be698f87b7b75595b44d0254e931bbd7cd745"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b28be698f87b7b75595b44d0254e931bbd7cd745"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[]}
