)]}'
{"id":"openstack%2Foctavia~869903","triplet_id":"openstack%2Foctavia~stable%2Fxena~I98cce51a2536bcf204e5753ce4984af57e7e69e3","project":"openstack/octavia","branch":"stable/xena","attention_set":{},"removed_from_attention_set":{"29244":{"account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"last_update":"2023-02-09 06:52:33.000000000","reason":"\u003cGERRIT_ACCOUNT_29244\u003e replied on the change","reason_account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}}},"hashtags":[],"change_id":"I98cce51a2536bcf204e5753ce4984af57e7e69e3","subject":"Fix listener creation with fully-populated API","status":"MERGED","created":"2023-01-12 07:36:25.000000000","updated":"2023-02-09 11:55:48.000000000","submitted":"2023-02-09 11:54:25.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"869903","meta_rev_id":"d1fd4cf3dbb241364cfdd070b4f3080d7f533782","_number":869903,"virtual_id_number":869903,"owner":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-02-09 11:54:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":2,"date":"2023-01-25 17:45:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-02-06 09:09:19.000000000","_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-02-06 09:09:19.000000000","_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-01-12 09:42:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-01-25 17:45:37.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2023-02-06 09:09:19.000000000","updated_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"reviewer":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"state":"REVIEWER"}],"messages":[{"id":"aba1a00df5f5132c4206e0138625ea77017770fa","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-01-12 07:36:25.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1052a8ffb1ab8689dc95fc586a539acc7741c712","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-12 09:42:40.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3792ec61c4d34597b461dd89798b5fa8\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c7a63cf611be42968a06e35011ac2aca : SUCCESS in 10m 09s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a4e28f5a2c01477cb31c6f29d6dde019 : SUCCESS in 7m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/29f45ed4a0ca4bcdbb61b36b31fc7d0b : SUCCESS in 5m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/043e606036da4b0e9ceefbae41af9196 : SUCCESS in 5m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/35580b6bd56245e6adfb79f9245d786a : SUCCESS in 4m 22s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c316769df513482a8016358147addf94 : SUCCESS in 4m 14s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/206749898b5f467c8384f5f8f2cfc702 : SUCCESS in 12m 56s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bb1637cf30f24304aace690b68bc05d1 : SUCCESS in 5m 33s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/275bf3e49a214f759ca7dca5d6e1bba2 : SUCCESS in 4m 33s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/e45c6775b83548558db186898fc431e2 : SUCCESS in 9m 12s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/c8f9bce6ab294dcda0154a2c0a4da158 : SUCCESS in 5m 05s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/85d11a568d5f4892bfd02679dcddc8f5 : SUCCESS in 7m 51s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/db148f5e81c04def9636ac50a63cbf46 : SUCCESS in 1h 21m 50s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/b0ca9a38421a47e0a23dc14e8f9893e8 : SUCCESS in 1h 43m 54s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/b7d302f679ca4533a26d4cdc7ca76313 : SUCCESS in 38m 23s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/e47b1c1bd9f34a56af060b255f735be3 : SUCCESS in 48m 44s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/de749b7390484c3d8ca8729d2a9e7255 : FAILURE in 14m 06s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/7baf88c61c4e4ff58488c4c8bdf4b671 : SUCCESS in 35m 44s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/6786632713bb425d9a4c56c2d8cd22c5 : SUCCESS in 1h 46m 22s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/cc8c277f28f7474582cb04f229313bc5 : SUCCESS in 2h 04m 51s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/55c7250ece814090a63b12799d416aec : SUCCESS in 1h 59m 25s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/03bcef69be9b43e5a6f0419c86da4cc8 : SUCCESS in 1h 51m 43s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/445c8f32f6a54ef297533ea4f50398b1 : SUCCESS in 1h 24m 11s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/8bf9fdbcc86f42d69df0de6f98b09fe0 : NODE_FAILURE Node request 300-0020155756 failed in 0s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"7be6a81a1ea9288c1001121199f62bd84997fe06","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-01-25 17:45:37.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"815577b9b1630d97efa7e97ed9ac4010349c582d","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2023-02-06 09:09:19.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"fde79efcfbd68395780e6a48990afdd7f31683d2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-06 09:10:13.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"141b3f5a070331421639dd70580cbd996477cb6d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-06 13:28:00.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/300dd351a44c4ba3b49aa803a80fc43e\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/cd16e66a3f7947eea5b9945c2e19728d : SUCCESS in 7m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7adbac8eef894c8db5b88da67dafcd28 : SUCCESS in 5m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/573620c9a86542e295870a857246fb7b : SUCCESS in 4m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/11bd99c4f3b748f0aac0aef23d982a07 : SUCCESS in 4m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dbc229512cda4e6d960696a8163f10f4 : SUCCESS in 11m 39s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9debc6c7fcb54c7c8debf5a75dc1b1bb : SUCCESS in 4m 35s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/ed2eb314f8b843748855086af3639fad : SUCCESS in 4m 35s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/44b4a9adf6c94820ac8615ad08bb7d9f : SUCCESS in 7m 58s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/5127d627dfd04759a3860c2719c1064d : CANCELED in 19m 55s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/3f72425da22b405fae5d2ea8d14e04b9 : CANCELED in 20m 20s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/356a6219e69c423c86dcf0dce1e96066 : CANCELED in 20m 19s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/04f24d2632074cfd9598532e11bdbc5f : FAILURE in 19m 28s","accounts_in_message":[],"_revision_number":1},{"id":"5e44a36648c2689bc2dada57194fc3857e1a4e6c","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2023-02-06 13:58:24.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"bbc2c05b43d9578bd404f5254f9b41054d195aa8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-06 16:07:05.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/304e6ac47fa248768e424e24bcf417fa\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/152a7fc031334a8f8fbfc1d30959ef0a : SUCCESS in 9m 46s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8945648f4f5a47c3bc0f55151e11727f : SUCCESS in 9m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a48d30f730f444d7b737ace2886a3cb3 : SUCCESS in 4m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/54f791bc928a42a4bbb12e583dffaf9a : SUCCESS in 4m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bfb14cf9b97b4b65abb400958ca40a5b : SUCCESS in 4m 43s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/699dce3449ff46c7823e1b09360b5b75 : SUCCESS in 4m 15s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/07996fe786734cbc9150ea59b619b9b4 : SUCCESS in 10m 32s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/420577df08644950b15761074aebf06d : SUCCESS in 5m 49s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/45290b19de8b492ea3bfef9fb1e65277 : SUCCESS in 5m 40s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/a8bda7039ac84a8c8773d674c62b8a30 : SUCCESS in 7m 25s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/ad6499397d5f4f59a77cf48a6de818a7 : SUCCESS in 4m 23s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/eee49d2ccf5c4a6e82ab21827949b05d : SUCCESS in 7m 46s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/dff8c6058ff2453aa5827bf4ad537f56 : SUCCESS in 1h 05m 57s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/5da990bbfce446018188869a58e526b8 : SUCCESS in 1h 44m 41s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/213e39d9daef45ad95a5c571e494475d : SUCCESS in 32m 06s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/449afb11c1704adea9267f57ae0d9557 : FAILURE in 1h 04m 56s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/2cddff8db0204d6ca0a4e89b08485859 : FAILURE in 10m 34s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/6e7c5ff2f5d04c4d9917ac89efdc453f : SUCCESS in 29m 03s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/45814252e76e443b847c644ed35c7718 : SUCCESS in 1h 36m 09s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/e600b8bf56ff4eb8a5b05b74b016de1e : FAILURE in 1h 05m 33s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/391a2622d972488ebc2f3f5b138993b7 : SUCCESS in 2h 06m 52s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/88a921c4151048adba7316279e8b4815 : SUCCESS in 1h 42m 41s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/559768d954e84396b8b6fd8a48b645d4 : SUCCESS in 1h 39m 07s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/e1bc8a5c89ba47b698eff6804c0006d0 : NODE_FAILURE Node request 300-0020372810 failed in 0s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"6b8e352452eddbbe66ffd3241bcf2fda3dc5f1aa","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-02-09 06:52:33.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"4b58d22c1c36ec7a22755afe31becb658faab665","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-09 10:14:26.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/345d3de338f94441887d71fcc3dd7aa7\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1b67a4f3db934b24bb43d8bc9fceaa40 : SUCCESS in 10m 10s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8b860fef2b96432a8ab3624af08bc27f : SUCCESS in 8m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eb2c70f999be4d348d017fe3e66bd53c : SUCCESS in 4m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f3b52fbbfe0d42739f3808677657f262 : SUCCESS in 6m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3c3d9b1f2ccd462e8c653bcf5e45b160 : SUCCESS in 5m 47s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0db0fd85833c4f70a67942f297ff2f48 : SUCCESS in 4m 38s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8dfc26ac7bbd442ea034abaf7d788b9c : SUCCESS in 14m 43s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4b1a3d4723764aeaa8bccfa192b11932 : SUCCESS in 5m 14s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/aebcaabdb15941e5ab2f4ceebfdf9c73 : SUCCESS in 6m 08s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/b0d97a3c211a4eb89d4bc729cb6863ab : SUCCESS in 9m 12s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/b2ffce2a6e0b497ba495bab91cb5b04a : SUCCESS in 7m 02s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/e85d47a6a162446f956a52feb2732c22 : SUCCESS in 9m 36s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/263f552618e94c86a383aebf49967622 : SUCCESS in 1h 22m 33s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/9eb5f6992ae14f758ac9a295fcbaad6b : SUCCESS in 2h 23m 18s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/9a73d3691c26495187d87e4f12f2c6e3 : SUCCESS in 39m 32s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/418f36a7ad6548caac6176fbc8cc90c8 : SUCCESS in 53m 32s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/8e0a62689e654ba4920023ac939bf1ec : FAILURE in 11m 39s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/3d1cf64621a14e11b813dc1238bbc152 : SUCCESS in 38m 36s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/ab895043f3d243c3a7ae7251a2987a4e : SUCCESS in 2h 16m 49s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/7b6919b4d22b4350a888cc5f92badd0f : TIMED_OUT in 3h 04m 04s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/c60c95c9b69e4a6e939dcbc779db0da5 : SUCCESS in 2h 20m 53s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/5df2ba0aeb2d477384255bdd64946d31 : SUCCESS in 2h 21m 54s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/2cd131b07d7c45749c02ffa2a2189e15 : SUCCESS in 2h 26m 15s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/13ee1f4051a14e0b9fac019bc3caeb84 : NODE_FAILURE Node request 300-0020403059 failed in 0s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"bc1a9380a3beddb3e99e2a2bd6ce9c76adb44515","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-09 10:15:27.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"767cdc4d009e9de840abff438d73105de2b786ea","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-09 11:54:25.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2c9285efc9d64984b0d2d1c399b97901\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ed68bf149cb94939b59611385843fd65 : SUCCESS in 8m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7b3fa3e3d2a3405bab258dab9ca91c3f : SUCCESS in 5m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/808146c73fc8472ca7f3ceaecd33f625 : SUCCESS in 5m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5463a8da26064d53b68fdee0bae4415c : SUCCESS in 5m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f288dc2f3a7942cc96c9f321387d6a30 : SUCCESS in 11m 53s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5bd01454a26c4575a8edc7bece81f603 : SUCCESS in 5m 45s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/76bdffad88ee4268aeec9d0549b6e246 : SUCCESS in 5m 03s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/da539475ce834f6591b087910fc79beb : SUCCESS in 7m 43s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/92dedcd8d2a84309a1c98809c12a4da6 : SUCCESS in 1h 26m 57s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/83d1b90af5e64d6383e877a334a73434 : SUCCESS in 1h 37m 48s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/9019d9e930ed4672a312baab20c569f0 : SUCCESS in 32m 32s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/b0716c99cb27402e932c56166d026bf8 : SUCCESS in 41m 05s","accounts_in_message":[],"_revision_number":1},{"id":"1cc33e15249f64f0332b177cb3573ee79d14eef4","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-09 11:54:25.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"d1fd4cf3dbb241364cfdd070b4f3080d7f533782","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-09 11:55:48.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7b51a19c085a42949b7d6ce7fdf727c3\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4bbf19d82e88432db37dcfd2bc0cc156 : SUCCESS in 54s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/02a06600f15b4885b2c7e484e29b3379 : SUCCESS in 40s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"7a7c81d32f4c248065421000050a9eed55b09800","revisions":{"7a7c81d32f4c248065421000050a9eed55b09800":{"kind":"REWORK","_number":1,"created":"2023-01-12 07:36:25.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/03/869903/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/03/869903/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/03/869903/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/03/869903/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/03/869903/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/03/869903/1"}}},"commit":{"parents":[{"commit":"c0b6f5c9ef07f777b10d952a70638c3e00e21c05","subject":"Merge \"Fix plugging member subnets on existing networks\" into stable/xena","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c0b6f5c9ef07f777b10d952a70638c3e00e21c05"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2022-12-22 15:30:57.000000000","tz":-300},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-01-12 07:36:20.000000000","tz":60},"subject":"Fix listener creation with fully-populated API","message":"Fix listener creation with fully-populated API\n\nCreating a listener with the fully-populated API was broken in SINGLE\ntopology, one required parameter was missing from the flow.\n\nConflicts:\n\toctavia/tests/unit/controller/worker/v2/flows/test_load_balancer_flows.py\n\nStory 2010488\nTask 47060\n\nChange-Id: I98cce51a2536bcf204e5753ce4984af57e7e69e3\n(cherry picked from commit d0a0f53dc3522cdc334afebbc42e70ac3f198530)\n(cherry picked from commit 4e9203adb6e7d612ebe16152cfa0b42c4401289f)\n(cherry picked from commit 66a62bc20153ba70fb1d6a87ecdbb96cbc143943)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/7a7c81d32f4c248065421000050a9eed55b09800"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/7a7c81d32f4c248065421000050a9eed55b09800"}]},"branch":"refs/heads/stable/xena"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[]}
