)]}'
{"id":"openstack%2Foctavia~870517","triplet_id":"openstack%2Foctavia~stable%2Fxena~Id328b27db1ec9c8e17bf18120259e41e75dab3b9","project":"openstack/octavia","branch":"stable/xena","attention_set":{},"removed_from_attention_set":{"29244":{"account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"last_update":"2023-04-03 19:17:30.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Id328b27db1ec9c8e17bf18120259e41e75dab3b9","subject":"Fix full graph loadbalancer creation if jobboard is disabled","status":"MERGED","created":"2023-01-16 09:42:40.000000000","updated":"2023-04-03 19:18:42.000000000","submitted":"2023-04-03 19:17:30.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"870517","meta_rev_id":"ab510f7d0a5521be6c66b785fce8224e76027286","_number":870517,"virtual_id_number":870517,"owner":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-04-03 19:17:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"recommended":{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},"all":[{"value":1,"date":"2023-01-16 09:52:02.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},{"value":2,"date":"2023-03-23 16:09:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-04-03 16:56:04.000000000","_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"all":[{"value":0,"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-04-03 16:56:04.000000000","_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-01-16 09:52:02.000000000","updated_by":{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},"reviewer":{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},"state":"REVIEWER"},{"updated":"2023-01-16 12:04:03.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-03-23 16:09:52.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2023-04-03 16:56:04.000000000","updated_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"reviewer":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"state":"REVIEWER"}],"messages":[{"id":"28aa624e13507dc7d01f3821cf03ea5219440d0b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-01-16 09:42:40.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ad6b8be584e7c12488a4f908ed8ee7f8b21d835f","author":{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},"date":"2023-01-16 09:52:02.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"aea741462e395f6b635dfc4ad95653ae47dc6502","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-16 12:04:03.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ce93e0835b1a403e9b4dc7268967e79d\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8c689825cdba417cb28a27d15218484c : SUCCESS in 10m 09s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/bb99f2ee0f8648dcbf0a503bc7047f7a : SUCCESS in 8m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9f24883f54254a5cbc8070e8d5afd7ae : SUCCESS in 5m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/aa31c57f29ce4120a13ddcad3ee32f40 : SUCCESS in 4m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/329c9369a9c24a13b8b39e257cd0527f : SUCCESS in 4m 33s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/14007f2f11074ca5848b3eeb4b10073d : SUCCESS in 4m 54s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/19e71d61e9a744d5b48465bf1f46c8b5 : SUCCESS in 12m 14s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/214bb34fd60c485794f33aee76ad9505 : SUCCESS in 4m 45s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/b70cebbbe3904716a91110d7d695bf50 : SUCCESS in 5m 10s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/71d239744f82483cb0ea6b82e6754d63 : SUCCESS in 8m 07s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/babe289c19c44b97b73cb1ea721918b5 : SUCCESS in 4m 51s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/cd35f292f20f469c8103cebb4c99733d : SUCCESS in 7m 46s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/6c1643b047aa4686be4aebb135bdde41 : SUCCESS in 1h 33m 27s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/486e5089726b4f198c6a20cf8cbb0a46 : SUCCESS in 1h 50m 02s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/f9c61aefe09748c2aaa928ecee539a7f : SUCCESS in 35m 06s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/cfae127bd18744d1a843072270805ca7 : SUCCESS in 46m 53s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/159e8e06a02c4e1d9a4a6883b16180b0 : FAILURE in 9m 49s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/cc2dd2d5cf464ebe82673d1705cf7220 : SUCCESS in 33m 12s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/2830f38f534541859d2a58b329848a51 : SUCCESS in 2h 04m 47s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/d2a3d5e7ac9f476cb28447d64aa5d7b1 : SUCCESS in 2h 19m 44s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/089078a71a734430bbb2a807a0f34ef6 : SUCCESS in 2h 08m 29s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/415081f2703446c4a109973985d0691c : SUCCESS in 1h 56m 44s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/262cbe0fede84fcb8eba48a38aebca71 : SUCCESS in 1h 35m 24s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/8d9c53bdda7d46abb2291a0258cfefca : NODE_FAILURE Node request 300-0020183846 failed in 0s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"94e1cadfce6a309ae10832cf409dd8b3db8192a9","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-03-23 16:09:52.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"de3065aaea184ac1ae20f3287dd9894f65b0169f","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2023-04-03 16:56:04.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"08fa51cc1d136f0cd30f3dc8027f76d46e16bd08","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-03 16:56:46.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b9ccdac984c7c5430a5e94f72b871d0638c9c8d5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-03 19:17:30.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a299589cbde34004bb72a762f2bc505c\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/369f0f995bf94dc79acd4522814fbc51 : SUCCESS in 7m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9bea415a3b2d412294eb35596c2c1adc : SUCCESS in 4m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f672a70491cf4b49b0299703e18b60ea : SUCCESS in 6m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2638ca351f454768b1d5fd489646e250 : SUCCESS in 4m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fe0650cd64114388ad0db9f48da1910e : SUCCESS in 12m 42s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a9d8e6c5a1de43f8aade6e5c9bb95bfc : SUCCESS in 4m 43s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/fc600e0da43749958788501bbfc47ec9 : SUCCESS in 4m 46s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/2cc269ff54764faf9c8dc2d73373533f : SUCCESS in 8m 29s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/9c5ffb51a1754c8699bee7388dd2bd3c : SUCCESS in 1h 21m 12s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/4420ced786284a67b6a2ea92d698a945 : SUCCESS in 1h 38m 22s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/8480360b8d874c17adce55f8e1954e26 : SUCCESS in 38m 20s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/576a52b745db4125a2a69477eb3034ea : SUCCESS in 43m 44s","accounts_in_message":[],"_revision_number":1},{"id":"1cec7aa0eaefe97c6099279a7ba5c718cfb345c5","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-03 19:17:30.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"ab510f7d0a5521be6c66b785fce8224e76027286","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-03 19:18:42.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1be942a978a143559777be0d6237237d\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/107e3f5707be4935b1f39a307c655c03 : SUCCESS in 48s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3ec7d5bf0b9e4aa5b6b44307c1d90b27 : SUCCESS in 37s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a0ea8a900ca53a1e2d421e9a7defaf0d0f3354d5","revisions":{"a0ea8a900ca53a1e2d421e9a7defaf0d0f3354d5":{"kind":"REWORK","_number":1,"created":"2023-01-16 09:42:40.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/17/870517/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/17/870517/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/17/870517/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/17/870517/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/17/870517/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/17/870517/1"}}},"commit":{"parents":[{"commit":"c0b6f5c9ef07f777b10d952a70638c3e00e21c05","subject":"Merge \"Fix plugging member subnets on existing networks\" into stable/xena","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c0b6f5c9ef07f777b10d952a70638c3e00e21c05"}]}],"author":{"name":"Anton Kurbatov","email":"Anton.Kurbatov@acronis.com","date":"2022-09-28 18:48:45.000000000","tz":0},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-01-16 09:41:45.000000000","tz":60},"subject":"Fix full graph loadbalancer creation if jobboard is disabled","message":"Fix full graph loadbalancer creation if jobboard is disabled\n\nThis patch fixes the creation of a full loadbalancer graph in case the\njobboard_enabled config value is False, when no \"listeners\" parameter was\npassed to the get_create_load_balancer_flow function call: use kwargs as the\nflow creation function parameters and handle \"store\" separately.\n\nStory 2010335\nTask 46462\n\nSigned-off-by: Anton Kurbatov \u003cAnton.Kurbatov@acronis.com\u003e\nChange-Id: Id328b27db1ec9c8e17bf18120259e41e75dab3b9\n(cherry picked from commit 216cce39c9c22831d725703f41c54b9305d65dc8)\n(cherry picked from commit c662e42b849aa962a814fe756e109950e794277c)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a0ea8a900ca53a1e2d421e9a7defaf0d0f3354d5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a0ea8a900ca53a1e2d421e9a7defaf0d0f3354d5"}]},"branch":"refs/heads/stable/xena"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
