)]}'
{"id":"openstack%2Foctavia~870518","triplet_id":"openstack%2Foctavia~stable%2Fwallaby~Id328b27db1ec9c8e17bf18120259e41e75dab3b9","project":"openstack/octavia","branch":"stable/wallaby","attention_set":{},"removed_from_attention_set":{"29244":{"account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"last_update":"2023-04-03 19:17:36.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Id328b27db1ec9c8e17bf18120259e41e75dab3b9","subject":"Fix full graph loadbalancer creation if jobboard is disabled","status":"MERGED","created":"2023-01-16 09:43:00.000000000","updated":"2023-04-03 19:18:49.000000000","submitted":"2023-04-03 19:17:36.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"870518","meta_rev_id":"9030879032ee6af4156658d451d7240b0455e138","_number":870518,"virtual_id_number":870518,"owner":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-04-03 19:17:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"recommended":{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},"all":[{"value":1,"date":"2023-01-16 09:52:08.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},{"value":2,"date":"2023-03-23 16:09:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-04-03 16:56:46.000000000","_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"all":[{"value":0,"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-04-03 16:56:46.000000000","_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-01-16 09:52:08.000000000","updated_by":{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},"reviewer":{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},"state":"REVIEWER"},{"updated":"2023-01-16 11:49:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-03-23 16:09:38.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2023-04-03 16:56:46.000000000","updated_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"reviewer":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"state":"REVIEWER"}],"messages":[{"id":"8eb37a911cfb1fc7ecc76aa8a9ccea19ac160679","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-01-16 09:43:00.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"581f65e96000f826dc11785c5b3e46452fc2743d","author":{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},"date":"2023-01-16 09:52:08.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"9f8673edf4f47e4817bf2c2578a7392403230339","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-16 11:49:22.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9a997ed396e64a8f931d129cb1431faf\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fe3354b408dc4fa383f23c53653217a6 : SUCCESS in 8m 56s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8a22c1d1b45f46bf9e136a0f2e8e92be : SUCCESS in 8m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/aeecb987209b4875927009c4bd2d7d68 : SUCCESS in 5m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/36733bf3aa2e4a0dbd7a7924380118be : SUCCESS in 6m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/18cdf2197f6c405993062b78aeed1d4d : SUCCESS in 5m 03s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/514c7a933730420b82b5f2b3d87bbead : SUCCESS in 4m 48s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ceb90a1b5f7f4e35befbf5262f46e7cd : SUCCESS in 10m 46s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/900d3753b9df4bafa85e0e59171457a5 : SUCCESS in 4m 45s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/a945e2eb081741b4903c20232306292d : SUCCESS in 5m 13s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/62071b7c2c91496fb2627ed8fcd7b3fe : SUCCESS in 7m 13s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/35f8e36aae134a59b4e9ffdeedb70aa6 : SUCCESS in 5m 00s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/aa4a0fc3be64468eb4342ecc787c5e77 : SUCCESS in 7m 23s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/a0819032f41d4338a7800c2673b0d887 : SUCCESS in 1h 27m 03s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/e84e8107260347b3bfeccf3de9ddd39d : SUCCESS in 1h 44m 58s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/2e29a8b76c434003a314e92a5de03072 : SUCCESS in 35m 30s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/d3abe6a4dcb14d5a883a8e45bfead09d : SUCCESS in 30m 17s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/8b19086a5b034571ae4a391667fbd26d : SUCCESS in 34m 12s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/d81c416656a14a24a4ae6979e2722e5c : FAILURE in 9m 36s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/bad3413d397f4b74b4fc5016cad74ef0 : SUCCESS in 1h 37m 24s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/bc91030551b24f618c22ca9e408cd1ff : FAILURE in 29m 08s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/d8d7c7243bec48cf879e9760e5ad5e7c : SUCCESS in 1h 43m 19s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/94fa1bc9b7db4649aea6dc1bbc9af526 : SUCCESS in 1h 50m 06s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/e5e51c66f26046e19a1ce6169671ce9b : SUCCESS in 2h 02m 40s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/b830dd8bb589441eb051282165b16655 : NODE_FAILURE Node request 300-0020183870 failed in 0s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a05922d96b86b0ec95d46ad0d696a8112cd71a47","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-03-23 16:09:38.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"d5ce1685618b263d5091362f5dd407321d10a7ca","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2023-04-03 16:56:46.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"2ccad5a16f992d2b340de5e1a8ad10f6d6eae6c0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-03 16:57:30.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"c672e4902dfa33bc33944da5aa3d79057a0f3834","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-03 19:17:35.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/928e32bae4d44ea5937d289bd8ab9ab0\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9247c689819e4a92901b38d14fb75ae5 : SUCCESS in 7m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f1ea6fe9dcfc4ef1a3a180d4fa53a67c : SUCCESS in 4m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7ef77e1ba9e94f8ba20a0456cc867a88 : SUCCESS in 5m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/31d1aa597f4c465182e8b846af572e2b : SUCCESS in 4m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c8539dc87cd446a79a45bc0f53f915b4 : SUCCESS in 11m 03s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0d7ea967178146aeb202c5e4595e5ba4 : SUCCESS in 4m 33s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/55e6ffdce8f0440dbd159921509dea74 : SUCCESS in 5m 24s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/d782af1f06544703b5de7b5697131fc0 : SUCCESS in 7m 28s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/14ed252f413a49b29890280b187b09e6 : SUCCESS in 44m 35s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/8a885b6d6e3e44bea11fda194418f786 : SUCCESS in 1h 45m 52s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/e6e392795ea7403f98ac9a6053b437b0 : SUCCESS in 36m 02s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/cdc7d80cc4044863a38a52d739ca69bc : SUCCESS in 33m 45s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/dfd1e96ac88d4cda9c25c80270c7aa9f : SUCCESS in 37m 42s","accounts_in_message":[],"_revision_number":1},{"id":"1da1e386d9756d3374de9a0a30d7ccdea522a9f8","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-03 19:17:36.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"9030879032ee6af4156658d451d7240b0455e138","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-03 19:18:49.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c4e741d8fda8410a8c55c970b851a032\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1039b5336f9147baab0a69f8a2ee6043 : SUCCESS in 52s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7c15704a44de4d69b768432f39e70b15 : SUCCESS in 45s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"2c80f89bb6d96240676bade01f2a9ea04481796c","revisions":{"2c80f89bb6d96240676bade01f2a9ea04481796c":{"kind":"REWORK","_number":1,"created":"2023-01-16 09:43:00.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/18/870518/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/18/870518/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/18/870518/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/18/870518/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/18/870518/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/18/870518/1"}}},"commit":{"parents":[{"commit":"b0379d6a8411ae65413579bdd7927bfa07e51b5b","subject":"Merge \"Fix plugging member subnets on existing networks\" into stable/wallaby","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b0379d6a8411ae65413579bdd7927bfa07e51b5b"}]}],"author":{"name":"Anton Kurbatov","email":"Anton.Kurbatov@acronis.com","date":"2022-09-28 18:48:45.000000000","tz":0},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-01-16 09:42:49.000000000","tz":60},"subject":"Fix full graph loadbalancer creation if jobboard is disabled","message":"Fix full graph loadbalancer creation if jobboard is disabled\n\nThis patch fixes the creation of a full loadbalancer graph in case the\njobboard_enabled config value is False, when no \"listeners\" parameter was\npassed to the get_create_load_balancer_flow function call: use kwargs as the\nflow creation function parameters and handle \"store\" separately.\n\nStory 2010335\nTask 46462\n\nSigned-off-by: Anton Kurbatov \u003cAnton.Kurbatov@acronis.com\u003e\nChange-Id: Id328b27db1ec9c8e17bf18120259e41e75dab3b9\n(cherry picked from commit 216cce39c9c22831d725703f41c54b9305d65dc8)\n(cherry picked from commit c662e42b849aa962a814fe756e109950e794277c)\n(cherry picked from commit a0ea8a900ca53a1e2d421e9a7defaf0d0f3354d5)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/2c80f89bb6d96240676bade01f2a9ea04481796c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/2c80f89bb6d96240676bade01f2a9ea04481796c"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
