)]}'
{"id":"openstack%2Foctavia~871468","triplet_id":"openstack%2Foctavia~master~I2df8a49851feb1445b5128ce99b880ddb77782ad","project":"openstack/octavia","branch":"master","attention_set":{},"removed_from_attention_set":{"29244":{"account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"last_update":"2023-01-24 19:23:57.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I2df8a49851feb1445b5128ce99b880ddb77782ad","subject":"Filter out details from taskflow logs with v2+jobboard","status":"MERGED","created":"2023-01-23 12:52:04.000000000","updated":"2023-01-24 19:25:11.000000000","submitted":"2023-01-24 19:23:57.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":3,"has_review_started":true,"submission_id":"871468","meta_rev_id":"26f477d2994cea0ace074261d0c7e9e2e4a56e80","_number":871468,"virtual_id_number":871468,"owner":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-01-24 19:23:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":2,"date":"2023-01-24 17:42:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-01-23 13:51:34.000000000","_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":1,"date":"2023-01-24 17:42:41.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-01-23 13:51:34.000000000","updated_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"reviewer":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"state":"REVIEWER"},{"updated":"2023-01-23 16:41:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-01-24 17:42:41.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"}],"messages":[{"id":"5a3df6aff0c87eb67c94314960adcc841a9e8dea","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-01-23 12:52:04.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4cf52ea4d9847c13cf59db9f8984bc9e087027ad","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-01-23 13:48:01.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"0a8350662d74e9de682ac6bf98f54965b565096a","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2023-01-23 13:51:34.000000000","message":"Patch Set 2: Code-Review+2\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"fd48fbfb68e17f66cb9044bbdf72e85d05351f9c","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-01-23 14:32:07.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"aa037d0a3c7723ebdb15d5959cc5e5f28c503dce","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-01-23 14:35:17.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"71007a9f3a348c4b0cccd61ba9419147d2ee1eed","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-23 16:41:27.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d4f134ef4028461ab5603d658380615b\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fba89a54c5654c76b015a4d84ceb9a0f : SUCCESS in 14m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c35b54ca504b4c93a283390f84fcf0f2 : SUCCESS in 5m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/83af0493491445e0871d262065b9c958 : SUCCESS in 4m 17s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5e6941db9b5a491e989109c6433ade2f : SUCCESS in 5m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3e4eb9ff02684198ac4e0ce8cd4990df : SUCCESS in 11m 53s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ded5699da7d041f9a88dd32739266fda : SUCCESS in 4m 54s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/b831887f17ba4c7b84e1e656da0ef171 : SUCCESS in 5m 03s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/c27576040042449ab237bee086865487 : SUCCESS in 7m 34s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/5c3c50a86ba640fea4abae6f42207aa3 : SUCCESS in 5m 26s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/cf669e47e6de4a798ad47152c8eacd25 : SUCCESS in 7m 21s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/cd1a557101a546e3bcc4102dcc80695d : SUCCESS in 7m 22s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/b99ea41bd187409a8470c499ae8a7d25 : SUCCESS in 7m 45s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/92714166cda94e0d8f959cd6fd9c4198 : SUCCESS in 1h 30m 43s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/f2be078f517d4a22a8115c4a3cf4db7b : SUCCESS in 2h 30m 13s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/2e8470933aab4f9787f31da85a07119e : SUCCESS in 40m 18s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/3804f21fb7a3463e98c0908c0d1cf6b6 : SUCCESS in 1h 13m 19s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/c389e0da6c964ca5934ba27f771eec02 : SUCCESS in 1h 17m 20s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/739af3c4b5a04541aade5bd0fbefa4a0 : POST_FAILURE in 57m 19s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/8c95e31f431c4121a73973e3ff2fa40b : SUCCESS in 2h 11m 34s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/13add2c2d62849ce8ad8f143f52726c7 : SUCCESS in 2h 00m 23s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/eaef8e96f71c4608b5bb205990a60cd8 : TIMED_OUT in 2h 33m 26s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/e749cfc3a3494a658770beb81b9f672b : TIMED_OUT in 2h 32m 54s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/74f217888c4b4c75ac643df33ed39988 : SUCCESS in 1h 49m 00s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/6a7c5d43fb2e4a3b8538f737143778e8 : NODE_FAILURE Node request 300-0020252482 failed in 0s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"c07e99ca9b46920cf3cbb0bc944cc51ee00ba773","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-01-24 17:42:41.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"b123d401f99865f80b5fe46f9aa7fd87acdc64bb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-24 17:43:28.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"5dfc6b6a127cb181b6906f095604c2a0346b72ba","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-24 19:23:57.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4cb31107fc4043239f2c4a146b67526d\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e649bf75510845f580118639c146592c : SUCCESS in 6m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3698e833195f4e5f96d89e6899891254 : SUCCESS in 4m 27s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/03ffd684adc24636a20be95765f366cb : SUCCESS in 4m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/590e6728934e48ddac90d75eba628897 : SUCCESS in 10m 45s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5006b1ca36b4417e8d27c483ac6235a0 : SUCCESS in 4m 25s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/881bc379b1db4e64beb9e8cc31dfdd1d : SUCCESS in 4m 55s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/57d305606add4b608dbb42e4c2b7c6da : SUCCESS in 8m 55s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/837b3aaa318747a4bd666a96d5009b65 : SUCCESS in 1h 21m 22s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/5734c3851aa848f7be118f59aa296248 : SUCCESS in 1h 39m 40s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/1fd4b2f173fa4fa1814e3a0b38b5c684 : SUCCESS in 35m 34s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/1b6b0cad805748148041edb45d06b172 : SUCCESS in 51m 39s","accounts_in_message":[],"_revision_number":2},{"id":"0730ac74e09da820aef6f8eb32d2ea6181352e7f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-24 19:23:57.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"26f477d2994cea0ace074261d0c7e9e2e4a56e80","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-24 19:25:11.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9abe850779f74ab68aa3dac636240581\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/959d0b6787c042cb9cc37ad2e7c85a41 : SUCCESS in 54s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d93e05e30b2c4adcaaf1593daa27d730 : SUCCESS in 40s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"6c731fa2fdcb4f316fbf65b8d89b24f4c4f277bc","revisions":{"13ae04fa8cbaadfc664d1e1d2a78849acd6313ff":{"kind":"REWORK","_number":1,"created":"2023-01-23 12:52:04.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/68/871468/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/68/871468/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/68/871468/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/68/871468/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/68/871468/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/68/871468/1"}}},"commit":{"parents":[{"commit":"eea6ac62fd29616ffc6a8c6b5c8d1e9ca0efb734","subject":"Merge \"Optimize HAProxy SSL cache size\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/eea6ac62fd29616ffc6a8c6b5c8d1e9ca0efb734"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-01-22 07:41:26.000000000","tz":-300},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-01-23 07:54:22.000000000","tz":-300},"subject":"Filter out details from taskflow logs with v2+jobboard","message":"Filter out details from taskflow logs with v2+jobboard\n\nWhen enabling INFO-level logs in taskflow and using jobboard with\namphorav2, taskflow prints the string representation of a job when it is\ncompleted. It includes the parameters of the flow, which might include\nprivate information from TLS-enabled listeners and pools such as\ncertificates, private_key and intermediate certificates.\n\nThis commit filters out the private information from the logs by using\nlogging.Filter, it replaces private attributes with \u0027***\u0027.\n\nStory 2010523\nTask 47125\n\nChange-Id: I2df8a49851feb1445b5128ce99b880ddb77782ad\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/13ae04fa8cbaadfc664d1e1d2a78849acd6313ff"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/13ae04fa8cbaadfc664d1e1d2a78849acd6313ff"}]},"branch":"refs/heads/master"},"6c731fa2fdcb4f316fbf65b8d89b24f4c4f277bc":{"kind":"REWORK","_number":2,"created":"2023-01-23 13:48:01.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/68/871468/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/68/871468/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/68/871468/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/68/871468/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/68/871468/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/68/871468/2"}}},"commit":{"parents":[{"commit":"eea6ac62fd29616ffc6a8c6b5c8d1e9ca0efb734","subject":"Merge \"Optimize HAProxy SSL cache size\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/eea6ac62fd29616ffc6a8c6b5c8d1e9ca0efb734"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-01-22 07:41:26.000000000","tz":-300},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-01-23 13:28:50.000000000","tz":-300},"subject":"Filter out details from taskflow logs with v2+jobboard","message":"Filter out details from taskflow logs with v2+jobboard\n\nWhen enabling INFO-level logs in taskflow and using jobboard with\namphorav2, taskflow prints the string representation of a job when it is\ncompleted. It includes the parameters of the flow, which might include\nprivate information from TLS-enabled listeners and pools such as\ncertificates, private_key and intermediate certificates.\n\nThis commit filters out the private information from the logs by using\nlogging.Filter, it replaces private attributes with \u0027***\u0027.\n\nStory 2010523\nTask 47125\n\nChange-Id: I2df8a49851feb1445b5128ce99b880ddb77782ad\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/6c731fa2fdcb4f316fbf65b8d89b24f4c4f277bc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/6c731fa2fdcb4f316fbf65b8d89b24f4c4f277bc"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[]}
