)]}'
{"id":"openstack%2Foctavia~871683","triplet_id":"openstack%2Foctavia~stable%2Fxena~I35c5976c6bdb8828e54bcde00052622a7b6bc96a","project":"openstack/octavia","branch":"stable/xena","attention_set":{},"removed_from_attention_set":{"29244":{"account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"last_update":"2023-02-09 06:54:08.000000000","reason":"\u003cGERRIT_ACCOUNT_29244\u003e replied on the change","reason_account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}}},"hashtags":[],"change_id":"I35c5976c6bdb8828e54bcde00052622a7b6bc96a","subject":"Fix PING health-monitor with recent haproxy releases","status":"MERGED","created":"2023-01-25 08:06:20.000000000","updated":"2023-02-09 11:22:22.000000000","submitted":"2023-02-09 11:21:14.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"871683","meta_rev_id":"a96771104f940d37ed1836e2354fea7dc42281f9","_number":871683,"virtual_id_number":871683,"owner":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-02-09 11:21:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":2,"date":"2023-02-07 15:43:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-01-31 10:36:59.000000000","_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":1,"date":"2023-02-07 15:43:05.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-01-25 10:29:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-01-31 10:36:59.000000000","updated_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"reviewer":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"state":"REVIEWER"},{"updated":"2023-02-07 15:43:05.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"}],"messages":[{"id":"0b7f602b6d61f671740bc849da49a24e608ddbb5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-01-25 08:06:20.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2a0d0e81c08223d357e2a816fe3abbe2ac715390","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-25 10:29:52.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/af7fa3f10aec44139721139ef4519d38\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/56d286ad1b0c45e29fb5552a68a53d08 : SUCCESS in 9m 34s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a8995ec277374010a87feb58772d2447 : SUCCESS in 8m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/76d82027f5954852b76f26d22df5d931 : SUCCESS in 5m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b985f7285b814d71948b5b5b5a602845 : SUCCESS in 4m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/03a0af27fa2442899cfa29b68bd7a5f0 : SUCCESS in 5m 06s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/75c7b317e5234795b9a5620352b46bdd : SUCCESS in 5m 15s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/398092b6beca45ae9c0739e5b89d05ad : SUCCESS in 11m 51s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9e10489ddc214ad493bc6fcd453b9dfa : SUCCESS in 4m 45s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/0d61b1708147457aa8a38e8f91c873d2 : SUCCESS in 4m 54s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/12dd11ef051449559afa493462c02010 : SUCCESS in 9m 09s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/8638195bb5a24ae8a460217bfadd7554 : SUCCESS in 4m 25s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/7015294933d944aba0a517f58d27a925 : SUCCESS in 10m 07s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/17dc689ed537450f9cb96afd2d784b35 : SUCCESS in 1h 28m 16s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/6b8a7a4b3c144e65b5a70978d54736c6 : SUCCESS in 2h 22m 01s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/2124faa17dcb4b35a70569c68e3e628c : SUCCESS in 39m 19s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/539a3a43e8c84ddd95d72bc71204a877 : SUCCESS in 47m 41s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/6ddfcaccee7149bda9f5317b12c90b4d : FAILURE in 10m 51s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/4f30587d175349c488a93a99027a568f : SUCCESS in 30m 00s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/d1454bea7c6245e0915461e3f76920dd : SUCCESS in 1h 56m 06s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/3214300cddcb411ab830db61d6d2b284 : SUCCESS in 2h 11m 33s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/923218279e894bf787f2742fe588fa1f : SUCCESS in 1h 50m 34s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/32e83aa61b8843038f757d3903f8bb9c : SUCCESS in 1h 53m 18s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/12556fceb97b43eeb4f9cf0af0d7bb8e : SUCCESS in 1h 44m 39s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/ebec2d693f7644768269be59ed7c74de : NODE_FAILURE Node request 300-0020271197 failed in 0s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"fecb6c310b88478503aaffb76ec61e48c6170cc5","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2023-01-31 10:36:59.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"eaa3c95275bfcde0529ade683fbb0988f1b0f28d","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-02-07 15:43:05.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f6741de7c62d81195e8618fb9bc7bb492f1b17b3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-07 15:44:03.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"185196b1982879914b3767703bacf2a4bface4b0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-07 19:04:34.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/58060930bc4143e9846255882e80e261\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1d5e5c979b984d309d772a4a3fc4c9da : SUCCESS in 9m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/150ec68fb8ff42f9a81746aca54445a4 : SUCCESS in 4m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0de446fab08f4bbf86e9df995594de57 : SUCCESS in 6m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7317a5ebe7ca4a319c253d4cb68ae836 : SUCCESS in 5m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d0ea4ddddb634170bdc2d4c56359e0a9 : SUCCESS in 15m 37s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a2e1cad0fa2343e6a6c3ae1192e6ff53 : SUCCESS in 6m 58s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/867043425b2446cbb6cc081a9580b91d : SUCCESS in 4m 29s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/6f66ff8c82f747c4a11921d69c8f7012 : FAILURE in 12m 14s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/6b4125502047438db55897c9bbcd019a : CANCELED\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/f4b7d6af19d84296ba0da0c1952b0b86 : CANCELED\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/7e1d7b6e817d47a1a7e5b38146764158 : CANCELED\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/8fb185e372fc4d1096598e5d6e14d87b : CANCELED","accounts_in_message":[],"_revision_number":1},{"id":"3e3340ec162f92bb3ce2923c0c1b162c27a189b8","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-02-09 06:54:08.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"b708f9cec412ff7ccf2599a4652b754ac201bac5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-09 09:20:19.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9db16097e1754cbc82ebdad57ae9003d\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2a31a1c392b242bb8313ce446f34d269 : SUCCESS in 12m 26s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/11380e0461b64abea96c98813b819b7f : SUCCESS in 7m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b7d6111cfdff4444a1984e52b5dede38 : SUCCESS in 5m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/73cd17d1643f422cb41af551e8223687 : SUCCESS in 5m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0a11d7bd44cd467ca5642b552695d704 : SUCCESS in 5m 32s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/affd4b10016340eb9a596a33a98fe8e4 : SUCCESS in 4m 58s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b89809ac034d4c30a995bb0c6c05c399 : SUCCESS in 10m 59s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ef5fb18c128b429b8b19f80dab52bce1 : SUCCESS in 5m 25s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/5eb07bb423cf4e4084f785854b2f74dc : SUCCESS in 5m 05s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/81d22c39259d4ba4bdf2ff9bbe27f0f4 : SUCCESS in 9m 01s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/a7bff4ac42e945cc8218a271e5e2fec2 : SUCCESS in 4m 51s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/6822db4adce74ce785ccccd658d39782 : SUCCESS in 9m 02s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/201c2bc70f964e7b922a4461271f0161 : SUCCESS in 1h 02m 33s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/616c18b64575476dbcb7eaf1c21d14b6 : SUCCESS in 1h 31m 32s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/53716b7b788d48dabcc48f5c0a0db24c : SUCCESS in 35m 49s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/f47ee23b69014866a02e4af22258ee70 : SUCCESS in 38m 15s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/706c73e460cc4ce6b5d61ff9ec5ae6af : FAILURE in 13m 31s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/c9075ac958c64b12a3136e0ab68e821b : SUCCESS in 30m 56s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/9740cd1fcd6c4fac9f6c6526b346f7a1 : SUCCESS in 1h 56m 46s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/922b6be073674f11b344c9fd358914e4 : SUCCESS in 2h 24m 20s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/92a143f0cb1e41188cf36d04491d0ac3 : SUCCESS in 1h 47m 28s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/bd43f7072ebc497bbb78908386d1d90f : SUCCESS in 1h 43m 50s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/48a8d44442274e66937598c5dd39aef1 : SUCCESS in 1h 34m 03s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/1b0d3454a23b4c34939a6c08b27e7e54 : NODE_FAILURE Node request 300-0020403110 failed in 0s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b910b6eaf8a31b75df88c38457d9e2bfee2acbf6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-09 09:21:09.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"6bc11a01aa270922752135fe551f03eea91932c0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-09 11:21:13.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e3264da2b5d648148874c7b3e95224d1\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ed521452f101474e9e2574008d734995 : SUCCESS in 6m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b6132487468245cbb843f91542ef0719 : SUCCESS in 4m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ad6e79dafba540f495b271cadae976c1 : SUCCESS in 4m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bedd8f9d08b1443582b4ec167a0f1688 : SUCCESS in 4m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/af820def836c428581773f72ca66b372 : SUCCESS in 9m 39s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0baa6f4ef54644ccafc75b423e90e135 : SUCCESS in 4m 22s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/eb28a18c5c3348b0ad14fdaa14038166 : SUCCESS in 4m 54s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/5d915698832d44f891ddf79c394d8f01 : SUCCESS in 7m 59s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/4acfa754158e49ca8e69b019626788ba : SUCCESS in 1h 56m 49s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/936095ca26394c638a2a008f33d95092 : SUCCESS in 1h 34m 28s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/fdd95b5c74374f96aa75346d60de1138 : SUCCESS in 29m 58s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/a9edd93dd8df4673ab886c203bbade68 : SUCCESS in 40m 55s","accounts_in_message":[],"_revision_number":1},{"id":"1a6ec5308d1a83d228d0a02844236f1ff3fde04b","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-09 11:21:14.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"a96771104f940d37ed1836e2354fea7dc42281f9","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-09 11:22:22.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a68e39e190434cea902be64ea4d2fd4c\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/49b8c8fb45ce4f0f950af77ca4dd871c : SUCCESS in 46s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/20e7b132f49944e9a5aa8155d8d5014c : SUCCESS in 38s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"ad65672f1f1b1c51c7614af5eaf0cae91dd0c785","revisions":{"ad65672f1f1b1c51c7614af5eaf0cae91dd0c785":{"kind":"REWORK","_number":1,"created":"2023-01-25 08:06:20.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/83/871683/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/83/871683/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/83/871683/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/83/871683/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/83/871683/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/83/871683/1"}}},"commit":{"parents":[{"commit":"c0b6f5c9ef07f777b10d952a70638c3e00e21c05","subject":"Merge \"Fix plugging member subnets on existing networks\" into stable/xena","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c0b6f5c9ef07f777b10d952a70638c3e00e21c05"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2022-04-01 04:56:54.000000000","tz":120},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-01-25 07:56:20.000000000","tz":60},"subject":"Fix PING health-monitor with recent haproxy releases","message":"Fix PING health-monitor with recent haproxy releases\n\nhaproxy 2.2.x requires an \"insecure-fork-wanted\" global option when\nusing external-check.\n\nThis commit also fixes a function that builds the list of the features\navailable in haproxy based on its version.\n\nStory 2009953\nTask 44900\n\nConflicts:\n\toctavia/common/jinja/haproxy/combined_listeners/jinja_cfg.py\n\toctavia/tests/unit/common/jinja/haproxy/combined_listeners/test_jinja_cfg.py\n\nChange-Id: I35c5976c6bdb8828e54bcde00052622a7b6bc96a\n(cherry picked from commit 1ac781812872d4219cf245cf18ca3e93ec1bb035)\n(cherry picked from commit d4d8d597d53dee95f0cc589d5cad07acdeeb10f1)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/ad65672f1f1b1c51c7614af5eaf0cae91dd0c785"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/ad65672f1f1b1c51c7614af5eaf0cae91dd0c785"}]},"branch":"refs/heads/stable/xena"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[]}
