)]}'
{"id":"openstack%2Foctavia~871684","triplet_id":"openstack%2Foctavia~stable%2Fwallaby~I35c5976c6bdb8828e54bcde00052622a7b6bc96a","project":"openstack/octavia","branch":"stable/wallaby","attention_set":{},"removed_from_attention_set":{"29244":{"account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"last_update":"2023-02-09 06:55:51.000000000","reason":"\u003cGERRIT_ACCOUNT_29244\u003e replied on the change","reason_account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}}},"hashtags":[],"change_id":"I35c5976c6bdb8828e54bcde00052622a7b6bc96a","subject":"Fix PING health-monitor with recent haproxy releases","status":"MERGED","created":"2023-01-25 08:07:02.000000000","updated":"2023-02-09 11:22:31.000000000","submitted":"2023-02-09 11:21:20.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"871684","meta_rev_id":"e18f4983e0d844d0fc3f77191e108c9607fee6d0","_number":871684,"virtual_id_number":871684,"owner":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-02-09 11:21:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":2,"date":"2023-02-07 15:42:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-01-31 10:39:45.000000000","_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":1,"date":"2023-02-07 15:42:47.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-01-25 10:23:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-01-31 10:39:45.000000000","updated_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"reviewer":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"state":"REVIEWER"},{"updated":"2023-02-07 15:42:47.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"}],"messages":[{"id":"cf6d88b1e7884147be5d92d46d70e07827f6e2f7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-01-25 08:07:02.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"bb9524492b0ac7f84cdb01478661ee8367a4d90f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-25 10:23:31.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/72bb5c740bea4a9f866a3d2e99424694\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3872c2bc8418492a82b67fb1ad9325fb : SUCCESS in 11m 35s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/87c33bc6b3f049e0b5d5ea8288b24d37 : SUCCESS in 6m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/142f6d1bfe364418ae75cf88e43fcfb8 : SUCCESS in 3m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a05efd88904b4780bfeec4e52fd3597f : SUCCESS in 4m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7e001bc30a414df48c7bf8f38562a19b : SUCCESS in 4m 19s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2b63d7b6236b4f648d509396b82fd4d1 : SUCCESS in 4m 41s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9291a90f2b994b20be5440b682abf84f : SUCCESS in 10m 04s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/594b3f77d0e5431f841c5d3aa77f66e5 : SUCCESS in 4m 44s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/caf0485134cd4f66b0bf9f04af50049e : SUCCESS in 4m 35s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/bce3c122d3764e2b94f750ec8fb89cf4 : SUCCESS in 7m 16s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/9b04e7c6fec747a796a6370b6d03a017 : SUCCESS in 4m 18s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/eacb3a603427416984f2880f5f420dfe : SUCCESS in 8m 53s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/9b84fc7d6a41417f8fc31b66dcb210cc : SUCCESS in 1h 01m 29s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/e4ef18a69da74f859d7d3e034eb7a761 : SUCCESS in 1h 30m 48s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/e9d44798efca4f8abc062c6ebdfa6a1e : SUCCESS in 48m 57s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/92b4c81b6a784500ab6c662abbc84f39 : SUCCESS in 47m 05s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/889e03780a81420e86ff2615f306c12b : SUCCESS in 30m 46s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/b7c803a6ee4943d294bbfa621514ad9a : NODE_FAILURE Node request 300-0020271216 failed in 0s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/d4f45446d48c4735ba459e99228456c7 : SUCCESS in 1h 53m 57s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/4b446db3f55f495cb04f93822be5d12e : FAILURE in 28m 48s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/d8a721421edb41859d525938640ad672 : SUCCESS in 2h 14m 47s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/037656d37a0b41b1a078bbfc797e5c56 : SUCCESS in 1h 44m 25s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/1e86282de2ac45f591f9669d01503084 : SUCCESS in 1h 59m 34s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/4895a70661104878b529a686572b6212 : NODE_FAILURE Node request 300-0020271222 failed in 0s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b95f5ba76c3c8d2b8f64b91e4c2f02fe2ec81bd1","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2023-01-31 10:39:45.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"f79f30a01d6080a02774564154534f5d55dc0ddd","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-02-07 15:42:47.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"eaf995a8e52d176ebe7c2f7a40c51efea461e31d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-07 15:43:32.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"810a345977b155d07ec24dc14df32064dc4250b9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-07 18:19:10.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/3fe5556e723d4e1b9f32134308b30bf5\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/bdfda5e1e5864bdf9d886e9eaf1f11f9 : SUCCESS in 8m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a239c8231fec46ce9af54bdee27f40d7 : SUCCESS in 4m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3c2a081d89c1496cb19ef720ae0486bc : SUCCESS in 4m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6fad673c7c054adf93667f8a7b87c069 : SUCCESS in 4m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/008c29e251994de7b8d36b5e9dc1f226 : SUCCESS in 10m 39s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/56b5e93be1ca49cc8eae87982a4ee6f8 : SUCCESS in 4m 36s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/13bc5412966c4f5e8ffc8fcad70317ac : SUCCESS in 5m 17s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/06739b3f2fa2419ea5f1e7fc7e27e1b7 : SUCCESS in 7m 50s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/a5ff78d44609407586597ec788d041a8 : SUCCESS in 1h 19m 38s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/cc56d541c738442ca51afd6c3363d9ce : TIMED_OUT in 2h 33m 56s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/cd0a082033a046d6832919e28dae19c1 : SUCCESS in 46m 22s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/f53e43b8f6c0460387b3e8cb1ce83f25 : SUCCESS in 29m 51s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/150411c557f746daaada444b3da1bb50 : SUCCESS in 31m 01s","accounts_in_message":[],"_revision_number":1},{"id":"fd219d70eb266c282742611b3e6631ad480bf47a","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-02-09 06:55:51.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"23a3d6b353feb0967172c8ffff69efe0a70368c7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-09 09:42:20.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4263924e988c4989abe67a29d48b51d8\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/508e8fc1ac5c4e499eb1f327f23f2141 : SUCCESS in 8m 21s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/53165ee63f8245eea9627c8a5a45a1c2 : SUCCESS in 7m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2b6ef471ea054d088b524195aeec019a : SUCCESS in 5m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b1fd93442146415395fc8b5212d4aaf2 : SUCCESS in 4m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/532848234ce8422d986b237091e3feea : SUCCESS in 4m 10s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9895fce4e630483892b5bf08323c475d : SUCCESS in 5m 08s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a2eab03a9fcc47a18a27ef44f55c0e9a : SUCCESS in 11m 03s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0ad99461d19740f696e0a0d89aebf31f : SUCCESS in 4m 09s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/7d2f4f76905e4255b925156a70fae9c7 : SUCCESS in 4m 59s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/c15fb85627c44cc09c44ded466511227 : SUCCESS in 6m 47s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/560bc272b54246848827046bbbe9e5de : SUCCESS in 4m 45s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/c07ec68083d344bb9a74c1fb68882c60 : SUCCESS in 7m 51s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/007a1021775545eab2d260c0ea125589 : SUCCESS in 1h 00m 25s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/195069e0825141318bbd5bd337a3f5b4 : SUCCESS in 1h 24m 38s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/91e4c0cf51b746eebe54578ce8e8bdee : SUCCESS in 32m 49s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/a598a7251cb549a1bbf87d5cd03def20 : SUCCESS in 30m 01s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/6140a79198264d7bb7541d1f4e00b366 : SUCCESS in 31m 50s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/f74a87a51d4b45979ede8a1a7235f19f : FAILURE in 12m 20s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/77edbeba1ff54036856cd5bb5eaef16b : SUCCESS in 2h 12m 55s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/d2388c92bf4f460580dcda7b2cc0957d : FAILURE in 36m 16s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/4aeca2b9661d4ed6a618acc4a2a16b85 : SUCCESS in 2h 14m 32s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/56235efa5e5c4bd7a9453c1ad954e3f1 : TIMED_OUT in 2h 33m 37s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/4593af63eb2f4bd7b602158ceb67e566 : SUCCESS in 1h 39m 35s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/2b8ce12d653641d28946b43c8da0e6eb : NODE_FAILURE Node request 300-0020403193 failed in 0s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"833d19e42c0a271a940d802f2f7c792af1dca112","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-09 09:43:11.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"9584817734d52aa9889d45e0a3ed3a0d0d0362f7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-09 11:21:20.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e2e1a3f470334223a5c2ee499f0f0ef4\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/102b5c5773b64131834a0eaf2276f623 : SUCCESS in 6m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0c67f76fb7c04e7a9fbae8b8d0c3c0d9 : SUCCESS in 4m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8f7961e926f24f7da99d4a403f638798 : SUCCESS in 4m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b0de9fa0161c4a1485c2e66a8fdc4795 : SUCCESS in 4m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8341d06862b2489f8890b9ba1029e5a3 : SUCCESS in 10m 34s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/50582bfa7a1649f2a1dcbcb47205b440 : SUCCESS in 4m 37s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/40aa374c966f41abb45ca87c44ad6a9f : SUCCESS in 4m 19s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/da2836cd5964449ea94b43ec72fbaff5 : SUCCESS in 7m 39s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/e6098596f2434f5daa9a87455daa0139 : SUCCESS in 1h 09m 54s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/018650712352433bacab125e925ec15b : SUCCESS in 1h 29m 16s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/5362cab99350461e877822d6990112a3 : SUCCESS in 35m 00s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/2073944151d848d8ac714b054008b1c6 : SUCCESS in 31m 36s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/b1a90d1d8877487dba601b0c7378a703 : SUCCESS in 31m 53s","accounts_in_message":[],"_revision_number":1},{"id":"ba6b84e7555fc1ced8a65f700907f77280c3c7e4","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-09 11:21:20.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"e18f4983e0d844d0fc3f77191e108c9607fee6d0","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-09 11:22:31.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b5aa1ed529d64660af9ed821ef15b8bd\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/750e61532a3a44538d4a6ef9fb8393e3 : SUCCESS in 47s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3bec1841362e4997bd6d7f6247c87fe5 : SUCCESS in 39s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"55275a2e5f82b7e3106ef5ff4fc628171fdcd325","revisions":{"55275a2e5f82b7e3106ef5ff4fc628171fdcd325":{"kind":"REWORK","_number":1,"created":"2023-01-25 08:07:02.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/84/871684/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/84/871684/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/84/871684/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/84/871684/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/84/871684/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/84/871684/1"}}},"commit":{"parents":[{"commit":"b0379d6a8411ae65413579bdd7927bfa07e51b5b","subject":"Merge \"Fix plugging member subnets on existing networks\" into stable/wallaby","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b0379d6a8411ae65413579bdd7927bfa07e51b5b"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2022-04-01 04:56:54.000000000","tz":120},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-01-25 08:06:30.000000000","tz":60},"subject":"Fix PING health-monitor with recent haproxy releases","message":"Fix PING health-monitor with recent haproxy releases\n\nhaproxy 2.2.x requires an \"insecure-fork-wanted\" global option when\nusing external-check.\n\nThis commit also fixes a function that builds the list of the features\navailable in haproxy based on its version.\n\nStory 2009953\nTask 44900\n\nConflicts:\n\toctavia/common/jinja/haproxy/combined_listeners/jinja_cfg.py\n\toctavia/tests/unit/common/jinja/haproxy/combined_listeners/test_jinja_cfg.py\n\nChange-Id: I35c5976c6bdb8828e54bcde00052622a7b6bc96a\n(cherry picked from commit 1ac781812872d4219cf245cf18ca3e93ec1bb035)\n(cherry picked from commit d4d8d597d53dee95f0cc589d5cad07acdeeb10f1)\n(cherry picked from commit ad65672f1f1b1c51c7614af5eaf0cae91dd0c785)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/55275a2e5f82b7e3106ef5ff4fc628171fdcd325"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/55275a2e5f82b7e3106ef5ff4fc628171fdcd325"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[]}
