)]}'
{"id":"openstack%2Foctavia~871687","triplet_id":"openstack%2Foctavia~stable%2Fxena~I499bdd5edc23b602495a9b2b9f22baf9547261d5","project":"openstack/octavia","branch":"stable/xena","attention_set":{},"removed_from_attention_set":{"29244":{"account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"last_update":"2023-01-31 12:43:24.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I499bdd5edc23b602495a9b2b9f22baf9547261d5","subject":"Fix PING health-monitor with recent Centos releases","status":"MERGED","created":"2023-01-25 08:11:16.000000000","updated":"2023-01-31 12:44:41.000000000","submitted":"2023-01-31 12:43:24.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"871687","meta_rev_id":"d2d479b71005ac2cd62789a3eb20b320f88ed160","_number":871687,"virtual_id_number":871687,"owner":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-01-31 12:43:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":2,"date":"2023-01-25 15:28:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-01-31 10:31:05.000000000","_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-01-31 10:31:05.000000000","_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-01-25 10:57:23.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-01-25 15:28:25.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2023-01-31 10:31:05.000000000","updated_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"reviewer":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"state":"REVIEWER"}],"messages":[{"id":"b372d128e52364cce92c7b78e1845ede2e37855e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-01-25 08:11:16.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a9c06f83bd4f3296364bb26302b8d19feeab854e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-25 10:57:23.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f0c7f7b5c75f4048aecbef5f7d3aac37\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cac73ecfdfd346c794423cc237c2db47 : SUCCESS in 9m 43s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1b34c52b3c5f4efc85b64137646394bf : SUCCESS in 8m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4f8c8f1e336f42a4996a225012d7edcd : SUCCESS in 4m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9826e8315ada4b86acfe92cf09b70f79 : SUCCESS in 5m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3197de2ab55e4fabaf37eff02515b770 : SUCCESS in 4m 04s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f161cb2ebb80412597f266efdc8b5a35 : SUCCESS in 5m 21s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5204646f2d91438d8a94523bdbce99c2 : SUCCESS in 11m 54s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4a36a225c5514bb8ab6e2a2394067c26 : SUCCESS in 5m 00s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/2bbf2fef59a34103b2a3f7bc7edf6209 : SUCCESS in 4m 59s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/2d4143f02c62412dbdf2c72cb2ce53f9 : SUCCESS in 8m 43s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/da2b7bca95f44e36b1dc3e87d8a980ed : SUCCESS in 4m 46s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/8d4dc09404a34afcb762578548867eba : SUCCESS in 7m 54s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/0adaed80edd94f51889b5850b3b5270e : SUCCESS in 1h 32m 14s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/966c36f677054fe585745f084f4a75a8 : SUCCESS in 2h 20m 50s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/917a4dc26a724cfb8354e077ea2691c8 : SUCCESS in 31m 50s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/475a356c4dde4b8095682e38f7757f42 : SUCCESS in 47m 28s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/6b2006655d3d4bf690f78a7458758457 : FAILURE in 11m 32s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/22433c0c88e547ffb02fc51401d5cf65 : SUCCESS in 29m 20s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/746544b7d889455c9b3f8d841e3b8d53 : TIMED_OUT in 2h 33m 09s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/a708df3fbba14bfb9d4643640206046e : SUCCESS in 2h 28m 44s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/f613a14b76454a0fb646ee659b6ff6d8 : TIMED_OUT in 2h 33m 12s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/83eae72c6d7a44acbb6d5fd816b70f9e : SUCCESS in 1h 42m 27s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/a31cd56144664246b7071e5528eba5c2 : TIMED_OUT in 2h 33m 06s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/5804796737004eaf8684be1dc5a13665 : NODE_FAILURE Node request 300-0020271316 failed in 0s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"2512087fa34a24d609d14ae91e0d072d9320f9a2","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-01-25 15:28:25.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"24a4de8dd46e5c8d96027de0202d469127809e7f","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2023-01-31 10:31:05.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"e81c54681061744caf0d4d238dbe457991bf638d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-31 10:32:10.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"9e17059f1e0ca7366fbd42990bce6d4c1be1ba76","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-31 12:43:23.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b2ca1f4ee85d4a6e8f7f9d23329d171a\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d736bd8461a04aa7a99f787a22a42ce3 : SUCCESS in 7m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/93994f3e66144d43aeb2525c77fd2576 : SUCCESS in 4m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f17d06b3c7704cd3ae5ce54f23313282 : SUCCESS in 5m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3f3a7b48d45245ab91da42879329b9ff : SUCCESS in 4m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/31ca5dac85d14dc8a127fea5092d210b : SUCCESS in 12m 02s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1c680784e2fc4aca94c0380de4ca889a : SUCCESS in 4m 42s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/2950104aa0a243f89751d7e81945bdeb : SUCCESS in 5m 13s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/fdf21a84e53e4509a74a08853d9fe213 : SUCCESS in 7m 48s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/afb9ae8e1a6149a782dd0bc93944b57c : SUCCESS in 1h 35m 49s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/e35a742a44d34ab1a67ea0d387630e0b : SUCCESS in 1h 31m 56s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/93fc39a8d2d440a4baca025594565617 : SUCCESS in 31m 23s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/16975764db974b21b977230b8310ab46 : SUCCESS in 45m 48s","accounts_in_message":[],"_revision_number":1},{"id":"41e42ab5a7a7cc17abe8ab3aa08410a013ce6ea8","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-31 12:43:24.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"d2d479b71005ac2cd62789a3eb20b320f88ed160","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-31 12:44:41.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ee20c7329b514fdc84e53f5c395e76f2\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9908615f55184b408bfc9674fd8ae7ae : SUCCESS in 51s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/89b496a5cb994418afed58822c94999a : SUCCESS in 49s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"71acdb46604287130c6d7fc9a376f8289267b93a","revisions":{"71acdb46604287130c6d7fc9a376f8289267b93a":{"kind":"REWORK","_number":1,"created":"2023-01-25 08:11:16.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/87/871687/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/87/871687/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/87/871687/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/87/871687/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/87/871687/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/87/871687/1"}}},"commit":{"parents":[{"commit":"c0b6f5c9ef07f777b10d952a70638c3e00e21c05","subject":"Merge \"Fix plugging member subnets on existing networks\" into stable/xena","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c0b6f5c9ef07f777b10d952a70638c3e00e21c05"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2022-04-01 04:56:54.000000000","tz":120},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-01-25 08:10:46.000000000","tz":60},"subject":"Fix PING health-monitor with recent Centos releases","message":"Fix PING health-monitor with recent Centos releases\n\nCentos 8 Stream removed the capabilities to open raw sockets from the\nping binary and added new sysctl default to allow an unprivileged user\nto send ping request.  This sysctl was not inherited by the\namphora-haproxy namespace. Add it explicitly to our sysctl list.\n\nStory 2009953\nTask 44911\n\nChange-Id: I499bdd5edc23b602495a9b2b9f22baf9547261d5\n(cherry picked from commit cbf79c5627e1888611f9212a539b56ceaba1bfeb)\n(cherry picked from commit 8cd697e8bf7b91d1d010a3968a440358e8e12dbb)\n(cherry picked from commit 2e5bc334942b873d6ba7926f45b71ac80d545a47)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/71acdb46604287130c6d7fc9a376f8289267b93a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/71acdb46604287130c6d7fc9a376f8289267b93a"}]},"branch":"refs/heads/stable/xena"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[]}
