)]}'
{"id":"openstack%2Foctavia~871689","triplet_id":"openstack%2Foctavia~stable%2Fxena~I00e327e1d94e46aa13a38120df9865ec34eaa593","project":"openstack/octavia","branch":"stable/xena","attention_set":{},"removed_from_attention_set":{"29244":{"account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"last_update":"2023-03-23 18:08:20.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I00e327e1d94e46aa13a38120df9865ec34eaa593","subject":"Add a newline when writing the server state file","status":"MERGED","created":"2023-01-25 08:35:57.000000000","updated":"2023-03-23 18:09:27.000000000","submitted":"2023-03-23 18:08:20.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"871689","meta_rev_id":"3a19fea74a9b58f7b8678920cfbbc3454100ba0b","_number":871689,"virtual_id_number":871689,"owner":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-03-23 18:08:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":2,"date":"2023-03-23 16:10:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-01-31 10:22:35.000000000","_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":1,"date":"2023-03-23 16:10:36.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-01-25 11:38:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-01-31 10:22:35.000000000","updated_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"reviewer":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"state":"REVIEWER"},{"updated":"2023-03-23 16:10:36.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"}],"messages":[{"id":"c8f08b68ec97ec47504f09d0be967ec087f28c96","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-01-25 08:35:57.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d2700791a26ea2f50b692c48eedcc3cbc8f9ee02","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-25 11:38:17.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/958350be234f443fbbddbc7bd4c13518\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cd965594f18a4aa4a3e269a8f2983d9d : SUCCESS in 9m 08s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/394c97ab7cb14cb6a54120114fc933a7 : SUCCESS in 8m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f7a12498d828416d8faf6c65001c9884 : SUCCESS in 4m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0298c60f1ddb400ba1f990d4a3b3d7ac : SUCCESS in 4m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5f98ac367d1246e1b5ca2fc368621ce4 : SUCCESS in 4m 53s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/484f0979a81e407e89ac92994b3924de : SUCCESS in 5m 03s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/772ea6748de34cf4a47ef4463de55475 : SUCCESS in 10m 42s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/aeccddbff23e44c19c93552cbcd4d848 : SUCCESS in 4m 29s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/ce92bb4f710b46d5bef3b53fe21e5305 : SUCCESS in 5m 01s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/b5c0ce388e7c41d8b1957b26898ec217 : SUCCESS in 9m 55s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/edfe50154eda478783875d755ff26e8f : SUCCESS in 4m 59s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/322226b9ede14a07a22a5b7b6485fe30 : SUCCESS in 7m 39s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/ca66aceda4574d249008f9180c43d33f : SUCCESS in 1h 25m 49s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/59d629fa731547c28a8a118c9293d2c0 : SUCCESS in 2h 26m 04s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/4066a6c9a41841f98cab69239684118b : SUCCESS in 37m 20s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/afd557fff6c04f1a8216b53f204347c5 : SUCCESS in 58m 03s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/4fa947336c4a4dea8078df910c00041b : FAILURE in 11m 22s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/1b77b60ab4584f0aaea5b489ebd1f659 : SUCCESS in 33m 00s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/c9891c4f3574494d9646b6c32adb13a6 : SUCCESS in 2h 31m 26s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/f2ee5843469f435b87c492629d3515bb : SUCCESS in 2h 57m 22s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/6895a02a4e6e4ae49fa284519972bf53 : SUCCESS in 2h 21m 35s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/3101d19d03464892a5987880335f2c79 : SUCCESS in 2h 06m 49s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/0f87f66d089a4aaa941eeeb98d3a35d9 : SUCCESS in 2h 00m 35s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/b1009d6e3ad54f60b4b8316005f84a77 : NODE_FAILURE Node request 300-0020271582 failed in 0s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"6a29684cf76508dc94088a9db9ec8cc60642d4e6","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2023-01-31 10:22:35.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"3580e5343bd1cd63b2c8f0b415b3fb64113bfb89","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-03-23 16:10:36.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"0e0eaaf94b3319b4ea91d26bbfa2b30511a1e722","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-23 16:11:24.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"299b23feeb1f24c14ac6f957ccbcdee156551a69","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-23 18:08:19.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a7dbe61f884942f3b8848944a0564fa6\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/dd49ef3832e74abfae2211ade4150eb5 : SUCCESS in 8m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5ea1dda0e16f43fda466080a8c0fcecb : SUCCESS in 5m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c5653122fea241258df490d44a0a2513 : SUCCESS in 5m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c023f0e1730d4156be0db9817ea6640a : SUCCESS in 5m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/311e1bbdd1554232ab1d6af3a7bec710 : SUCCESS in 15m 14s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c6610d1b58ce4786a15af68518365c5f : SUCCESS in 5m 13s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/9ead52201aa84ce0bf026cedcdc5edbf : SUCCESS in 5m 53s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/a7b290973bff496598870b053203015d : SUCCESS in 8m 22s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/8484cea9c62a49be848f74b3af3811b9 : SUCCESS in 1h 41m 43s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/f061af7cc1fb467f8bef0d9148445ee0 : SUCCESS in 1h 38m 36s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/d78aeb8e7e054f61bb4d67650da70fe1 : SUCCESS in 35m 01s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/35ae6a0cfb2b45efa0b1b90c7c487011 : SUCCESS in 52m 28s","accounts_in_message":[],"_revision_number":1},{"id":"0be5779726455078ff335bd1291f2ab7e30bcfec","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-23 18:08:20.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"3a19fea74a9b58f7b8678920cfbbc3454100ba0b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-23 18:09:27.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a5a8be2287de4445a02ade9717142363\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2d74189541034ada89b03357929a97cc : SUCCESS in 48s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/faba7f0a8232462c917f2f76c2ca5ae8 : SUCCESS in 43s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"353b55c58806a650af894aba1967b2e392bc221a","revisions":{"353b55c58806a650af894aba1967b2e392bc221a":{"kind":"REWORK","_number":1,"created":"2023-01-25 08:35:57.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/89/871689/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/89/871689/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/89/871689/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/89/871689/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/89/871689/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/89/871689/1"}}},"commit":{"parents":[{"commit":"c0b6f5c9ef07f777b10d952a70638c3e00e21c05","subject":"Merge \"Fix plugging member subnets on existing networks\" into stable/xena","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c0b6f5c9ef07f777b10d952a70638c3e00e21c05"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2022-11-17 16:56:03.000000000","tz":60},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-01-25 08:35:47.000000000","tz":60},"subject":"Add a newline when writing the server state file","message":"Add a newline when writing the server state file\n\nBefore restarting haproxy, the amphora-agent dumps the state of the\nservers in a file, so when haproxy reloads, it can recover the operating\nstate of its backends\u0027 members.\nBut with haproxy 2.4 (in Centos 9 Stream), it fails with a \"corrupted\nglobal server state file\" error when reading this file. It seems that\nmost recent haproxy releases have a more strict validation of the format\nof the file and ensure that it is terminated with a newline.\n\nThis commit adds a newline to the file (all the responses to haproxy\nqueries are stripped in the amphora-agent). It fixes the issue on Centos\n9 Stream (and turns off a warning on Ubuntu).\n\nStory: 2010442\nTask: 46873\n\nChange-Id: I00e327e1d94e46aa13a38120df9865ec34eaa593\n(cherry picked from commit 3605689ac9048cf1126d337833b5cf12ff37b88d)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/353b55c58806a650af894aba1967b2e392bc221a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/353b55c58806a650af894aba1967b2e392bc221a"}]},"branch":"refs/heads/stable/xena"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
