)]}'
{"id":"openstack%2Foctavia~871692","triplet_id":"openstack%2Foctavia~stable%2Fyoga~I2df8a49851feb1445b5128ce99b880ddb77782ad","project":"openstack/octavia","branch":"stable/yoga","attention_set":{},"removed_from_attention_set":{"29244":{"account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"last_update":"2023-01-31 12:00:34.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I2df8a49851feb1445b5128ce99b880ddb77782ad","subject":"Filter out details from taskflow logs with v2+jobboard","status":"MERGED","created":"2023-01-25 08:40:28.000000000","updated":"2023-01-31 12:01:44.000000000","submitted":"2023-01-31 12:00:34.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"871692","meta_rev_id":"ab2a0aa27264783acdce82df33a8734d7269f431","_number":871692,"virtual_id_number":871692,"owner":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-01-31 12:00:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"recommended":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"all":[{"value":1,"date":"2023-01-26 11:11:56.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":2,"date":"2023-01-25 15:25:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-01-31 10:13:44.000000000","_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"all":[{"value":0,"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-01-31 10:13:44.000000000","_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-01-25 11:22:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-01-25 15:25:13.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2023-01-26 11:09:58.000000000","updated_by":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"reviewer":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"state":"CC"},{"updated":"2023-01-26 11:11:56.000000000","updated_by":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"reviewer":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"state":"REVIEWER"},{"updated":"2023-01-31 10:13:44.000000000","updated_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"reviewer":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"state":"REVIEWER"}],"messages":[{"id":"99648e25c824406f6e5d5be6da60935deb37c1e0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-01-25 08:40:28.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"873ee842d04e02ccd224492cd599d30624052e00","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-25 11:22:05.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8770ce93670744999d6d83c6e9e33d5c\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e37510cae0be4cd49d128005788dcc53 : SUCCESS in 15m 13s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f0a6576354f746d58a24cd24c5edeac1 : SUCCESS in 9m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/78ca1640eafc4045a05c6e2448e4da1f : SUCCESS in 5m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/aec85836ce2a42949f8a894dc587b317 : SUCCESS in 4m 53s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d30769037fcd4246b4315558feabee7e : SUCCESS in 5m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/42034819a210441eb1b63574663ed1a2 : SUCCESS in 11m 51s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/215eb0df582b4146ba0055b9d51e7ced : SUCCESS in 5m 34s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/6a1d579501564f7ba7253f5eecf1c361 : SUCCESS in 5m 11s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/f9436761f28c47818098db3bdf9694fa : SUCCESS in 5m 16s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/cccc03ee07fb482382f84a9b49cae947 : SUCCESS in 4m 49s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/ab471196549241c4bade9344793f25cc : SUCCESS in 8m 10s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/ea4a429442574860a39fdca88047983e : SUCCESS in 1h 36m 30s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/bef3bf7d696943d88663bfbfaaeede94 : SUCCESS in 1h 40m 51s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/04743cc8023d4cb0aa0c2342305d6ede : SUCCESS in 33m 40s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/4faabf4e58cc4ba28c707b25edbba534 : SUCCESS in 1h 01m 24s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/f9ff978f9fb14190a938b78ee168cd0d : SUCCESS in 1h 16m 37s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/96c21525bdd84b10abf8a471c76253a1 : SUCCESS in 40m 31s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/692a44cfcde34a15a2017cc71c515a17 : SUCCESS in 2h 23m 25s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/46bf614e73b64de18de37fa63290dfee : FAILURE in 26m 11s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/be7ebf00cad34f9f858a4a905a1da3e6 : SUCCESS in 2h 30m 56s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/abf78bfb64ea4b0bb27cfe55b3dad15f : SUCCESS in 2h 17m 42s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/e581fa6ae20c47d4a882635627fd3c61 : SUCCESS in 2h 31m 19s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8-stream https://zuul.opendev.org/t/openstack/build/e1075e454c4142a98377cb4dc6977d86 : FAILURE in 33m 56s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"28fc7b3cf6af77ae15d61a92bb2bf68d20dfcd84","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-01-25 15:25:13.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"9e9b52a49f11d609fe63c3dd511ac0af96a365c6","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2023-01-26 11:09:58.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"407f0971e8c0d2b781d7b161dc77cbe18d76aaef","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2023-01-26 11:11:56.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"5fa63420a82becb5846a13507fa68639b8d04f54","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2023-01-31 10:13:44.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"1acb3285a4f0acb665823c83590b55d3b02e5140","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-31 10:14:24.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"30d811f5caf4dd8034efe4593148ec3226f06df9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-31 12:00:34.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/adc6c5ec734c4d868aa385a2dc6c899c\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5076ec34d4e94e63b9bde45e1b29eb1d : SUCCESS in 11m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8e8eb818cadb4d16b6159ca8534dffb6 : SUCCESS in 6m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f6e9c36a160e4198ba8b6692598d4fc8 : SUCCESS in 5m 09s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a2ce0a43d7f148bda7ba07f6f8b24016 : SUCCESS in 5m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bb3b66841b5e4801bb88c70635b3a1ec : SUCCESS in 11m 29s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cb090d25f82340a285e4d367cd109a89 : SUCCESS in 4m 47s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/d89db919c26243078db7f7b196a832f4 : SUCCESS in 5m 29s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/a246022aae7c4540a9a5c4c695bbbc6d : SUCCESS in 10m 20s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/dfc0068e443d46ea8b5105cfa99d4dee : SUCCESS in 1h 18m 35s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/9013cc03cac145f084aea043b0aaf9e8 : SUCCESS in 1h 41m 53s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/ae40ef0e8cff4ecdb7bfe6d0e4954378 : SUCCESS in 32m 13s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/f86ba437609e4146a19eb5b8cc26146a : SUCCESS in 42m 24s","accounts_in_message":[],"_revision_number":1},{"id":"94b2db5f8f016e83b6eb6445d1145d41d7435e2d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-31 12:00:34.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"ab2a0aa27264783acdce82df33a8734d7269f431","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-31 12:01:44.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/726e2f1fe67740acbeed64bc31a1c105\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/680061f4ff2041ad8a28dec69c49495f : SUCCESS in 48s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/29af589c73634ba2be79244c256adf02 : SUCCESS in 41s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"46f774ad14a0e9b949962a037870ef26d28cb859","revisions":{"46f774ad14a0e9b949962a037870ef26d28cb859":{"kind":"REWORK","_number":1,"created":"2023-01-25 08:40:28.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/92/871692/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/92/871692/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/92/871692/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/92/871692/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/92/871692/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/92/871692/1"}}},"commit":{"parents":[{"commit":"c662e42b849aa962a814fe756e109950e794277c","subject":"Fix full graph loadbalancer creation if jobboard is disabled","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c662e42b849aa962a814fe756e109950e794277c"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-01-22 07:41:26.000000000","tz":-300},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-01-25 08:39:50.000000000","tz":60},"subject":"Filter out details from taskflow logs with v2+jobboard","message":"Filter out details from taskflow logs with v2+jobboard\n\nWhen enabling INFO-level logs in taskflow and using jobboard with\namphorav2, taskflow prints the string representation of a job when it is\ncompleted. It includes the parameters of the flow, which might include\nprivate information from TLS-enabled listeners and pools such as\ncertificates, private_key and intermediate certificates.\n\nThis commit filters out the private information from the logs by using\nlogging.Filter, it replaces private attributes with \u0027***\u0027.\n\nStory 2010523\nTask 47125\n\nConflicts:\n\toctavia/tests/unit/common/test_base_taskflow.py\n\nChange-Id: I2df8a49851feb1445b5128ce99b880ddb77782ad\n(cherry picked from commit 6c731fa2fdcb4f316fbf65b8d89b24f4c4f277bc)\n(cherry picked from commit 4022aaf7810bad7c6622eed7deda6459bfabeb06)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/46f774ad14a0e9b949962a037870ef26d28cb859"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/46f774ad14a0e9b949962a037870ef26d28cb859"}]},"branch":"refs/heads/stable/yoga"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[]}
