)]}'
{"id":"openstack%2Foctavia~871693","triplet_id":"openstack%2Foctavia~stable%2Fxena~I2df8a49851feb1445b5128ce99b880ddb77782ad","project":"openstack/octavia","branch":"stable/xena","attention_set":{},"removed_from_attention_set":{"29244":{"account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"last_update":"2023-01-31 12:17:15.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I2df8a49851feb1445b5128ce99b880ddb77782ad","subject":"Filter out details from taskflow logs with v2+jobboard","status":"MERGED","created":"2023-01-25 09:04:56.000000000","updated":"2023-01-31 12:18:36.000000000","submitted":"2023-01-31 12:17:15.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"871693","meta_rev_id":"ab388b07dd17ab29a69aeac5f069083ef0050592","_number":871693,"virtual_id_number":871693,"owner":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-01-31 12:17:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"recommended":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"all":[{"value":1,"date":"2023-01-26 11:11:38.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":2,"date":"2023-01-25 15:25:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-01-31 10:14:01.000000000","_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"all":[{"value":0,"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-01-31 10:14:01.000000000","_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-01-25 12:11:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-01-25 15:25:54.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2023-01-26 11:11:38.000000000","updated_by":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"reviewer":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"state":"REVIEWER"},{"updated":"2023-01-31 10:14:01.000000000","updated_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"reviewer":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"state":"REVIEWER"}],"messages":[{"id":"9b3c68999a21a7246e20acc9a46ffbf6a83e3366","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-01-25 09:04:56.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ba9f600f470a06ba8b095b99a1ce8be2e5d2f3c9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-25 12:11:46.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/ad9979999e0949efa4f38862c9657b7d\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b1d8870b169f4d77a6a283c17b689693 : SUCCESS in 11m 09s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6f97cdd2f20e411083e918e0d41ade9e : SUCCESS in 7m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f46dcbbdb4b3470cb8bf601786277b64 : SUCCESS in 5m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3d1be8a147e24abbb87f323d46a7af84 : SUCCESS in 4m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b9750fb9588d48d9ab1c193f0a2165d4 : SUCCESS in 4m 07s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/855438bf31334127936f5c352dc507b7 : SUCCESS in 5m 24s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d36bc382766b4e50a609fde2729afd8a : SUCCESS in 12m 38s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/00c083d432334ab599cf4bfcd75f4bcb : SUCCESS in 5m 05s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/4d5d1c4eee66414288cc59f0130c1c9e : SUCCESS in 5m 47s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/98549980bce740d39eff8e4c70244124 : SUCCESS in 10m 23s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/94e988650086489990aafdd2ad537a3e : SUCCESS in 5m 32s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/4b10ace8adee4cc0a45d0431bf643237 : SUCCESS in 7m 14s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/e5f71858cfec43aa9e656de76317b5f4 : SUCCESS in 1h 41m 18s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/84823f029eae4298975df238bab6cb7c : TIMED_OUT in 2h 33m 51s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/ccc54b54d0a749659d63bdd49000487d : SUCCESS in 34m 22s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/192d86ca3f2b48d69352549a6ad17544 : SUCCESS in 1h 07m 37s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/ce3d2894772d44c78c2cbe4b2e71769f : FAILURE in 11m 38s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/3bccf521cfcd40cd84958fa9e6196ad6 : SUCCESS in 33m 04s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/44bf534dfdd0454aa89fd1884c98e767 : SUCCESS in 2h 00m 25s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/c145a72f314347da8deb00238932bd52 : TIMED_OUT in 3h 03m 27s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/50564ea7269b459bbeb3bd6d94878a1f : SUCCESS in 2h 32m 16s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/33edf8f234294de38f75b0d71eb347a2 : SUCCESS in 2h 22m 46s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/0f18d63742ac4319b218b486f11ec719 : SUCCESS in 1h 51m 32s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/5489954db9bd4d869d8c2478a61bc188 : NODE_FAILURE Node request 300-0020271759 failed in 0s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a3f9ece7c3354c417ce51835f09100d9ae2f815b","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-01-25 12:19:00.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"2503e066c3f667cd75a5ed19e473be23b9ef6632","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-25 14:25:13.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fc017411be574f2d8f58cb50b5e14365\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9255c76fb9664318b02ce4bf88fc3a0a : SUCCESS in 11m 10s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/faad382d18994ee2829e842434a5b215 : SUCCESS in 8m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/452d110ca5a84acaba162138ea934650 : SUCCESS in 5m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2410c6eba25f4e2bb1e31eb6f307c505 : SUCCESS in 5m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1f505d1a5b344a6281c5d72736015122 : SUCCESS in 5m 00s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2c81e971135a431a820f3006c43ec083 : SUCCESS in 4m 56s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2315ce62e175415b9192da7c2af5b1f6 : SUCCESS in 11m 50s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bbf79e3075404449b8488ccae724b1c2 : SUCCESS in 4m 58s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/d51e122d043949768feaaefa736f8117 : SUCCESS in 4m 55s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/2c7e4a82cf2d40ed93cd934ab181043e : SUCCESS in 8m 20s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/2925076c5bab47ef92e7ffff65434165 : SUCCESS in 5m 05s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/bd1e67f4be864d51beca9600fe79c426 : SUCCESS in 8m 42s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/72ccc6d5c52f4a7698c0dce35a041fab : SUCCESS in 1h 22m 13s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/3ba310ac6d3e401a8844c48418729f4c : SUCCESS in 1h 43m 46s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/3ff089b14e5449b9984c20f37513f2f0 : SUCCESS in 32m 28s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/0279715777764d3e9d528ef3b910b48f : SUCCESS in 48m 55s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/a8a3a6f22f7940fe969362702d26f611 : FAILURE in 14m 52s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/45382f0de3d6403996ea51e4548d5e9b : SUCCESS in 30m 21s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/8ad3a20e145e4ee8885abebb495530d0 : SUCCESS in 1h 36m 51s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/b7422dbe008648088fe507603ab4a36c : SUCCESS in 2h 03m 39s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/851f800a3adc486da0299918fb0196f2 : SUCCESS in 1h 51m 31s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/c8aeb98f3ab443e4afa43244310a41b9 : SUCCESS in 2h 00m 50s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/ee7796a5b8dc4e0190db087cbcb9cc66 : SUCCESS in 1h 34m 22s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/66e4098da36949e29416e4fb19c14660 : NODE_FAILURE Node request 300-0020273625 failed in 0s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"02a3158693219fdd9c493c665602422848d4ed10","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-01-25 15:25:54.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"09ec02f6f9332f16cc747ee68968c2eaa3eecf37","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2023-01-26 11:11:38.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"8b84059a8ef7791618fdca137ddaddd3f6445008","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2023-01-31 10:14:01.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"47bf96f7e14cc0e1c64eab3657c65eb1f6017dd0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-31 10:14:44.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"f825cb71d369c500355b1a11b790e7fb74dc5704","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-31 12:17:15.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/28c581c0f52b4cfb83ce145813f4c1e0\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3a7fa53873b64701b172894c377bef5e : SUCCESS in 7m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e4647223710b45778d10988c8e3fb13d : SUCCESS in 4m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6b1cae4b103d4b7da216560dabe2a8b4 : SUCCESS in 5m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/921c3a2f0bd6471288cd6bf833d18d4a : SUCCESS in 3m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/05529ae4e44d4ccda9babe913588e74c : SUCCESS in 10m 18s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5eea4839d5ac48558094c2d576dc54c3 : SUCCESS in 4m 14s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/d89b667abd734d97a9d56eebfdd4950a : SUCCESS in 4m 12s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/fe6efe38df28446992388ebc309abf5f : SUCCESS in 8m 18s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/8b402315656345df84ac225532ec364b : SUCCESS in 1h 34m 58s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/13ca024d233846c0b4ea7a6cb9e6c407 : SUCCESS in 1h 39m 00s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/5a2c8066a88943058109001e9a4b4895 : SUCCESS in 31m 28s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/a3b2f0ca3c494e8496652a3d01a9ed85 : SUCCESS in 46m 35s","accounts_in_message":[],"_revision_number":1},{"id":"091cf129020f62d7dc306e64d4c59f351d8ef2b8","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-31 12:17:15.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"ab388b07dd17ab29a69aeac5f069083ef0050592","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-31 12:18:36.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a5d06f7960ba45a68beecf4163ca26d5\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/84861bcd92414d75acd46e386a30a39f : SUCCESS in 1m 02s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e83a50115ddf4ebcb3e34d0f78709767 : SUCCESS in 47s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"6f1975627819deaa6a6d3b0e4ae91115579d0fc6","revisions":{"6f1975627819deaa6a6d3b0e4ae91115579d0fc6":{"kind":"REWORK","_number":1,"created":"2023-01-25 09:04:56.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/93/871693/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/93/871693/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/93/871693/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/93/871693/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/93/871693/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/93/871693/1"}}},"commit":{"parents":[{"commit":"c0b6f5c9ef07f777b10d952a70638c3e00e21c05","subject":"Merge \"Fix plugging member subnets on existing networks\" into stable/xena","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c0b6f5c9ef07f777b10d952a70638c3e00e21c05"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-01-22 07:41:26.000000000","tz":-300},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-01-25 09:04:46.000000000","tz":60},"subject":"Filter out details from taskflow logs with v2+jobboard","message":"Filter out details from taskflow logs with v2+jobboard\n\nWhen enabling INFO-level logs in taskflow and using jobboard with\namphorav2, taskflow prints the string representation of a job when it is\ncompleted. It includes the parameters of the flow, which might include\nprivate information from TLS-enabled listeners and pools such as\ncertificates, private_key and intermediate certificates.\n\nThis commit filters out the private information from the logs by using\nlogging.Filter, it replaces private attributes with \u0027***\u0027.\n\nStory 2010523\nTask 47125\n\nConflicts:\n\toctavia/tests/unit/common/test_base_taskflow.py\n\nChange-Id: I2df8a49851feb1445b5128ce99b880ddb77782ad\n(cherry picked from commit 6c731fa2fdcb4f316fbf65b8d89b24f4c4f277bc)\n(cherry picked from commit 4022aaf7810bad7c6622eed7deda6459bfabeb06)\n(cherry picked from commit 46f774ad14a0e9b949962a037870ef26d28cb859)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/6f1975627819deaa6a6d3b0e4ae91115579d0fc6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/6f1975627819deaa6a6d3b0e4ae91115579d0fc6"}]},"branch":"refs/heads/stable/xena"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[]}
