)]}'
{"id":"openstack%2Foctavia~871694","triplet_id":"openstack%2Foctavia~stable%2Fwallaby~I2df8a49851feb1445b5128ce99b880ddb77782ad","project":"openstack/octavia","branch":"stable/wallaby","attention_set":{},"removed_from_attention_set":{"29244":{"account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"last_update":"2023-02-09 06:02:51.000000000","reason":"\u003cGERRIT_ACCOUNT_29244\u003e replied on the change","reason_account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}}},"hashtags":[],"change_id":"I2df8a49851feb1445b5128ce99b880ddb77782ad","subject":"Filter out details from taskflow logs with v2+jobboard","status":"MERGED","created":"2023-01-25 09:05:19.000000000","updated":"2023-02-09 10:21:36.000000000","submitted":"2023-02-09 10:20:20.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"871694","meta_rev_id":"3d3d88901353e1b7136b85d1f8334d097b182b99","_number":871694,"virtual_id_number":871694,"owner":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-02-09 10:20:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"recommended":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"all":[{"value":1,"date":"2023-01-26 11:09:51.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":2,"date":"2023-01-25 15:25:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-01-31 10:15:01.000000000","_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"all":[{"value":0,"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-01-31 10:15:01.000000000","_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-01-25 11:08:44.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-01-25 15:25:05.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2023-01-26 11:09:51.000000000","updated_by":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"reviewer":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"state":"REVIEWER"},{"updated":"2023-01-31 10:15:01.000000000","updated_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"reviewer":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"state":"REVIEWER"}],"messages":[{"id":"88f66c00a2d4f027b8a6a9e20d2ab1b8d6119b9d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-01-25 09:05:19.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c15e97a7001218fc7d2f804e26a241be35869725","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-25 11:08:44.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/86c96c893e9a480798bc46cfcc3aa635\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ab45b55c2e884489a4c63c6cd0415f22 : SUCCESS in 10m 50s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e05a4982dee546cf9f4f4ef6edc6cc71 : SUCCESS in 8m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/44b6b03587684f58b5c9679b8cdf12b0 : SUCCESS in 5m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f1696520b8b54b529bc25e237a5648fe : SUCCESS in 5m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c0f76d8924e241f089a182e3a945e94e : SUCCESS in 4m 42s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c5161c509da74a47aa1dcc2622ec12d0 : SUCCESS in 5m 48s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6e07fe9aaf0c45619c281b2d055c0be3 : SUCCESS in 13m 02s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/af91fa4819fd48c4a33ece2dc3da3940 : SUCCESS in 4m 40s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/21de74e6c7c84be3b4af3c187c478a24 : SUCCESS in 4m 59s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/175c846af7754be6ba3bc960e366ac4a : SUCCESS in 6m 34s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/6575f0997d684162a9cef5fbe63ccb20 : SUCCESS in 5m 10s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/ae75d4974838456d94a6a29d65015638 : SUCCESS in 7m 53s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/13fda39caef14dc89e7f89cdd20c1bf1 : SUCCESS in 1h 06m 06s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/e94d34aacba947d9af8ffe3c1f772315 : SUCCESS in 1h 32m 02s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/dca291f19a1c4469b2e26091323bb09e : SUCCESS in 34m 55s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/f956d391ef564cadbb984d90c777306f : SUCCESS in 30m 07s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/054070caf6db4deba42b47e739a47584 : SUCCESS in 32m 56s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/06246064b0064e23931c5e00838f161c : FAILURE in 14m 44s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/4fbeba7085b74227a51645ec8dc7d238 : SUCCESS in 1h 26m 38s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/0c6f9251332b48b58cca72c1b0466f77 : FAILURE in 27m 28s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/2b20c064fdcb4754a34d5eebb99e47a6 : SUCCESS in 1h 35m 41s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/3f950ed4aad946ad977f1a454dd504b0 : SUCCESS in 1h 39m 24s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/4408447ae89d48cb870ffd155c3f9c6d : SUCCESS in 1h 59m 21s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/fbfa0910df0d41dd8c60013aaf834878 : NODE_FAILURE Node request 300-0020271784 failed in 0s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"e5aecd818031373d10fed39798619640a2397fab","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-01-25 15:25:05.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"958ede485e9ba8b21b6b885e81b5b7b1138e35ab","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2023-01-26 11:09:51.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"312389c7507ba1e2d5edabf3b18ba914fee03ced","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2023-01-31 10:15:01.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"d3218282f710f9ddfe7f2e71f73eba29b7fca430","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-31 10:16:00.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"033dedb1c690d3a55b701782a8ccbf207254bba5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-31 12:41:59.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/2bf94595d72e4d9dbd20a98306e9b87f\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/aceeabfc124c4a94a8a1e66da116154c : SUCCESS in 8m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3cf66fa9a06347389448340977568e06 : CANCELED in 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b7efe3bf7b594d54aadb04efcdb104ca : SUCCESS in 5m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/58b29b24a98d4e229862dcc33d5e6f0f : SUCCESS in 4m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bc923a19e6c14299ace69d4fb522c066 : CANCELED in 10m 14s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/38737f0cd70b4f6bb6f9a9d22f6315f9 : SUCCESS in 5m 22s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/61e862b0864842c9b51daa83f49d3fa6 : SUCCESS in 5m 14s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/e48f56028a3f41aeaa9fe0bfd74a2711 : SUCCESS in 7m 15s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/b0de09dc62684040aa23f414c0a4c387 : CANCELED in 25m 57s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/b39dd8b9d9cd4391a52cf781e86ae89e : FAILURE in 24m 43s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/584cb81748ea417196f6f5febc2b81dd : CANCELED in 24m 38s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/0c37a5ae357d484391f8a12ab2c228d2 : CANCELED in 24m 47s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/8aee0097dd284835bca91cbe999a2375 : CANCELED in 25m 19s","accounts_in_message":[],"_revision_number":1},{"id":"1b712c0bd3fa8d39087b71abbc38dc5040f820e1","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-02-06 06:40:17.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"2ef5148e283ecdb2d9ecad5d16b307c6000c0537","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-06 09:12:57.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/9b4aa0a9d66b4914abc6a05a62602b88\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4b1cf413894a4c728eb001ec4172f57e : SUCCESS in 8m 34s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/408f3306ed0540788043ab2a1b676747 : SUCCESS in 6m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7d1f05fdc0434041a871f2f9afd27832 : SUCCESS in 4m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d07b467931e74ebe9c763d2ef68e86e4 : SUCCESS in 4m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1f8848a3d2da4a2aa44a79d9772ced80 : SUCCESS in 4m 14s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/42459c540a83493681561e3242f5ff53 : SUCCESS in 4m 12s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0c02553d31eb4df28a10ed03b6eb23f6 : SUCCESS in 12m 26s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/72650e6cac284ebf9ffd3f42cd29de0c : SUCCESS in 4m 33s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/6bc9ac0ca6354ecc9ca88080b04bc664 : SUCCESS in 4m 43s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/d2d67600ff09494394701a0e29c55697 : SUCCESS in 7m 02s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/8927bd3d51fb4f98bc2ee3713cd95740 : SUCCESS in 4m 48s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/0486a66d6d3142cbbc7904587e6037b1 : SUCCESS in 6m 42s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/0dbaaa92512444b6a00abf81288512ff : SUCCESS in 1h 01m 38s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/4b3d4092fed44cd1aad9ccc879fa0e33 : SUCCESS in 1h 58m 03s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/43555d426f014d23801fa090d4a4b8d6 : SUCCESS in 35m 51s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/3d2b80feb6ef404d98c7d0e911ecd0e4 : SUCCESS in 32m 57s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/2b985b1906b34ebb9cb64e523935e337 : POST_FAILURE in 35m 46s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/5644c7f6b49b476891ca0d497c560883 : FAILURE in 9m 05s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/4bc5fa47ee6b4bcb81ec9d1981d3c452 : SUCCESS in 2h 23m 34s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/9e134975ebd446caa9c2cbcd72b48282 : FAILURE in 43m 26s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/ac69cb46adbc47f69c48a7c47c37e562 : SUCCESS in 2h 02m 58s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/a161767ef45e4522bb56512bab59066e : SUCCESS in 2h 23m 59s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/aa9728313dba49418756e06e3ff0bc4f : SUCCESS in 2h 12m 40s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/fe0ddd1e1baf4a24a63f93788ef425f0 : NODE_FAILURE Node request 300-0020368914 failed in 0s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b2bc7d7402e84dd02ffd96bb28adac8e6bfec242","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-02-09 06:02:51.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"afa995fe8803c0d2bb658abb2c58761acab2fa97","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-09 08:11:26.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/68515835a6a54978aeeddaaaa868c72a\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3162e462997c43e5b6c5d6c7b62d0715 : SUCCESS in 10m 22s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/cdc6c43b3fe14e708ebb51bb4bb64003 : SUCCESS in 7m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1f4f3e61df764ade9c515589757ecef0 : SUCCESS in 5m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/479ab2522514460c8e684a871e80ae3c : SUCCESS in 5m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/97caaad2edd44d8d9807074724c1e9f4 : SUCCESS in 4m 50s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8b14d45fc6074400bdfa6db283c6e1b7 : SUCCESS in 4m 41s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a48ae6bbee3b41ffaa29453d5e3f25a7 : SUCCESS in 12m 23s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/05b1474e1157417d90c365ddfadd04fc : SUCCESS in 4m 12s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/af7460367af146b3b7489943db706af0 : SUCCESS in 5m 49s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/49f93dbb00104b279013e76ce4b203b4 : SUCCESS in 7m 38s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/b418e95aadb24b47bbc6193f3ac72911 : SUCCESS in 5m 25s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/59ef6262193b430b9223584d48df31a0 : SUCCESS in 8m 34s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/b0d93a678c09423498b34b4939c57ef4 : SUCCESS in 1h 17m 13s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/bc05dc578a88400288fbceed4e47764e : SUCCESS in 1h 47m 04s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/a863bdbb8e144f0e86e2af5861ea2901 : SUCCESS in 37m 04s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/40d19fd5f48b435a8eb922802f2bc860 : SUCCESS in 35m 29s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/cf03c70776084604804d5c64e46cd4a1 : SUCCESS in 35m 03s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/d8194874c8f84fa3b51d74ee861c8514 : FAILURE in 10m 54s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/21c782b7c9e14c20995136be4387a5a6 : SUCCESS in 1h 32m 47s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/7621c80f0e484b2495bac2e4bcfebb5c : FAILURE in 31m 26s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/0b3320af3f074cb3b8021fa22e0fa7e3 : SUCCESS in 1h 38m 52s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/a1021b89b9ae457d8df15896f0aa6f4b : SUCCESS in 2h 06m 19s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/59cc38b22d8d482cb575554d3cfda721 : SUCCESS in 1h 38m 13s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/2e578e71a2a74a1688c605c05beb47d2 : NODE_FAILURE Node request 300-0020402732 failed in 0s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"28ba90c0b7d48e9600547dbacb50524329523f37","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-09 08:12:07.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"f6a2e1fdb29699359cd7330e37e34f52e7dc3ae4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-09 10:20:20.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5cc35c1f9dba46ac9b2305397ba5ccb7\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/222b4dd8ea094b27bbff96741f106cdc : SUCCESS in 8m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/13372e631aa9484a92961d7d56d82f06 : SUCCESS in 4m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4f595b0bbfb845a8a749a39c9709532b : SUCCESS in 4m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e4a0ce040366462fa1aa8ffa323f6b5b : SUCCESS in 5m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/55279f7bfcad49489578367022cb9234 : SUCCESS in 11m 49s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/541b2443449f4ff294db39a8a935b59a : SUCCESS in 4m 35s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/4d33e273c367433fbf91cbe6fcc1fb45 : SUCCESS in 5m 13s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/76c8fd6523574447ae8e07b2b9c0b788 : SUCCESS in 6m 53s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/1ae662975eff44e0b1d4daed4ec2e6cc : SUCCESS in 1h 11m 51s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/27f8e62249cd43fcb6713ef0852e7507 : SUCCESS in 2h 05m 31s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/1396e94132ea42bc8f553f8e19739e26 : SUCCESS in 38m 42s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/00bf6693111a40fa98b54223c29de194 : SUCCESS in 39m 15s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/68c0755a97bb48b5a7f1baa2c6f1831a : SUCCESS in 36m 52s","accounts_in_message":[],"_revision_number":1},{"id":"0278c69579fe787961372a089c6f688342647ae5","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-09 10:20:20.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"3d3d88901353e1b7136b85d1f8334d097b182b99","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-09 10:21:36.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/46a05bb803b1498d85bac156671aec7c\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/345b2bf243644092a08eea699e0ece19 : SUCCESS in 51s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/499c113e22fb484eb6728b63043dda06 : SUCCESS in 47s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"deba44d022395851ab3f354a96b5e15c659244a8","revisions":{"deba44d022395851ab3f354a96b5e15c659244a8":{"kind":"REWORK","_number":1,"created":"2023-01-25 09:05:19.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/94/871694/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/94/871694/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/94/871694/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/94/871694/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/94/871694/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/94/871694/1"}}},"commit":{"parents":[{"commit":"b0379d6a8411ae65413579bdd7927bfa07e51b5b","subject":"Merge \"Fix plugging member subnets on existing networks\" into stable/wallaby","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b0379d6a8411ae65413579bdd7927bfa07e51b5b"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-01-22 07:41:26.000000000","tz":-300},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-01-25 09:05:13.000000000","tz":60},"subject":"Filter out details from taskflow logs with v2+jobboard","message":"Filter out details from taskflow logs with v2+jobboard\n\nWhen enabling INFO-level logs in taskflow and using jobboard with\namphorav2, taskflow prints the string representation of a job when it is\ncompleted. It includes the parameters of the flow, which might include\nprivate information from TLS-enabled listeners and pools such as\ncertificates, private_key and intermediate certificates.\n\nThis commit filters out the private information from the logs by using\nlogging.Filter, it replaces private attributes with \u0027***\u0027.\n\nStory 2010523\nTask 47125\n\nConflicts:\n\toctavia/tests/unit/common/test_base_taskflow.py\n\nChange-Id: I2df8a49851feb1445b5128ce99b880ddb77782ad\n(cherry picked from commit 6c731fa2fdcb4f316fbf65b8d89b24f4c4f277bc)\n(cherry picked from commit 4022aaf7810bad7c6622eed7deda6459bfabeb06)\n(cherry picked from commit 46f774ad14a0e9b949962a037870ef26d28cb859)\n(cherry picked from commit 6f1975627819deaa6a6d3b0e4ae91115579d0fc6)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/deba44d022395851ab3f354a96b5e15c659244a8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/deba44d022395851ab3f354a96b5e15c659244a8"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[]}
