)]}'
{"id":"openstack%2Foctavia~871916","triplet_id":"openstack%2Foctavia~stable%2Fwallaby~I971a22e397617829ba71b6b43343390df5af6ed6","project":"openstack/octavia","branch":"stable/wallaby","topic":"wallaby-pin-tempest","hashtags":[],"change_id":"I971a22e397617829ba71b6b43343390df5af6ed6","subject":"DNM: testing tempest pin for stable/wallaby","status":"ABANDONED","created":"2023-01-27 02:08:59.000000000","updated":"2023-02-02 03:50:46.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"13a055ae6d5cd5751c1cd4c728c9ac843d9ff7a5","_number":871916,"virtual_id_number":871916,"owner":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2023-01-27 04:53:11.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-01-27 04:53:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"0d000966a17a04d8a6311417e5b0503fc0104ab1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2023-01-27 02:08:59.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"54121e83777d1e28422b03af3cdf39d4cb1fea38","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-27 04:53:11.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7762335186b9460eaba2114aec3db698\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a1069e88831b4ffb976012162eb42203 : SUCCESS in 8m 55s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/cbb489aa49384e05be8c8dd526902c50 : SUCCESS in 8m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/20129a88166048f2a331997025d4e622 : SUCCESS in 5m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/87b4bc5396654a1d85dfa713ee42cead : SUCCESS in 4m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a576cc9ff95f4bdbbb1a7e39f1858780 : SUCCESS in 4m 34s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e098dc6ccc024dc78603d131a21372ab : SUCCESS in 4m 38s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/868a5c2e4b424ceb89b1645da753d67b : SUCCESS in 11m 56s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/682f952c16294733a9ff12483d879a2d : SUCCESS in 4m 54s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/118a2af79b1640f881d52bfb1bd886eb : SUCCESS in 7m 13s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/c5e79f31f2934dbd8e12da4f0b0dabee : SUCCESS in 5m 58s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/238102aef37d466492ae6561322608b6 : SUCCESS in 7m 16s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/58bbd492315e45a7b04b7c3db4342ce8 : SUCCESS in 1h 10m 10s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/68677145f5d7432687829191dc811e0c : SUCCESS in 1h 31m 33s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/c80dd286ba744630b970aec5823ed80b : SUCCESS in 35m 52s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/7b1f2c818aba47bcac18292f4ef28d08 : SUCCESS in 32m 54s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/24c2cf03ee364e03989158cfc856e3f9 : SUCCESS in 44m 56s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/6927a3eeee5d479b987c18b254a2cd13 : FAILURE in 13m 00s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/c2c243a382654283925728aaea98cead : SUCCESS in 2h 10m 34s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/e7b7b334327b41baaac03a928f903130 : FAILURE in 35m 58s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/eb3b9783431c4cd4aad113066a9d7d13 : SUCCESS in 1h 45m 27s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/845d8177b536453895c407e4d0c7ce8c : SUCCESS in 2h 24m 59s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/a2571a18350244199d60e4b4a6abceef : SUCCESS in 2h 17m 06s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/a05280d1bb954792a615308352206632 : NODE_FAILURE Node request 300-0020290194 failed in 0s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"92ac0e8ffd8e53987a1bc6041f15c8a668b1c93c","tag":"autogenerated:gerrit:abandon","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2023-01-29 02:56:33.000000000","message":"Abandoned\n\nall good here","accounts_in_message":[],"_revision_number":1},{"id":"7c5e692f5f4738acb9d90288116ddb7f176519ef","tag":"autogenerated:gerrit:restore","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2023-02-02 03:49:55.000000000","message":"Restored","accounts_in_message":[],"_revision_number":1},{"id":"13a055ae6d5cd5751c1cd4c728c9ac843d9ff7a5","tag":"autogenerated:gerrit:abandon","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2023-02-02 03:50:46.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"c94dfbef90c1dad44e2b8643c65900424e2d73e0","revisions":{"c94dfbef90c1dad44e2b8643c65900424e2d73e0":{"kind":"REWORK","_number":1,"created":"2023-01-27 02:08:59.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/16/871916/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/16/871916/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/16/871916/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/16/871916/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/16/871916/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/16/871916/1"}}},"commit":{"parents":[{"commit":"b0379d6a8411ae65413579bdd7927bfa07e51b5b","subject":"Merge \"Fix plugging member subnets on existing networks\" into stable/wallaby","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b0379d6a8411ae65413579bdd7927bfa07e51b5b"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2023-01-27 02:08:32.000000000","tz":-360},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2023-01-27 02:08:32.000000000","tz":-360},"subject":"DNM: testing tempest pin for stable/wallaby","message":"DNM: testing tempest pin for stable/wallaby\n\nDepends-On: https://review.opendev.org/c/openstack/devstack/+/871782\nChange-Id: I971a22e397617829ba71b6b43343390df5af6ed6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c94dfbef90c1dad44e2b8643c65900424e2d73e0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c94dfbef90c1dad44e2b8643c65900424e2d73e0"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
