)]}'
{"id":"openstack%2Foctavia~872729","triplet_id":"openstack%2Foctavia~master~Ifb1490c7da28941ed493cd249e582e24cd52277f","project":"openstack/octavia","branch":"master","topic":"dsr","attention_set":{"29244":{"account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"last_update":"2023-02-06 09:24:22.000000000","reason":"A robot voted negatively on a label"}},"removed_from_attention_set":{},"hashtags":[],"change_id":"Ifb1490c7da28941ed493cd249e582e24cd52277f","subject":"WIP Direct Server Return implementation","status":"NEW","created":"2023-02-06 06:38:25.000000000","updated":"2023-07-27 16:31:19.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"23b018f73d52b8c1159b56c093d654092f468903","_number":872729,"virtual_id_number":872729,"owner":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2023-07-27 16:31:19.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"}],"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-02-06 09:24:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-02-07 22:00:15.000000000","updated_by":{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},"reviewer":{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},"state":"CC"}],"messages":[{"id":"09c1c0fd4bb16bf9261fe4a71f99739129ef406c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-02-06 06:38:25.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e6ad7d50a2c67931910e7cb4a21585f5b82cad34","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-06 09:24:22.000000000","message":"Patch Set 1: Verified-1\n\n(3 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/db487cfdf7ac4547ab4a704b3a83e6d1\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bc3107737597475faa5fbfc46cae87ca : FAILURE in 5m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5bd3c7ac069844179498f38e3d7ab6bb : FAILURE in 3m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0dcc6e8191dd49e2be7e70955f93832d : FAILURE in 4m 10s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3f7e9a64581448a7ad5af1492109be30 : FAILURE in 4m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f46798bbfc4447bab76969e8318d3efc : FAILURE in 5m 01s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/8dd9e026f7df402898ddbabe2f97b59f : FAILURE in 5m 26s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/fba28310fc014080b503054b768da126 : FAILURE in 17m 51s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/7f25ab7061bf4a03891961aaee539d94 : SUCCESS in 4m 11s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/403fff924edf4dc3aa80baebed3485f9 : FAILURE in 6m 07s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/f7758def6cb94f0f9a585c80c906e938 : FAILURE in 7m 02s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/14214c9b6aad4dc083087c132a4aa2a2 : FAILURE in 6m 06s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/39f7edf588b34d21835d16c6f7db8124 : FAILURE in 1h 34m 55s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/8bc00267f340495babdae38557303ec5 : FAILURE in 22m 15s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/c84e8dfe64d746c7b2982dd4e922e096 : FAILURE in 26m 35s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/9bf64f87d2d24b33ad64badff4af5ab4 : FAILURE in 55m 07s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/253f08325fb14a7b87cd297e059b2fd1 : NODE_FAILURE Node request 300-0020368860 failed in 0s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/c4eeb20f582d43f3ac11014d01ab4637 : FAILURE in 22m 35s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/07625bdea8914d298cc50fe742d9aaad : FAILURE in 24m 22s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/9cf4bacba94d424ebd3e66dbbc282ba5 : FAILURE in 39m 17s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/e8fb1d4119f147f5930efd524c39e886 : FAILURE in 23m 44s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/e98fc3114c394270b320599527641d0d : FAILURE in 25m 00s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/9bce2da2c6394ff08e3a9a07a44ef260 : FAILURE in 24m 05s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/9597dbaf4a3a49209494aed64045cb24 : TIMED_OUT in 2h 33m 49s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"44e36f9c2ba22d0b57f9578996a48fd09847c7f2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-07-27 13:27:02.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"23b018f73d52b8c1159b56c093d654092f468903","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-27 16:31:19.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/d4d2d2b5724b4ffa952de3f22dae01dc\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4390ab251dcf4ba58dcc963328b17b7e : FAILURE in 4m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/323af8a1e83e4b5c81ba01827de5b7c2 : SUCCESS in 5m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5e972b5da476495d8e07c83e5de9dcb5 : FAILURE in 3m 55s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0625727aeeb5485c846587cf764be108 : FAILURE in 5m 35s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d5359063ddf6489aa3f25c62c26b4cb8 : FAILURE in 5m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ad1aa93ae0774edf8802d5d60858d168 : SUCCESS in 10m 37s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/7cf1e61a3521494589b9b827c22bdb24 : FAILURE in 5m 50s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/e7fd30d15d804052b09630d6b3370ba7 : FAILURE in 8m 02s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/a37e025072ef4305b6676b5c2ff925b3 : SUCCESS in 4m 45s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/d955970309364ac9b15e1440b678eea0 : FAILURE in 7m 02s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/92942019a3844c45abe51157f24d8ee5 : FAILURE in 7m 17s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/db4b07f01a0641fcac2b463e4125e486 : FAILURE in 6m 53s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/3f9f8cf675364a9ba9d06ec98aabff27 : FAILURE in 1h 28m 46s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/80b5e515fc11457189911098e0924e97 : FAILURE in 1h 58m 45s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/871972b48e7f4e6bb45826a8b444aaef : SUCCESS in 1h 09m 33s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/3bf66613ed2e49ffb45e8c071f60525b : FAILURE in 1h 59m 32s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/b2a6cb19e74d437191ed30feb0566a19 : SUCCESS in 1h 29m 29s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/e6383556df4b40e49c403ab332cda784 : SUCCESS in 39m 29s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/f877b406e12e4044af7db3685324c238 : FAILURE in 35m 20s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/e32931ac2c5846a58ab5249b561ab828 : SUCCESS in 1h 04m 11s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/5e21a9abe57f4041b163e46d3ff36e2e : SUCCESS in 42m 27s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/c86f5db11cf74f978f557b0a48d08922 : TIMED_OUT in 2h 32m 49s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/4a4965be00e14685b0039604fd0f7e31 : TIMED_OUT in 3h 02m 58s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/c17eab0a218f4fb59739c69a08a8feb3 : TIMED_OUT in 2h 33m 35s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/dc9fece85a8b410ca373a62eca4da47f : SUCCESS in 2h 28m 47s (non-voting)","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"b014db1287ee56984fef0e5cf3da188856d39177","revisions":{"35866921c3f6649b0e8be44dbe577e5d4ee0197e":{"kind":"REWORK","_number":1,"created":"2023-02-06 06:38:25.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/29/872729/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/29/872729/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/29/872729/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/29/872729/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/29/872729/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/29/872729/1"}}},"commit":{"parents":[{"commit":"fe65e1efd76944860d0f874e56a418532818cf57","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/fe65e1efd76944860d0f874e56a418532818cf57"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-02-05 06:03:30.000000000","tz":-300},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-02-06 06:16:02.000000000","tz":-300},"subject":"WIP Direct Server Return implementation","message":"WIP Direct Server Return implementation\n\nChange-Id: Ifb1490c7da28941ed493cd249e582e24cd52277f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/35866921c3f6649b0e8be44dbe577e5d4ee0197e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/35866921c3f6649b0e8be44dbe577e5d4ee0197e"}]},"branch":"refs/heads/master"},"b014db1287ee56984fef0e5cf3da188856d39177":{"kind":"REWORK","_number":2,"created":"2023-07-27 13:27:02.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/29/872729/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/29/872729/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/29/872729/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/29/872729/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/29/872729/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/29/872729/2"}}},"commit":{"parents":[{"commit":"12d8e0de5de9ca754859542bcffd2311f5145d61","subject":"Merge \"Update doc for the OVN provider\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/12d8e0de5de9ca754859542bcffd2311f5145d61"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-02-05 06:03:30.000000000","tz":-300},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-07-27 13:19:17.000000000","tz":-240},"subject":"WIP Direct Server Return implementation","message":"WIP Direct Server Return implementation\n\nRequirements for the LB:\n- same network/subnet for the VIP and the members\n- stateless SG on the VIP port\n- same port number for the listener and the backend members\n\nRequirements for the members:\n- AAP with the VIP address\n- VIP as an anycast address (ex: ip addr add \u003cvip\u003e/32 dev lo)\n- echo 1 \u003e /proc/sys/net/ipv4/conf/eth0/arp_ignore\n- echo 2 \u003e /proc/sys/net/ipv4/conf/eth0/arp_announce\n\nTODO: add constants to octavia-lib\n\nChange-Id: Ifb1490c7da28941ed493cd249e582e24cd52277f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b014db1287ee56984fef0e5cf3da188856d39177"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b014db1287ee56984fef0e5cf3da188856d39177"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
